)]}'
{"id":"openstack%2Fswift~962542","triplet_id":"openstack%2Fswift~master~I0f0c9f02cf89ee34c2adf72f6febed20f49a890a","project":"openstack/swift","branch":"master","topic":"cohanaja-add-skip-commits","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-09-30 17:51:15.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I0f0c9f02cf89ee34c2adf72f6febed20f49a890a","subject":"test: verify find_and_replace with pending commits","status":"NEW","created":"2025-09-29 19:01:18.000000000","updated":"2025-09-30 18:57:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a140ed7383de1f2af9f0ef3cdb1771d0c9b72a0a","_number":962542,"virtual_id_number":962542,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-30 18:49:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-29 19:53:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-29 20:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"251f5215fe4c89d8bf822d54756af4ab9b6d4e6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-29 19:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"877096a53a0d7b0684adb049a649b9344f66d628","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-29 19:11:36.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c6192d61143557c9adb37339674995d6c0398b57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 19:53:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2134297d080743feb9faefcc2dd0752c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7dd729d1f50e4f5fa6b73c891d21c839 : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c23b1d22a0b948a405dcd9acfc43c6384e0261ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 20:08:37.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50524619ab874844adc75911b0e08235\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51bf5bf4e707435ead7d9c8179220c79 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/97eaf894ea4f458db6909ff4e040ea21 : SUCCESS in 1h 02m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/76b07dbf2b194f9f92b2e5f00eced54a : SUCCESS in 1h 05m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/02290bbe0f1a4a9897d14ff1e398bee7 : SUCCESS in 52m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf220ba87b8541188e2c9fa4a99778b1 : SUCCESS in 1h 02m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b574f65fe87c432e9703bdf708383621 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6342c4e13da48af8dbba6790396de51 : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/daa2757611144f16b2b81bd18de9b2d9 : SUCCESS in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5392fe8f609444f0a97246a507a6fa79 : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a0cb1c4fba2e4e66aad7b03803860d01 : SUCCESS in 23m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c134a2f1f4aa485f9a265b9cc0e7d859 : SUCCESS in 22m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f8d0b1aac68f4b4598b939d0327940ad : SUCCESS in 21m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/11a9728fc84a4245b24997c122438773 : SUCCESS in 19m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d329ba6549ad4643aa948a530af2f59f : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8a9aaba0d3f1449e8bd1d7cb7873992c : SUCCESS in 8m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b6cb979cdc3c4d719670d30b1c89a3aa : SUCCESS in 13m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7fc30f71e81843c690ce0384b4244517 : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f38f85ed2794cd7a402bb1e20f2a8e3 : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/54387427f28a4bf08a32eca36b5ecfbb : SUCCESS in 10m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3555d49e5fc64654824cec9a08a9bb55 : SUCCESS in 51m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/47cb16cecb6e4c999e22bdae3edbcb56 : SUCCESS in 54m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71f38ac3c5bd411fb55217d0629ef92d : SUCCESS in 55m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e9640c0c9814f02bad8b1125f51f1ad : SUCCESS in 27m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5f978259e3432b992ef1684ccb5c90 : FAILURE in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce44a00051554ce29e20df0b119a9e3c : SUCCESS in 17m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f9f5fca232a43b099391a9541f68fe9 : SUCCESS in 49m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ee43a67056c45edb4cc632424e378f5 : SUCCESS in 39m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f3074f1e4fcd4f33caa832bf06591091a31f33e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-30 17:49:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a2a821d1d8ec82cc7993dfc15501279c50925c4a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-30 17:51:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44de970e9b8b85684f79fc958d45b4bd7f531936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 18:49:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80fe59b4e20f400db7e3478f65914ccd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49fe243da3f34a7ab41f5a8b2d3fa3ae : SUCCESS in 10m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e081597a69b4425826739198ce958f2 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec537aa2537a4031b830a0198fd17017 : SUCCESS in 10m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/814ba652e8fe4266b2aa1b0c73a8b078 : SUCCESS in 14m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/845c74d0f2be446a822d9016ff9ab85c : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/acfadda879ca4eefa1d3bb3d2a4417fa : SUCCESS in 23m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6595c318a6f14478a49dbba0282d6985 : SUCCESS in 18m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/81f482d854194aaaad5ec9708842bc87 : SUCCESS in 10m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3f78a5355fe348578dc4e5d67292d6aa : SUCCESS in 30m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1fe5ba476bcf4c629c8eca867707df08 : SUCCESS in 11m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8f83d5fda83244a1b274e52458332e73 : SUCCESS in 7m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ab38e37e34e64502a54869fb51f171a8 : SUCCESS in 8m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37797b7826d34257ada26754961fda3e : SUCCESS in 54m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7bc702020baa4411832804c64e3c30ed : SUCCESS in 54m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a632641fed94a7ebc0417baee26f1c6 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f72f67d576714254ab52c8cd9f96dcf3 : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00264288f03f4033885d8f0604d6927d : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a140ed7383de1f2af9f0ef3cdb1771d0c9b72a0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 18:57:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee822e00f9994cbda28acb1f698bf3f2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/928e24e5d55b4d69951d547fd1f81710 : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"af73ac2b1cc1c3c835b656f9f8d09a66c771e6fe","revisions":{"8f864a318ac44077b90d341be8f50c0e3e99216f":{"kind":"REWORK","_number":1,"created":"2025-09-29 19:01:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/42/962542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/962542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/962542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/962542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/962542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/962542/1"}}},"commit":{"parents":[{"commit":"70329f628322fcd359f41f058d502f1d183a4284","subject":"replaces --force-commits flag (should be default true) with a --skip-commits flag to determine when to commit object updates in smsr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70329f628322fcd359f41f058d502f1d183a4284"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-29 19:01:00.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-29 19:01:09.000000000","tz":-300},"subject":"sq: better argparse?","message":"sq: better argparse?\n\nChange-Id: I0f0c9f02cf89ee34c2adf72f6febed20f49a890a\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f864a318ac44077b90d341be8f50c0e3e99216f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f864a318ac44077b90d341be8f50c0e3e99216f"}]},"branch":"refs/heads/master"},"af73ac2b1cc1c3c835b656f9f8d09a66c771e6fe":{"kind":"REWORK","_number":2,"created":"2025-09-30 17:49:41.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/42/962542/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/962542/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/962542/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/962542/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/962542/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/962542/2"}}},"commit":{"parents":[{"commit":"ba1ab9d11c27faa8607e42908949cfcd08eba5c7","subject":"Adds --skip-commits to s-m-s-r","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba1ab9d11c27faa8607e42908949cfcd08eba5c7"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-29 19:01:00.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-30 17:48:54.000000000","tz":-300},"subject":"test: verify find_and_replace with pending commits","message":"test: verify find_and_replace with pending commits\n\nRelated-Change-Id: I6de041f5c12dca2618d22d1271efe242b2f35258\nChange-Id: I0f0c9f02cf89ee34c2adf72f6febed20f49a890a\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af73ac2b1cc1c3c835b656f9f8d09a66c771e6fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af73ac2b1cc1c3c835b656f9f8d09a66c771e6fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
