)]}'
{"id":"openstack%2Fswift~962551","triplet_id":"openstack%2Fswift~master~I0314e9b78449cb71d7700663520c4ddf030ec047","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-06-22 17:35:30.000000000","reason":"Change was abandoned"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-06-22 17:35:30.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-02-03 18:08:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I0314e9b78449cb71d7700663520c4ddf030ec047","subject":"tests: Work on more filesystems","status":"ABANDONED","created":"2025-09-29 21:24:59.000000000","updated":"2026-06-22 17:35:30.000000000","total_comment_count":21,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"ba8db663d7d40fac3115047363e649528870a565","_number":962551,"virtual_id_number":962551,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-03 19:17:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-29 22:14:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-29 22:34:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-06 01:42:19.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2026-01-21 14:53:40.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"1a86e62b1de9574dc998eda69e001efccec7a601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-29 21:24:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3bf2bb29e2bf8f17c5803061293df62798adbf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 22:14:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b296c3a9084247c9834b1908e49a3721\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6907cc08400842c7a788d1a169bf881f : SUCCESS in 48m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f60911a1cbf22888630a2d0972a300c1275c77ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 22:34:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdc7eccc37d345a793c9024f73335825\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3648d47d89e4c288d8888a359c66a6d : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/b4c9c0651b704fa88bd030251de49974 : SUCCESS in 27m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/176ecd6ebcb74c90ae116832f4798b0a : SUCCESS in 57m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e29f0f60938540009ac198548b8a4a4e : SUCCESS in 49m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d39ddcc06074b388cb15c727bc61d6b : SUCCESS in 1h 04m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d33c82584c694517b359d266aaa19eb6 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ffdc0bba372413f83ee5be16a2522eb : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27324df81ace4458b64410c8b135554d : SUCCESS in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b7aafa8007c4495b5edfbf97a4215f8 : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/079eb3fe7fda40b08abc4092ec398c1c : SUCCESS in 15m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7e1ecb71c3054fa2bd7d958349010311 : SUCCESS in 14m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bd9360cef86f4a948f4fb180671ed841 : SUCCESS in 28m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/58a7743275714bc39a3f5cee602e6501 : SUCCESS in 22m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/31484d7a041b4c69a036006e96a5ed85 : SUCCESS in 7m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2def2272c32d47969043eaf6407c2337 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b577d342bb9d41198eaf8afadb3a6751 : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5f669853850c461f9cca142934675095 : SUCCESS in 4m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/565e84bdaf62409da8938615393f6a8f : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6157ba130774b6ea93cdaae929c4b84 : SUCCESS in 5m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7f955debba2417ca1f4b4dfaa585f0f : SUCCESS in 1h 01m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26c717b46afb4e10af9365298a89d021 : SUCCESS in 34m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c4adb5a0d564806adb9d9adb8a99fe0 : SUCCESS in 32m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78af6dea7d2d471c9fc538cbcf853bc5 : SUCCESS in 24m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e61e1e7a032b4940bf7e1eac3fcb77a7 : SUCCESS in 2m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6504de2f4b041089ae117f38d9f7032 : SUCCESS in 8m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f96d17ab355945718e43e33ad8c82e0d : SUCCESS in 34m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efed7b7114544dae99e200651b064aca : SUCCESS in 40m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b8b449022651807d4718c35a1834c147061dd68d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-01 17:20:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b170c81866d3df182dd1091ef2f53e7e9b64d17","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-06 01:42:19.000000000","message":"Patch Set 1: Code-Review+2\n\n(4 comments)\n\nLooks pretty cool and I like if does give up a little more control.\n\nI wonder if we should think about making it a contraint so it can be configured.. or is that too much of a shoot ourselves in the foot.","accounts_in_message":[],"_revision_number":1},{"id":"da8b116b26d4e7f206b34c939136b52ab0cf9e48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-20 16:35:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d27f3d0c24da99181d8c9a50b93faea851e6468a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 17:30:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c60190f7ef44899b54a7424a1d311a0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eceedf787aa049bf90645139ae59a0a7 : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":2},{"id":"4ae8f9242552c8d2f10bed5ba0702368ceba29fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 17:44:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2397ebb703ca49b6a4eb75244c61ae42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/661635bb3f444704b0c63b195417f9e8 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/411a4a65868c484f8ab20a83fa29ba1b : SUCCESS in 1h 07m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb398ae0ff2448cbb3f0c137bed72a6c : SUCCESS in 31m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b2f2694a6f64ef9bd59057314e36279 : SUCCESS in 30m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c64a8ce16b5476ea8d6f5922df0d977 : SUCCESS in 1h 03m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd63af1c7bfb41dda2e932a48dd6d8df : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5b563fc5be54a71969fdd8972a0a82a : SUCCESS in 10m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92be33b0397c4dd393147cbfd2033bc6 : SUCCESS in 11m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/62a2abef082c4b87b7655a432d3ae8b8 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e05c1d77a4374b5fb771d2e800f48087 : SUCCESS in 30m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8cd59a793e1d40db908d3e491189d0f0 : SUCCESS in 21m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a53796086234ec7b16d3bbed906a6f3 : SUCCESS in 32m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ff63089e735c4b10a1b469c757aa8a21 : SUCCESS in 30m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3e8bca1fa74e45afbd69247bc1bb5420 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3a8c59cc0ef649e190fed5dcee516c06 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5e9a9d06c03642059041ea5ce99c20df : SUCCESS in 9m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/85b6271b62444d24887ea27b2070449c : SUCCESS in 8m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b5701b6a8056417388cdda246bd33952 : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/21ee67c3d14144a3ac6c232ee68f1cd4 : SUCCESS in 8m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef0556c5758f495e8c8a38cfe869339b : SUCCESS in 58m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5331ffba91714d3e96db3f1ccbb880d2 : SUCCESS in 22m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ec483d1bd6714050871446f830f86b15 : SUCCESS in 26m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44bd9faa19614eda8e3596076a0bbd19 : SUCCESS in 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7abd580deb6b4b0fb3bf3421338cb82f : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06bcef7e4ace478f8ea7de6607573003 : SUCCESS in 21m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88520058c14243f590852c099e876403 : SUCCESS in 41m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31b640cf18a44f4a8610883ca3d8be82 : SUCCESS in 34m 46s","accounts_in_message":[],"_revision_number":2},{"id":"51dc848c3827f9220f7aad876d087bc212f7bdd7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-21 14:53:40.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"50e4c6869c7f7faa4ad36eb190953be8616bd859","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-21 14:54:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9ee0c8fb8a0dc7607e221018bf6b5127ed9a843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-03 18:07:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f30fddeef1910c4b0af96a34231a23e9c9645d64","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-03 18:08:04.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c353b8bf17c44b2f368ffbf19cde4c77ccef5397","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-02-03 18:14:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a2649be507597f3df8844c80e1d1fe2ce317fa51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 19:17:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb119a418af341daa2fe5dd0d66f0d39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66866940490a4194b3d75793716d6829 : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/5aba1d4e3f174fb7b2e87851525ea8e6 : SUCCESS in 1h 03m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6673a0567d4d473f8a66a9816fbe2c56 : SUCCESS in 43m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2482e13c30ef4015b28fb65c6add893d : SUCCESS in 53m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0985846ae5e94243b123de9dc093888c : SUCCESS in 35m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0dec82f16f724484a3c3277d6117a96b : SUCCESS in 10m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6886ce6345334b1690fec3a2f423e355 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cddc329716d4426ab137a17e8325434 : SUCCESS in 20m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/43f4cd0fecc3484fb3d681a794b64660 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4fe192f9fdc04ab0a6c474888ff2d3bd : SUCCESS in 20m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6b235035713f40569c4c75fb98bdf2b9 : SUCCESS in 21m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f644b3ffca4747608ae2ddbe03cd1fbf : SUCCESS in 26m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/375e5f2902fb4c93b733daabdf347043 : SUCCESS in 31m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/661a971cf8a34ab4919fc3dda6d817e4 : SUCCESS in 12m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/da54afaab29d465f85064deb68e00a3c : SUCCESS in 12m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ba2c1bf8fe5d4166a1300aabd383e1ab : SUCCESS in 14m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89a3c51834e446d4a0e3dd06fb31d5c1 : SUCCESS in 5m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d58e17b45396468b95c2c060e2a8f0a0 : SUCCESS in 13m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/630ea0d60bb347f8911900ded3e9c55e : SUCCESS in 8m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4f09a4f1de54826b018d7cf8aa413f6 : SUCCESS in 54m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f34834e3ed84253baab1fb07e48eaa6 : SUCCESS in 35m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/117c5b6a6a6d4a429793e788e4803ada : SUCCESS in 49m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f96b91e67b2b47d4b1877c7cf1d61cec : SUCCESS in 28m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d264e96a36342d6a453d8ec1fe3c026 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d9da65013df54067877ab0b2ad8a984e : SUCCESS in 9m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/028aecd65c2f43699c865fb8cfa08533 : SUCCESS in 39m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2bc72e0be374301b6975797fd18fcb2 : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":3},{"id":"de16704cd922a4cb5b9b3b1e200b8abd1dad2e79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 19:20:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a727a0b974fb4f6f8c7e0b9b4e5b060f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3f37bc6b6a904d149e4da81fa6f49f39 : SUCCESS in 1h 07m 43s","accounts_in_message":[],"_revision_number":3},{"id":"ba8db663d7d40fac3115047363e649528870a565","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-06-22 17:35:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"155dc0b0149408eac723a41ddf731100b705b0b3","revisions":{"6996486497978dc591926877452f62b96adb8940":{"kind":"REWORK","_number":1,"created":"2025-09-29 21:24:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/962551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/962551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/962551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/962551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/962551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/962551/1"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-29 19:21:47.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-29 21:24:49.000000000","tz":-420},"subject":"tests: Work on more filesystems","message":"tests: Work on more filesystems\n\nThis gets more tests passing by lowering the per-xattr length for tests.\nSeen good effect on tmpfs on Linux and apfs on OS X; ext4 on Linux bumps\ninto trouble in some crypto tests, though, as there\u0027s a cap on overall\nxattr size, regardless of how we try to split it up across keys.\n\nChange-Id: I0314e9b78449cb71d7700663520c4ddf030ec047\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6996486497978dc591926877452f62b96adb8940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6996486497978dc591926877452f62b96adb8940"}]},"branch":"refs/heads/master"},"99e0914b957c3481c7f323caa4d940be22e88ccd":{"kind":"REWORK","_number":2,"created":"2026-01-20 16:35:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/962551/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/962551/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/962551/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/962551/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/962551/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/962551/2"}}},"commit":{"parents":[{"commit":"732fd859d8e0a0b795fbb42ed8390ab4cd494227","subject":"Fix flakey db_replicator unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/732fd859d8e0a0b795fbb42ed8390ab4cd494227"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-29 19:21:47.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-01-20 16:34:40.000000000","tz":-480},"subject":"tests: Work on more filesystems","message":"tests: Work on more filesystems\n\nThis gets more tests passing by lowering the per-xattr length for tests.\nSeen good effect on tmpfs on Linux and apfs on OS X; ext4 on Linux bumps\ninto trouble in some crypto tests, though, as there\u0027s a cap on overall\nxattr size, regardless of how we try to split it up across keys.\n\nChange-Id: I0314e9b78449cb71d7700663520c4ddf030ec047\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99e0914b957c3481c7f323caa4d940be22e88ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99e0914b957c3481c7f323caa4d940be22e88ccd"}]},"branch":"refs/heads/master"},"155dc0b0149408eac723a41ddf731100b705b0b3":{"kind":"REWORK","_number":3,"created":"2026-02-03 18:07:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/962551/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/962551/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/962551/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/962551/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/962551/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/962551/3"}}},"commit":{"parents":[{"commit":"732fd859d8e0a0b795fbb42ed8390ab4cd494227","subject":"Fix flakey db_replicator unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/732fd859d8e0a0b795fbb42ed8390ab4cd494227"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-09-29 19:21:47.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-02-03 18:07:01.000000000","tz":-480},"subject":"tests: Work on more filesystems","message":"tests: Work on more filesystems\n\nThis gets more tests passing by lowering the per-xattr length for tests.\nSeen good effect on tmpfs on Linux; ext4 on Linux bumps into trouble in\nsome crypto tests, though, as there\u0027s a cap on overall xattr size,\nregardless of how we try to split it up across keys.\n\nDrive-by: Swap the xattr_size kwarg to write_metadata for a module-level\nconstant; no callers used the kwarg anyway.\n\nRelated-Change: Ie22db08ac0050eda693de4c30d4bc0d620e7f7d4\nChange-Id: I0314e9b78449cb71d7700663520c4ddf030ec047\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/155dc0b0149408eac723a41ddf731100b705b0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/155dc0b0149408eac723a41ddf731100b705b0b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
