)]}'
{"id":"openstack%2Fswift~962553","triplet_id":"openstack%2Fswift~master~I4d0f01157018405deb0f27bfb542fec3704c1313","project":"openstack/swift","branch":"master","topic":"cohanaja-add-skip-commits","attention_set":{},"removed_from_attention_set":{"38368":{"account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"last_update":"2025-10-03 04:15:36.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_38368\u003e by clicking the attention icon","reason_account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}}},"hashtags":[],"change_id":"I4d0f01157018405deb0f27bfb542fec3704c1313","subject":"replaces --force-commits with a --skip-commits flag to determine when to comit object updates in smsr","status":"ABANDONED","created":"2025-09-29 23:24:16.000000000","updated":"2025-10-03 04:16:49.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5c73876e386af7ba65873612ad90b163f950f336","_number":962553,"virtual_id_number":962553,"owner":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-09-30 00:33:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":-1,"date":"2025-09-30 05:31:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-30 00:18:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-30 00:33:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-30 05:31:39.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"95cf7d6b884b86ebe5b08ce3e60d08cd56188939","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-09-29 23:24:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3a660fdbf1828789a699456e6c8815677ab1442","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 00:18:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2e42600fa9f4b0ab2073fe07dbe0093\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/499578491c574b86a6f951ede25299d3 : SUCCESS in 52m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a496f97bdafeb06130b2ad2372649c84d0ad6bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 00:33:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0c9a43d39043c095f87ab51f840b44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c4d9a42f2d4f88863974637a665113 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/51b1e921134b4496bb7e0b36ea67a24d : SUCCESS in 1h 04m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/244ff77d252b44e0b4a79ed55ff55a65 : SUCCESS in 57m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ac7ce92e23e14a0383f9e3babbbc1d06 : SUCCESS in 46m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75c7a86fd2bd4f3180d0ad9d2b25715e : SUCCESS in 58m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53b0f5620ebe4ceda111d6a49cf36bfc : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b8dddbeca294da696db8430db9ae15f : SUCCESS in 12m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe4ac2deb3da4f6aa5ff9fa6126bf81c : SUCCESS in 12m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb72ff13cc2546a89e29524b9a02b8c3 : SUCCESS in 4m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/57962287d9a8421d9f90380f497a38d5 : SUCCESS in 23m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3ba522c1339f45e0855af31b9e488a94 : SUCCESS in 8m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a6e898fa28042ab97e4bb9a8fff8175 : SUCCESS in 27m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/749a490e2e2d43478972a1dd1d798c76 : SUCCESS in 18m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/86183ca5e2324b9396149ba3e5bc11ec : SUCCESS in 14m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9e86eb1113454bd98aeaa420b24f1652 : SUCCESS in 12m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2a462f999c7349759aa6b5e316fe8350 : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/643d57e5482246809ea6bf87f4590a1f : SUCCESS in 6m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb835201b0fc4538b2fa44acef570f88 : SUCCESS in 13m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/105fced0007f4629a8f0b6f5bae57871 : SUCCESS in 10m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/392eecf63ac841f68693c7b1c314ac31 : SUCCESS in 33m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/219a3e94b3d5420ba418190b700a40fb : SUCCESS in 56m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f14f6abc16294943ba54e53b8ec10730 : SUCCESS in 21m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beb6b53ec3864fc9916f655d7a937644 : SUCCESS in 31m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a21f4c1a537848988b0705f9bb3fcf18 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa7df7e11ba14af6848c37df44a10e61 : SUCCESS in 19m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bb9e49b4a8514bcda796a43c82cb749b : SUCCESS in 53m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/535052998c544185a7730fe37300b364 : SUCCESS in 53m 45s","accounts_in_message":[],"_revision_number":1},{"id":"d9879f94d6abf327feb0e74aacc517fee7e64189","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-09-30 05:31:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c73876e386af7ba65873612ad90b163f950f336","tag":"autogenerated:gerrit:abandon","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-10-03 04:16:49.000000000","message":"Abandoned\n\nAccidentally ran git commit -s --amend which might\u0027ve modified the Change-ID and made this duplicate change, apologies!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9297a4c297e9588c7436da4b7039d852b542ea8a","revisions":{"9297a4c297e9588c7436da4b7039d852b542ea8a":{"kind":"REWORK","_number":1,"created":"2025-09-29 23:24:16.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/53/962553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/962553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/962553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/962553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/962553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/962553/1"}}},"commit":{"parents":[{"commit":"b035ed1385ee729411adbd459bb03033e90bb13a","subject":"add Christian O to AUTHORS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b035ed1385ee729411adbd459bb03033e90bb13a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-26 22:52:02.000000000","tz":-420},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-09-29 23:24:05.000000000","tz":-420},"subject":"replaces --force-commits with a --skip-commits flag to determine when to comit object updates in smsr","message":"replaces --force-commits with a --skip-commits flag to determine when to\ncomit object updates in smsr\n\nChange-Id: I4d0f01157018405deb0f27bfb542fec3704c1313\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9297a4c297e9588c7436da4b7039d852b542ea8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9297a4c297e9588c7436da4b7039d852b542ea8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
