)]}'
{"id":"openstack%2Fswift~962608","triplet_id":"openstack%2Fswift~master~I5b25b7e830a31f03d11f371adf12289222222eb2","project":"openstack/swift","branch":"master","topic":"token","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-10-01 15:51:44.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I5b25b7e830a31f03d11f371adf12289222222eb2","subject":"doc: specify seconds in proxy-server.conf-sample","status":"MERGED","created":"2025-09-30 17:34:14.000000000","updated":"2025-10-01 16:34:28.000000000","submitted":"2025-10-01 16:33:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"962608-token","meta_rev_id":"0bb7b4daf4ccfb7ebe47a30ba922520dd3b078de","_number":962608,"virtual_id_number":962608,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-01 16:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-10-01 15:51:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-01 15:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-01 15:24:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-30 18:21:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-30 18:52:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-01 15:24:08.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"0f3841a617add2af40f197a1aee9dc4a70041b5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-09-30 17:34:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84b30dd5de515dd551979d61654e47e4cf2d2bb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 18:21:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8a86b13ef3049f29a9a5ac2228341d6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ccf74173be994d1eac18ac7955f735d7 : SUCCESS in 42m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c051d7746b00d6c807f1ba38a2aff1eb608a5101","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 18:52:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b63dc1a5cfe04f23b816ac540ffc12be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0635e2c783784626b3bd745ff1cc4fa5 : SUCCESS in 6m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/ee0ab0900ec543fca7781f90ea550a71 : SUCCESS in 31m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5cc0a55c126d4f9a9903cf7d93952a7c : SUCCESS in 52m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ec10953238740e98adae89afb8ef724 : SUCCESS in 33m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b26d2f029184d8d9b83541bd553858b : SUCCESS in 1h 04m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a724e68ec32469296b2bf2375ab1600 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09ebf75b219c4b10906654bfe51e673d : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/337283f9f6764bef94a2c12e62211c0e : SUCCESS in 14m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e8ce116cc3e24cd8918c7d5f0926e831 : SUCCESS in 8m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d4c42662f7314371b590f8b49edf06d2 : SUCCESS in 21m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/95afbe2c746e4784a393ca6fc667c59b : SUCCESS in 12m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/40ed6ab952d149fdb790754c40224eb8 : SUCCESS in 10m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5d167dcbb72743b2a1f541e56b5ada55 : SUCCESS in 17m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fca6182a391c46428fc0d1172d16c3a5 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b4874031e6ae477a9569072cf66dcc6c : SUCCESS in 14m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c114965f248547138546e7decea8a49c : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d4931bad127a46caa9162985de30fef6 : SUCCESS in 3m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c88cf95f2d2b47a19a8bc632abc27099 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/335a653b671b4e998fc838c16b517f99 : SUCCESS in 10m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48a7c2fd4de84ebca2bd523c406dbc50 : SUCCESS in 1h 12m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e79bd2c1e6134a2bb2dbd49193d4ae0c : SUCCESS in 41m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/25d4795b2b504e2f80d0bd8540339955 : SUCCESS in 57m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f4d56bf986c46158205c6077aca407a : SUCCESS in 29m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf07a80554a8441ba6c7c6fd6c107808 : SUCCESS in 19m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33c7304781504c04b819c5b2ee87dbe7 : SUCCESS in 47m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b81d50155ab4ad3a16013105a7cb53c : SUCCESS in 45m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7f649609aef3a7c177b8b57382397affcbbbb513","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-01 15:24:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e18d16a679bc1b6024a51bf3bed80350aa26e514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 15:24:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b070f6bd9ddf01b0773b0c974b91e93efd107cf7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-10-01 15:51:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ff518d8b049310b22de5e49a7e1b55506e53e03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 16:33:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/058905d184f941e5813764c66ac7c76c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e991625317f64f24b7e7e0b4ffeac795 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/dafdf19b13e3426c85fde7d46483d8d6 : SUCCESS in 1h 03m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/328b42afae8143219321db99fabba020 : SUCCESS in 35m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/54283c4b0fd24143b952368bd138b81e : SUCCESS in 33m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04ed5ac3492641b7a91a5931db601d67 : SUCCESS in 1h 00m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/31d7c26dce00481f9c1280272fe261b7 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b210f25a545f4fe1aefe67dfabcfae92 : SUCCESS in 10m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/292ff35626564df5b97daa1d9c6751af : SUCCESS in 23m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/905fc527eaeb4ce28abdbea50ec345c0 : SUCCESS in 32m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/11b67ccfc05f448192eb0fda905f33b2 : SUCCESS in 21m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f8053897d8f143b593f74d4ff64fb941 : SUCCESS in 12m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6b0d88da05964c188f79563f8154917c : SUCCESS in 15m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/28ace0b80cf54d938be484882fb09f2c : SUCCESS in 15m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8a1e71b52c05462caae12c82e0c29e9f : SUCCESS in 6m 17s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ebe81167e7ad4101a0c997b5bb10b1cc : SUCCESS in 6m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a88471645c141d0bc8203ce9c0d4d52 : SUCCESS in 33m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/355c9f0a12434e539ec4cf659cfc9f5d : SUCCESS in 48m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab4d77bfbbfb415b8d9883c3f82f4f4e : SUCCESS in 36m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df1e138dd1594caf92df472b616a06e0 : SUCCESS in 13m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c84c52f90ca465fb8709faec8392932 : SUCCESS in 16m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b53c2a6aa86b462785ac7889d13211f1 : SUCCESS in 41m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e7bc01b0207444b846d4eddadb4698d : SUCCESS in 54m 39s","accounts_in_message":[],"_revision_number":1},{"id":"860daa8fd64bf44980f9df4e7ffc43dd4a6a4b65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 16:33:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0bb7b4daf4ccfb7ebe47a30ba922520dd3b078de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 16:34:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cbf3a025fbf49b5abc8de7be1907460\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc2ed6c08fb46109f5dfcb5ec1ec7bf : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/050c9b905bbc46b98f1d5b069236a2ce : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"389747a8b2419dd14bc672afd395bb8e65262c60","revisions":{"389747a8b2419dd14bc672afd395bb8e65262c60":{"kind":"REWORK","_number":1,"created":"2025-09-30 17:34:14.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/08/962608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/962608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/962608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/962608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/962608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/962608/1"}}},"commit":{"parents":[{"commit":"d9883d083409baac3db44e1db14bf3c79a75f411","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9883d083409baac3db44e1db14bf3c79a75f411"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-30 17:32:18.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-09-30 17:34:05.000000000","tz":-300},"subject":"doc: specify seconds in proxy-server.conf-sample","message":"doc: specify seconds in proxy-server.conf-sample\n\nMost of swift\u0027s timing configuration values should accept units in\nseconds; make this explicit in the sample config for values that did not\nalready do so.\n\nRelated-Change-Id: I38c11b7aae8c4112bb3d671fa96012ab0c44d5a2\nChange-Id: I5b25b7e830a31f03d11f371adf12289222222eb2\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/389747a8b2419dd14bc672afd395bb8e65262c60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/389747a8b2419dd14bc672afd395bb8e65262c60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
