)]}'
{"id":"openstack%2Fswift~962874","triplet_id":"openstack%2Fswift~master~I3b19eff8e64fbc81a997b3cb413b1b82ef35a8fb","project":"openstack/swift","branch":"master","topic":"realtime","attention_set":{"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2025-10-02 20:10:16.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I3b19eff8e64fbc81a997b3cb413b1b82ef35a8fb","subject":"Test for invalid and non-existent bucket names","status":"NEW","created":"2025-10-02 19:58:04.000000000","updated":"2025-10-09 00:12:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cc4ee0607cc21bfe2dddd4862ef0a27781ebfaf8","_number":962874,"virtual_id_number":962874,"owner":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-08 22:29:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-02 20:10:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-02 20:54:57.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"309bee573b81f3a4d7b2b5b1f0467d19e29ca28f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-10-02 19:58:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e57142c05d5692786d73a0d6fbabe7601b24702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 20:10:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1fdc184afe649c1a5344d43227dc884\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3083b1d31d4676931e9c5fed11abdd : SUCCESS in 5m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b22534f7463a48da87bc867f04055b70 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f847fe164eef4d669b61cc8691401aa5 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc5fc58d9c445d1a8374d5816a869fb : FAILURE in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"920fbc192d4d603a679407261c7fdd3a0b7ce8dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 21:13:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360c110a45db47aaadbd84c29dfda080\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7585acc5d0e4acc9628f7bace8627cb : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e627399b6951947b99b7a76c05d926c99abfdf24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-08 15:48:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_35790\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"_revision_number":2},{"id":"bcd512a90b2193e85c9d89987da99fff756cb558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 16:02:48.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94e6dbb472a3437ba97d494fe3e2017d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005ebcd946fc4667bda7b8256f043e82 : SUCCESS in 6m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1719ddb9245042479ab2f2a7ff6de7aa : SUCCESS in 6m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4aa09a875847426fb32f65edb4bee94a : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bddbc882cfb64c81be9e10c1b3153be6 : FAILURE in 4m 46s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/proxy_logging.py\n  Comments left for invalid file swift/common/middleware/s3api/utils.py\n  Comments left for invalid file test/unit/common/middleware/test_proxy_logging.py","accounts_in_message":[],"_revision_number":2},{"id":"623305c1c6e3d1aca4eb93edc012866dab6a2e32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 17:12:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f593cba73ad44f88555347d12544fec\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d9d8ff73287a4b4b83be296ddcbfaa50 : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":2},{"id":"84408a3b732daa7b530efa920903426486aa1757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-08 22:14:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_35790\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"_revision_number":3},{"id":"7ee4e35b75ee17fe7781883366ff1220af415933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 22:29:04.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c46f6d1feba4e0c961a6597d968d78b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e2e7b8a74d49a79376e2cdcf24a9e6 : SUCCESS in 9m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b533f34ba1a4843b9ed5656c8454e72 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ab2dbe9cf0fd4ccda7af8ef7e35d6482 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f803f675a3b94bc29e04b6a671380970 : FAILURE in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"cc4ee0607cc21bfe2dddd4862ef0a27781ebfaf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 00:12:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22cd7e506063425abfe2b9ca09651ed1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ff94ef38a3084a949b220b7ff3de332e : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"749eef6fee908de4437c6aa8e518871e5e9cae6e","revisions":{"546d2cbee0c4110e117b5497b21fc6d89c795455":{"kind":"REWORK","_number":1,"created":"2025-10-02 19:58:04.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/74/962874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/962874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/962874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/962874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/962874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/962874/1"}}},"commit":{"parents":[{"commit":"1e61e565c14700ee7df5490fb6ebbed672045479","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e61e565c14700ee7df5490fb6ebbed672045479"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-02 19:55:00.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-02 19:57:24.000000000","tz":-300},"subject":"Test for invalid and non-existent bucket names","message":"Test for invalid and non-existent bucket names\n\nChange-Id: I3b19eff8e64fbc81a997b3cb413b1b82ef35a8fb\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/546d2cbee0c4110e117b5497b21fc6d89c795455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/546d2cbee0c4110e117b5497b21fc6d89c795455"}]},"branch":"refs/heads/master"},"5fb11bba302ae3061c1f8003c6aa01c655c5827c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-08 15:48:52.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/74/962874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/962874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/962874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/962874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/962874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/962874/2"}}},"commit":{"parents":[{"commit":"0f7719469d92e067c30c5f64e77f2c6aeb0abae7","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f7719469d92e067c30c5f64e77f2c6aeb0abae7"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-02 19:55:00.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-08 15:48:52.000000000","tz":0},"subject":"Test for invalid and non-existent bucket names","message":"Test for invalid and non-existent bucket names\n\nChange-Id: I3b19eff8e64fbc81a997b3cb413b1b82ef35a8fb\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5fb11bba302ae3061c1f8003c6aa01c655c5827c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5fb11bba302ae3061c1f8003c6aa01c655c5827c"}]},"branch":"refs/heads/master","description":"Rebase"},"749eef6fee908de4437c6aa8e518871e5e9cae6e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-08 22:14:29.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/74/962874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/962874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/962874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/962874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/962874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/962874/3"}}},"commit":{"parents":[{"commit":"0a724c02fbdf5c065f6d0a72b6ed2753379ee095","subject":"proxy-logging: Add real-time transfer bytes counters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a724c02fbdf5c065f6d0a72b6ed2753379ee095"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-02 19:55:00.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-10-08 22:14:29.000000000","tz":0},"subject":"Test for invalid and non-existent bucket names","message":"Test for invalid and non-existent bucket names\n\nChange-Id: I3b19eff8e64fbc81a997b3cb413b1b82ef35a8fb\nSigned-off-by: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/749eef6fee908de4437c6aa8e518871e5e9cae6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/749eef6fee908de4437c6aa8e518871e5e9cae6e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
