)]}'
{"id":"openstack%2Fswift~963168","triplet_id":"openstack%2Fswift~feature%2Fmpu~Iafc87fe7fc17dad155de171688ce24935eecfd89","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","hashtags":[],"change_id":"Iafc87fe7fc17dad155de171688ce24935eecfd89","subject":"merge master to feature/mpu","status":"MERGED","created":"2025-10-06 11:31:47.000000000","updated":"2025-10-06 13:45:57.000000000","submitted":"2025-10-06 13:44:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963168-merge-master","meta_rev_id":"afc1581657851292a69cd664fe4eb33bc58646d7","_number":963168,"virtual_id_number":963168,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-06 13:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-10-06 12:54:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-10-06 12:54:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-06 12:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8710cb17f3a732d944ec03a3d8a84764895142bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-06 11:31:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5e7d0881c44bcc0b7683abb4a2039b57ae670b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 12:37:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c93b3dd6c3b14909b20b9e8d952341ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d6ed0375f74657bfdd1afe345f93a6 : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ba2f96cd82e4a019bc3e0c27279e7f1 : SUCCESS in 3m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6530769546d94da686cfcdcb16a28a12 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a5bd3eb7b0d45c284dee4b96dbb086d : SUCCESS in 21m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e957d3cb54d4b2a970875ef9a03905a : SUCCESS in 15m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34245970e35940c984a6b19bf5c43928 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/62ccb404676044bbb9e50f5af05c129c : SUCCESS in 11m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6c4711ce36974faeb19d0422256f661b : SUCCESS in 13m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0061bae8456940d3a8d81f7badb4ccac : SUCCESS in 19m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f0f3e9f7eaa34f8d99366a5d36c641f8 : SUCCESS in 18m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bd8fe5ccf468483780e6e7f698de6858 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b1655c74498f4c139b3d6674a156a02c : SUCCESS in 15m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/daf1218dee8641808f6329e82e823aeb : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9945ee82dde24d2eadd380a1ef70b3f8 : SUCCESS in 6m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc1f357519b44cc4b53fd83ec2365be9 : FAILURE in 15m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/16b4e48839164d9f8181481e0d830c1e : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/62ea031551cb46b1866daf9d6eb7c8d3 : SUCCESS in 50m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b87227d14bb46e8be1bee4507d46752 : SUCCESS in 27m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/764c03d7344e4012907c9c595e392da1 : SUCCESS in 7m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/effc4a56b4964c9782155cc9ccb1506e : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ca75cf399a526c2195555c6506f5956065b58b32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 12:41:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/019cf57f9d3a402cab20cc94e944947d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8e9fd8748bb649ea99e922859ebbe241 : SUCCESS in 1h 07m 03s","accounts_in_message":[],"_revision_number":1},{"id":"635883d98ba69dfb44f7f59c17b73fd02fa0e48a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-06 12:54:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d239b74dfc0f5733ee91c49326c7a934b75c08bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 12:54:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ff44e49b74baaf3d57e800d349be2ac706ba633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 13:44:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4f776b51bc7457e99654d6f4888080f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c065484c924fe1b3aaf153bb95de49 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c25c81a78acf49a98ea22693ea4ff68e : SUCCESS in 7m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ef3e48e753cf4c7f82289a4ac6f2059a : SUCCESS in 7m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f4d855748cc5440892120f3612fccf96 : SUCCESS in 19m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5aa39303543d44938b29ba9529afad62 : SUCCESS in 21m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dd596b92c3ec4705b2247881d278e1e8 : SUCCESS in 29m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/41d6c2ca9b3e4f519a06e81e5c23ac3c : SUCCESS in 35m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e09ef300598f47b6874aedd1ce3a5557 : SUCCESS in 10m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8f6d3c79d8c4408fb9e141a591d1d17e : SUCCESS in 9m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ce1a66f26f67412eb84dcc802b8b4009 : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/efa203a8185c47c795012f88a00f3ab8 : SUCCESS in 5m 32s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/04255e9c401444328b60322aaa89eef8 : SUCCESS in 9m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9bfae029825410aa924644979fdb978 : SUCCESS in 44m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0473a89ae5b54a729a6a121ca482fd04 : SUCCESS in 19m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5bbc3b6da9447fc9484adb4cc48eb5a : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7af06b16d8834f21a416699f4a454fdd : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6cc0217e5e35b6f935acf61577b36bffb6a86190","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 13:44:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"afc1581657851292a69cd664fe4eb33bc58646d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 13:45:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa592a9dbcb4456f8a254cf76481f3bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41660d2a022a4638ba4715dc220fde65 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1333cc9e2e7b4274991ce03eac331689 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/88a9b7f1cf1446e98bfad15b7e2bf3e2 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c64899419720c3428fe0e0d5478da813f2ae2e3","revisions":{"1c64899419720c3428fe0e0d5478da813f2ae2e3":{"kind":"REWORK","_number":1,"created":"2025-10-06 11:31:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/68/963168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/963168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/963168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/963168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/963168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/963168/1"}}},"commit":{"parents":[{"commit":"34277497a4bf7723885300e53c6135afa38cb931","subject":"mpu: more compat tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34277497a4bf7723885300e53c6135afa38cb931"}]},{"commit":"ac5c783d652e66f2f19cdb1d8ba29b5dc3c51710","subject":"Merge \"Assert metadata of SLO PUT from container sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac5c783d652e66f2f19cdb1d8ba29b5dc3c51710"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-10-06 11:31:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-10-06 11:31:40.000000000","tz":60},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: Iafc87fe7fc17dad155de171688ce24935eecfd89\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c64899419720c3428fe0e0d5478da813f2ae2e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c64899419720c3428fe0e0d5478da813f2ae2e3"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
