)]}'
{"id":"openstack%2Fswift~963325","triplet_id":"openstack%2Fswift~master~Ie0efee33e684b1c5bad6ee2191c187bb680de5f1","project":"openstack/swift","branch":"master","topic":"fix-swift-dir","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-10-14 23:48:18.000000000","reason":"Change was submitted"},"2622":{"account":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"last_update":"2025-10-14 23:48:18.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-10-14 23:48:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0efee33e684b1c5bad6ee2191c187bb680de5f1","subject":"Fix swift_dir setting in WSGI servers","status":"MERGED","created":"2025-10-07 19:43:48.000000000","updated":"2025-10-14 23:49:33.000000000","submitted":"2025-10-14 23:48:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963325-fix-swift-dir","meta_rev_id":"ac90fd07a89b97d01ec34719291ae40b18675e9c","_number":963325,"virtual_id_number":963325,"owner":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 23:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-10-07 23:02:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-10-14 22:12:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-07 20:51:18.000000000","updated_by":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-10-07 20:51:18.000000000","updated_by":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-10-07 20:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-07 21:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-07 23:02:45.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"0d6ba44dd5fdf452c874cac1d2a43b79dfaba7ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2025-10-07 19:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f17506704315704424ebe2994b14861ec3854be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 20:53:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c365c6f7bae04f6daf22bcf93eba7b08\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac818ab53f8149a68432d9bdc400fa1f : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":1},{"id":"62a498f7b29aa78b4936afeb56dde08f7e4ff9e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 21:03:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38d792c14e584aa4869888195449449d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03263c095ef411c956757c98d0ddc51 : SUCCESS in 10m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/c1e7c966ff9f474ca5499d871f196fbc : SUCCESS in 1h 01m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/05d999fe921e4c2d84fdc63959723851 : SUCCESS in 51m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d5854192220340709510636b41521aec : SUCCESS in 28m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/487343389de5400d8324a5d209b6cb2a : SUCCESS in 50m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76beb2c3200a4b7cba141faa918b5c5c : SUCCESS in 10m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f65e107cac745369e87acc5ff30df20 : SUCCESS in 10m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7f026f3cd8e43bba33e16936ebb332b : SUCCESS in 17m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77235d5c946142d3b597ffe6f5d302f3 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bab76f173e9c4974816def40337ef3a5 : SUCCESS in 26m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3fe22c2d7ecf462681a0c16eb6a3e701 : SUCCESS in 20m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b6113f31901f4ebba785e41e861b1c70 : SUCCESS in 27m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/29f6a7d986fd47e79d8f1a2e138cee1f : SUCCESS in 30m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/36f78b9326ce47fe9c79d6fc3005bd0c : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5c65bf1d78aa454ea9c1f372750b5915 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/35490f646d3246f5a679f588c96d55fc : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8b6c4b8b3c77471f94116765190171a2 : SUCCESS in 6m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/67da06821d1a4e96bac87dd0d4345b1a : SUCCESS in 14m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4b1a5e24cf874d66ae687b9a2d406437 : SUCCESS in 11m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2b55b92648245db93fddfce86e75f8e : SUCCESS in 54m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bda47f4167524fcdb01968589a25f576 : SUCCESS in 54m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/efdfc1c328fb4020a188e7096a4a5752 : SUCCESS in 52m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aff1a3ca41544c92869fe6a08e481b08 : SUCCESS in 28m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc1808c768f64e1ba80dea4733a83210 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9051bf515aee49df8f14135c8547f9a0 : SUCCESS in 20m 14s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fe3f2a05833455a8f8a89bab8bae25c : SUCCESS in 54m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0331301bf75a4c9199d35cb2ae749107 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":1},{"id":"da1027deb3411e2464c6a62c2fafc8f640f9ecea","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-07 23:02:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68bcf6feb0d11e1d81db5b2cea2afdfc806bc902","author":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"date":"2025-10-08 18:47:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69a63bc83e4005457a9eaafb967dabe05dc12aba","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-14 22:12:05.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bde1a5bd6cee5df2b451232827069a58f9e33a5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 22:12:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f283b0148d984fbbcd9504937d2c65182ca6df0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 23:48:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5a9e9f395ad47158c6d47ea3a94362c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5986831ec7814f99ba7818d72714b419 : SUCCESS in 11m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/5c87c460c62c4cf6a6dc262868e89e0e : SUCCESS in 45m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9c5259b135f94072bd50a5ccf1ae2204 : SUCCESS in 33m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/598fdcc3083e408a8bca066d52e6a314 : SUCCESS in 39m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e4314c485ca4c0db9560c0219dfb190 : SUCCESS in 32m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/03b3b7e0c4f94c3b9e34a03da0522d94 : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aa660ef4adff4f629eb347fe2b6e9781 : SUCCESS in 28m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd25241345b94722bd9adbf739ff8fc7 : SUCCESS in 17m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/49c1269ebb2b4128a70281e4f25669e9 : SUCCESS in 11m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f829a44a8daf4f2ab82932b4d8323213 : SUCCESS in 22m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/62e24da6a8144ccdb53aba52950f3992 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/22bfbe917b8b435d855573ae055accd1 : SUCCESS in 8m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2a393a1ebe27468db80feae2c61628d7 : SUCCESS in 8m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93f25a270875465ab7cde89f2794adac : SUCCESS in 4m 07s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f6c0220dcee843a2a03e46711b114745 : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/843baad5ccb5425ebf5f7fefba9b8824 : SUCCESS in 54m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e14ea1e41bd4f4a9365a9f77c39090b : SUCCESS in 22m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e9ee13f6d8c4321a5b0d9e68b42e256 : SUCCESS in 24m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf2993cdade04bc68a9ea0898de24673 : SUCCESS in 27m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8af383bd8684b4cbc9da94c7dcffd1e : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a22e6c46e89a4fb7a94cb7d378c93af5 : SUCCESS in 18m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cd5197826da47a7903b92ff618f866a : SUCCESS in 38m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61d0a504fce044d78a08aaa6a667b642 : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f1037329a6d7063e81f14cf465be29470eb7f874","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 23:48:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac90fd07a89b97d01ec34719291ae40b18675e9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 23:49:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f19e2e9d1634673a257079e487d42de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f442ee55e3a4b1faeec5895c741fb6e : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8238deb30f4346388ee7fce9cfcb2942 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5568dd09b5d1bd2c7d04c9bf7d0f670a9df298c6","revisions":{"5568dd09b5d1bd2c7d04c9bf7d0f670a9df298c6":{"kind":"REWORK","_number":1,"created":"2025-10-07 19:43:48.000000000","uploader":{"_account_id":2622,"name":"Samuel Merritt","email":"spam+launchpad@andcheese.org","username":"torgomatic"},"ref":"refs/changes/25/963325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/963325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/963325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/963325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/963325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/963325/1"}}},"commit":{"parents":[{"commit":"97e00e208fc6a1076bef41ba28a17cda8974aeb0","subject":"trivial: reorder AccountBroker.path methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97e00e208fc6a1076bef41ba28a17cda8974aeb0"}]}],"author":{"name":"Samuel Merritt","email":"smerritt@nvidia.com","date":"2025-10-07 19:13:37.000000000","tz":-420},"committer":{"name":"Samuel Merritt","email":"smerritt@nvidia.com","date":"2025-10-07 19:43:31.000000000","tz":-420},"subject":"Fix swift_dir setting in WSGI servers","message":"Fix swift_dir setting in WSGI servers\n\nTheoretically, the various WSGI servers should be able to operate on a\nsystem without /etc/swift/swift.conf. However, this doesn\u0027t actually\nwork. WSGI servers call utils.validate_configuration() before looking\nfor a swift_dir option, and that validation reads swift.conf from its\ndefault location. Even if you set swift_dir\u003d/some/where/else, the WSGI\nservers require /etc/swift/swift.conf to exist.\n\nThis commit makes the WSGI servers call utils.set_swift_dir before\ncalling utils.validate_configuration.\n\nMotivation: I\u0027m working on testing some client software against actual\nSwift, but my CI environment doesn\u0027t have /etc/swift at all, so the\ntest suite can\u0027t start the Swift daemons.\n\nChange-Id: Ie0efee33e684b1c5bad6ee2191c187bb680de5f1\nSigned-off-by: Samuel Merritt \u003csmerritt@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5568dd09b5d1bd2c7d04c9bf7d0f670a9df298c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5568dd09b5d1bd2c7d04c9bf7d0f670a9df298c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
