)]}'
{"id":"openstack%2Fswift~963647","triplet_id":"openstack%2Fswift~master~I12d52e84aa7a4a833bfba2654a2b176d969c00bd","project":"openstack/swift","branch":"master","topic":"cohanaja-cycle-time-recon","hashtags":[],"change_id":"I12d52e84aa7a4a833bfba2654a2b176d969c00bd","subject":"sq? more tests","status":"ABANDONED","created":"2025-10-10 03:27:27.000000000","updated":"2025-10-11 03:28:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a97cc6589a732cb329e5cabdf7a8fffc785fe703","_number":963647,"virtual_id_number":963647,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-10 04:14:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-10 04:14:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4fe8d1d095026dd31aeed24586390d5cee16e04b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-10 03:27:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c063dff8e7bf48e18d6cee0c83dfe7531fea8963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 04:14:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95a5f73adb2a46de83cf84844af3b836\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4812e272c1f4d9bad6496c90d5b647f : SUCCESS in 8m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/030bcb204d674f86a620b9aa5cd4792e : SUCCESS in 10m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd93cfe32b3e4e1788c0fc678105d15d : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1830d815c39a48379835965a681f4d11 : SUCCESS in 18m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b07ec7db01f47dd8ef9fb5842a15b97 : SUCCESS in 7m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3c7d7c7b61494d48bbb97bbddba2970e : SUCCESS in 20m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/90425cc1066a4b95a7785f92ab50fa55 : SUCCESS in 20m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5beeeface9a84ff4bdb4d39cf7f7b04c : SUCCESS in 19m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b9f7b8b10823402b82643ec3e266b0bd : SUCCESS in 11m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ac3cf67ac79c4a52a99030f5c5f7f265 : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dcf342f741484574a167dd626cc06fbc : SUCCESS in 14m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6c26119b0a8b46f78174d7d477d6cb1d : SUCCESS in 14m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cfd1b39f26c44ed0a6a90bc193436fe6 : SUCCESS in 47m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ce4438b0c8d4da5b5023d2aaa6b5c1b : SUCCESS in 32m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4df2abf9f2154b5c809af2b1c26dea29 : SUCCESS in 25m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f09d5a5d56f404b9ad55030c5399944 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e741917bc17a46f784305bda4b8de903 : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7533f348928a4123a0dcb17d75b2520442d5b8d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-10 04:45:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbc4ce6898ca4f818a4ac8710c3c8824\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2d3b4e7d17f492cb9fc5313add7a550 : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a97cc6589a732cb329e5cabdf7a8fffc785fe703","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-11 03:28:25.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/962305/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64a8222122f7e90f5a28bbd7c46e43c977d52dd2","revisions":{"64a8222122f7e90f5a28bbd7c46e43c977d52dd2":{"kind":"REWORK","_number":1,"created":"2025-10-10 03:27:27.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/47/963647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/963647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/963647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/963647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/963647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/963647/1"}}},"commit":{"parents":[{"commit":"a67a64d0de25e6c42ae7d957da2db2f66996b4e5","subject":"Adds processing time metric to sharder recon logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a67a64d0de25e6c42ae7d957da2db2f66996b4e5"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-10-10 01:37:12.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-10-10 03:22:08.000000000","tz":-420},"subject":"sq? more tests","message":"sq? more tests\n\nChange-Id: I12d52e84aa7a4a833bfba2654a2b176d969c00bd\nSigned-off-by: Jianjian Huo\u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64a8222122f7e90f5a28bbd7c46e43c977d52dd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64a8222122f7e90f5a28bbd7c46e43c977d52dd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
