)]}'
{"id":"openstack%2Fswift~963877","triplet_id":"openstack%2Fswift~master~I1e53d171faff02e2dbcbcc779ad3a47506b26853","project":"openstack/swift","branch":"master","topic":"comn-db-log","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-10-14 11:21:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e53d171faff02e2dbcbcc779ad3a47506b26853","subject":"test-db-replicator (trivial): just one tmpdir","status":"MERGED","created":"2025-10-13 17:46:45.000000000","updated":"2025-10-14 11:22:55.000000000","submitted":"2025-10-14 11:21:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963877-comn-db-log","meta_rev_id":"9ef98f705e9beaa055f583f301cd048421e5467e","_number":963877,"virtual_id_number":963877,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 11:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-10-14 09:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-10-14 09:13:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-13 18:42:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-14 09:13:05.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"ee34a5a9b3e697932e8cfeb05c9c0d5bfee797d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-10-13 17:46:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb6d664014beafac2ec71fa733bc171e6b6aacf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 18:42:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2960a7d0e01741e1b1e1e7260adf9644\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572ffdce7c6942f4b9417f08482ca07c : SUCCESS in 11m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/718d6d4ce33c45cabbdbafd48e87ca31 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36fb3cf889274b46b6d94eb7c78f7ab3 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d78b254d3c649f98a30af9ed5f532bf : SUCCESS in 10m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/76e8fd92a75e45b6b330218389b8d16d : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/938f332d23354502bb06f25715c9c697 : SUCCESS in 21m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6468afd17d2640c39a135cc867be83e9 : SUCCESS in 23m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/81d626a698524ebe8a586d498bc31ed2 : SUCCESS in 29m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f27a3d474f654d468fbf97eab2f34384 : SUCCESS in 11m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3127b91cba314aa3b1dc712d9a1f1850 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dddc857e53ab46328121753bee1ebd28 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b1c6eeb213f34f88ad36375eaba16452 : SUCCESS in 15m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d99f1daaefb84944a5a9c91829a84166 : SUCCESS in 50m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa95ec114cd84f20a283d5ec93f4fa09 : SUCCESS in 29m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02674bf290414e418bcd63a5931def0f : SUCCESS in 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1060492dfc1244b0a58359bc96c51c30 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ccaf80fbeb6c48de9073637cefca364f : SUCCESS in 15m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b5d2fbc125e67f26febc20f20dfd0f0fc6288d7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 18:47:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4677731782d143e08a0720d25bffae32\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e1b900e42a074aaea40778cc54e52e2a : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":1},{"id":"59eb8b5edfc01f8365c72fe35117a7ee276fc133","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-14 09:13:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4ac682f1d98ce601180e794fa426f1f97b5e7ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 10:18:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eca451ac40a2d51d4b5ab472dd5605c997ad2ea4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 11:21:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e8da79a541041c0a018f3b462bb5a58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48911ccde0d249cbbddd3583dd54d444 : SUCCESS in 11m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4ab9a97ca22d4ba993fe8e0670538776 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/36e975e77ea14f1b8b04327b856e7825 : SUCCESS in 25m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d43ff716cf88473eacff5a57e6ed4a74 : SUCCESS in 20m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1510f78cd951425a99f5fc74b9248c6c : SUCCESS in 33m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a8bac99c5aed4201bfe82c36725b74c2 : SUCCESS in 27m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b3a89b6c6c55474e890aca0c313460a3 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a188bdf395e44222adc3f91373a12868 : SUCCESS in 13m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/779ae99515784b67a5090c32018c555e : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4dd3fc5ef02f40a689bb7037bf0e05d2 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1de4c58d2e6346f9b2dc87e5228ec31c : SUCCESS in 7m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26f41ae625ed4ec4ab69a6a04af5b4f6 : SUCCESS in 55m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7c3ffbe821d54b0b9177f3cc75e6f8a8 : SUCCESS in 47m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2020e6b4d7a442298942f6dca56f7ccb : SUCCESS in 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a17336b0a445c89c6f23614aa760f7 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0839d91fe51145ca8c733caedbb8350c : SUCCESS in 19m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1f0543e9ecdf1d2959aa67a4c59e8351b61b4095","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 11:21:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9ef98f705e9beaa055f583f301cd048421e5467e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 11:22:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4656c94136714db1aa28f092164b0b17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca96dadd448e4fa0987770dcc489adc2 : SUCCESS in 38s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/16a4f5723bfc44a590b055847864fd83 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fac55ced3c6dd15b3f3d64633d297a9e2e381fe2","revisions":{"fac55ced3c6dd15b3f3d64633d297a9e2e381fe2":{"kind":"REWORK","_number":1,"created":"2025-10-13 17:46:45.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/77/963877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/963877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/963877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/963877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/963877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/963877/1"}}},"commit":{"parents":[{"commit":"c0fefe80b3114383a2d89ae43ee502534f9ccda5","subject":"trivial test_[db_]replicator cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0fefe80b3114383a2d89ae43ee502534f9ccda5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-10-13 17:45:45.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-10-13 17:46:28.000000000","tz":-300},"subject":"test-db-replicator (trivial): just one tmpdir","message":"test-db-replicator (trivial): just one tmpdir\n\nChange-Id: I1e53d171faff02e2dbcbcc779ad3a47506b26853\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fac55ced3c6dd15b3f3d64633d297a9e2e381fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fac55ced3c6dd15b3f3d64633d297a9e2e381fe2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
