)]}'
{"id":"openstack%2Fswift~964060","triplet_id":"openstack%2Fswift~master~I6e8d66d1db2e2459cff6ec37db9114510648b89d","project":"openstack/swift","branch":"master","topic":"fix-db-test","hashtags":[],"change_id":"I6e8d66d1db2e2459cff6ec37db9114510648b89d","subject":"tests: be consistent with mocking daemon","status":"ABANDONED","created":"2025-10-14 19:02:03.000000000","updated":"2025-10-28 14:57:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"c4a6b5ccc470195b129d203c7639180c84e65739","_number":964060,"virtual_id_number":964060,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-14 20:33:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-10-14 20:33:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-14 20:33:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d10d661ad1c77f1369aded84a42023882047ff3b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-14 19:02:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9b4610466e081df9a79a8ca9febb11c77c10e4f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-14 19:23:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"de9a067a98136be1a2acd3f6b63feb5b3bda3e04","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-14 19:35:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"939a89341c1a4db945765f0b2c4363329c1394e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-14 19:36:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"607a804deeab405898383d4a18905b028797a592","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 20:33:01.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a312a59249340128d963e18f62ed7f9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0512bab6c77a4945b9a819d56706545a : FAILURE in 55m 10s","accounts_in_message":[],"_revision_number":4},{"id":"e5da0dbafbd82ae2738e1d5a6bb666c3e46124f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 20:33:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95209b5e3184457b9c7658523eb085d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32bd6f6b1184443b330ba1723f16289 : SUCCESS in 7m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8958ad88bd02495ba74f3535c5ce1240 : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adccd5033ae046bcbffaa4ab055535a5 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39235e57a118416197e0d6e9fd122a35 : SUCCESS in 20m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c85eab86841643d9abbc890e8b21f9ce : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8432ede18eeb4089804d104f91652acd : FAILURE in 10m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b7f7d4257f340e5822cf6bbd603e51e : FAILURE in 21m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/50dc516b202e4f449e19f82ce9bce83d : FAILURE in 21m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/daefe751981548938e24f6cdfb2027e6 : FAILURE in 19m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d49410fe62cb452abdde089795c17904 : SUCCESS in 7m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f21b1da749c04f73a79a01e17d618ee9 : SUCCESS in 9m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3f69c4b347814c14b19dc127977185ce : SUCCESS in 16m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d4b01dbe4c7441587db30d87013aea3 : SUCCESS in 50m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1757f7711d7149938b6632b7b4cd5d09 : SUCCESS in 55m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4621cd952b1e46ecb59f510ac4c670df : FAILURE in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a282570d7ee462283cb1273fadd2716 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46a1e7034b3842a9a9d8c157482241bc : SUCCESS in 25m 39s","accounts_in_message":[],"_revision_number":4},{"id":"c4a6b5ccc470195b129d203c7639180c84e65739","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-28 14:57:16.000000000","message":"Abandoned\n\nwe have https://review.opendev.org/c/openstack/swift/+/964080","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1768cd10f2a71010bbf1522bf80e8e5a226fa6b6","revisions":{"325054a7e967304e551873325d57e9bd93a17d56":{"kind":"REWORK","_number":1,"created":"2025-10-14 19:02:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/60/964060/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/964060/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/964060/1"}}},"commit":{"parents":[{"commit":"48a5d5e42f017ab1438fcb257cfc53d3818f1b02","subject":"Merge \"test-db-replicator (trivial): just one tmpdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48a5d5e42f017ab1438fcb257cfc53d3818f1b02"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-14 19:01:33.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-14 19:01:54.000000000","tz":-420},"subject":"tests: be consistent with mocking daemon","message":"tests: be consistent with mocking daemon\n\nChange-Id: I6e8d66d1db2e2459cff6ec37db9114510648b89d\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/325054a7e967304e551873325d57e9bd93a17d56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/325054a7e967304e551873325d57e9bd93a17d56"}]},"branch":"refs/heads/master"},"8ee68c518a5d1c41e190f78871fab031fb892595":{"kind":"REWORK","_number":2,"created":"2025-10-14 19:23:44.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/60/964060/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/964060/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/964060/2"}}},"commit":{"parents":[{"commit":"48a5d5e42f017ab1438fcb257cfc53d3818f1b02","subject":"Merge \"test-db-replicator (trivial): just one tmpdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48a5d5e42f017ab1438fcb257cfc53d3818f1b02"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-14 19:01:33.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-14 19:23:31.000000000","tz":-420},"subject":"tests: be consistent with mocking daemon","message":"tests: be consistent with mocking daemon\n\nChange-Id: I6e8d66d1db2e2459cff6ec37db9114510648b89d\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ee68c518a5d1c41e190f78871fab031fb892595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ee68c518a5d1c41e190f78871fab031fb892595"}]},"branch":"refs/heads/master"},"f890952cc547db23f72f820e162c9de63eb4ab60":{"kind":"REWORK","_number":3,"created":"2025-10-14 19:35:43.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/60/964060/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/964060/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/964060/3"}}},"commit":{"parents":[{"commit":"48a5d5e42f017ab1438fcb257cfc53d3818f1b02","subject":"Merge \"test-db-replicator (trivial): just one tmpdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48a5d5e42f017ab1438fcb257cfc53d3818f1b02"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-14 19:01:33.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-14 19:35:28.000000000","tz":-420},"subject":"tests: be consistent with mocking daemon","message":"tests: be consistent with mocking daemon\n\nChange-Id: I6e8d66d1db2e2459cff6ec37db9114510648b89d\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f890952cc547db23f72f820e162c9de63eb4ab60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f890952cc547db23f72f820e162c9de63eb4ab60"}]},"branch":"refs/heads/master"},"1768cd10f2a71010bbf1522bf80e8e5a226fa6b6":{"kind":"REWORK","_number":4,"created":"2025-10-14 19:36:19.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/60/964060/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/964060/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/964060/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/964060/4"}}},"commit":{"parents":[{"commit":"48a5d5e42f017ab1438fcb257cfc53d3818f1b02","subject":"Merge \"test-db-replicator (trivial): just one tmpdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48a5d5e42f017ab1438fcb257cfc53d3818f1b02"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-14 19:01:33.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-10-14 19:36:12.000000000","tz":-420},"subject":"tests: be consistent with mocking daemon","message":"tests: be consistent with mocking daemon\n\nChange-Id: I6e8d66d1db2e2459cff6ec37db9114510648b89d\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1768cd10f2a71010bbf1522bf80e8e5a226fa6b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1768cd10f2a71010bbf1522bf80e8e5a226fa6b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
