)]}'
{"id":"openstack%2Fswift~964244","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ie524a132b3afc383534f96a0272d515430ec1040","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-hidden-rows","hashtags":[],"change_id":"Ie524a132b3afc383534f96a0272d515430ec1040","subject":"InternalClient: support passing headers to iter_objects","status":"MERGED","created":"2025-10-16 18:12:12.000000000","updated":"2025-11-17 18:40:39.000000000","submitted":"2025-11-17 18:39:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964244-p-mpu-hidden-rows","meta_rev_id":"f035fe25385334c91cbc3c39ce183712b2eeb886","_number":964244,"virtual_id_number":964244,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 18:39:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-11-17 15:46:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-11-17 15:46:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-16 19:07:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-16 19:07:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56969350d041f18ebf70ef783ebeb2d677419618","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-16 18:12:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b8c8a34af29799fb7eff52f18c0cc466f9371c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 19:07:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe21c0d8f09a4278883fdf943747234b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ae1b33d27be54c429cfd70902da3f05a : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8970dadfab2bdf337820894a67677949fb407e5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 19:07:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac583e07f3fa448193237189e36f9c3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e36c42acc64a69b7c866cb12eb7eda : SUCCESS in 10m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2bfcda65dfc41eeaa5554ee7681a599 : SUCCESS in 16m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ef1a82e3f0745febd633c203504a2f8 : SUCCESS in 26m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df1bf0e4f5494a718f6799b1276d7b06 : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9359fa2c2f144b319d37bbef08dfbbbd : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/671cbf5522b147089bc5a2bbea235353 : SUCCESS in 27m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/37a4556f1ae947bbae22328a0e04893e : SUCCESS in 20m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/05ff984389a94c8a8cac93aba2852629 : SUCCESS in 33m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f032df269b6c49029db3429f74bba03c : SUCCESS in 18m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6b20ccfc16be44cebdcf7081e059a623 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f0b0aa2c8d044021af701ae0886db088 : SUCCESS in 18m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/78fa77e647a2419da5533fd8af82ed06 : SUCCESS in 18m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dfc28ff7c3754b468c19c7b50b760764 : SUCCESS in 11m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/29cf88cbb1174a8fa9943f827ea0c025 : FAILURE in 16m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bd3abef161254fb29171df9604a978ab : SUCCESS in 9m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/324b3db475054717b0e320126e78cc1f : SUCCESS in 54m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47a683afa8cc4205b9d5fc5dcd6c4cbc : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/552e763308584c55b1a878e52c26a541 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b278ddbfacac4e3ea2a1ae547c39a736 : SUCCESS in 20m 56s","accounts_in_message":[],"_revision_number":1},{"id":"6d0c1f84a0d5ac3cf585969319f0e3e77e4574a8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-17 15:46:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d55a3bb0d656673f0cc094fc783898b584c33514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 15:47:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e09ed2ceb4ad5b34505ce38b5819e781a19fd1eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 18:39:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf5fac1e057744c8a44031e076bd1973\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b7d4b9056224ed9b6d1bf45a6a7ac4d : SUCCESS in 8m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f3d67d64263040268dc542f2b9187e53 : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1bc72854173d4a02b06c5690cfe7e0ce : SUCCESS in 31m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9ed983f492a645c68f918e1334e08704 : SUCCESS in 14m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/916ff33311604f1ebe86161d7a942297 : SUCCESS in 14m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8dc42fbea7ed4d19b5dfe51dc568ece3 : SUCCESS in 11m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/44af20842b084b748dcdf8225799aa50 : SUCCESS in 10m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6c51dfc8765e43009387ea6f44ad3aed : SUCCESS in 16m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/11ada30eec6a4f84a6b989b3ac5ead0a : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8910e8d7c7654433990957dde617e1c3 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cd7ebe3aabde4be49c45b9b6d187aaf5 : SUCCESS in 8m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/315fe3f8d57f4700a0047dc56e74ec48 : SUCCESS in 50m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ced19769e8ea4b9281ae181e384504e7 : SUCCESS in 34m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f665c463a1d44f2e8997c029e881f889 : SUCCESS in 2m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac873fe2339f43039a74f3c31f28524f : SUCCESS in 17m 53s","accounts_in_message":[],"_revision_number":1},{"id":"824a6831dc0b0c1302cd1e8db3ae61a1f267e9a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 18:39:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f035fe25385334c91cbc3c39ce183712b2eeb886","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 18:40:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91d7384a8dd440819b2ae0d507798a54\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9453c81de50e4943aeb6980ec68a46d6 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6d36902b9ee9431da256fc09b957945c : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"52baa48245718d4f596e3e00329c04ee8b3e388e","revisions":{"52baa48245718d4f596e3e00329c04ee8b3e388e":{"kind":"REWORK","_number":1,"created":"2025-10-16 18:12:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/44/964244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/964244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/964244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/964244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/964244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/964244/1"}}},"commit":{"parents":[{"commit":"9e49d2026d0a375c3d4085df2fbaab6ff52775bb","subject":"mpu: use object event history to detect orphans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e49d2026d0a375c3d4085df2fbaab6ff52775bb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-10-15 13:37:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-10-16 18:09:25.000000000","tz":60},"subject":"InternalClient: support passing headers to iter_objects","message":"InternalClient: support passing headers to iter_objects\n\nChange-Id: Ie524a132b3afc383534f96a0272d515430ec1040\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52baa48245718d4f596e3e00329c04ee8b3e388e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52baa48245718d4f596e3e00329c04ee8b3e388e"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
