)]}'
{"id":"openstack%2Fswift~964354","triplet_id":"openstack%2Fswift~master~I552ae6ae2aee29be94683e9e01b3ab7b10a4f42e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-10-31 15:26:22.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I552ae6ae2aee29be94683e9e01b3ab7b10a4f42e","subject":"Move Pete Zaitcev to Core Emeritus","status":"MERGED","created":"2025-10-18 20:08:13.000000000","updated":"2025-10-31 16:33:50.000000000","submitted":"2025-10-31 16:31:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964354","meta_rev_id":"ac85cc8ff0c2347ab730f46de6b7fa48b9f441f2","_number":964354,"virtual_id_number":964354,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-31 16:31:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-10-20 20:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2025-10-18 21:02:21.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2025-10-20 12:07:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2025-10-27 17:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-31 15:26:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-31 15:26:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-18 20:44:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-18 21:02:21.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2025-10-20 12:07:09.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-10-20 20:19:43.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2025-10-27 16:49:04.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"CC"},{"updated":"2025-10-27 17:10:28.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"e41e36400789351a055916ef4bd3cd7d18d30aef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-18 20:08:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02c4d42d16246023bbb535cfda6328996862e596","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-18 20:14:56.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"877223a3e4d62fd00292a6e9d8bc4897d2165447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-18 20:44:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c20836a39a346a9b688d5fe823a131c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/166b503d32c64a6c89a1db254292b67a : SUCCESS in 5m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be680a6a758c4c29afe31439b981ad04 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8db902c7557e4203b062434d187beeaa : SUCCESS in 18m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8d9b995c405249a3b62d6e7e1d883c85 : SUCCESS in 20m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/690234556f044ed086a1a9dd4c8a82ef : SUCCESS in 19m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e003cfe213cf4306b2f8209aabddf875 : SUCCESS in 31m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58d5eec2fefd4531abadaaa42f24e0a5 : SUCCESS in 26m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e024be022474c3fb0e0eb6c1e1e4962 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1d67032ba4cc3a8092870fd9c499e3428b579d19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-18 20:56:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eda4b2aab9b4983a14fe7ce6b0e32d3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c511d9cf00cf426d888b9d61db0010f2 : FAILURE in 46m 42s","accounts_in_message":[],"_revision_number":1},{"id":"bc9bd47126111f66a55d72ee57275de6d9d5df2c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-10-18 21:02:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d106a7357361e1f467e0d75f0889541d0d6e29f4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-20 12:07:09.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04b0f4da9405f45c7d95891a58f04011327ea92a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-10-20 20:19:43.000000000","message":"Patch Set 1: Code-Review+2\n\nBoo, we\u0027ll miss you Pete","accounts_in_message":[],"_revision_number":1},{"id":"5a05e50c385c24916f73621cdfcdece8f61397f4","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2025-10-27 16:49:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e1ad01ce254a5f8c3ac48de81fab8b67cbfd4da","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-10-27 17:10:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6380d09dd8e604b0bb040d942c185ee4c3214154","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-31 15:26:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c1af317919aa852d903854b7eb1fbb1dc243c91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 15:26:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4195899b145e855865cb4b250560f265013aa3f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:31:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d837a66131f84c85a31aeaa2796656d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f59b274d4b433dbb9e1f17824c71b1 : SUCCESS in 9m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6306b7059d934888ad844a8f8614cf9d : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d4e58c120c69498e962c83034da2b36d : SUCCESS in 21m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b8d87d54f4e41aeba9c1dfe2c410d50 : SUCCESS in 10m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2e859fcfb4d44927a8da1b65de2a436c : SUCCESS in 30m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c20d7d2bd8274ec1aaad22f05a63c11d : SUCCESS in 14m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7fb15b9d2836433da19e366aa5e58034 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9c2c6da6f10c482ca485f1693118ad40 : SUCCESS in 17m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ab2a7f92bf234d018fd894a81a91d591 : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/204023399d4b4250b7af2bae35404713 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8cfcd613d2644ba588c5b98bc784be49 : SUCCESS in 8m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0eb9bd080ae143138a39135f20368fc5 : SUCCESS in 29m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c8d3c2365446bea4d7aa82f8d0cb5f : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"01a6e390929cac2b94d279fea5106afe175a8be8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:31:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac85cc8ff0c2347ab730f46de6b7fa48b9f441f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:33:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7ed62b484324a0fa3b06c64ba05f312\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/669cab1db7e745fb9e40a5237edc0148 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/42096cab593442d3972adb126ccf9f77 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7744e4a43a565d0ac92cfdd3380fe48cab750cc9","revisions":{"7744e4a43a565d0ac92cfdd3380fe48cab750cc9":{"kind":"REWORK","_number":1,"created":"2025-10-18 20:08:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/964354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/964354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/964354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/964354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/964354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/964354/1"}}},"commit":{"parents":[{"commit":"9232350af06336760de6323b50fd6765b322e81f","subject":"Merge \"Fix swift_dir setting in WSGI servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9232350af06336760de6323b50fd6765b322e81f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-10-18 20:07:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-10-18 20:07:57.000000000","tz":-420},"subject":"Move Pete Zaitcev to Core Emeritus","message":"Move Pete Zaitcev to Core Emeritus\n\nHe offered his core resignation earlier this week.\n\nChange-Id: I552ae6ae2aee29be94683e9e01b3ab7b10a4f42e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7744e4a43a565d0ac92cfdd3380fe48cab750cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7744e4a43a565d0ac92cfdd3380fe48cab750cc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
