)]}'
{"id":"openstack%2Fswift~964356","triplet_id":"openstack%2Fswift~master~I5525a7e615e12496fbe52a3a8767d325efc869ed","project":"openstack/swift","branch":"master","topic":"964356-patch6","attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-01-02 16:53:28.000000000","reason":"A robot voted negatively on a label"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-06-11 21:31:34.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-06-12 18:06:01.000000000","reason":"A robot voted negatively on a label"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-01-02 15:37:51.000000000","reason":"Vote got outdated and was removed: Code-Review+2"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5525a7e615e12496fbe52a3a8767d325efc869ed","subject":"Object Updater: Log when all primaries return 404","status":"NEW","created":"2025-10-19 01:06:33.000000000","updated":"2026-06-16 22:19:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":51,"unresolved_comment_count":17,"has_review_started":true,"meta_rev_id":"5a7df72fdd0f6a5bd0cac988e3e530e959b18ed9","_number":964356,"virtual_id_number":964356,"owner":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-12 18:06:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-19 01:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-19 02:15:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-20 04:40:18.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-10-21 20:11:36.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-01-02 15:43:57.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"7568230b8b00f63503b754e7f81a13db9f743d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2025-10-19 01:06:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"900605f6f394c7c329401ee828df8023c4c2b5f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-19 01:55:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99c3efbd578f4f1d9d007efd97b6538e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec38d4da99f94251a2c1ee0ed4fea0c6 : FAILURE in 47m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e7fdbd44da66d95ee7568448d341dd4a86e7df0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-19 02:15:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b33338ba705a4ffca5defc2f946e1047\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0630b7d1196a413586f5e001b0662660 : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/c8680294703a4882a31eeb10504fd4ec : SUCCESS in 58m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff0207708bc84f8ca16fa4c3cac04851 : SUCCESS in 1h 03m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0e2b7f02a59148659e7efd37587cd1cc : SUCCESS in 58m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/783879c579bc4ef693d881468760f498 : SUCCESS in 43m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/442a4082afd34570872d08210ff6dc97 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1656add1e0834a49bf01b9739c0db3e5 : SUCCESS in 11m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/424b537f526340a7b497b9328bc6ca38 : SUCCESS in 21m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bbf80d8c7c1f4c93ae617950ca897ff2 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/decfef12a42549438badfb27cf826027 : FAILURE in 16m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9375123bc7ef4878a0416fd0919039bf : FAILURE in 8m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7ac606c086ef4ba7a231c8bad6521965 : FAILURE in 28m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fece97e4e6444aa0853c1cef8a7f0463 : FAILURE in 29m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f03a3e1e7fe24a79a183ffdb9669ae06 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f8db5f883307439a91a7a7d56b384160 : SUCCESS in 11m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7f26a56fa84845ac837dcb36fc0367c4 : SUCCESS in 13m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3fe46884901340cb8e518fe1d6e7bb17 : SUCCESS in 6m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc43e4e1aaec4e698f5e6f68ae2308b2 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fde846de2241488d9008915601ad474d : SUCCESS in 8m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1be6e2f60b444339857dca4988c8fbd6 : SUCCESS in 46m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f75a14a0bf44e0792db638829fe8089 : SUCCESS in 54m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6010d0031ac14d168251833152b708a4 : SUCCESS in 32m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70d35a0e7f9b430895c074002462c631 : FAILURE in 23m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2970449472014bdbaf8bb8004faa7b0e : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/299c1ef4fb524f5991ae850c70110d52 : SUCCESS in 8m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56011ce83f404e6f82e283f7b861e8e6 : SUCCESS in 48m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a815a47f1354b718f042de2b9cd522e : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":1},{"id":"461851d938e9618bfa7f7182e4247e208241bed1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2025-10-19 02:51:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e9cf4b8a69dd16da4525dc64f83c2fa8467f1a4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-19 03:50:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b730c23a1d1248cd8de22a89ca39e1b5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3e732f18c2242cb808439b9179f6683 : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ce7534543bdc1d7993db7cb205dc689ea43d7724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-19 03:57:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7e9b4708be446dd80332ebf0eed0b28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011024eb641a444fbf2a635f1cee0a0f : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f6b7a95d9c1447cfa93ca3245cb64541 : SUCCESS in 57m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c07ee4c4e50c4d979e49350c4cf55597 : SUCCESS in 1h 02m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1764212f11cb4722a42edad9d59fbaed : SUCCESS in 38m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe871d271e6743e0aa2f300109ba5fd5 : SUCCESS in 1h 03m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff0f16b5712f41d292be9598e827a903 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d81df562d3044ac7b97da8337b2f8c0a : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8eee364946ce47718af6157aa4c6a6a5 : SUCCESS in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7efe6cbdf774415bb42ccf44712e9ba8 : SUCCESS in 7m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cd5afb86ca15434db7687ef6d36373da : SUCCESS in 21m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/907fea585f55472f86d9ff53bd6f384b : SUCCESS in 18m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dfb9bb7505144286bbb1b5d38e94c336 : SUCCESS in 34m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2e5caf3e3d20424f9b7e10089d8a3a80 : SUCCESS in 26m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5705b5f506dd48ceaeac02fefc901322 : SUCCESS in 13m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/594fa0241b0d465f9d7046f8ed7a8e23 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/31efdbd9ea5c447281aea3c8ec2e22db : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/25ae3f3337dd42d5ba530ae4a8c3a6f8 : SUCCESS in 6m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5564aef9328b4f2d931f2924bf24b1c3 : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6de1cbc60fbf4f6f91cd46b046d1ff89 : SUCCESS in 11m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c48dc49e17e743f3b4cd12ac66580e16 : SUCCESS in 59m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/96d4756e999a43b9a90fb48eb922cb08 : SUCCESS in 22m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/29adc9091c884e76aee8dcd575a41f5b : SUCCESS in 48m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b0efe47d165458d99b1d501def4fbbf : SUCCESS in 29m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01fbf723d8d740e1b2876c67855e2cca : SUCCESS in 2m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01e0470950df4a959995c8dfbdec30d5 : SUCCESS in 17m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16c95b0ebc5941f29f0d20743256e7b2 : SUCCESS in 55m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70475b6c8d5e4006bf185e2d910d926d : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":2},{"id":"c742470b56ede4196cc0a2ed60a4693c2f592cb9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-20 04:40:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e94ad9ca2fcef07eecf9e849162cf3728061969f","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2025-10-20 11:52:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9db6b89768e6d21b63dead7d2bcba97bde4dd14f","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2025-10-20 11:52:29.000000000","message":"Removed Code-Review+1 by \u003cGERRIT_ACCOUNT_38496\u003e\n","accounts_in_message":[{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}],"_revision_number":2},{"id":"0966c34ccd29a639f1938ea1a422e7fca8b36f95","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2025-10-20 13:22:27.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e5e10cafed938be5412a4c262e3642063095e301","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2025-10-21 01:45:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ece7bc5a84d2d700c1c949167be156b8358b3ab0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 02:48:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2966c32b2cf345308caba57739cc97c8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62671794a15948c986145d3108ac39da : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":3},{"id":"f3a40a321a657e1f5cb18217f1be720e4e7dfab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 02:50:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c74c5bff4249dfbf5a8874e1a53628\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6dc803a1f794b21b2acd0e35f5ea693 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/c599b84e36a042c1ab435572835ad3e8 : SUCCESS in 59m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aee6785f10fe4bd1917141f6ef66176d : SUCCESS in 1h 00m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a002cfcd72714dcfaf37e3cc1d723d8f : SUCCESS in 55m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e55ae9a4db04be09aee5017e418a3fe : SUCCESS in 45m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6732ef16f40a4a2a94a1b6c8fd918d25 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad0aef47ec28494d9a269441d5a9f289 : SUCCESS in 22m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65bde684da0b442bbe6b0f6081b3dd8c : SUCCESS in 21m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/890aceb8762d4cfb84fd75653d9e94f2 : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9612791091e34cc5af017128d63a83e6 : SUCCESS in 25m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/323c21975f8d48b189cf533e4d267bc5 : SUCCESS in 18m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/188dd23e3a464eaabcbeeba9895e401f : SUCCESS in 33m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a34e12f85aa6469fbe949715c7b29f6a : SUCCESS in 30m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0615983a17964c5d9b192ea1706d6ed7 : SUCCESS in 7m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1ca555bfe8f64c2eb5faf27d4a35f121 : SUCCESS in 14m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/46b9b70566a641b98a43a2d0018eaf7e : SUCCESS in 16m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0543bf2d81c14dbcba9a4c3cb1160ba8 : SUCCESS in 6m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef811630b4f148b0b6f438c64c13f550 : SUCCESS in 16m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/713b777671ba469283a7f68edc170a6c : SUCCESS in 8m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/47f5b86062084bf5b0cdefc6349c152b : SUCCESS in 59m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/da4946d5e64b4d9e9a5e981a2f89685c : SUCCESS in 53m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c037456fb4e646f5bd97d17bb575831e : SUCCESS in 51m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a360d39d61e48e68d145b41650fee1c : SUCCESS in 22m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d016d90db8a741c7a80792a0e83dae5d : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad5b5fc0a89345788a6948e32bd16ae5 : SUCCESS in 20m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4435c9f5ca4a4ad696f51d33d3ee01d5 : SUCCESS in 40m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/305af559a18c442f946505c7d2fde477 : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":3},{"id":"21a5d508d8697c7d0c530fc5131ebd8b9d894033","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-21 20:11:36.000000000","message":"Patch Set 3: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"12cca733ec8e74c86d02f184b807692c846910ea","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2025-10-23 00:29:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"777c6f7b2314fa0c560d9d0de0d9501ae094383e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-25 22:16:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"30a500a9140a4d538608ef3302285167d3a46fc7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-05 20:07:32.000000000","message":"Patch Set 3: Code-Review+2\n\n(16 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ca275756aeb6575d3c0274e37cc733618fdf4209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2025-11-08 14:43:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2b4a457588f9de6f8e9d30797f654b5253a0de5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-08 15:34:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee6030947f1044c0909d5d27d8f3bf67\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42e6a283b1bc41859a882acae360a88c : SUCCESS in 49m 03s","accounts_in_message":[],"_revision_number":4},{"id":"2b01a28550e4801612c232641e4a97a937e037e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-08 15:51:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a594723059c4e64a1ed046de490ffa4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2cb4440e4364cc099044924f20c9f38 : SUCCESS in 10m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/a2360f978ecb480f93901fd532ea1e5c : SUCCESS in 1h 01m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dcbed8bd60b24488b82c0e29665eb6d4 : SUCCESS in 59m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef4323e63b874e54b331377789c51708 : SUCCESS in 55m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/241824812e384531a970f45acd61d26d : SUCCESS in 45m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09fbf3fbe7d14bb5b4433ac066cff00d : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bab25ca34a274ef9a2cc02b9a19e77f0 : SUCCESS in 11m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d082db52ed084c69917a1f490bb7ffaa : SUCCESS in 15m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2f1db178cd974006ba2b24cc73dfc85f : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d7fa30fcc8f94327a54805652d85d2c4 : SUCCESS in 27m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bf3bee03ada445f49e7b4e145403e547 : SUCCESS in 22m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/05ef3b4e74294267bcedf0b0c7526645 : SUCCESS in 27m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1bc454099d3b43d6a6f4935598e4e042 : SUCCESS in 30m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/57755b2068784761b0f01b42fcb90e5d : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/301b6dae35874a6bae5872011a60c07b : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d377f5777db341cd8c99141d63fd4b75 : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a154799400874b6e974d7740965605eb : SUCCESS in 6m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/90705e46595045f8b4ce4d584f3028b1 : SUCCESS in 14m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e3a5ddb8937c4ad9b769dd38cb8060d3 : SUCCESS in 7m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/894bd480836e4cb38278c25e96caf533 : SUCCESS in 43m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/614f61cfdd0e4cf68feee82a64f2c9d3 : SUCCESS in 54m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9b85a8167fa405ebfc3c8523988b38f : SUCCESS in 24m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbccd853da294b609e6e6e00b27739db : SUCCESS in 21m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6255b257d7de485d88847ee61c8d82bd : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6241835919d5411d8512d999aa32ed69 : SUCCESS in 18m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d80ae584d0746a5a5e1f15967b6d654 : SUCCESS in 54m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/219d476c542241e3b8a4539da6c94924 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":4},{"id":"bb8642e6fe10ceac751f74a6821e59500c3cfc10","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2025-11-11 14:03:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d8ddc10bdc97f6a89983324cf1e824a5bb86b74f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-19 15:11:44.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"052d3f574c641c52d267fb6003cc3391308fe0a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-01-02 15:37:51.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"af91fd48c0d32e6f447cd45cf323a6f43359a215","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-01-02 15:49:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ac1023a8a0a02193eea24f476ec3738fe9046364","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 16:13:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e212f18895343bba040524602470f81\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/518cfa529df74e39a62d19eec6627dac : SUCCESS in 27m 29s","accounts_in_message":[],"_revision_number":5},{"id":"1dd7afc66026e7bce9a5bd9c24d9fd2767a90e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 16:53:28.000000000","message":"Patch Set 5: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1546307e57841dbbe8f2a39b55a4095\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c058465c7674cf887d2aff70d2b0054 : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/051f7f8456984b63b57adc468eb2542a : SUCCESS in 58m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/872ffbb11b704a63af10355ccdc2bc0c : SUCCESS in 1h 05m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4fa322fad5ad4d76bf4cd0ac7e81c33d : SUCCESS in 37m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13728ea37d824988baafcd60fbd601f1 : SUCCESS in 1h 00m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44cb24338e9d4a1a8f6839999b22aa8d : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c73c756bb3d542b2a8bef456543215f6 : SUCCESS in 17m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f7649f0ec3148ddaaa9ce4e3ab88648 : SUCCESS in 10m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef2ca26dfb9c4d888216ce2c59f779f3 : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2a5e9af4c5ea4baab5981e3c096106a6 : SUCCESS in 11m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/333260238d92486a966bd4bcb9006a5e : SUCCESS in 20m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0d07f7602d09473aaf1edc945c4939dc : SUCCESS in 27m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c5021349157b4f7c9bd639b1dfab014a : SUCCESS in 12m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/903e74547c2a4447987c0ba41e5fa27d : SUCCESS in 11m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cd544ee1d5f24f4bb29b46edf4781ff2 : SUCCESS in 14m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8c9210100443491f8991edc969b1d8e0 : SUCCESS in 18m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b42f3e16e8d41f8a7daa2dabfc9934f : SUCCESS in 5m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78eea00bb96e43e8a13f72a58c559e3a : SUCCESS in 15m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a44e344371ea4d2e8639ca7583e404e1 : SUCCESS in 7m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2196ffc414e14859bbf07c379e0dc0f9 : SUCCESS in 1h 04m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8c6004af44e4a9da5cf1247aeded793 : SUCCESS in 21m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e74e8870b6a45dda28ef53926e01e63 : SUCCESS in 45m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d668521a2d634291afa7fe74389b24a7 : SUCCESS in 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d682f06f87547b89df18746806e930c : FAILURE in 3m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/496bf179271146f08839dc357debc261 : SUCCESS in 15m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03948f27157f47e981bd207bc3f36611 : SUCCESS in 50m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56f3da0286c9423ebfcad16b76932c65 : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":5},{"id":"45a2eef4e804f8ebad220530a58b449bfd5e5fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-01-02 17:29:01.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"bce28101c4205761cd89187b6a9f4caa8b13a139","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 18:24:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0983f3ec4d9429292ac98df77dacdcc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0019825ed3f24baf9358d77b3a11621e : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":6},{"id":"a688534ee47b91d61207d285046578d7e8b77aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 18:35:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b365a521604cbb939c560093a021e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4bbb1bb958142229c874c95705a40c1 : SUCCESS in 4m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/4197425f14124704a8dee42c8174d836 : SUCCESS in 35m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d9c4fb571f3d473db5ec3864c7366181 : SUCCESS in 55m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8f990e5986d94b6389c6b3d182d6f143 : SUCCESS in 37m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6dc81925cdd14ea783129e7db7a1a3a6 : SUCCESS in 31m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93b566ffeba1468b9872e06d293c8253 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2595c1af1890437fb6d566445f918a39 : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7548ed22ede74651a05a3c951cae8d43 : SUCCESS in 17m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c131b863d6d4f34a89145a6db5884c2 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7670159b871d4631a12ec328f77278d5 : SUCCESS in 24m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5aff3c45ac874afcae8d0bd3caed98bd : SUCCESS in 21m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6e6eb36a48384ee8b7bdd6491f7f9d50 : SUCCESS in 18m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f5bf7b51d4a54a01b9184b55f2853125 : SUCCESS in 29m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d1c9a3cdefa64aca88e39614f44b3134 : SUCCESS in 13m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f5596900c93f451f872869d870c577c5 : SUCCESS in 15m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c355937aaf1d43a0839282c6a83ae397 : SUCCESS in 14m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/60596841f5ab4e4fb713878ae07de114 : SUCCESS in 14m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a919bdb8beb426aa3502232475fe16e : SUCCESS in 14m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3cd6dcdb8d204e54a2e8e82e941450bf : SUCCESS in 6m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3aeb0394a1b4eeb9eb5b87c82b754f6 : SUCCESS in 55m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9998b5df58944f6988651f5583484195 : SUCCESS in 51m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f38d37f8708426dbcc5f60285b479f9 : SUCCESS in 58m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cb13b868b714580a1e08fe139f8663b : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7146746d8344d7995dd83f9cf949c7 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3df61447d1f841d69e889e019e42bad2 : SUCCESS in 9m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b38451fc360400292c020d2348f02f2 : SUCCESS in 47m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09df29ad94b04c788a07a00865f4f043 : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":6},{"id":"59c75b8600472f73ed959a18ca34d4fced89f64b","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-09 13:13:34.000000000","message":"Patch Set 6: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2580ca9ae7c00948d6ae6f539dc903affb09a5d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-06-11 21:31:34.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"8773e0cd433b80192c8fa48dc592a90a77c7a428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 22:49:42.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8dc0cfd466741cc8390886224ba00f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b5aa53c18741c59986493793943556 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/306e5868ae674f0abd320a209b48884d : SUCCESS in 6m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/6895397d239546db9a90257aa37f9f0c : SUCCESS in 46m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/948a1be012394203acf685652aee6cdb : SUCCESS in 33m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d8ca19ccb1e49c2b6c2580f3c133f2b : SUCCESS in 58m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa7deab4dd9347b5a05e674f23162f79 : SUCCESS in 37m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bcef4d1287914ce6b57f63128612fc5e : SUCCESS in 14m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48a6ba65305448ee8ae7daf216e92ff1 : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba8e64f9742f4d2b8e92b0486c6c5844 : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca3bdf17a6c64ef1b32b7eb727aba91f : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7a6dccff864f441490099d0a0966b993 : FAILURE in 16m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/144a1ee038ba417eb59aafca9acf498b : FAILURE in 14m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c5f776d257e041d0bc262c2dd269fe5a : FAILURE in 17m 50s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5784a749ed214241b6ac3fd9ea8cd41b : FAILURE in 20m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e2c5fdcd44bd4d65b4cca955785ea248 : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2e7c20f703bf4d00a00d893c13eb26bb : SUCCESS in 15m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5f2137b9542645cbade19123c39fbd0e : SUCCESS in 17m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e5701b35ef9543b48dc4ae7784b42aa4 : SUCCESS in 5m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae458f4134394fd29d16b88830fccc3c : SUCCESS in 22m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e707b72b0a74ab98cf21fa186a93621 : SUCCESS in 5m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c17c60c9e144c85bb79bc788386e682 : SUCCESS in 57m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5b35667c27740669f2e5daa23707bdd : SUCCESS in 50m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b7de4cec05c420abe8508a987f25788 : SUCCESS in 1h 02m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5754bfde7ff4b6d9ee7da79e9f1cc88 : FAILURE in 28m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c223920ded9f43f88d5fbe522bff6995 : FAILURE in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ecf9eb35e304e79904089ad2dfcb6a8 : SUCCESS in 27m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/375a2072163041fbbf502fa0378c3754 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":7},{"id":"908b1e7f15b5f360ffb5f4e4fe5567098bc72a8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-06-12 16:47:11.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"ac88d3b65e5b232c3870987288896f863afda180","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-06-12 16:52:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5087b854623591c32e56637bf737f10173e87af4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 18:06:01.000000000","message":"Patch Set 8: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e556098307c4edb98f0ccd45f81c77d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d414481e37b3447bb085abf5d5515937 : SUCCESS in 9m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31ccbd45a1fd444795beb39e032d96ac : SUCCESS in 9m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/f2dd1b440f9a4da9a9180fe342c20e06 : SUCCESS in 1h 13m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abd8825f676d4ad78093ea0c31538538 : SUCCESS in 1h 08m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3c7f1e6384e4435a8a47e3cbfa950ecb : SUCCESS in 56m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63bd60cabc644213ae3f22e6be302b30 : SUCCESS in 39m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92d4a77aba2045d58811e62e010ff45d : SUCCESS in 27m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c1eba5e285d4146b634dcd948e8b135 : SUCCESS in 23m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34ce835fadd54c44ad927eef5ed46f67 : SUCCESS in 20m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b664c02178754550949e458baa71f646 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6a5843b14af4414ea7de3f560a64a013 : FAILURE in 28m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2df4721983e1436585f9e1ff242478cf : FAILURE in 10m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/997885c3fe14403fa48fead329aae9aa : FAILURE in 34m 19s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a47d9d48e9304532becd8363ea485376 : FAILURE in 21m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ad9831049f064e67827be0e737e20ba7 : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b9a5f2cb74584b9892bd790d8d560ec4 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2fb9450fff254ba284e4f3a0a285ab33 : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83671ff175aa4c94a78bbcc3b1ca7630 : SUCCESS in 6m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e69b0c485e254cafa133ae1bd3d78842 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a551af553b5c47afb53ccb2dd5c4e8e8 : SUCCESS in 10m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1398beeb2113443cb8a8b29dad11bec7 : SUCCESS in 1h 00m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b62922f13efc4896845a1df74e75b3aa : SUCCESS in 55m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f4c567727564ce091c97b44a4b4b3a9 : SUCCESS in 53m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f52aad4bb8dd46c2a6f3fc2e5e558c33 : FAILURE in 28m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c78cd84d8be45918e226529e850a9e5 : FAILURE in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87d08528b59343ecaac7fd232f7226bc : SUCCESS in 20m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c94f2e2abb8747bba10ef707b48a77f4 : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":8},{"id":"5a7df72fdd0f6a5bd0cac988e3e530e959b18ed9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 22:19:12.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d584426af7cc466fa30834f1f405791f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/950822cb26644f7ea765d53f1f25af12 : POST_FAILURE in 47m 21s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"4e3d26aca67ee7b427e00420e4d48ba0130c6d65","revisions":{"b75e676f910438b183dcd38f809dbe52c138d581":{"kind":"REWORK","_number":1,"created":"2025-10-19 01:06:33.000000000","uploader":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"ref":"refs/changes/56/964356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/964356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/964356/1"}}},"commit":{"parents":[{"commit":"9232350af06336760de6323b50fd6765b322e81f","subject":"Merge \"Fix swift_dir setting in WSGI servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9232350af06336760de6323b50fd6765b322e81f"}]}],"author":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-19 00:54:52.000000000","tz":-240},"committer":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-19 01:05:47.000000000","tz":-240},"subject":"  Object Updater: Log when all primaries return 404","message":"  Object Updater: Log when all primaries return 404\n\n  When the object updater attempts to update container listings and\n  all primary container servers return 404, log a warning message to\n  help operators identify potential issues with missing container\n  databases.\n\n  The warning message includes the object path and the list of node\n  IDs that returned 404.\n\n  Closes-Bug: #1647840\n\nChange-Id: I5525a7e615e12496fbe52a3a8767d325efc869ed\nSigned-off-by: Andressa Cabistani \u003cacabistani@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b75e676f910438b183dcd38f809dbe52c138d581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b75e676f910438b183dcd38f809dbe52c138d581"}]},"branch":"refs/heads/master"},"54d70ea0692630b5347c39298d1a2e24849b735c":{"kind":"REWORK","_number":2,"created":"2025-10-19 02:51:36.000000000","uploader":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"ref":"refs/changes/56/964356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/964356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/964356/2"}}},"commit":{"parents":[{"commit":"9232350af06336760de6323b50fd6765b322e81f","subject":"Merge \"Fix swift_dir setting in WSGI servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9232350af06336760de6323b50fd6765b322e81f"}]}],"author":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-19 00:54:52.000000000","tz":-240},"committer":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-19 02:51:24.000000000","tz":-240},"subject":"  Object Updater: Log when all primaries return 404","message":"  Object Updater: Log when all primaries return 404\n\n  When the object updater attempts to update container listings and\n  all primary container servers return 404, log a warning message to\n  help operators identify potential issues with missing container\n  databases.\n\n  The warning message includes the object path and the list of node\n  IDs that returned 404.\n\n  Closes-Bug: #1647840\n\nChange-Id: I5525a7e615e12496fbe52a3a8767d325efc869ed\nSigned-off-by: Andressa Cabistani \u003cacabistani@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54d70ea0692630b5347c39298d1a2e24849b735c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54d70ea0692630b5347c39298d1a2e24849b735c"}]},"branch":"refs/heads/master"},"16e4c0ac5eb1e4ef7517d4b03a4715ce6c5084b8":{"kind":"REWORK","_number":3,"created":"2025-10-21 01:45:18.000000000","uploader":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"ref":"refs/changes/56/964356/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/964356/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/964356/3"}}},"commit":{"parents":[{"commit":"9232350af06336760de6323b50fd6765b322e81f","subject":"Merge \"Fix swift_dir setting in WSGI servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9232350af06336760de6323b50fd6765b322e81f"}]}],"author":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-19 00:54:52.000000000","tz":-240},"committer":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-21 01:45:04.000000000","tz":-240},"subject":"  Object Updater: Log when all primaries return 404","message":"  Object Updater: Log when all primaries return 404\n\n  When the object updater attempts to update container listings and\n  all primary container servers return 404, log a warning message to\n  help operators identify potential issues with missing container\n  databases.\n\n  The warning message includes the object path and the list of node\n  IDs that returned 404.\n\n  Closes-Bug: #1647840\n\nChange-Id: I5525a7e615e12496fbe52a3a8767d325efc869ed\nSigned-off-by: Andressa Cabistani \u003cacabistani@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16e4c0ac5eb1e4ef7517d4b03a4715ce6c5084b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16e4c0ac5eb1e4ef7517d4b03a4715ce6c5084b8"}]},"branch":"refs/heads/master"},"ee9a4d304bdf284b67f863405417425088a24652":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-11-08 14:43:11.000000000","uploader":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"ref":"refs/changes/56/964356/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/964356/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/964356/4"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-19 00:54:52.000000000","tz":-240},"committer":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-11-08 14:43:11.000000000","tz":0},"subject":"  Object Updater: Log when all primaries return 404","message":"  Object Updater: Log when all primaries return 404\n\n  When the object updater attempts to update container listings and\n  all primary container servers return 404, log a warning message to\n  help operators identify potential issues with missing container\n  databases.\n\n  The warning message includes the object path and the list of node\n  IDs that returned 404.\n\n  Closes-Bug: #1647840\n\nChange-Id: I5525a7e615e12496fbe52a3a8767d325efc869ed\nSigned-off-by: Andressa Cabistani \u003cacabistani@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee9a4d304bdf284b67f863405417425088a24652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee9a4d304bdf284b67f863405417425088a24652"}]},"branch":"refs/heads/master","description":"Rebase"},"be01c21c220b4f1f70b5ca65a90d4e25702f7036":{"kind":"REWORK","_number":5,"created":"2026-01-02 15:37:51.000000000","uploader":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"ref":"refs/changes/56/964356/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/964356/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/964356/5"}}},"commit":{"parents":[{"commit":"9232350af06336760de6323b50fd6765b322e81f","subject":"Merge \"Fix swift_dir setting in WSGI servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9232350af06336760de6323b50fd6765b322e81f"}]}],"author":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-19 00:54:52.000000000","tz":-240},"committer":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2026-01-02 15:37:35.000000000","tz":-300},"subject":"  Object Updater: Log when all primaries return 404","message":"  Object Updater: Log when all primaries return 404\n\n  When the object updater attempts to update container listings and\n  all primary container servers return 404, log a warning message to\n  help operators identify potential issues with missing container\n  databases.\n\n  The warning message includes the object path and the list of node\n  IDs that returned 404.\n\n  Closes-Bug: #1647840\n\nChange-Id: I5525a7e615e12496fbe52a3a8767d325efc869ed\nSigned-off-by: Andressa Cabistani \u003cacabistani@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be01c21c220b4f1f70b5ca65a90d4e25702f7036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be01c21c220b4f1f70b5ca65a90d4e25702f7036"}]},"branch":"refs/heads/master"},"7a5b1b4ad12580638505035da5cd92ee7fdd3d3a":{"kind":"REWORK","_number":6,"created":"2026-01-02 17:29:01.000000000","uploader":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"ref":"refs/changes/56/964356/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/964356/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/964356/6"}}},"commit":{"parents":[{"commit":"9232350af06336760de6323b50fd6765b322e81f","subject":"Merge \"Fix swift_dir setting in WSGI servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9232350af06336760de6323b50fd6765b322e81f"}]}],"author":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-19 00:54:52.000000000","tz":-240},"committer":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2026-01-02 17:28:47.000000000","tz":-300},"subject":"  Object Updater: Log when all primaries return 404","message":"  Object Updater: Log when all primaries return 404\n\n  When the object updater attempts to update container listings and\n  all primary container servers return 404, log a warning message to\n  help operators identify potential issues with missing container\n  databases.\n\n  The warning message includes the object path and the list of node\n  IDs that returned 404.\n\n  Closes-Bug: #1647840\n\nChange-Id: I5525a7e615e12496fbe52a3a8767d325efc869ed\nSigned-off-by: Andressa Cabistani \u003cacabistani@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a5b1b4ad12580638505035da5cd92ee7fdd3d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a5b1b4ad12580638505035da5cd92ee7fdd3d3a"}]},"branch":"refs/heads/master"},"769bbc2dd34b4644a2304193f23e0c23c7501056":{"kind":"REWORK","_number":7,"created":"2026-06-11 21:31:34.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/56/964356/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/964356/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/964356/7"}}},"commit":{"parents":[{"commit":"a4fcc02614e9b47c35da234eb0c6c5b2c3b46a4a","subject":"Merge \"ssync_receiver: clarify log message annotations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4fcc02614e9b47c35da234eb0c6c5b2c3b46a4a"}]}],"author":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-19 00:54:52.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-06-11 21:29:18.000000000","tz":-300},"subject":"  Object Updater: Log when all primaries return 404","message":"  Object Updater: Log when all primaries return 404\n\n  When the object updater attempts to update container listings and\n  all primary container servers return 404, log a warning message to\n  help operators identify potential issues with missing container\n  databases.\n\n  The warning message includes the object path and the list of node\n  IDs that returned 404.\n\n  Closes-Bug: #1647840\n\nChange-Id: I5525a7e615e12496fbe52a3a8767d325efc869ed\nSigned-off-by: Andressa Cabistani \u003cacabistani@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769bbc2dd34b4644a2304193f23e0c23c7501056"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769bbc2dd34b4644a2304193f23e0c23c7501056"}]},"branch":"refs/heads/master"},"4e3d26aca67ee7b427e00420e4d48ba0130c6d65":{"kind":"REWORK","_number":8,"created":"2026-06-12 16:47:11.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/56/964356/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/964356/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/964356/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/964356/8"}}},"commit":{"parents":[{"commit":"9232350af06336760de6323b50fd6765b322e81f","subject":"Merge \"Fix swift_dir setting in WSGI servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9232350af06336760de6323b50fd6765b322e81f"}]}],"author":{"name":"Andressa Cabistani","email":"acabistani@gmail.com","date":"2025-10-19 00:54:52.000000000","tz":-240},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-06-12 16:46:57.000000000","tz":-300},"subject":"Object Updater: Log when all primaries return 404","message":"Object Updater: Log when all primaries return 404\n\nWhen the object updater attempts to update container listings and\nall primary container servers return 404, log a warning message to\nhelp operators identify potential issues with missing container\ndatabases.\n\nThe warning message includes the object path and the list of node\nIDs that returned 404.\n\nCloses-Bug: #1647840\n\nChange-Id: I5525a7e615e12496fbe52a3a8767d325efc869ed\nSigned-off-by: Andressa Cabistani \u003cacabistani@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e3d26aca67ee7b427e00420e4d48ba0130c6d65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e3d26aca67ee7b427e00420e4d48ba0130c6d65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
