)]}'
{"id":"openstack%2Fswift~965004","triplet_id":"openstack%2Fswift~master~Ib54172131f1e3abd9b0419aa0370930c6fc82ba9","project":"openstack/swift","branch":"master","topic":"fix-common-db-teardown","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-11-04 11:34:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib54172131f1e3abd9b0419aa0370930c6fc82ba9","subject":"tests: belts and braces idiomatic patching","status":"MERGED","created":"2025-10-27 18:24:52.000000000","updated":"2025-11-04 11:36:11.000000000","submitted":"2025-11-04 11:34:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"965004-fix-common-db-teardown","meta_rev_id":"93e04883df60815b7d1961aba24650e7fe6925c5","_number":965004,"virtual_id_number":965004,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-04 11:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-11-04 10:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2025-10-28 15:10:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-11-04 10:49:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-27 19:27:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-28 15:10:28.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2025-11-04 10:49:14.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"bea3e908a1e32b27d86e0966fc693aebd2c23891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-10-27 18:24:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6ce09b3503ef1a7e9a94a0f67fb38960fc8b5c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 19:27:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27859131848c4818b26bf9e846d73d69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5283f1197c7b4ba79c5bfc77957facd1 : SUCCESS in 10m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c66850fa10924871aaf7095b8a81cdb5 : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e03f659c82704d0ea981b0398e8fcb2d : SUCCESS in 10m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b11956a5e80447fb19a442c1c953812 : SUCCESS in 21m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e79814a98f3404ba86761e994919f1e : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5651c2b0a1064de6b26803e972ebcd6a : SUCCESS in 20m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7870259a0553493eb9995364e930fb8e : SUCCESS in 8m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/861959e77f1e4f05a13d68e230e1ccce : SUCCESS in 27m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/29639a6a98e44a8d9edd765c1d4fd7c4 : SUCCESS in 11m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5b5cb7186a21475ca915f9e867c7eab2 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0982e51187504f7c82944d60ecd9e5bf : SUCCESS in 15m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/62c34dc52d3c41518b8933cd150979ec : SUCCESS in 15m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b43e76bdc0ff4fd0822860fbebf21734 : SUCCESS in 54m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77df3157beed43cc92fb4209e662abde : SUCCESS in 56m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6781eb43261420b8faa1c7eb7afb539 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91fc25bcc66f4535828e3bed8b4b0f46 : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b9a4a36e0444367bf6d1a797e151275 : SUCCESS in 26m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f62576d617e6aa14f1d5e67f564f7f0e35d97837","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 19:30:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/743457642e64458cbe40ba70b72a532b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2ad93d4015d94091a8360412e71561e0 : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":1},{"id":"bcff180c60141bca211ed7434c8409fedaac1a08","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-10-28 15:10:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78c86b8a49b19d41e7fcd9ae66de7355a7a63e4a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-04 10:49:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"15e9a8525355ae5de55cc143ecf0f52445ec9b8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 10:49:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d0a40d981bd1ebec1d07feb74e4ed374dffab1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 11:34:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bda6a4fa2ced469bad07a80a9110dad9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/953d9790da6444a8a25411c152e37a95 : SUCCESS in 10m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5d3ca2ec8fc9479ab5df45d69df2cfdc : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fda8f52b45c543a5821afe08f4e6b285 : SUCCESS in 25m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c39b22932ad481f9bb76ffaac57657d : SUCCESS in 20m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c3888c4085d241188d168f78c25f78ac : SUCCESS in 21m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/41e12bd28a85499685d68c36aadb2944 : SUCCESS in 28m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/27950e61919440c7a93a8184aae73010 : SUCCESS in 8m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/19ed7cb469bb4e94ad6eecf78f6a05ee : SUCCESS in 14m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fdcb59f22f7845b7b236a111550c0434 : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c42fbbd4d854328965664e80e5bd7b7 : SUCCESS in 5m 37s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/92cb881120514d28ba79145f5e08b9fa : SUCCESS in 8m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5a380b43aaa4b58ac3eac10d591327b : SUCCESS in 41m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee534f9a7ad44bafbb6553c7e09957e2 : SUCCESS in 25m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12fcf8621c2b4797b987b284ff16583c : SUCCESS in 23m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9430390fcb8f4d63b48a8b6d95d31048 : SUCCESS in 3m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14d5280f2f0d4461acd18c32157c5b06 : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a1132a59d8d4fc98609d0421e5960d9dce2d3d03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 11:34:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"93e04883df60815b7d1961aba24650e7fe6925c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 11:36:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b04f481da8cd4efeb7f592c88af58d04\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a0992a28ab4dd8829a5f96b5d8c53d : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a884ccd3b6ea4b0185aaa5f1d2f07e3a : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e450b8c670dcc769f8b64d5f4e0c182518592dee","revisions":{"e450b8c670dcc769f8b64d5f4e0c182518592dee":{"kind":"REWORK","_number":1,"created":"2025-10-27 18:24:52.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/04/965004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/965004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/965004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/965004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/965004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/965004/1"}}},"commit":{"parents":[{"commit":"89e9ced9e56c0f3c5f7cdf75f40ff8c5f0dc04c3","subject":"tests: idiomatic ring patching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89e9ced9e56c0f3c5f7cdf75f40ff8c5f0dc04c3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-10-27 18:09:02.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-10-27 18:22:03.000000000","tz":-300},"subject":"tests: belts and braces idiomatic patching","message":"tests: belts and braces idiomatic patching\n\nAFAIK this is un-needed once we ensure test_db_replicator un-patches the\nobject with a cleanup, but in isolation a test patching a global and\nrelying on someting else to clean it up looks messy/lazy - or at best\njust not a pattern/style that can easily be copied elsewhere.\n\nRelated-Change: I66d4df1e2dba058b7c719a4a932234b3fc10b554\nChange-Id: Ib54172131f1e3abd9b0419aa0370930c6fc82ba9\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e450b8c670dcc769f8b64d5f4e0c182518592dee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e450b8c670dcc769f8b64d5f4e0c182518592dee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
