)]}'
{"id":"openstack%2Fswift~965531","triplet_id":"openstack%2Fswift~unmaintained%2F2024.1~I4f3320abab3c1d461c16d3c820a332f0754241a7","project":"openstack/swift","branch":"unmaintained/2024.1","topic":"create-2024.1","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-11-01 20:27:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f3320abab3c1d461c16d3c820a332f0754241a7","subject":"Update .gitreview for unmaintained/2024.1","status":"MERGED","created":"2025-10-31 11:46:56.000000000","updated":"2025-11-01 22:03:36.000000000","submitted":"2025-11-01 20:27:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965531-create-2024.1","meta_rev_id":"ca4a26858fadc9bff401b4d0f0cdd2d747ad1dda","_number":965531,"virtual_id_number":965531,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-01 20:27:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-11-01 19:24:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-11-01 19:24:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-31 12:38:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-31 18:31:56.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-11-01 19:24:30.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"7d15687424f6784add9c32340e0506f50a3a6280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-10-31 11:46:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7af1a12e3fa9c68490c0b0104b6c59444a70a771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 12:38:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fb8e1735aa84333b3bcaa253a4bf5aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8347e424fdb4c4f8c5feb32e8541281 : SUCCESS in 4m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e64207cd9c83434b8602323bc32bd816 : SUCCESS in 15m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b8dc92d747a746d8852b07860def401b : SUCCESS in 10m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca47b483fa2c47abb4abe140183f8de3 : FAILURE in 7m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61ce5e4a9c6e44b49d6187483a765902 : SUCCESS in 19m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f7c3ab10aa924491ae88283fb9df2a1d : SUCCESS in 14m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/39fe40b864aa47b9aa6525f1bef404e6 : SUCCESS in 17m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6867d707a114cea98e3f50f13039616 : SUCCESS in 28m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9515a1afc0ab44189151ea5fb3026df2 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f6c13b707c16b078caf22c5572b450704d3f4374","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 14:07:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1008725a94264621bc20060a73712d32\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed43de14a3154c2cadcfb62977866165 : SUCCESS in 44m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3bf9c409636f4b1490c3296824b13436 : SUCCESS in 28m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c6a19054ff345a7b47a5a1592357321 : SUCCESS in 29m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a210b229787b45bdb5c2be3cad6eda8c : SUCCESS in 33m 51s","accounts_in_message":[],"_revision_number":1},{"id":"65dd3b3a5133b7a3a6e9092b1c6ca573a51e8bc6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-31 18:31:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"caa16f96477160d149ea0973b5b06bdb8f3481b2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-11-01 19:24:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14c5da0b6f7ca5b9b68fbbe944618553196022e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 19:56:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1a9b65f963240cda85dd15116bc2b6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f253372a3e248e6ac7805b86d6bdeba : SUCCESS in 8m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3dfb57d6d1b413bbcb69558f4f77c58 : SUCCESS in 12m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e29a49de405a4e018997430b00d19e36 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f52fa6040814e1aba14ab3b2c8ab084 : SUCCESS in 25m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/30fe61bcc90c4a7d8a79df1b3f8b811e : SUCCESS in 20m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d13f6e7d89104d378c0d132aa9063389 : SUCCESS in 18m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/09c0cc82881b4d2084239529bf6ba913 : SUCCESS in 17m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f67bb3a5e1c497290361a70f39bb661 : SUCCESS in 26m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e036f48737d5421285f20941f166c2f8 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a0e56f5ef38dc0b1dd1b24f7cce91f7d68a0abed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 19:56:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58ddc3557f0e8524d60a4f45302520bbcaff9926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 20:27:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01a6d50b25954f18874be7483b0164ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7788e1e729674c6390ed2b03d9820811 : SUCCESS in 9m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d5cef7d1ceeb44ab944d0c4593de74aa : SUCCESS in 13m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/76256a1ac59b4637a8de3853f1c56a5d : SUCCESS in 7m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cb536ecff5ca4de5a85cf09dbc4eb8ec : SUCCESS in 24m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac5986e32dc6470387324a0017fd52bf : SUCCESS in 18m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/567991f93b6c4712bd5edb2f965a0d30 : SUCCESS in 20m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/542259afe6134b3a8af13c8cfbee8faf : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a3b77ec7ea04914967ffc56e942981b : SUCCESS in 23m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d57337039dfa41eb99edd40a2c39c4c6 : SUCCESS in 21m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7e540b0039cf428bb05b74d85c2986a5 : SUCCESS in 23m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0011880d6d9047b288e19758ae3f99ac : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2c0e3aa56f5e4aa9a77c7e62dca4cc28 : SUCCESS in 11m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/67866e89052649149817914b8151bc64 : SUCCESS in 19m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29364cc2c6814d648f696b91300c5e62 : SUCCESS in 7m 16s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/039cdee7530548e99ccacc38da853e17 : SUCCESS in 9m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27d604990de54027b7be18d47f671f1f : SUCCESS in 25m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a24850e4814462fa5f8cca31a347358 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e6448625bf6a1ec5eb9de37b8f9572a05af3f44b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 20:27:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"942ee5ac8a505e6d531844c92d86f52ccebbec10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 20:27:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8265bab744b4cf2a91a2eef7c97d784\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ccd3fdbb8e647b383838b94d7342f8b : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7e93654d9df7458e83bb2c238053b568 : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8843c53907114112b42e361e9b18d825 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c690eb74e989a91d353c1f986cbcd4478ce6d40d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-01 22:02:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca4a26858fadc9bff401b4d0f0cdd2d747ad1dda","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-01 22:03:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9264209f52510dfa85aeb0f22722b716c8893af6","revisions":{"9264209f52510dfa85aeb0f22722b716c8893af6":{"kind":"REWORK","_number":1,"created":"2025-10-31 11:46:56.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/31/965531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/965531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/965531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/965531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/965531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/965531/1"}}},"commit":{"parents":[{"commit":"38b7cbb29a87e0a6df55f446e518f91a2681a122","subject":"CI: Fix py27/py36/py37 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38b7cbb29a87e0a6df55f446e518f91a2681a122"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-10-31 11:46:55.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-10-31 11:46:55.000000000","tz":0},"subject":"Update .gitreview for unmaintained/2024.1","message":"Update .gitreview for unmaintained/2024.1\n\nChange-Id: I4f3320abab3c1d461c16d3c820a332f0754241a7\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9264209f52510dfa85aeb0f22722b716c8893af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9264209f52510dfa85aeb0f22722b716c8893af6"}]},"branch":"refs/heads/unmaintained/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
