)]}'
{"id":"openstack%2Fswift~965830","triplet_id":"openstack%2Fswift~stable%2F2024.2~Ib95a3368ae2beae06b206d400cc358e5fa6637e7","project":"openstack/swift","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-11-01 22:04:06.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ib95a3368ae2beae06b206d400cc358e5fa6637e7","subject":"[stable-only] Pin pyperclip\u003d\u003d1.9.0 for py2","status":"MERGED","created":"2025-10-31 17:25:08.000000000","updated":"2025-11-02 00:24:31.000000000","submitted":"2025-11-02 00:23:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965830","meta_rev_id":"ff9127550614fe8fc61cc247e7f50bf4950818c2","_number":965830,"virtual_id_number":965830,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-02 00:23:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-31 18:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-01 00:11:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-31 18:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"719f9c8d2b6a3a85be64f2230315fe3c70d067d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-31 17:25:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26a4ae2197f1e017da769ca25d671f744dd868d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 18:28:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc364632b5f24c67bb587a982ef58628\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c82de3c5ae14328915f72686414b8a4 : SUCCESS in 5m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/d9bd0014b6624dff9350920b69ddfd91 : SUCCESS in 31m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a1a9dd44d44941efbae5336b22735d4f : SUCCESS in 27m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8f03845fa93421aa7d78375b60aa59e : SUCCESS in 55m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a073df05c334c84845132a5d919a842 : SUCCESS in 12m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2417732a79cb443e91afedb001824469 : SUCCESS in 22m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/614eb4f39cf74bc9bde6ca34c6512849 : SUCCESS in 22m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f6f5a308557349b395e4bd017df2702e : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68c24702200347ed85abdaeff4cf5535 : SUCCESS in 23m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3830c274ddf14fde8e7a7bac6a018274 : SUCCESS in 10m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/17913c392e3d4c06a75b5c3563db3c93 : SUCCESS in 20m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/137ad2efa81a4c52a7932fc7b1c5fbad : SUCCESS in 22m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/be5396b98185492dadac64199f0e7204 : SUCCESS in 27m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf64c2186dc14152aca13cf29266d8fb : SUCCESS in 23m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ba9462bb53d64bb4a2a87288137c30f5 : SUCCESS in 21m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/492f33913cae4368ba95dfc34d01f51b : SUCCESS in 23m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/208714305f6a4decad09fec6cd668d8d : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7b3a00c344ba43bea8399491c762384c : SUCCESS in 17m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b6f054a742f8499db1617cf237b85e49 : SUCCESS in 10m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1bd39ea364974872bd58b18bc2ee5840 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/740c2fb23a4e4ba28399bc8e6bbea18a : FAILURE in 13m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/94b15b4a05554686ad094d90a8804bb1 : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/21de6198c6404c58a68282b70313ca90 : SUCCESS in 47m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cdfe4faeed1c413d9225efeaebd6a8b9 : SUCCESS in 45m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a158879d67a04e70bb8f6b4f88e2951f : SUCCESS in 50m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6b2f188184545eb9d96c9da56b6d38a : SUCCESS in 22m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b838a6c18c4c83b614c57fc540d90e : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc07e397a79b4ddba96e19dc895ad038 : SUCCESS in 29m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbebd1dc79d74921963001337576f8bc : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":1},{"id":"67c4932998ecbbd8fb28a689456770f033a0843a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-31 18:29:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dd469ad8a0021c8d06054e181747bd9e0a02d38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 19:49:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5392a9a13d6f43dcac94e658114638c6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f996d4c6f3cc4f8c9a78ecbd41c8904d : SUCCESS in 38m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e320062f67b84f168c33921e89c7edde : SUCCESS in 28m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/73bdd808439b44c7986137375c264526 : SUCCESS in 36m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c73171c097b4f418c9655b8c734aeca : SUCCESS in 55m 25s","accounts_in_message":[],"_revision_number":1},{"id":"12e58dfff4dfdfa78cd34e00218667a44a284f14","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-01 00:11:39.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd605961f3911ee53aa1f3d6be6a8a5298a53b57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 00:12:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8b659932a54f08c95ec9e68cbd2879d167ebce2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 01:49:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f797e9e40994f05b5daee57f280c04b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8ccbdf212c94e6e9819cf3fcbb7ea86 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/0df9483c6f5e4241bc55b2c42a6c1bcc : SUCCESS in 36m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b39ca77a5cc84e86944d3c3fbc3a7b9f : SUCCESS in 27m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/485eb51ecb9a4f33b5213406db435065 : SUCCESS in 43m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e607e81c85e947bca6c7c14ca908eebd : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/10c071a738e24943914afbabb2f1c39e : SUCCESS in 27m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6104d603395c47b691ba301b555bbf9e : SUCCESS in 7m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4498f412a4dc4e96b0d9f9df5d81011c : SUCCESS in 23m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b3d135a44663440a816b797c88337846 : SUCCESS in 21m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6d16c525f3df49a4aafca080bed0a433 : SUCCESS in 21m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/23befedb07fd4be0b322964d70d4fe68 : SUCCESS in 21m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/81c8f6a55cda4bb3801470c464eb8c88 : SUCCESS in 12m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/249b020a90664a6da3982d22edbd4056 : SUCCESS in 23m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/30563f3dee314e4e9f00ecaeed87f7ae : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2891e78cc3aa42efbbf8c2f1e6e52745 : SUCCESS in 9m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c85587d34be84b35aac8f6c8bdb4c6c4 : SUCCESS in 13m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3daa6f5138d4f9296c3a72dd8ab8334 : SUCCESS in 8m 25s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/087046c37367440dbc767278fcc4edaa : SUCCESS in 4m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/13f80ddfde8549d29cb13709f406d734 : SUCCESS in 38m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/922476d09c0a4d5c9065d8fb94f17e42 : SUCCESS in 24m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6701ac33eaa1459d9285885fec7f633c : SUCCESS in 24m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bb24f3818694632bfeee46b15b943e5 : SUCCESS in 11m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2de8045d9547bba9af8d52102fc483 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01025a0c40494b5da8d305b7c199bbf9 : TIMED_OUT in 1h 31m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/658bcb2107fb46089c6b1e671fcc1fbc : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"371903fe03a85bc62d38fb3a7a99d3ed1f4d8090","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-01 22:04:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cdf1da4f5af3e118b59e08ae51106c52ba8def69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 22:49:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/275b6294d0414b618135c452923f6ccb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8557bda778b44c3bb8c149d43f6bf3b3 : SUCCESS in 43m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0eaeb504c264fa6a779164285a678cd : SUCCESS in 26m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/746e5948a416443a8b6f1e13f00d6d5e : SUCCESS in 27m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/efa3e9e770b14c5b9193899939ecc955 : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"09232c2da229c03c80e4a476aab87665b80a7913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 23:15:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcce7f41f8ff4d58a385253f36696cef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333356e1c4f14f04a477f85da51df756 : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/06d6b21f0d5343828360898ca87717ff : SUCCESS in 58m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/36b7b6679ed7406ab061a535ffd46e59 : SUCCESS in 47m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0199470233274bb29e3ca2056230d10e : SUCCESS in 1h 05m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/821132ecf36f4d6bb745030c34101ec4 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b544aa6e144844ba8992eac996b6c659 : SUCCESS in 16m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cff969e174f94fb097e2f0daf15f2f51 : SUCCESS in 21m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fbe36aa998c0464fb1761e93dd759114 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2651a72f2eda421ba4f81d2111baf689 : SUCCESS in 19m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52da370165614b82b91b3c4ee35129ad : SUCCESS in 20m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b2e07bf97f4a42ee901e9e64056fdafb : SUCCESS in 20m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ada79302bcf343bd8db4754b5509db34 : SUCCESS in 21m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/551134f030084299b742e51240cb41af : SUCCESS in 27m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cbbb80640d2243b3aa070e9e6241604b : SUCCESS in 22m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/27b2a5accee4452f8e6a67422e6c007c : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/64cf8e8e222c4d6092c09a9a0027a499 : SUCCESS in 18m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b2d3e065b9604e1ab4822307a7f6a739 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2db0df64c48546e8ac1032579c81ee02 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d28b80120102429d83be3985bc1ed27e : SUCCESS in 18m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/df4b168fd6c34bf3900e803a0b4b4325 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b125a23acbe04824acb1a09fb411e057 : FAILURE in 16m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5780bb5a944b4979a99a6a0d9246c91e : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b386eb19e5a4767ab3c3e1c7c3d85d1 : SUCCESS in 47m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8259e19e167640f5b7b934cb58ab3c13 : SUCCESS in 31m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13bd3ac3179441d59e63fe9dfa7d39db : SUCCESS in 51m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f57a285649c04f4096057698f42c3096 : SUCCESS in 24m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32ea5ee3fe354713b762c14430a043a9 : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ebd828e38684311a8a08ed692e784d1 : SUCCESS in 14m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6965ea7e9c6445cdbd61b8a34258b5c3 : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a7d911ff1e82fc1178944ca9992c8304d5974e7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 23:15:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3e3c9674f1888246012aaacfade1011ab2a3835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-02 00:23:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e655a8d9ea19427e816defbbc6775381\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe9254babf34a4da538eb1f2f5c1fd1 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/3b77d4269a094944b09d5567fcad079c : SUCCESS in 40m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7362e2ed05846d7be92ad0f563196d3 : SUCCESS in 51m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ead8fe6f011a4ab2b9a3b4b5206497fe : SUCCESS in 1h 02m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c64382c3df8741088dc2e5ff6259bfec : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/900f1789bd2a4fd8a33cb81b88bb1474 : SUCCESS in 27m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/79dd09b9d94a45bfb3aa7dfdc8f1328a : SUCCESS in 22m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ceafe23be37141a3978347c2f8c51b57 : SUCCESS in 16m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a616e6c1182a443eaa00f74686a4232b : SUCCESS in 18m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5041bbe0f73e4fd5bba8454667b81ae6 : SUCCESS in 21m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5fe22c78fce840348d244d30bf25771b : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a8ff6eeaabbc42fd98cfe315eae1d976 : SUCCESS in 23m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/33961a48ae90440fa94b93145c420db3 : SUCCESS in 24m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c3f715daaf1b46b4b8e44fd371d8e70a : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/850d11cbeb594981a9a416b5c58bc623 : SUCCESS in 18m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a57df71f25c34076a3939e39b98170ba : SUCCESS in 18m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2223c1e38b0c41098fa596291a043146 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/45def88a39bd408cae57152290c0f6a8 : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b0737548d6f40edb83bb7bb86406e21 : SUCCESS in 51m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae5bd28c715a4d2b860a8405b72f4668 : SUCCESS in 56m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e106c1f048364545ba230a9a06569a21 : SUCCESS in 47m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0200d89d1af44f0d9dffecec52b504e4 : SUCCESS in 24m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3375e134483e4596b72355045abeb3d3 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a436ecbd2bcc4012a7e2575a92ec7029 : SUCCESS in 19m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2feb803295874f92aafa91fe78438863 : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":1},{"id":"405a394342d3913a8bba603242c416d86972122b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-02 00:23:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff9127550614fe8fc61cc247e7f50bf4950818c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-02 00:24:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/891c41b81fd141fe840f5f41ea54cfdc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae87577b3784c75bc2f6ead7421a692 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/150684b58eb44207a9e7caa96c6ff85b : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3ee5952f66d2daefbd0f4abca1569a3e661a134","revisions":{"f3ee5952f66d2daefbd0f4abca1569a3e661a134":{"kind":"REWORK","_number":1,"created":"2025-10-31 17:25:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/30/965830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/965830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/965830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/965830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/965830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/965830/1"}}},"commit":{"parents":[{"commit":"9439a6257b1acce192d9a6ebb92ba65d637ddd34","subject":"CI: Fix py27/py36/py37 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9439a6257b1acce192d9a6ebb92ba65d637ddd34"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-10-31 17:24:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-10-31 17:24:42.000000000","tz":-420},"subject":"[stable-only] Pin pyperclip\u003d\u003d1.9.0 for py2","message":"[stable-only] Pin pyperclip\u003d\u003d1.9.0 for py2\n\nApparently this slipped past us when setting constraints until now.\n\nChange-Id: Ib95a3368ae2beae06b206d400cc358e5fa6637e7\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3ee5952f66d2daefbd0f4abca1569a3e661a134"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3ee5952f66d2daefbd0f4abca1569a3e661a134"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
