)]}'
{"id":"openstack%2Fswift~965863","triplet_id":"openstack%2Fswift~master~I0f9784b7142ae44ddc3e1001036211b0956fc2b1","project":"openstack/swift","branch":"master","topic":"post-404","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-11-04 00:12:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0f9784b7142ae44ddc3e1001036211b0956fc2b1","subject":"fixed two more cases:","status":"ABANDONED","created":"2025-10-31 20:57:08.000000000","updated":"2025-11-04 00:12:45.000000000","total_comment_count":7,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"754a8a7f0984849f1c99b0183d04d3e015c0269e","_number":965863,"virtual_id_number":965863,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-03 22:09:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-31 21:32:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-10-31 22:45:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-04 00:11:28.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"3101ecdcc726a848523224dd2237d632ddfbce49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-31 20:57:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e10e8d248c78a4e2a61e94def798f5982e51f7eb","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-10-31 21:04:16.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"479826f7aa33d93ee5372ae46981b73932bcdf51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 21:32:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dc7a3015e7e4bcdb18b8e1b25d80442\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6330924ceb884e3298363dc78430c1e2 : SUCCESS in 33m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1546fad36a837589810086664d39d4cbe6fefe22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 22:45:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4157326138644404a47486b36310284e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72c477f64dbd485fa1a73c8a93e92788 : SUCCESS in 11m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0f46c9c122e5487e813b8099973ec70f : SUCCESS in 1h 09m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86187068023b4d09aea46b1b6cffc866 : SUCCESS in 48m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db3be94b783940b28e0d887a921b7edd : SUCCESS in 49m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80b12f8a1ce9480abbe5ffd9ff4c2765 : SUCCESS in 1h 05m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9355078f3ae84b969b1c0c2a8142c9c3 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/897abe3ed1c94343989d5e626506555e : SUCCESS in 11m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7865033a31774135b49bf1c3cfe107e9 : SUCCESS in 19m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cf8534e98f745e993ce1f0010c583ce : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f84c2b4984234652a9a078052cfe160a : SUCCESS in 19m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a2298871d54456b89725425f5b97cf0 : SUCCESS in 21m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bfd10d078501482fa361aceaf6e35d16 : SUCCESS in 20m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1b4deebc05f84734b6b5e458ed094710 : SUCCESS in 21m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e1a5bdd5ef9d4dcea5840b90582bb586 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3c6a70dfc01b48caa83564353b9273ce : SUCCESS in 14m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fc27f09e2fac4613a39dc9531e448284 : SUCCESS in 15m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/773eb58549f94fc49ff515325a25a044 : SUCCESS in 6m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3623d37bf57d4aca90be0036de7b5ee4 : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1669c888b31e4116baf4886d5ec75501 : SUCCESS in 7m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c6f418228d2493485c27306d616c781 : SUCCESS in 57m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c814dcd749543268a38fa8525061949 : SUCCESS in 44m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0072f06934fa439ba5946704e03fde3f : SUCCESS in 42m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/437aff88011745d3bfb9179bdb48b55d : SUCCESS in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1862b391371a40c9afb0b0df17784b41 : SUCCESS in 2m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/367cd01b7f7e493bbee5ea27f8010f12 : SUCCESS in 10m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfd54df71a4d4d68b53b7d45062db2b3 : SUCCESS in 37m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa19b6b5d9234dc7a78a639c5297f564 : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":1},{"id":"dd2e2ad6f8fa985626fa8dab2b67d775284eff25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-03 21:00:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2d424de5f6de584f29d52f31e876a5e83d7ac2d2","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-03 21:05:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"59aab5631b76bc6b0f2196d23f8dbf4cc549fe5b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-03 21:30:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"38264038458ac95a7dcac18863b5fd08de4cb2b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 21:55:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8fbb9e0fac34ce5a96d1189c35f7071\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8da90802680d48508f42d3c8da086351 : SUCCESS in 53m 03s","accounts_in_message":[],"_revision_number":2},{"id":"03a0cb9d6d2133792501eddba1d9e7647de83dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 22:09:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ccfdecf15284649b66a68310a6e489d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e89ff9e77764974aa3db699eab3f99b : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/645b4669939344f7b0948a30a050c8ce : SUCCESS in 49m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44d42bcc9146485f910642eeeb99dd39 : SUCCESS in 1h 01m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/153187b40626482199def8a13ad9843d : SUCCESS in 36m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d2eaa9ef54e40019bc2d5be70bd9aaf : SUCCESS in 1h 05m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05710f8864db4d0c8247c51144af95fe : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8faaed6f419d4fcfb682d76706a2efa2 : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/452105b5668b4171870985cd9eb6b343 : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/42aea4ca805d41939e62b6ecd22ae50a : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ecb8b116622f4a138268f02f9fb1242f : SUCCESS in 11m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9db3e6b76118463aa151518f9a51504b : SUCCESS in 22m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/50e4ca4d78434cb08a2bbb45ab46328a : SUCCESS in 20m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2be266d811c34c0aabf9aeae910db17d : SUCCESS in 29m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/598e1df5caef416382934e42a508d2aa : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e9d9d9834073432199f3e1c416d43f0a : SUCCESS in 16m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5b327f4e8f7c4f2bb8e5d8146106982d : SUCCESS in 9m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/acd8851526e04554a4214ca39a09477f : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/923bcace40e8491d825985ac329fae65 : SUCCESS in 11m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7c9a08fd4bfe402a8834b182cd726a16 : SUCCESS in 8m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/458c810bc5e443f1873974e73cee7ea7 : SUCCESS in 53m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a35fd92317f44285842c47806c19a7ee : SUCCESS in 54m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/96168bcf8a2a469cb06ae921ff4b04d4 : SUCCESS in 50m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf2d78e947c9479bb86b20eec16429e0 : SUCCESS in 22m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba15576a683448b6a8966903748ddaa9 : SUCCESS in 3m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/19cb9f482407463abb3b26aff3ab921a : SUCCESS in 10m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ada7c1d82b347a8906998e80bb1db84 : SUCCESS in 44m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce9fca0686a940d0b45bac8603a732be : SUCCESS in 42m 52s","accounts_in_message":[],"_revision_number":2},{"id":"ad9dc4005febe61bb1b68193ce329275dbb9da40","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-04 00:11:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"754a8a7f0984849f1c99b0183d04d3e015c0269e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-04 00:12:45.000000000","message":"Abandoned\n\nsquashed to 956411: Return 503 for POST request with mixed 202/404 responses | https://review.opendev.org/c/openstack/swift/+/956411","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9e02e2edc53e86700968be2657903c7846cac6a5","revisions":{"2fee42125df91918b8b1f022c3e3e3b1622233cb":{"kind":"REWORK","_number":1,"created":"2025-10-31 20:57:08.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/63/965863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/965863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/965863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/965863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/965863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/965863/1"}}},"commit":{"parents":[{"commit":"86f255751f35918b0305a89a2c5baf1f109bd571","subject":"Make extra POST to handoff on mixed primary 404","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86f255751f35918b0305a89a2c5baf1f109bd571"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-10-31 20:54:12.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-10-31 20:56:47.000000000","tz":-420},"subject":"fixed two more cases:","message":"fixed two more cases:\n\n   [202 404 404] + [404 404] \u003d 503 (instead of 404)\n   [202 404 404] + [202 404] \u003d 202 (instead of 404)\n\nChange-Id: I0f9784b7142ae44ddc3e1001036211b0956fc2b1\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fee42125df91918b8b1f022c3e3e3b1622233cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fee42125df91918b8b1f022c3e3e3b1622233cb"}]},"branch":"refs/heads/master"},"9e02e2edc53e86700968be2657903c7846cac6a5":{"kind":"REWORK","_number":2,"created":"2025-11-03 21:00:29.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/63/965863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/965863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/965863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/965863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/965863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/965863/2"}}},"commit":{"parents":[{"commit":"86f255751f35918b0305a89a2c5baf1f109bd571","subject":"Make extra POST to handoff on mixed primary 404","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86f255751f35918b0305a89a2c5baf1f109bd571"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-10-31 20:54:12.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-11-03 20:58:04.000000000","tz":-480},"subject":"fixed two more cases:","message":"fixed two more cases:\n\n   [202 404 404] + [404 404] \u003d 503 (instead of 404)\n   [202 404 404] + [202 404] \u003d 202 (instead of 404)\n\nChange-Id: I0f9784b7142ae44ddc3e1001036211b0956fc2b1\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e02e2edc53e86700968be2657903c7846cac6a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e02e2edc53e86700968be2657903c7846cac6a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
