)]}'
{"id":"openstack%2Fswift~965895","triplet_id":"openstack%2Fswift~master~I53318892af799b4888484d5939f76c000154e5b1","project":"openstack/swift","branch":"master","topic":"cohanaja-replication-mode-config","attention_set":{"38368":{"account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"last_update":"2025-11-15 22:52:43.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-11-10 21:05:14.000000000","reason":"\u003cGERRIT_ACCOUNT_38368\u003e replied on the change","reason_account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}}},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-11-15 22:52:43.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"hashtags":[],"change_id":"I53318892af799b4888484d5939f76c000154e5b1","subject":"Adds rsync/usync/auto replication choice config","status":"NEW","created":"2025-11-03 09:13:16.000000000","updated":"2026-09-29 22:16:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"84b097c4b16201adabc95e6fab7e0b5d9038f5f0","_number":965895,"virtual_id_number":965895,"owner":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-29 22:16:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2026-09-29 21:05:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":1,"date":"2026-09-29 21:05:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-03 09:47:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-03 10:27:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-03 15:45:12.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-11-03 15:45:12.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"910ef0e827ab454dea7d704c2fb8814f91128225","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-03 09:13:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc16b504cae4e0c5e4122b8b07239a8d3ec27636","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 09:47:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dd00597af5c4d30ac3c4c2dffe05c2a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/45531175dfd9438394b4c84764e4a115 : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e365b2dcbfaaf19cfb4d0b8d770db9854e5f3720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 10:27:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/724fe018f0b04260a40880308a5b32ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ea2f32eb204a4189248f0f65808c24 : FAILURE in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/5eb2d63320a64a909f81610d81a0a543 : SUCCESS in 1h 03m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/168cc9f83a6949838ffc9b4facd63633 : SUCCESS in 59m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/42de4618ae654445a4428fdc045e057e : SUCCESS in 52m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcaed4719b7f42c9899eaca12808797e : SUCCESS in 1h 00m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b2a5c82b7fb40dbbe078d1f1b663fa5 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f66fc30a0d15457886beefc1864a5392 : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9046772f6daa4e0d91516a675c74f689 : SUCCESS in 21m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/41f0a0fba52e43569356208ee21ef7ff : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b5f09617367740f296edd0ade3d54b1f : FAILURE in 21m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/685a221f159f4b2989d1e7e5578dc507 : SUCCESS in 15m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cd09aea118174a61975619d8cd739ff4 : FAILURE in 32m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a4ceba2cb2c441808a8f1f28881b55a7 : FAILURE in 29m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/26ec785c963943bf8bc3e1215c2c95a7 : SUCCESS in 12m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3c05a47094354a2291d9fc3c42447a4b : SUCCESS in 12m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8048e6b8226d4ad8aa029ec34c3676cc : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ff54b632f5a24da8820309b650945daa : SUCCESS in 6m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f74a07420ec74f0c9896dd447c98d7d6 : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/35b35429d6634943bd70b43b968753e5 : SUCCESS in 11m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b103cc7309e4c69adbf71c3c81e6719 : SUCCESS in 1h 09m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/54a19790452b4b2eb10bed435dcda83f : SUCCESS in 57m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f97d2cee533f43b7b68114397dea2007 : SUCCESS in 33m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50b9f70baf5548af8dab095b8f0ed277 : FAILURE in 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c68837ca3224ec5b6b7209810d2fa6c : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f2bd4a170f645e3b86d48607fe622b7 : SUCCESS in 18m 12s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bffdcc3951de466292f20b84c33a44ef : SUCCESS in 52m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8efb81b863bf44969f5ff3ffa3a8b7c4 : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":1},{"id":"dd1f8c17d9676052d58338614329ab52a3c19419","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-03 15:37:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"850ddd4de21a314954c9a4c38761c7080cc9c90c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-03 15:41:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f69d54788b390f6b50111620c55a8e38bad6cbbc","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-03 15:45:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"715aba9e03bf1701db099d68d3a410b7aa2bc206","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 16:34:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6e697a9c34a498abd93773dfa8d422d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/458e604629a5485c9bd776e06a783235 : SUCCESS in 50m 56s","accounts_in_message":[],"_revision_number":2},{"id":"76c4eda308d83f3d2e7d2c6dbcd84da6aff25ae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 16:51:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dda6d952ac774ca79c21a8121ee4ae2a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea77c23382341458e482966fcad55e9 : FAILURE in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/07248024a74b469da5bbd0b739e659bd : SUCCESS in 1h 03m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/469d05dfd6a14c9aad8a796fb7927ef2 : SUCCESS in 35m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b3d97c6686b1419fb809b51982027765 : SUCCESS in 54m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ba5e1e07d1b401481dd37e509fc3159 : SUCCESS in 37m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03a09a9bc65c4dba87e75bca014a1284 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9759e3f1e3cb4803bcd18ea94e37a3e3 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f90bd3d62d24eea9a61e23c106a8a69 : SUCCESS in 18m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07c51e2ac7d14ab988d62f0ffdeaf673 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/baa778cbb12147ae91c802600193807a : SUCCESS in 28m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d0991b9a78249b7bf07237af1b8d51b : SUCCESS in 22m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/56bfff6bca774ba38e35621e10802745 : SUCCESS in 31m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/82ec036fc0e441e19f58b70fd862a408 : SUCCESS in 20m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/868852af2f1647448db7af830997fdb8 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6cbcba76387d4fb4aa32968edf413155 : SUCCESS in 14m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/743fa33eb8af47cd9bb12841f983f2c8 : SUCCESS in 13m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96f028dcbcd042d2b313029bde2a78a3 : SUCCESS in 5m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/99583317235943529a5ae10712fd9c24 : SUCCESS in 11m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bf5c15921cfc4d829025efe27e14c7eb : SUCCESS in 8m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e98f6dc2e46544fa899efc09726a60ff : SUCCESS in 59m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/93f351d0071c4216bef3e31314d7f7f5 : SUCCESS in 49m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/44207ea7569549bfa06246e1f39c61b6 : SUCCESS in 45m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2eb29b965a941ee87813d643ccae28e : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/890b7db59f07422e9b8fa6bd1214b47d : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f89904071f147608d58c2aed3a2f8be : SUCCESS in 17m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b90f599a58e4db9aac3781a2db6afc1 : SUCCESS in 44m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/668cb75d9d92402a9d9b65c9dd98f556 : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":2},{"id":"02ad0f20f130ef8f71f59e8869406971019b32fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-04 00:35:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e5053820478de01bff9e77ea480df325eb1b245e","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-04 00:37:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f72918af892b40837d1ca1b72f746ea71f21af63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 01:26:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeb183238b2a4754af7f01e58572222a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/90b9032f60714ae78cfc085617067ac7 : SUCCESS in 49m 58s","accounts_in_message":[],"_revision_number":3},{"id":"6103a453ae497601c8aa15657f46a17655b8e145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 01:45:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6bdb1b23ab04d6a8755a39c7fa2d586\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba339eabc2c4dab978f1bfcd40d8333 : FAILURE in 6m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/8ed17ccced4c4dfe8a53be967f3414f1 : SUCCESS in 40m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d60632c17a4f4ac6acdd720f1cbe7ec3 : SUCCESS in 55m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f9549c7718214f0bb495e82dc0a48fd1 : SUCCESS in 44m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3229cf063553409cb532d89540a9b394 : SUCCESS in 1h 04m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/967471a5eda94457a62ca64148f773e1 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3725e8dd0fe540b48271e21adbf7cf5d : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6349dc4c53f4d10b3b5c59772c7e5b1 : SUCCESS in 13m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d01dff6d4804cb4ba850d40a4010fdf : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/49ee155bbb2b43de8a90e16e30bc1d63 : SUCCESS in 11m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/623a6279314b49d9b1975aada0f5d799 : SUCCESS in 14m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c3ad1c7d9af343099ec483840528f900 : SUCCESS in 19m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/93e1f906559340d282b6d4a53cbe0390 : SUCCESS in 17m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d66754039710459a8382d750f88af481 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/54ba38574df24179910cb1e685ac4e1e : SUCCESS in 14m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2335c0e666c943faaba48f987bdf06a8 : SUCCESS in 11m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/abe302b1659f4810a19ed45dbde70948 : SUCCESS in 10m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/33678a0c89ae4df3ab4ad568e7b00e1a : SUCCESS in 12m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/81c8d9016bcc49d79d9f30af7dff274a : SUCCESS in 8m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bce5baf64e5d457883604223ce0f1680 : SUCCESS in 1h 00m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/32d73187202a40b6a319c8d03c28e25a : SUCCESS in 54m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17bfb62684ac4ab19a956a4f85b4b8af : SUCCESS in 34m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3ac138361294b8ca1b606c3983563c1 : SUCCESS in 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2cb0d97a9054dcd93cf1c8ca4291706 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5f7ac7314bde496ba720ccd633ca2d57 : SUCCESS in 18m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a79a5221af64ddca13ef1bd297870bc : SUCCESS in 37m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d897963f39764ca6b16ac9fc3b79b469 : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b2458ae8cecd3d7b5751a0e6ffdd051e2d9b01cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-05 16:48:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bc0d577385876e9fcd3e513b4d8b42d47a12d760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 18:05:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5aec695fa26042e6a9654c812a34a814\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f31089b0bb84c319813eb57013b93df : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/05feb0083d9e469080be909ba1ac76fa : FAILURE in 41m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9627b138ea144125ba2d7b1e5b14b5ad : SUCCESS in 47m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3f8e1fb15bc4cc7ac4a83e54463dd0a : SUCCESS in 44m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3c3b64b38e54f81b80f72f75af5336e : SUCCESS in 47m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71e59a02387946eeb35d8385b4168eb1 : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2ec1be9a09d48ef8aeaecfccadb1dce : SUCCESS in 12m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec3d1eee153e4f94bd2314717bf1d2b0 : SUCCESS in 13m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3d92d8c864a4bde9b321e5570149300 : SUCCESS in 7m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f010ce8d52e94d5d874260b489f065f0 : SUCCESS in 25m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/43fe2b9a32554b82aa63ec2b311aee93 : SUCCESS in 18m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/290f8a7bd05c452897dfd9bca21e751c : SUCCESS in 20m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b5f6bccc345444b0bbafa7d120d204c1 : SUCCESS in 25m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/04a379b6e8454489a2e2e7508316db92 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/db823d3d081347bdaa4757f9c13780cb : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/392090764ef747b8b496e03555020286 : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/98cbed41666646d8938994779cebf95b : SUCCESS in 7m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/67852b3d3d61405f95fe13af17ffeb8a : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fc8833e47ead4c7eb270a90c5c162107 : SUCCESS in 7m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1899e863974c46fcbd0c9261e28bd1b1 : SUCCESS in 42m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/353a791468354880aaa17daae75cc4c3 : FAILURE in 22m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/969e9870079144fab9c07185a5503a33 : FAILURE in 22m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3147e0326154485d8921837840eecb25 : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f423dcd0af54ca286ffe0e4b3119919 : SUCCESS in 3m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6fd3083b60ae44b68acfa713c2d48477 : SUCCESS in 20m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/372d1ba0052a4aa2bf8cf574e8f3e00c : SUCCESS in 56m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a304ba8368614efc9df03fddac50a943 : SUCCESS in 38m 42s","accounts_in_message":[],"_revision_number":4},{"id":"3b98d90ff55befcfca0681716c5081041ade2341","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 18:28:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/273f0194aecc4c72ad4723bc9f7f60aa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e8e0f15368f410686c9f34361ae0a3a : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":4},{"id":"ec18bdd1b54800b5e0e3ec9cfb65ddf229c83a1b","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-05 18:34:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"15756735bc304ab6d29551eaf4babbd3dc34c357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 19:46:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/512cec153f664944b16ce0123f02f0f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3f59b467f84028817f798423ef8740 : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/0c48c3f684584b8e9e27be097a683344 : SUCCESS in 43m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/03e1e871a1ca4e589f76a3e8eb6615cf : SUCCESS in 1h 05m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eed82b4de3024afbb8d4b8c540f15315 : SUCCESS in 51m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7640d556a134279a7667799db06e9a5 : SUCCESS in 1h 06m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf6debc2cf1744a18ded54274720e47b : SUCCESS in 11m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/241a7de0abb4412fa3ce2dc670d731e5 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2e18070ce2d4c85b3628aba68200456 : SUCCESS in 17m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/86d5d52de2a643cb8d57759f01af2a69 : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b483c1c57ac04ab1b261e40c94c09fc4 : SUCCESS in 21m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b6ac8a0f3e724ecc9209476ae4c27b9b : SUCCESS in 20m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/649a0859a8dc4821a45c3d6e44cce67f : SUCCESS in 26m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/87a2a1c128874c988feada097f1b3152 : SUCCESS in 24m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/646f0ebf5b6842b3a68fc6fdfb6d2f6c : SUCCESS in 9m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b836ee932adc408ca98d208ba53ce7ca : SUCCESS in 14m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2fc4b2acd35242849158f512d6f79a2b : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3057f283b4c44a108f393a4032ffe7e8 : SUCCESS in 6m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f55fb78ab55548e8a4bed01882a34c1c : SUCCESS in 9m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cd56815e77184739b3c1ca75ca0be699 : SUCCESS in 7m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6466562a6b35492b8ae141e2a629b8f8 : SUCCESS in 45m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/31762a870cd84cb8aa6c3cc04a462584 : FAILURE in 29m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d41788804eba4c6d911be651cffc76ee : FAILURE in 22m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dac11d0fcf84975801470ec11d6bdef : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c288c7dd277f44e68bbb9979f66f7c4c : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/27e71956bb454269bfb5f0b4e978e4b2 : SUCCESS in 18m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ce790efdb56426394d4dd2466e726e6 : SUCCESS in 55m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6714d20b8bf4409daa114bc341940cec : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":4},{"id":"e738dc244341b08606429d0b7a04d7a11a8c260f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 21:58:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e3c46ffacdc4f01b663f5601b35f870\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8627dd9dc4514e71a6805510bb070f35 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":4},{"id":"ce4cd7a53095715c6adf705a9a86de09e89378d4","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-07 04:54:24.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a8ef892f749e1d634058f02a07760ca1ab39485","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 05:48:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4df6be661e264770874e14c8520a8d45\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e74d1db9f98d4cf8ae769a3ca09bf6c0 : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":4},{"id":"fb448218df4724183433444dadf2a4e2f975a1bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 06:03:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e7bcc66705840b58157db1a4075c034\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e702cd942c4c59ad31dcb26f89d56b : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/451fdd123e5f4e339055b0b3953348bc : SUCCESS in 31m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3005486c580e4533b8e1a9758042de5e : SUCCESS in 31m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4f85c8986f14c089a0c2212dcde29fc : SUCCESS in 53m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ec66bbd8407400a85af3f78f4b53683 : SUCCESS in 1h 06m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29ca61abfd6c47f096ae190fda322276 : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b230261b9d524f9cafdbec6998923ee1 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c66e7bb274e4ff197486abd96867d76 : SUCCESS in 21m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d384181a0910427b8df5626191ea672f : SUCCESS in 7m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1dfbaff48b1477c9cee2d73612de1a2 : SUCCESS in 25m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d1d01180c1614163ad77b4a18ede5ef2 : SUCCESS in 21m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/429adeebd0d8435dbfa43b6f83db3fa6 : SUCCESS in 28m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a94201b4620440bda2dabc3c886afe12 : SUCCESS in 29m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/999d5374474e468990927f5772ac6d75 : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f75d4930143f4cb797a03ead574a67a6 : SUCCESS in 12m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cb3e19cfc5644e419cb168ab6c1cd12a : SUCCESS in 9m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/237630dfef524de29c378139f4772945 : SUCCESS in 6m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/383f1715d39f4c848d646862c729e1e3 : SUCCESS in 15m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5d603f8f2a38431596c2c7e862a94fa5 : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d311b000b32c4552b3d6907413cb1c81 : SUCCESS in 56m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9618f222d842460da41f88925919026b : SUCCESS in 45m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4ab444324633429c8bf947ab8b900d81 : SUCCESS in 33m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/728811e746784490a97c57dd3780e9e5 : SUCCESS in 26m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff3bb13b44af4aa79c7503e036f4e272 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24f250e84f7e4753bcd6adf03793af80 : SUCCESS in 15m 03s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a20221a83858462eb725e13a8314d86b : SUCCESS in 54m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/231805959b384d1c8c34b19be8be1b97 : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":4},{"id":"ee53abc2f2bb6822b4f499c547d7469c7c97d677","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-10 17:49:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cac3f36ba511501ee1563d5b34a18d7e8b8412db","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-10 20:15:27.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"aaf846ed31654cfc4092e082f8ba1b573e60da59","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-10 21:05:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c48ab0204d12ad6b9e29347ef8f625f425c6805f","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-11-15 22:52:43.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6745d824b8ed531b44bee06b44e44c574542f416","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2026-09-29 21:05:20.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"e821490e10a1c4b7015d45b0fda61387e56032de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 21:41:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04db45bb50514df49d7ab15488de32c2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a0be04f3182493ea8846c880da1a101 : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":5},{"id":"84b097c4b16201adabc95e6fab7e0b5d9038f5f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 22:16:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e61f22f43aab4b16b021d9c5e6b7c36d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031daebcc8dc4935b40708cc25000501 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3582d60f4d6e46f6802d1ec02cd3514f : SUCCESS in 5m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/e97749f3137b4e74ad739c551dd2a129 : SUCCESS in 1h 05m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/36cd423bc3cf448bad090f0f0d212bad : SUCCESS in 1h 02m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fd8bcbb973e7407a91af6ee94741d96c : SUCCESS in 48m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eccb172853fc40e88aa476fa7847387e : SUCCESS in 1h 01m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afff19da41874a63b83d13598a73cc56 : SUCCESS in 12m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c54a905c25448849349270d05ba886e : SUCCESS in 21m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1393a9b812984a51a2f7a904399a79e7 : SUCCESS in 22m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea12d7886f034f298dd1afd2154f9ebb : SUCCESS in 10m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2d4503a60db64444910b75d605353d95 : SUCCESS in 26m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/afdc30a39bea4f1dbc12b180bbcf7f7b : SUCCESS in 22m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7f2c0c7cb0b148e09500cb92b6781ba8 : SUCCESS in 31m 54s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5d7af9ed98964d9395b60b763e359210 : SUCCESS in 17m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/52e49a3db96c4571ade671ebe9ad0b2d : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/08aabbee16bb47e4b7b63b3cf11a583d : SUCCESS in 13m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8dbacd0d154748e39e667b8c50d7a30f : SUCCESS in 14m 15s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0f4dd976e4941b087af47ff2445a4aa : SUCCESS in 19m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9f73dcee54447aca09e379a6a99218d : SUCCESS in 5m 55s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1f5b51e95d82485b899b9e462b1cfc3f : SUCCESS in 8m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bea36c329e94cc0bb677ce8896c53e2 : SUCCESS in 59m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/194d048afe56429ca6aa94c6c841ddca : SUCCESS in 56m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dd6d5db8f8c141f9a92e12227f3b9e65 : SUCCESS in 51m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/daeee51c9f854794a25fdcb63dc93f66 : SUCCESS in 27m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c4e42cb8f94214b440f8815e1d13ba : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23f4afe1fa254b6ca3b0c06dc69af14d : SUCCESS in 20m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c60c9c85e4a7474693c156959957998f : SUCCESS in 59m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7c441320ef5c7fed2ed1ae64fbb5e2b19c76661b","revisions":{"4619412fe5270367cf9acd4bce771b54a1f93c45":{"kind":"REWORK","_number":1,"created":"2025-11-03 09:13:16.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/95/965895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/965895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/965895/1"}}},"commit":{"parents":[{"commit":"3809085f8931be3b160357ce2bab92d742a0524f","subject":"Merge \"proxy-logging: Add real-time transfer bytes counters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3809085f8931be3b160357ce2bab92d742a0524f"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-03 09:01:46.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-03 09:08:28.000000000","tz":-480},"subject":"Adds rsync/usync/auto replication choice config","message":"Adds rsync/usync/auto replication choice config\n\nThis change adds a new config to the db_replicator and its inheritors\nthat determines how choose_replication_mode selects which replication\nmode to use. When not specifically set to usync/rsync, it falls back to\nthe present-day automatic heuristic.\n\nChange-Id: I53318892af799b4888484d5939f76c000154e5b1\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4619412fe5270367cf9acd4bce771b54a1f93c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4619412fe5270367cf9acd4bce771b54a1f93c45"}]},"branch":"refs/heads/master"},"02d9ccf8e26a7bc4b5cc2bd26abec75ba7fbcd45":{"kind":"REWORK","_number":2,"created":"2025-11-03 15:41:18.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/95/965895/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/965895/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/965895/2"}}},"commit":{"parents":[{"commit":"3809085f8931be3b160357ce2bab92d742a0524f","subject":"Merge \"proxy-logging: Add real-time transfer bytes counters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3809085f8931be3b160357ce2bab92d742a0524f"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-03 09:01:46.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-03 15:41:09.000000000","tz":-480},"subject":"Adds rsync/usync/auto replication choice config","message":"Adds rsync/usync/auto replication choice config\n\nThis change adds a new config to the db_replicator and its inheritors\nthat determines how choose_replication_mode selects which replication\nmode to use. When not specifically set to usync/rsync, it falls back to\nthe present-day automatic heuristic.\n\nChange-Id: I53318892af799b4888484d5939f76c000154e5b1\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02d9ccf8e26a7bc4b5cc2bd26abec75ba7fbcd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02d9ccf8e26a7bc4b5cc2bd26abec75ba7fbcd45"}]},"branch":"refs/heads/master"},"7daa313fabd3c853e6af4b25544d9253b152f0f0":{"kind":"REWORK","_number":3,"created":"2025-11-04 00:35:11.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/95/965895/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/965895/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/965895/3"}}},"commit":{"parents":[{"commit":"3809085f8931be3b160357ce2bab92d742a0524f","subject":"Merge \"proxy-logging: Add real-time transfer bytes counters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3809085f8931be3b160357ce2bab92d742a0524f"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-03 09:01:46.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-04 00:35:04.000000000","tz":-480},"subject":"Adds rsync/usync/auto replication choice config","message":"Adds rsync/usync/auto replication choice config\n\nThis change adds a new config to the db_replicator and its inheritors\nthat determines how choose_replication_mode selects which replication\nmode to use. When not specifically set to usync/rsync, it falls back to\nthe present-day automatic heuristic.\n\nChange-Id: I53318892af799b4888484d5939f76c000154e5b1\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7daa313fabd3c853e6af4b25544d9253b152f0f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7daa313fabd3c853e6af4b25544d9253b152f0f0"}]},"branch":"refs/heads/master"},"4586c0bdc3b35bee6897953c2b3d136d448a3562":{"kind":"REWORK","_number":4,"created":"2025-11-05 16:48:21.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/95/965895/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/965895/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/965895/4"}}},"commit":{"parents":[{"commit":"3809085f8931be3b160357ce2bab92d742a0524f","subject":"Merge \"proxy-logging: Add real-time transfer bytes counters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3809085f8931be3b160357ce2bab92d742a0524f"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-03 09:01:46.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-05 16:43:47.000000000","tz":-480},"subject":"Adds rsync/usync/auto replication choice config","message":"Adds rsync/usync/auto replication choice config\n\nThis change adds a new config to the db_replicator and its inheritors\nthat determines how choose_replication_mode selects which replication\nmode to use. When not specifically set to usync/rsync, it falls back to\nthe present-day automatic heuristic.\n\nChange-Id: I53318892af799b4888484d5939f76c000154e5b1\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4586c0bdc3b35bee6897953c2b3d136d448a3562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4586c0bdc3b35bee6897953c2b3d136d448a3562"}]},"branch":"refs/heads/master"},"7c441320ef5c7fed2ed1ae64fbb5e2b19c76661b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-09-29 21:05:20.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/95/965895/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/965895/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/965895/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/965895/5"}}},"commit":{"parents":[{"commit":"3a8cdee9b26b93b853e9163fe46850d1a0d31d13","subject":"tests: Fix diskfile tests on systems without EUCLEAN errno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a8cdee9b26b93b853e9163fe46850d1a0d31d13"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-03 09:01:46.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2026-09-29 19:54:08.000000000","tz":-420},"subject":"Adds rsync/usync/auto replication choice config","message":"Adds rsync/usync/auto replication choice config\n\nThis change adds a new config to the db_replicator and its inheritors\nthat determines how choose_replication_mode selects which replication\nmode to use. When not specifically set to usync/rsync, it falls back to\nthe present-day automatic heuristic.\n\nChange-Id: I53318892af799b4888484d5939f76c000154e5b1\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c441320ef5c7fed2ed1ae64fbb5e2b19c76661b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c441320ef5c7fed2ed1ae64fbb5e2b19c76661b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
