)]}'
{"id":"openstack%2Fswift~966216","triplet_id":"openstack%2Fswift~master~I65e6d1f17eecc4763eb1ac453591f6d6cb4d91c8","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-06-12 16:52:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I65e6d1f17eecc4763eb1ac453591f6d6cb4d91c8","subject":"dnm (sq?): log when all attempts 404","status":"ABANDONED","created":"2025-11-05 19:43:06.000000000","updated":"2026-06-12 16:52:53.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"190443bfeea304a377fd5e436bde915c147b0da6","_number":966216,"virtual_id_number":966216,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-11 23:01:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-05 20:49:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a20df281e83fd614adcdd8d9e342f8d12a0d0adc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-05 19:43:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21e8891a877e21119c72fc28df6f8c57a6779981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 20:49:44.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07b0b6665e384849884bbb17469c8717\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9295ecdc4464bae865df1c8706f09d3 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/04038849bf1949cb880f4bb770cd70ba : SUCCESS in 58m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2ec87b049cb04324be57f8ca26f6fc12 : SUCCESS in 44m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e63655c61fa845a8b54ef9261179581a : SUCCESS in 46m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fca99b1b7ed24f219fd39dc48e66d123 : SUCCESS in 1h 01m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/abfef53b876941f5bc372286c44eb4a7 : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db8320b9936a4d7e96bdc502e334aa84 : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19039a5330294a51be29512834818d12 : SUCCESS in 20m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5287652aeba84f628692b66648981f19 : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a029ba6d6cff4a45b37f66f26c740dd5 : FAILURE in 24m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/569a2d7ff70f4bf49f96314d314f0279 : FAILURE in 20m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6dd79f201ce74563b262c8c1a0fa438d : FAILURE in 24m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6f765dc15e1e42e09dd7b4dd4eb03de8 : FAILURE in 28m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/02462a494b3a458e8140cc9ae6332fc8 : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3e579d506f4a41918dfe833e0112f22a : SUCCESS in 15m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5ba3aa09754d4e51b367aa186085f56e : SUCCESS in 14m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bb2a2a7ffd634abe9de4b5d74c243478 : SUCCESS in 6m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7b6e290eb42449c984e421cf52737159 : SUCCESS in 14m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7bd969bb523d410eac6a10e2ee19d95e : SUCCESS in 11m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d682f77468f246a398579c91476a6221 : SUCCESS in 57m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aeb14f2816f7487baebfa2dbb0f1da00 : FAILURE in 29m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7620e14490c44933a643297ae30f3b11 : FAILURE in 28m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34a989087b8e441e9047b73e31c9cf22 : FAILURE in 22m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e37297cdcd74197b7f55cbc0806fecd : FAILURE in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f9cdf1f4ead246799b96ed4ca35729e4 : SUCCESS in 18m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75e4089bc82a40efb072cef0e68b5fc6 : SUCCESS in 51m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16a4f4176afe4053855c0dad74bb9345 : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d68043366d015a3978f2bb36dfb2b242fc383353","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 21:06:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e607a93ae7bc4fce9e107c132b70eb85\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aa6f17c63a43421eb542d00396dc2369 : FAILURE in 1h 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"478d7c4785555cca3a7c8d70c5a1d078175814bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-06-11 21:31:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"1d2c9892772b10c2997c4c8c40b192ebfe409b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 23:01:00.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cef7f70d97f9476593ec31f9ca0efb77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9338ed24ee8a496d9f7665ad44f8730b : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c979edde6c9498c815a68468b1fec2b : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/d4e58e026f5349dda7174ba4584a4f6e : SUCCESS in 1h 06m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9493c6c3ed64a55bdaef95c691037fa : SUCCESS in 1h 22m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8d245494e814070a4866c774c1b038e : SUCCESS in 57m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/095a7dfbe7a54fdf957aa23c1f25f97c : SUCCESS in 1h 08m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d40afe7144ce4d46913cf75e1571f03c : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e0f2c55606348fabc2464ae8add3fc0 : SUCCESS in 11m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/104773a1e7404117af79ee9ace6d9afa : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c1bea51e3104282a71423ce314c4a1a : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5866d0a908e1472cb18654b65b3bcbe1 : FAILURE in 27m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0257326516954e9e95c2e0584581acfc : FAILURE in 20m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/80b555307d2341b08313eb42dd631007 : FAILURE in 30m 11s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/476378afb0d2400e8fbe8def73bba9ad : FAILURE in 20m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e506f0d61a6745e58cbddd4e0b6bf174 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/20022a5f87ae4d28b544410711be5332 : SUCCESS in 14m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/943cd4d16eb642d1a8105ea96adeba59 : SUCCESS in 16m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c7c1a1796ecb4de4ab5ce721755620e4 : SUCCESS in 6m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b39e5e0c697d474da59a71af83469f1d : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fa0057fb5a2b48dbbf51106cbcafdabd : SUCCESS in 11m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/025b392bdc8f49989191e4d0d4632339 : SUCCESS in 1h 02m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50d75fc64d8349d9bbd579f3eaa3a077 : SUCCESS in 52m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b223e2f72b74385b045963492e6ac57 : SUCCESS in 53m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49e685ab829a4fad91348440265305e3 : FAILURE in 26m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0b545c5b9e443b186564f32d84b3442 : FAILURE in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff21d4b65ba04dd8aa626e23b8b597bd : SUCCESS in 19m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db6760e495cd4062bf7d5fb8f90c3330 : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":2},{"id":"190443bfeea304a377fd5e436bde915c147b0da6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-06-12 16:52:53.000000000","message":"Abandoned\n\nbetter as 993154: sq? Split object-updater 404 logging | https://review.opendev.org/c/openstack/swift/+/993154","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"12125f3573e7de8819b8388dab8ca4f70f886e91","revisions":{"1b78c6c1dd92fb559305e5242a9035c5637bdb28":{"kind":"REWORK","_number":1,"created":"2025-11-05 19:43:06.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/16/966216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/966216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/966216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/966216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/966216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/966216/1"}}},"commit":{"parents":[{"commit":"16e4c0ac5eb1e4ef7517d4b03a4715ce6c5084b8","subject":"  Object Updater: Log when all primaries return 404","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16e4c0ac5eb1e4ef7517d4b03a4715ce6c5084b8"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-11-05 15:01:22.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-11-05 19:42:05.000000000","tz":-360},"subject":"dnm (sq?): log when all attempts 404","message":"dnm (sq?): log when all attempts 404\n\nDrive-By: use replication\u003dTrue in node_str\n\nTODO: fix new tests broken by drive-by\n\nRelated-Change: I5525a7e615e12496fbe52a3a8767d325efc869ed\nChange-Id: I65e6d1f17eecc4763eb1ac453591f6d6cb4d91c8\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b78c6c1dd92fb559305e5242a9035c5637bdb28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b78c6c1dd92fb559305e5242a9035c5637bdb28"}]},"branch":"refs/heads/master"},"12125f3573e7de8819b8388dab8ca4f70f886e91":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-11 21:31:34.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/16/966216/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/966216/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/966216/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/966216/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/966216/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/966216/2"}}},"commit":{"parents":[{"commit":"769bbc2dd34b4644a2304193f23e0c23c7501056","subject":"  Object Updater: Log when all primaries return 404","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769bbc2dd34b4644a2304193f23e0c23c7501056"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-11-05 15:01:22.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-06-11 21:29:18.000000000","tz":-300},"subject":"dnm (sq?): log when all attempts 404","message":"dnm (sq?): log when all attempts 404\n\nDrive-By: use replication\u003dTrue in node_str\n\nTODO: fix new tests broken by drive-by\n\nRelated-Change: I5525a7e615e12496fbe52a3a8767d325efc869ed\nChange-Id: I65e6d1f17eecc4763eb1ac453591f6d6cb4d91c8\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12125f3573e7de8819b8388dab8ca4f70f886e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12125f3573e7de8819b8388dab8ca4f70f886e91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
