)]}'
{"id":"openstack%2Fswift~966255","triplet_id":"openstack%2Fswift~master~I82792579f9bd15a898eaab8a6e2a43b24acbe027","project":"openstack/swift","branch":"master","topic":"p-test-warning","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-07-10 17:16:32.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-07-10 17:16:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I82792579f9bd15a898eaab8a6e2a43b24acbe027","subject":"tests: rename TestableMemcacheRing to prevent warning","status":"MERGED","created":"2025-11-06 11:05:19.000000000","updated":"2026-07-10 17:18:58.000000000","submitted":"2026-07-10 17:16:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"966255-p-test-warning","meta_rev_id":"e310d86887d16fda28b30fe5817d5a64e3762a9d","_number":966255,"virtual_id_number":966255,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-10 17:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-07-10 14:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-07-10 14:42:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 11:53:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-06 16:11:47.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-07-10 14:42:12.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"db1222e64d6319549b40244b2d504ac24df1aa4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-06 11:05:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d70f7bf02761fe53e7e81fb29c8fa318d11a987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 11:53:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/524d9cdf4c0f45149f8208be23f3f903\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30e1bd3d9534dbeb2bcd9e39ed434d1 : SUCCESS in 5m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/176d505348364fff910eff0f30ab0b31 : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a31f3a0b23c463aa8df59c29d6cc3fe : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c91e7dbb2ad34c3e9fbda590f67ced12 : SUCCESS in 22m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4053cfbef25d427fae209cdda16987fe : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fd398ed5360a4bd78e160fa4d85e4994 : SUCCESS in 25m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e7cfa614ab045b6938fbf81bd6a3027 : SUCCESS in 18m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6727ae490ed146c59dc06660ab0e320c : SUCCESS in 21m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/af1a599304ad4d97b6ed2c99af649de4 : SUCCESS in 27m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/48ed3a54c3674ec4b1e1defb77bfc071 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a99199eee98f4d388199e7838580b58e : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b12390490c6d4d1e86e4cc67a0d4963d : SUCCESS in 16m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/93f55ffbe1f94426b8f80fb611352e9a : SUCCESS in 36m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b00cc86d10f04041844a820738845e24 : SUCCESS in 26m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89efddcdcf984a17ae950e2de2b70176 : SUCCESS in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc14cbc02e040e5be16b6ecc12c11f5 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/22b5597bf4694af7ac13d4fa952b28e6 : SUCCESS in 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"57f1e70ec41c7d97e1451f35e86870db18bb1f9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 13:17:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e6d358fc8674c37807b6dcf899b551a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07a07e4e3f2945fa8295ac26ffb06f21 : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ad49d0261e0697f4343d6f9295a4ae4da74b613e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-06 16:11:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"657b2f2302d98d0613bca0e9c2785295986bdbdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-06 17:21:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"62da31ec45a5e887355fc277c5c033c14314ead7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 18:21:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c88a7d417fc40c9ad99add1f962ed17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d58a8169a3af49b79ab4585613d7518b : SUCCESS in 10m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a520c09119b04df9b24d977e79078b29 : SUCCESS in 24m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa1441141e994570add56020659c4159 : SUCCESS in 21m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/362a7333909149c682d938eade4cbace : SUCCESS in 15m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9bfa1c4557bc4e018121ba9a809f4663 : FAILURE in 4m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1b10b924936f4b4299b486dcbcc41604 : SUCCESS in 16m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ef059790212a42608be4e7c265cb3542 : SUCCESS in 18m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e7e9894bb2684002a76041978e42f723 : SUCCESS in 20m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9078f03290274c39bf594820521733c1 : SUCCESS in 32m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8d055d4d9d2e4bccbb8406a54b4890a4 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/83132495a9f74affa974944b4bb1f881 : SUCCESS in 14m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/43d222bcf8b548a79b181a867bb1251d : SUCCESS in 11m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/132e9169255047ce9547975b5e9fac49 : SUCCESS in 50m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/674acb2aadb94e139d6b6edd0684a499 : SUCCESS in 53m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a70d232a4521425a84369c88082b4679 : SUCCESS in 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20a2563fe0e0458ab56815fe84e9394c : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78cea081f11343f3a7340338ccd9e2a5 : SUCCESS in 19m 20s","accounts_in_message":[],"_revision_number":2},{"id":"89c50a49b7cfe27b2684952c53c1d08537bb6c3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 18:28:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf403f485334ec9b83aa0d7453a4d48\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7466dfe41e5c45a393af40deb9d2caad : SUCCESS in 1h 04m 11s","accounts_in_message":[],"_revision_number":2},{"id":"34f4f4a8f7df5cac305450cab08d4bc613e85c92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-10 08:01:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a5a83bcb30326b9bddaeda092ebe9545c113ce7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-10 08:03:47.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"824cdd1913bd4b632d06e7506085e4d35abf5c45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-10 08:04:26.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"666a811d7e7778753f5892824d129b5d0e2102a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 09:10:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eafa118c6b4e498f85e307d6a03d8f9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c8366d222914afa92c7dc1c665b20be : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/325a7df95b2649bab6ef3f45e6491b3c : SUCCESS in 4m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc16f4155eee4905b7bbf3d9554cb495 : SUCCESS in 11m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8219669718a4dcfa05d0871e6d0e7e3 : SUCCESS in 11m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb7a80809e894acf9a816b9ff71b458b : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3bc964649643461dab7f0abf27b91710 : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4be79d8462645a093223d543fdfe89e : SUCCESS in 26m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/19371017191a4ae384d5b19f9c31f44a : SUCCESS in 28m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5fcb0630ef8944118df8cd0633073fba : SUCCESS in 31m 05s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0566dff795a640bcac88c9fa4a685a79 : SUCCESS in 14m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b126d7affab342dfae7ddf169ae19bfe : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d2c8572b3289448daaf22e7a9fc5e334 : SUCCESS in 13m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4c161ddc9c45440d9f6f7bd44f17a2b4 : SUCCESS in 11m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1376b096e103421fa24d79ae080415fc : SUCCESS in 52m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/039494b593c647589889296d89dc786b : SUCCESS in 53m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/623628938bd4420e93f8ca0c69d31ca2 : SUCCESS in 33m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe35bc09667f4e8db8debd016d027300 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1c8244b88817483ca7b18841bb891e87 : SUCCESS in 19m 45s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"caf9a3af6330345424e51f36c74d624e2141792f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 11:35:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66fcc2155b3043beb15e4dba7ba7ac18\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4ac8c53d0fed42698e4e9f7205d4fca5 : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":5},{"id":"230cbec48d1a9cf58811864f4d30f5165819650a","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-10 14:42:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"df0afa84399f9e238e910ec566ee82bb4524f357","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 14:42:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4d12cd40eddc30e237f214b5da0380d8879e1729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 17:16:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc0db20bb66c4b8e9e03d8fc6aa204c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e47338c48e4543bead9cec30db58aa : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b17448642b7460fb4093c74145a9db9 : SUCCESS in 10m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2a8904b3a01d4b3aaa41102ae77295ac : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/826cd366553242439041e9a8584f167a : SUCCESS in 12m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/aed7c301f1d945f58e75f6d14347afcd : SUCCESS in 9m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c5582001a1544814bfd0fb720f1c9c6a : SUCCESS in 30m 33s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/251cb5952f5f4371b040ae7eccf87164 : SUCCESS in 10m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7a07bdccd0624af3a8a491fc57f3da3d : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/df0265ff3b8d497fa1ca349c401e63f1 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8cd57281d9154cfd92842a7ba8f9e15d : SUCCESS in 15m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2ed551a99c3d4f60a4e5d262ff114a57 : RETRY_LIMIT in 5m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/07b25d875abc47db944276c9843adfe4 : SUCCESS in 8m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a209947fff54e6abe88d6fee98658ac : SUCCESS in 50m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/73733f8cfe5d4cb7bd434d7ae44a13f4 : SUCCESS in 53m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9221bbc05c864cbba3c889e59bf7de33 : SUCCESS in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cace9c103014d85b2ae534edffa52af : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e1df3c272744950b3f090eeae2e8c4e : SUCCESS in 19m 10s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-upload-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"06f09d8855001ea45ed59ce25a0fdddeac3ad2de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 17:16:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"e310d86887d16fda28b30fe5817d5a64e3762a9d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 17:18:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dd8c9bf7a6140008a2eea3cbb70f2db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3c35ae0fea40b3acb9fb122d7c083c : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bf1518c4ee02427da5fd340fbf71faa0 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5f20c7609f85470c1b2fa865c39939567e7dea15","revisions":{"8a09a2a966735d5878c92e2511fae91a1d103e92":{"kind":"REWORK","_number":1,"created":"2025-11-06 11:05:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/966255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/966255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/966255/1"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-06 10:59:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-06 10:59:29.000000000","tz":0},"subject":"tests: rename TestableMemcacheRing to prevent warning","message":"tests: rename TestableMemcacheRing to prevent warning\n\nAmongst the warning emitted by pytest after running unit tests there\nis one containing:\n\nPytestCollectionWarning: cannot collect test class \u0027TestableMemcacheRing\u0027...\n\nThis warning is due to the class name starting \u0027Test\u0027 when it\u0027s not actually\ntest suite.\n\nChange-Id: I82792579f9bd15a898eaab8a6e2a43b24acbe027\nRelated-Change: I50ff92441c2f2c49b3034644aba59930e8a99589\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a09a2a966735d5878c92e2511fae91a1d103e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a09a2a966735d5878c92e2511fae91a1d103e92"}]},"branch":"refs/heads/master"},"f1d10b05848fddd47dfc69f5b98edc9fa9a3717a":{"kind":"REWORK","_number":2,"created":"2025-11-06 17:21:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/966255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/966255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/966255/2"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-06 10:59:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-06 17:21:28.000000000","tz":0},"subject":"tests: rename TestableMemcacheRing to prevent warning","message":"tests: rename TestableMemcacheRing to prevent warning\n\nAmongst the warning emitted by pytest after running unit tests there\nis one containing:\n\nPytestCollectionWarning: cannot collect test class \u0027TestableMemcacheRing\u0027...\n\nThis warning is due to the class name starting \u0027Test\u0027 when it\u0027s not actually\ntest suite.\n\nChange-Id: I82792579f9bd15a898eaab8a6e2a43b24acbe027\nRelated-Change: I50ff92441c2f2c49b3034644aba59930e8a99589\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1d10b05848fddd47dfc69f5b98edc9fa9a3717a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1d10b05848fddd47dfc69f5b98edc9fa9a3717a"}]},"branch":"refs/heads/master"},"37f469ef9d105692ded567a9d4aebbf1ca5966fd":{"kind":"REWORK","_number":3,"created":"2026-07-10 08:01:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/966255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/966255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/966255/3"}}},"commit":{"parents":[{"commit":"8be4f02ef61a09b044965134b9b7f95a5eb88bc1","subject":"crypto: Fix listings when pipeline was misconfigured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8be4f02ef61a09b044965134b9b7f95a5eb88bc1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-06 10:59:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 07:47:28.000000000","tz":60},"subject":"tests: rename TestableMemcacheRing to prevent warning","message":"tests: rename TestableMemcacheRing to prevent warning\n\nAmongst the warning emitted by pytest after running unit tests there\nis one containing:\n\nPytestCollectionWarning: cannot collect test class \u0027TestableMemcacheRing\u0027...\n\nThis warning is due to the class name starting \u0027Test\u0027 when it\u0027s not actually\ntest suite.\n\nChange-Id: I82792579f9bd15a898eaab8a6e2a43b24acbe027\nRelated-Change: I50ff92441c2f2c49b3034644aba59930e8a99589\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37f469ef9d105692ded567a9d4aebbf1ca5966fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37f469ef9d105692ded567a9d4aebbf1ca5966fd"}]},"branch":"refs/heads/master"},"945e9203b6d2b53718e25e11d3bd5653c8025244":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-07-10 08:03:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/966255/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/966255/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/966255/4"}}},"commit":{"parents":[{"commit":"8be4f02ef61a09b044965134b9b7f95a5eb88bc1","subject":"crypto: Fix listings when pipeline was misconfigured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8be4f02ef61a09b044965134b9b7f95a5eb88bc1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-06 10:59:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 08:02:54.000000000","tz":60},"subject":"trivial: rename TestableMemcacheRing to prevent warning","message":"trivial: rename TestableMemcacheRing to prevent warning\n\nAmongst the warnings emitted by pytest after running unit tests there\nis one containing:\n\nPytestCollectionWarning: cannot collect test class\n\u0027TestableMemcacheRing\u0027...\n\nThis warning is due to the class name starting \u0027Test\u0027 when it\u0027s not\nactually a test suite.\n\nChange-Id: I82792579f9bd15a898eaab8a6e2a43b24acbe027\nRelated-Change: I50ff92441c2f2c49b3034644aba59930e8a99589\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/945e9203b6d2b53718e25e11d3bd5653c8025244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/945e9203b6d2b53718e25e11d3bd5653c8025244"}]},"branch":"refs/heads/master"},"5f20c7609f85470c1b2fa865c39939567e7dea15":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-07-10 08:04:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/966255/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/966255/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/966255/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/966255/5"}}},"commit":{"parents":[{"commit":"8be4f02ef61a09b044965134b9b7f95a5eb88bc1","subject":"crypto: Fix listings when pipeline was misconfigured","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8be4f02ef61a09b044965134b9b7f95a5eb88bc1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-06 10:59:29.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 08:04:13.000000000","tz":60},"subject":"tests: rename TestableMemcacheRing to prevent warning","message":"tests: rename TestableMemcacheRing to prevent warning\n\nAmongst the warnings emitted by pytest after running unit tests there\nis one containing:\n\nPytestCollectionWarning: cannot collect test class\n\u0027TestableMemcacheRing\u0027...\n\nThis warning is due to the class name starting \u0027Test\u0027 when it\u0027s not\nactually a test suite.\n\nChange-Id: I82792579f9bd15a898eaab8a6e2a43b24acbe027\nRelated-Change: I50ff92441c2f2c49b3034644aba59930e8a99589\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f20c7609f85470c1b2fa865c39939567e7dea15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f20c7609f85470c1b2fa865c39939567e7dea15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
