)]}'
{"id":"openstack%2Fswift~966262","triplet_id":"openstack%2Fswift~stable%2F2025.2~Ief0a29c4300edf2e0d52c041960d756ecc8a2677","project":"openstack/swift","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-11-11 20:10:40.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2025-11-11 10:40:59.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-11-11 17:18:51.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ief0a29c4300edf2e0d52c041960d756ecc8a2677","subject":"s3token: Enable secret caching by default","status":"MERGED","created":"2025-11-06 13:39:07.000000000","updated":"2025-11-11 20:12:42.000000000","submitted":"2025-11-11 20:10:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966262","cherry_pick_of_change":951352,"cherry_pick_of_patch_set":4,"meta_rev_id":"460d6d8c1e1f687a87f1dbd6163dba4bc427aeb3","_number":966262,"virtual_id_number":966262,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-11 20:10:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2025-11-11 10:40:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-11 17:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-11 17:18:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 13:39:07.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-11-06 14:52:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-06 14:54:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-07 07:32:46.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-11-07 07:32:46.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"779a00eeb99d3c8d107fe437f9521dfd77e8afe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-06 13:39:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"aec0672dd9ac7f1fae6a23820bc348182542dd9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 14:52:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/937b7c9e7b0a47289aefa85578eb5d46\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a70f3f66c59436da54fb76c6a32c2f6 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"80a9ae11826fba6f675c4d7393dbb8e75e144726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 14:54:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc77a7b63e1403aab75d5d5a2871ebc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d29272e6c09b48e6a8a8be8ab6e615cc : SUCCESS in 11m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/c7ef7f5950f44d549c1dd25768340fb5 : SUCCESS in 1h 00m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/565060b96e244d2b98f052c66817ed7b : SUCCESS in 1h 09m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/684d317fabc64c98a0af7a1a13d2af7d : SUCCESS in 55m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fee98cfd16042d3bf3e9d1858dfb6e6 : SUCCESS in 58m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/444d1739b9d14d5dbd8895e6c29348a5 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d45839b264a4c1eba71226e41cfda92 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f9dd4ca7fe44ecd93a26779d15165e1 : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c5f432501fe4323820827288bd8fa98 : SUCCESS in 8m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/23d59437f9d045a788bd0dbf4e59890a : SUCCESS in 29m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6e254beae56241e7817beb5de8ba921e : SUCCESS in 24m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e854315f778b419684be1ec95b377b48 : SUCCESS in 34m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d6d84444f68043bf86fb9d4b25b6bb06 : SUCCESS in 26m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/002d0ed3e68c496db0d23228c5913c12 : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e7954e7943224faca750a8af97a33ed9 : SUCCESS in 14m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e2f512ad387a4c7fb1d6f3c24d6e55f1 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9b781c0ca5a24266a913207a6b4fff3f : SUCCESS in 6m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/736037d196e44275b8a689435ae503d5 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f28a92a45af84bfaaf5299517231f1db : SUCCESS in 6m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7704a7f7eb947b48fff08deac296b69 : SUCCESS in 56m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4fc99c9b60d48559ce9252325544074 : SUCCESS in 38m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c47d1eab4c834aa99906e0ed9d82a4f4 : SUCCESS in 51m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37c05e54d83f4849bee0775e3f54a856 : SUCCESS in 30m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45bf6ecc411a48c7b2493d8bb9885ee4 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5429c3ba003348a2a9b26f082cbdd45e : SUCCESS in 18m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/760bdc8561d04b05927953216da159c4 : SUCCESS in 37m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd35338d1193406c83bfbc341f4caafd : SUCCESS in 57m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8ce7929a6e70b08ffc2da337dc7ddec0c72d3678","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-07 07:32:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22d59d34de00ca11e71fca0edcf779c7749e0cf7","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-11-11 10:40:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"001e9792a1a7a48a98e90c165ded96bb5d392c88","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-11 17:18:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13193bd5775c8c671c6acf2fb7b7987b50d53090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 17:19:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38430164a3bd0bc570bbd8d7c43893665846ebdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 20:10:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3283380be3414c5e8da56d8c16d94f29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6642ce7e120466399ca94f088e1641f : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/ed2e7446097341448ce782d71a70524e : SUCCESS in 56m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd265a72335044d88e8f7b317c724e93 : SUCCESS in 39m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ab33202482c147a5be4640801406e83f : SUCCESS in 37m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/febeb47ba61a41e496b8ebb3be65f780 : SUCCESS in 1h 00m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0bd51649ecd34bc5be1c8bdfb326ee78 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/27cc6b35d0fa480b98793696e3dc1e82 : SUCCESS in 24m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/68ca9e3290c04728b3d56ede688d4992 : SUCCESS in 21m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a0c0b798e5524a1aa25d6f99b9ee6750 : SUCCESS in 20m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/46a68083cfd94f6abf40385bce69be3d : SUCCESS in 28m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2669a4298fde46fe9d6f63e7ad90cca3 : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/773c6b205c40444a8c149c80c88bdb1e : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3adb91c1f8e94eef8b79d75455dbe2cc : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78d2dfc0490345a8a8fed78b7c6785f3 : SUCCESS in 4m 55s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4b80c26342b54992a8ed2714f5456c48 : SUCCESS in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1ae6d47d2394c0294e46623b2706932 : SUCCESS in 53m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c6b9bd29a154378851c81eefa4a4ba4 : SUCCESS in 25m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/49cf9912cab142f7a1a24f0bd5a4abb4 : SUCCESS in 52m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1faaafdeec73440ea9cfd5e7472bd00e : SUCCESS in 29m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd24c010a04a4230bad3b2cfb9901a1d : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e78cee31d04426098d2b24d4eac1ff7 : SUCCESS in 19m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c141997084e74409a37b503d80b70d64 : SUCCESS in 51m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/232f53b24c0f4539ae37a69587b13161 : SUCCESS in 1h 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2a9a00982647b8cb833b8c442b99f236adf7d6e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 20:10:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"460d6d8c1e1f687a87f1dbd6163dba4bc427aeb3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 20:12:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4f7c14dfb104726a249684488f5919e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8459f6ac8a03418d806a1faa9ca9c37c : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/748c33c69a8c4b7d93235f7ac80e005e : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1922c2da2a642ec77ba0c74461a53f2e280cec7a","revisions":{"1922c2da2a642ec77ba0c74461a53f2e280cec7a":{"kind":"REWORK","_number":1,"created":"2025-11-06 13:39:07.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/62/966262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/966262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/966262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/966262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/966262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/966262/1"}}},"commit":{"parents":[{"commit":"00d84e8c00766dc6370c81869cb5c59d8ad24385","subject":"s3token: Pass service auth token to Keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00d84e8c00766dc6370c81869cb5c59d8ad24385"}]}],"author":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2025-11-05 23:30:21.000000000","tz":660},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2025-11-06 13:39:07.000000000","tz":0},"subject":"s3token: Enable secret caching by default","message":"s3token: Enable secret caching by default\n\nNow that we need to pass the service creds to keystone, we might as well\ndefault secret caching by default now that they need to be provided.\n\nThis patch also adds the required s3token configuration to CI so we can use the\nswift service creds to fetch s3api secrets.\nAs well as also configuring keystone users for cross-compatibility tests.\n\nChange-Id: Ief0a29c4300edf2e0d52c041960d756ecc8a2677\nSigned-off-by: Tim Burke \u003ctburke@nvidia.com\u003e\n(cherry picked from commit 5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1922c2da2a642ec77ba0c74461a53f2e280cec7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1922c2da2a642ec77ba0c74461a53f2e280cec7a"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
