)]}'
{"id":"openstack%2Fswift~966263","triplet_id":"openstack%2Fswift~stable%2F2025.1~Ief0a29c4300edf2e0d52c041960d756ecc8a2677","project":"openstack/swift","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-11-25 17:18:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief0a29c4300edf2e0d52c041960d756ecc8a2677","subject":"s3token: Enable secret caching by default","status":"MERGED","created":"2025-11-06 13:39:18.000000000","updated":"2025-11-25 17:20:12.000000000","submitted":"2025-11-25 17:18:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"966263","cherry_pick_of_change":966262,"cherry_pick_of_patch_set":1,"meta_rev_id":"b255fbbf2a11f74d7309c408787d8be170162971","_number":966263,"virtual_id_number":966263,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 17:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2025-11-11 10:41:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-11 17:25:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-11 17:25:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 13:39:18.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-11-06 14:50:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-11 10:41:25.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2025-11-11 17:25:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"46092a2cbd70e3e5ef9e77646f8dd79039f6c9bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-06 13:39:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"ea2f3bcd7eefb7aad8989d58f18540cf3c996dd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 14:50:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ae3873880fe41ae9e636c314cf34d5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb2fe90948d40d7ae128594e76bf109 : SUCCESS in 6m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/a871ee59fea34102b8e079f5e88f2bdd : SUCCESS in 54m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1ba2ee9df0244a24b2214f6120bb9e87 : FAILURE in 17m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84bc232d98ff457b8bc3c79487eae461 : SUCCESS in 53m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35299307cef343529247c706d6dbe6f0 : SUCCESS in 54m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4d75cb0ae4b4803a454bfdf3338e4b8 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a83d687932844b7aaa8e4fb6a1761d48 : SUCCESS in 23m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5680be7d6cd140a0a75818b18fe6b8f0 : SUCCESS in 21m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94a6ea6ef065451184787c3da049ea8d : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24293cc965c54f8098e499106d996d45 : SUCCESS in 21m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/67bd3ce3f69d4e118f35486fb0954a71 : SUCCESS in 11m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f194345bad734d9d83ac8c507ac3a56d : SUCCESS in 23m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/958c75ad3a144e98a47dc7a8e89fa89a : SUCCESS in 10m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1368386ebec14b8aa94738e34de96fbc : SUCCESS in 10m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ae778e4ce3284e8baa9aed3aabe83563 : SUCCESS in 13m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/61fb5d0311ce4fbf906dcfd1e1a46c5e : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7ac58740c7644f75b7f946d44c27196c : SUCCESS in 14m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aabea7af2fb14d338c14b1e9d25d4d82 : SUCCESS in 8m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5027a0ee512d404d9f47d4e3f125f5c0 : SUCCESS in 1h 00m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df7350471ae343d78dcbcc575b154590 : SUCCESS in 41m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c41a0f42f3cf4a6585a805e6d1b58567 : SUCCESS in 30m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18ffbecac6e24af5b8410e8a453cda4e : SUCCESS in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1526caf7fae4441a9c2e6c3759159510 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff41bf092b8c43408575416c550e7be7 : SUCCESS in 25m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d1f55ca122644e5bacb10ce86928dd3 : SUCCESS in 35m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/943de78027f14e0592fa946fda7dbf49 : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a4156c6ca4c1db4953279c287a4cd35978497f5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 15:04:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f287a6183574d9e87ce8fd63241c248\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fd2b5f7070b749c8a8ed21913583adc0 : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":1},{"id":"aec85475ccbd9f1d1cad85f9d5b15d9ee877c47b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-11-10 16:24:29.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ceccd9d05b97decabc13fe3c4e0184c336b72626","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 17:15:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd18ea7ccdb84811a950451dc412465c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1bd00d2abe5147fabe89fce98b8bc55d : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":2},{"id":"40fbf952fba89cd43942ad157e9506f920bf59a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 17:32:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfc6758d3cc640eca0f4be58b8336054\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f971252a0a645498997f7cecb2c42e1 : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/6d74500548d84147a1a95f03fe930b1b : SUCCESS in 58m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e84284603ef846768db7f1073176952b : SUCCESS in 1h 01m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/037ff3fce63548fab3c15de2bb32bab9 : SUCCESS in 58m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfbd1bd52ad949f69116caf1baaf8e22 : SUCCESS in 56m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbbfc46c3fe449d19970ad8521b2c283 : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f67e0e7ffaa64e05ba5fd28a24b9d0b5 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de55d954fe134fd6b8def28e9a26b15f : SUCCESS in 21m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/823ced3094d44ee2ac76147e3af00fe3 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3998c3db3a2841dd90a5b56d6ef3570b : SUCCESS in 20m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/28c253493e834f55a73753a53f217196 : SUCCESS in 17m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ff7b6c6d6f8342ccb8e25cbfe17fe49c : SUCCESS in 24m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9be750565e9348e9bc8e47518c2dd30f : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/efbe2ad7f38c46ac874c89f46d5793d9 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/34b3ba868b8e4184841c24a03c21f650 : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/495841137feb4639afc046a0a1f6f01c : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d27b5be670f249dfb3debd5cd793d953 : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38833e1a12a64c9b959eaac1ae2a1725 : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/75fb8f4a822941949a3adc25620a1524 : SUCCESS in 51m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c301e216778f48748ac92a0b4becbda0 : SUCCESS in 36m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d8b364368a74082bbc3252bbd19d0b7 : SUCCESS in 45m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed05a491e700420e8e49de9aa63790e1 : SUCCESS in 21m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6bbd6ba887a4d3c8c5cc33726eddcef : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c154c764678440048c26b707914a2144 : SUCCESS in 17m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/183846baed634e3e967b7ef38a363556 : SUCCESS in 49m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d4cccf8d026430aa18f6494ec9188f3 : SUCCESS in 48m 22s","accounts_in_message":[],"_revision_number":2},{"id":"9aa2838a7ab3d03325e09e3f7d4ce6798c325231","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2025-11-11 10:41:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0de77420206d7128a8bf76f7c76670d81b3e0f84","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-11 17:25:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7a7fc74f052557f143a5821a469116e373454d90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 15:13:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a2f2ee8d07ad44c13967c88882c39f9c9a767d85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 17:18:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e20fb3793c94a7d901c454ce43cc766\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/791fff9ac7c34057bbf1cc3d912db570 : SUCCESS in 6m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/d57372f00c4b465a8d65558f4802d957 : SUCCESS in 34m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aa89f70dff6d4a1187900d2ff1954b53 : SUCCESS in 58m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4304593f5d514c989569f0fb6a0fc0b5 : SUCCESS in 1h 50m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e1f02573f0b448991296b812cadc2fa : SUCCESS in 33m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1ed837eb5edc425fb767411e82092504 : SUCCESS in 8m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bfbfbf1318554cc3b282d583be19949c : SUCCESS in 19m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8e6bbe5b49c54af6b9ccad7fddeb0fd4 : SUCCESS in 8m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6dc86c44fd864864910ac2cb5afd4b51 : SUCCESS in 21m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6e67e220d14f4957b814e4d2ff5ae9ba : SUCCESS in 12m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/33a720dcf3b4411aa386bf054861943b : SUCCESS in 15m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8b3e3a381b3047c082390ebb6259b308 : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9d2e6354f0e4237a76a56844f1bbc1f : SUCCESS in 4m 38s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/461ca1f772b24043af1828bfc3b7aa76 : SUCCESS in 8m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca5fabfcbc0749e48757b165a021a9a6 : SUCCESS in 1h 03m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3921f02e58b04221b9791a92e9d9220e : SUCCESS in 50m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/214eef193e7442f08e13d327e3d17b39 : SUCCESS in 52m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0167a0f902224ea6a37190be8d2648e2 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e08e3c594246aaaba83b4bd9ac6c3e : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35fca20d96834597806d6dcbedd251da : SUCCESS in 9m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89c9c076eb8b4823ba1b7522a6851aa2 : SUCCESS in 40m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/225f5f1efcfb4b4c8c6b5cb8df32166e : SUCCESS in 29m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b423d9b80089a6889a345188538a652208fa76ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 17:18:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b255fbbf2a11f74d7309c408787d8be170162971","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 17:20:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e619aef242c44e9f8d6704fec3d98cfe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd356bbd0ac446e98136f8951d6510d5 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c2272a8726a04707ad03e906089c8488 : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06b34fdfa0f09f2bdb655ee5f6d1d1aa22bdec9c","revisions":{"ceaa8303bc241d5426e82a8eec88eee83619fb69":{"kind":"REWORK","_number":1,"created":"2025-11-06 13:39:18.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/63/966263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/966263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/966263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/966263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/966263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/966263/1"}}},"commit":{"parents":[{"commit":"76bee409dcb7faa19ad4cc83dbeb260e9e027461","subject":"s3token: Pass service auth token to Keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76bee409dcb7faa19ad4cc83dbeb260e9e027461"}]}],"author":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2025-11-05 23:30:21.000000000","tz":660},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2025-11-06 13:39:18.000000000","tz":0},"subject":"s3token: Enable secret caching by default","message":"s3token: Enable secret caching by default\n\nNow that we need to pass the service creds to keystone, we might as well\ndefault secret caching by default now that they need to be provided.\n\nThis patch also adds the required s3token configuration to CI so we can use the\nswift service creds to fetch s3api secrets.\nAs well as also configuring keystone users for cross-compatibility tests.\n\nChange-Id: Ief0a29c4300edf2e0d52c041960d756ecc8a2677\nSigned-off-by: Tim Burke \u003ctburke@nvidia.com\u003e\n(cherry picked from commit 5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceaa8303bc241d5426e82a8eec88eee83619fb69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceaa8303bc241d5426e82a8eec88eee83619fb69"}]},"branch":"refs/heads/stable/2025.1"},"06b34fdfa0f09f2bdb655ee5f6d1d1aa22bdec9c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-11-10 16:24:29.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/63/966263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/966263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/966263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/966263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/966263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/966263/2"}}},"commit":{"parents":[{"commit":"76bee409dcb7faa19ad4cc83dbeb260e9e027461","subject":"s3token: Pass service auth token to Keystone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76bee409dcb7faa19ad4cc83dbeb260e9e027461"}]}],"author":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2025-11-05 23:30:21.000000000","tz":660},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2025-11-10 16:24:29.000000000","tz":0},"subject":"s3token: Enable secret caching by default","message":"s3token: Enable secret caching by default\n\nNow that we need to pass the service creds to keystone, we might as well\ndefault secret caching by default now that they need to be provided.\n\nThis patch also adds the required s3token configuration to CI so we can use the\nswift service creds to fetch s3api secrets.\nAs well as also configuring keystone users for cross-compatibility tests.\n\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/963714\nChange-Id: Ief0a29c4300edf2e0d52c041960d756ecc8a2677\nSigned-off-by: Tim Burke \u003ctburke@nvidia.com\u003e\n(cherry picked from commit 5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06b34fdfa0f09f2bdb655ee5f6d1d1aa22bdec9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06b34fdfa0f09f2bdb655ee5f6d1d1aa22bdec9c"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
