)]}'
{"id":"openstack%2Fswift~966512","triplet_id":"openstack%2Fswift~master~Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910","project":"openstack/swift","branch":"master","topic":"cohanaja-cycle-time-recon","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-11-10 19:01:08.000000000","reason":"\u003cGERRIT_ACCOUNT_38368\u003e replied on the change","reason_account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}}},"removed_from_attention_set":{"38368":{"account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"last_update":"2025-12-05 00:17:40.000000000","reason":"\u003cGERRIT_ACCOUNT_38368\u003e replied on the change","reason_account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}}},"hashtags":[],"change_id":"Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910","subject":"Implements concurrent shard processing via cpool","status":"NEW","created":"2025-11-10 08:01:38.000000000","updated":"2025-12-05 01:25:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":10,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"4ec69c1b27307bb0a6c98ce88da1114fe4b579de","_number":966512,"virtual_id_number":966512,"owner":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-05 01:25:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"all":[{"value":1,"date":"2025-12-05 00:17:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-10 08:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-10 09:19:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-10 16:47:08.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"61e9e2a3e7194f84504cf17b9397aaabc9abe9eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-10 08:01:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88af8fbaf849d31cac7fd405b63789ab611ee6d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 08:52:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72fc14bc0ff3441d9cd881ec75f99aa6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/72158c3e08254b1594927268f9038376 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8959df7a0569f9499845dc59c1a586dfa1f32bd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 09:19:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/530fe718ef0b4e0eadbcd6517da67834\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31192b196ec644279f37434cac5f7aab : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/731df31f0ef34d2c8fc64564c16a1af5 : SUCCESS in 1h 01m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84ec1b04c76d431a8e36983a015731f6 : SUCCESS in 41m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb95c74a600245988d9d3a8385831d35 : SUCCESS in 52m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bce464b0e9864a46b5daa0403a88cd45 : SUCCESS in 1h 05m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d18048233fd427c8a976bbb8786f086 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/616d5f4be24a4df7926f35c9af3f8950 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8922ad342e1f43718838c911ed77bd93 : SUCCESS in 16m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/505494c63f8f45ca992a6f185b1390d6 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/51a35773e4394114bf46b82f957d3f7d : SUCCESS in 24m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/829c03cbbe8d4cc19d042f77e0ed99cb : SUCCESS in 18m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/76e51aafed5d4384b70ad8ed34dffee3 : SUCCESS in 28m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0cd12b360cc74734bab15cb946185931 : SUCCESS in 19m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6fa8fe47d2c443cbbdb65b971c59ed39 : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5740e852265b4013bd81d093e117899b : SUCCESS in 15m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e95cc2ce3f1047e3983db7c24d90bc27 : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/df4bf884294e4a8c914b2c0fc400f332 : SUCCESS in 5m 57s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a66f84c91ec148b88f8be263dbf092ec : SUCCESS in 14m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f1f9464a8338475db9ec33652fd20430 : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c7a61ce28e643c296ec34a10c141506 : SUCCESS in 44m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e000ec69f9504e09901d59b63c0dc90d : SUCCESS in 54m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de9ac344d7ce400b910fc4cf3ad94afd : SUCCESS in 46m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cb79e24fcdb479493e0258d34d755fc : SUCCESS in 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13bb82d24dc54239982a96c9f27cbb63 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93211fed037947abaa7978cc5ae4fe00 : SUCCESS in 14m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df969853472b486dabe72c062164a211 : SUCCESS in 1h 03m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a477b2d7c93c4517b67984db85a942d4 : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ab8e61c013d67bb105be98a006697e2b0da689a1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-10 16:47:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fa9367c60b722fd9b6ebd4f81cd9a95d4ea76fbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-10 19:00:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0e074ff603648becfae27b42c8dcec759858957e","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-10 19:01:08.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"178c2f58314f6bbe680d0670493edbffe50ad854","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-10 19:03:42.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"573f637e37ae2f3fa308d793fd8d409faf91e026","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 20:01:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdfa9d896cb243828fff1f64e4b80832\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4507172f25d342068d70e5d49b108d37 : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":2},{"id":"12888914321002af73492bbcbacec1d0568568c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 20:07:50.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cda500a67d04ef39329f2aafb5d3b76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6061ae04b364da5a5bda7fca49e8e17 : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/6bc5fa2a85f44fbcbc766af131bf415e : SUCCESS in 44m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7d5cb8acdb7a4bbabe1f0fc3f4addd93 : SUCCESS in 49m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e9d92ac8d094732a940540daf544139 : SUCCESS in 42m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c99ea0334afc403d893d1b49022c57d4 : SUCCESS in 1h 05m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc7b589ed25c4e8c80c5268b7a7abb29 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b4e0d73de7f4cf9b7470e5f2086ccd1 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b0914dd8be54e14acb1e654c8bc9ae1 : SUCCESS in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e7a09c7588b94e03a4bbb8e92b49f38d : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5cd193961c4e428bb68b20aeead788eb : SUCCESS in 21m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/16da2ccb74a747ac8fad62b2b8c4a06a : SUCCESS in 21m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f321f2b2d7084adeaab4ec4acb16815e : SUCCESS in 19m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3c7f16158ce744a591cf66db255b48e1 : SUCCESS in 30m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2fcb509c0c2a4bc09948970d86c6b109 : SUCCESS in 13m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d290ef81d192487bb6951c10dd75824a : SUCCESS in 17m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0bfe4739cfc94b1aba9c3d579df8741c : SUCCESS in 14m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f0fc4cdc0d054a7a871f2174890af79b : SUCCESS in 8m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ac55103b8009408891c4d2f749cbe143 : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c68442b23b6e4fc09d00507aaa828bf9 : SUCCESS in 10m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ee5f18e310945a18eef390a6a67626c : SUCCESS in 53m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4b83b894fa94598a6aab6af1f552aff : SUCCESS in 53m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61951c43adc8421f838a1c99ae4c0100 : SUCCESS in 47m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8549afed14b4b49968e41805816c3b9 : SUCCESS in 27m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8de183c7f2e410bb003bad534135187 : FAILURE in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/31e1508bf0fd4c6a8fb9c7064fb9758e : SUCCESS in 18m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fd9878105704d89a7068bf95f03ad34 : SUCCESS in 44m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e13f06bdab0848d0aeb90959f2bf888b : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":2},{"id":"39afd3b2a5581cc191bb20ab204240d2b11570f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-10 21:01:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"737add89fe22d12f115bce63a05df3ed8b8474df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 21:32:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a5172d6bfbf4f9097e5e6253722bba4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8bef4587fc1243508a6afadd35909ee3 : SUCCESS in 28m 41s","accounts_in_message":[],"_revision_number":3},{"id":"24946919f8015cfb73b624d00e45b363ee85c02e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 22:09:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2448f85b844094963b1c4ea93c39c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df958f2994c43a69f1e19f53c4abcfe : SUCCESS in 10m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/c15766d6347d4195b7e2834768883723 : SUCCESS in 59m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1b3f646fd6da4130bb7f7d333bf8014d : SUCCESS in 41m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/895f5afe38c748ab8b9bf09cfd15c0e9 : SUCCESS in 42m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17d7a8cd3d254eb89be61710cc955e20 : SUCCESS in 1h 01m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ed9fa66d13640f2807ba052c42bb628 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2d95a6a4ad64948b6c1aa27becbf074 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06a3fd00af6a4fab8ce230d63f92551b : SUCCESS in 20m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5587ca28bdba4e21b059b6569669ce5f : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/590e323bf8d248ee9459b49f8477d384 : SUCCESS in 29m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2e711e716af48e9b7aebfb428a0633d : SUCCESS in 22m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/08c8a566de074752b278927347b343f6 : SUCCESS in 19m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/aae9598bf0ef487986bdefb49755cb68 : SUCCESS in 20m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/07e51306921341a89aea430df81d61fb : SUCCESS in 12m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cdb6e2d2c5f24561a8938e33aa06299d : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3bb01320bd5547bba6b0025a73bd144b : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e49c30b14724a9ebfdced7ec445ddb5 : SUCCESS in 6m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1b378dac8d4049c98b29e712782665b0 : SUCCESS in 14m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f30544d5e3fb4b4b83b78a8e2aceb247 : SUCCESS in 7m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b505d91ad555413cba98ce850e9d1bfd : SUCCESS in 59m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc006e3a14ba41b2933b3f7aa1651a61 : SUCCESS in 37m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aed4b14e92f04ac8810e37146bbb55a0 : SUCCESS in 52m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f8484c8677e42b7afc55037c2bc7161 : SUCCESS in 27m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a99794eee824103baf35dc3a5750865 : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/22e12bb5ae8d4e9d8985802a4b95695b : SUCCESS in 15m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a79e8f3251a49879dccc54183ed0ffe : SUCCESS in 54m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87aa0e3d84f2499eb0107e65ecaca184 : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":3},{"id":"d0e95fd01b3776b9bd0413a651f30079fdba3f01","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-12 16:55:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"091b0d553d2aa1193fda6c2abad5d74e4f1fd9d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-20 18:23:14.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"344bb291c09b55c8f3e0d96eded2c5d82625cbf1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 18:33:38.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b08e6a8a3a5e4c52b1fa75c663c7a4f1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/51c5d02be49a44138b34b4327b4487ef : NODE_FAILURE Node(set) request c232a2f8aea94fb59ba3503ca7214483 failed in 0s","accounts_in_message":[],"_revision_number":4},{"id":"8eddc63e3acb73eea7030b99337d14e47668151d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 19:48:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/469858ed8ca44343b4bf90158d637f62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d9fc4a3b82c4177a5b8e93e582afafb : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/c82a7c75929749e78eb093b94316b3fa : SUCCESS in 1h 01m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1fb3cac8bf7e4cfdbbe131525250a5e4 : SUCCESS in 1h 11m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d901e88d9e64a4b9c39d408b5b8e943 : SUCCESS in 50m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b58b362874f4568b6a452ceb63799e2 : SUCCESS in 1h 02m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f08f30c852344be18156e0e6835dbdb4 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9da4a5092f64929a9ea2ea6cb8766a3 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc435dfc82c947beb277d2d3f0585d88 : SUCCESS in 19m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7570de872c3a4daea91db784bcd8f136 : FAILURE in 4m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5eacd23b89bc4c51ae25ed6afc0ce2f9 : SUCCESS in 26m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7cece31c29bb48c0bc0cba2d2b133b3a : SUCCESS in 21m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1dac1c8e9fa44cd4925abce990e41517 : SUCCESS in 27m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/925d0837143a4a10b6b9368e7340a562 : SUCCESS in 33m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3fa849175a3e448899ff5dc23c1a0343 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3fb129f7852844e586947376042ec46b : SUCCESS in 8m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/04d02042d6b7439183677fa99af631d7 : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/650c6e46559f4427a1fe8dd53f82a972 : SUCCESS in 7m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b579694c15944b0e97d285fc03f92566 : SUCCESS in 15m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b0ea612f612249a583eb1dfb25489a88 : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/62f8f039e55541979edeed1fbd4b77e0 : SUCCESS in 48m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a439069e4d94bbea1d7efacd54106df : SUCCESS in 32m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/28d5d337d71a41d0bfb8f293538e3cdc : SUCCESS in 26m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/568968889b7c470487575335d52bc229 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ef709db62d48bdb7a578aaa7223b81 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf0c45367ab04138afdce8aa21843afb : SUCCESS in 16m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98bdc78e5c6146e590f441b52087915d : SUCCESS in 46m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f72b66e4b7044bb18658f631f1c57014 : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":4},{"id":"65b630f80c603d6e3dad7dd8ecc204e44dcc20b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-20 19:54:42.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cd6d5b5228063c4eb8cf7b34d1c381abfe4e78af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 20:38:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c059bb1dc5c497c93bc74ed9896fd9d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7800642a420246a78ec3dac9485a747b : SUCCESS in 41m 27s","accounts_in_message":[],"_revision_number":5},{"id":"e6ad1a1d4f9a6e57188e34bf4e01667dedce297d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-20 20:51:34.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"8d9f087b91ba17a0654ee6b3cc9ebab4a3f3b2cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 21:34:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85160f911d2745e48de73c6c09fd065e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9cbe1c73c6be42b188a60458cbbc0d0d : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":6},{"id":"d332d00095c1cf5bda324ec9d5fe5e75f8dd33d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 21:55:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db402580d65c4514bcf5982301d59222\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4df02ab1c244c7b3c976b938ea1773 : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/d2a07864b29e4d5ab296ef9b9db67f2d : SUCCESS in 31m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83228fe89021438d8f48eaa259259875 : SUCCESS in 58m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2cc2f90798254da5bf4e432aec62dcbd : SUCCESS in 33m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d59a1be2d6d541bc9647f5a0f08984c7 : SUCCESS in 57m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83aa2ff3723e488ca2cb52843c88898d : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a723030ea36c4a2c86098c16c072fbb9 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/676341c3bb4b40aa907b2da38eb437b6 : SUCCESS in 12m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/59b0118a2c454487a01fde36b1c47774 : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/acc78754b23b4818b1e63d1cce4f3171 : SUCCESS in 12m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/42880ce698f940759bf7b291e6676509 : SUCCESS in 23m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6b6d0dcae00a43be8042a743a6d14fa7 : SUCCESS in 14m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6f3422431b324ea987bb962ae4103339 : SUCCESS in 28m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d2d7075a00e44062b5e67bc67e13b199 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a215d4cfb78a46aa8a665579a015bac7 : SUCCESS in 14m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cdc9c06669164f4cb48a4ba6d72d0069 : SUCCESS in 12m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cd52d57a754940d887c683dc02e5e2d5 : SUCCESS in 6m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f6c39e2ff8440d6aaf5283ba2045717 : SUCCESS in 14m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/224a6da1e2034c38b90dd4aa4c242a13 : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/94bc61c66ff14ed0be886f82fa66547b : SUCCESS in 52m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ad0598553974015b0d5851038452514 : SUCCESS in 49m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c9a573fd994845959715c057adb3633b : SUCCESS in 55m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b84a57eeef84697963f1287438463e9 : SUCCESS in 29m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb9c500f7c645efa22a552a1b597ec6 : SUCCESS in 2m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d01901b145aa42aab6315102d1bf0e01 : SUCCESS in 20m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0a26b495375495fb9586e45252772ae : SUCCESS in 57m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a1746f418b5454daaa33584cb4b140c : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":6},{"id":"53110a30e9fca6416ca0d0a634ec076e24cefd5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-20 23:27:54.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"ed5bc7359ed6dffe6bcd8199fec0ae75b9898902","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 23:38:54.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b418eeafbdc48acb26a33e992359c01\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2db06d02aade45a3b8ebab4c9bbeb5b2 : NODE_FAILURE Node(set) request 766d52662fb54c5fab96239ad68faf80 failed in 0s","accounts_in_message":[],"_revision_number":7},{"id":"8daecb05993ee7db1b64a252cd1f6bed139813f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 00:46:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94fe7577c57845eab5c188dc0d3641a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c972a8e5dca4f2a8d60be2bac511105 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/d77de71538fa48c1ba66f2de6981f42f : SUCCESS in 56m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5121698d6cd74f7a9d7a369f860e04b8 : SUCCESS in 58m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24a65ed4ce4941f982a6630275bee1d6 : SUCCESS in 53m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1940891f09845788781510ee7737c62 : SUCCESS in 32m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c2f62b30fab4dffadc1b4bc219ee937 : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/277a4bea6fb14acfadc86de4f58ad2e0 : SUCCESS in 12m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f09ed7b7b37849079407387273d1118c : SUCCESS in 19m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53d31b6582b14ec9858c1e33bb37aa98 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1cdecd2bffc14af39983ee9f1fd3e222 : SUCCESS in 24m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/780b2dde2574406b959201d108c9cfb6 : SUCCESS in 19m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/741f0ce283f540ee80b9b2d305fd2968 : SUCCESS in 13m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/475626b3ede64867a6ba1c0b40d5079d : SUCCESS in 11m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/84b76004005f41acba790beb0ebeba7d : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a885557c1b514642a714787f2642d22f : SUCCESS in 10m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/58b4ad0e440f479da14756e1b50429a5 : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ad8304562d94709a3963e7e9d25b6bf : SUCCESS in 5m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1a48e9eca3b54d39bea73eae1ed67a2a : SUCCESS in 13m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/66482f79d5fd4ad6a3e332b10d48757d : SUCCESS in 5m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/68780b777164489eac53df8573bb6c6a : SUCCESS in 40m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f49674cecf5642ddb0b3b0fee47c64b0 : SUCCESS in 53m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/38b928acb63649f29de8fd448bf6414e : SUCCESS in 49m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45c52f096dea439eb37a474ba2657c73 : SUCCESS in 22m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9037007bd8141f7a6630a1cbe889dfd : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e3d0059d13eb422dbf6a06771d3a1b92 : SUCCESS in 16m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42172b7dea84471e8509571c8ca123a8 : SUCCESS in 32m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5533592c9f93424dafc01df5d8128643 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":7},{"id":"5503cba5b6cd376a02c7349d67d00be5332ebfc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-22 04:24:49.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"7445662fdec81651dd6d889609d4bd5348ec794c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-22 04:40:14.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"89b28f1a24948198752c7b865bd6d8b395a96b51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-22 05:07:57.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"b505245b2e81a94fcbe7b941188af628b3a0d0f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 05:49:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d57fd8950b9d40f2aa04da79957e21e9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b26fa7564014c8c8d634c06eada4440 : SUCCESS in 39m 07s","accounts_in_message":[],"_revision_number":10},{"id":"0397a5e50edc5255c6a508cab0b985d8d8ff696c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 06:15:45.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/898192b66b8d44a380f0698170662b58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0e463749d5845f6a2fb3966a22eb0d8 : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/42a56fb07c804b8fa2f15f5b811d7c8f : SUCCESS in 59m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/58ca987a8c544f1b801132b35e92b4da : SUCCESS in 30m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c65a75bf69324f46bec3d594318c8298 : SUCCESS in 50m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3fc6f9556c44930b6a46b8b444e1cab : SUCCESS in 1h 01m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40c90fd3af314042bd5590b5cc32ccaf : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2939918f16e742b4abf9e3364cae66c7 : SUCCESS in 11m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/569b9f79867b4f46be9c0875168180ba : SUCCESS in 17m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a7f1745a7ece45aca1c43bd27c07e0a3 : FAILURE in 4m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a57802a88141497fabe47f94905762da : SUCCESS in 24m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0875846345a64a4c9ecc603397292b7a : SUCCESS in 21m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2cabc9bc0fcb442085302f8a4cee748a : SUCCESS in 12m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b1c38cf3a97c4ad98e5dc5947e9b9956 : SUCCESS in 17m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/37d96fd33ff5427bb34280def6f47b02 : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e826ffcdbf9c4c45aecaa2f11c2d7c69 : SUCCESS in 11m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e11a15eca7cf44138e9122bc4b1e9e62 : SUCCESS in 16m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fbc9fe8d1db7427782979d8c7d4e4338 : SUCCESS in 4m 16s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/de2b337b6bec4d3195a6bbbd62c5afda : SUCCESS in 15m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/964e161f556343668b62b64db8dc564e : SUCCESS in 4m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/58c3cec3450a418b8272eeb59989600e : SUCCESS in 47m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4348418e24354596a5bf85ab621859b4 : SUCCESS in 35m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb74ae7f29ef49dc9dd1dc3e4b9cbd54 : SUCCESS in 49m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f106a80e1254e3aab448650615d9eed : SUCCESS in 29m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c15aebac1244bcb98e05355ef2590ec : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/69c56c69ca454afe8632b7c8b5ad8e26 : SUCCESS in 17m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/549ea52687014d53a0b44e4ecea6adda : SUCCESS in 50m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6979e4a2b7f44ac08472bcd2640af503 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":10},{"id":"51d21cca3d9a61510a8cb6c99742609dc5eaca70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-26 19:47:28.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"775de32a9772cb9f1773a8e41582505f2e1164cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 20:30:10.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/882de655b6fb4552afd463fca6f39bca\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62d4c859d7cb42a884deb7a4466e4263 : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":11},{"id":"c151a5a44ef520b839b318feb4ed85d47c9c5ad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 20:55:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d707d706d1e446f9d325ffa9fd52d54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4b1d0c43ba4b2aa858c26a6f5ffe2f : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/29c62225739d4d5a8ca3a0b513d48b36 : SUCCESS in 33m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/da82e27ddc004b3899093cc2bfddc034 : SUCCESS in 57m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bbdb13a50be3443db6331eb54c4b4dc9 : SUCCESS in 53m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7244cf0f286f44f486940570717a833d : SUCCESS in 50m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5af3ceab64f646ec98c053260e5f1f62 : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d7c18e4905e4ab3b58b19289c5abf99 : SUCCESS in 10m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44eaec0ddf27490d8a25dab694a259db : SUCCESS in 17m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c57baf94797457b9b5993ba8086232b : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f7fcf6f034a04824a538148498caad57 : SUCCESS in 20m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2c283865464b4766951a21b28e675f7e : SUCCESS in 20m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/44bcbb652e034c818c70dcbaa7306bcf : SUCCESS in 28m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0f7d458b1292445a9a70e4f3876437db : SUCCESS in 13m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dac93fc5314c47af9e9084853e161761 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/772b3084206f4976b7f0d636d10ee3c1 : SUCCESS in 8m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1d98d2ce151042f9a4769a3db12fad82 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e61946e5e4994bfd850382cbda9bfc0f : SUCCESS in 6m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a48710b84b21403892308a37570e4851 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/099efd27461944a68ec1fbfca046caf8 : SUCCESS in 7m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/462edc2f9ee54c4b86f0fb87f2e3c6b0 : SUCCESS in 1h 00m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb231f5e4fe94a57ab79651d3246e8b7 : SUCCESS in 21m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d39b268e2664e09a397d1295634fde3 : SUCCESS in 50m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e1906eb56c542e2ac12d9204c044a8f : SUCCESS in 28m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f0fdbc765b04f9b9bf79116ce97474d : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f412709bb4a4a6fbc3ec5feabdd4b9e : SUCCESS in 18m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f7a08a40cd24af18088f1877be11fd5 : SUCCESS in 49m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acf39239a68d4cf5846ed502e2f51e3e : SUCCESS in 34m 28s","accounts_in_message":[],"_revision_number":11},{"id":"d3ecad99eae322e97e64b585f5aa7bdabec08c1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-26 21:20:47.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"cdfeae28188396302b6cc368206e77ec72d3de5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-26 21:40:00.000000000","message":"Uploaded patch set 13: New patch set was added with same tree, parent tree, and commit message as Patch Set 12.","accounts_in_message":[],"_revision_number":13},{"id":"bff900fb7200990ceb6534ca4e7c6b4920728c22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-26 21:46:29.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"e4b0d018a6b191c67d7dcf8ab3976ba9b9302f46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 22:43:45.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d55e0a5700d4cdda4cbae31ff47d73e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d543eb3d64dc4350ae9005662eb7ba7c : SUCCESS in 5m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/129f21c7b759425998c7d933f59fea82 : SUCCESS in 30m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d2c808044fc8424b8211e2d873dd687f : SUCCESS in 31m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/184ce502662e4430aae8721fa7285d5a : SUCCESS in 32m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6766f4c6463f4a099fb8c44a9688dc9a : SUCCESS in 33m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2fe0c3f51a2421da53249a432d281c3 : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db20d553fb304619b94c4e7010b75da7 : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a60cdd9ef0e4c41b1c7a1aa21819ddd : SUCCESS in 19m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be8b6d923b324dc586dd20c6d10c89ea : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2b17ffc9be1349f5881a8b9f281c4810 : SUCCESS in 12m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36bc561799ad44eeb84977a9c808bfa3 : SUCCESS in 21m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6e0b326d12864b059bb8ef49dc36a339 : SUCCESS in 11m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/47dfcc2c392c472ea05fba6979b10136 : SUCCESS in 30m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e3323c9b70a9437baaf1cd83d7e547f9 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/42ee289bd7fc4f8f8c655034bacc5d41 : SUCCESS in 14m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8b72728dea0e4f4b8004739b6b939f8b : SUCCESS in 8m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d75ec12681c24f718f1791d440b97e00 : SUCCESS in 5m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6082c800241549a7ba8623292c4d644b : SUCCESS in 8m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e07c0e5617e14fd6a8f5210826863107 : SUCCESS in 7m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c728ef671b7c419c88aeb0553d8a83e5 : SUCCESS in 52m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de69bff3f0fa42ffb756227f61d482e5 : SUCCESS in 22m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c36f545322d54860abf8d773f36e5d9e : SUCCESS in 49m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22ed03a4dd1d49d394c90ca0785b8fac : SUCCESS in 22m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363037af072c4403ac6b1a04d9a831fd : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/413845d461c447f79bab4b80845f4814 : SUCCESS in 18m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0f5e311a58e49ff9332743b2fbddc4c : SUCCESS in 36m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4596724906a5408fbfaf8b814892790e : SUCCESS in 34m 13s","accounts_in_message":[],"_revision_number":14},{"id":"682514496a9d7d56510f233f968eacb06f024f31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 23:52:28.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a43247b14f4440cb9bdae8211ff06b6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b03b603561c04b2b84b9a3359f35b5b1 : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":14},{"id":"e4f5369f24fd4e123fa1fa0370bd22e5a7625d4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-12-05 00:17:08.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"7dac4d754942b93a89120ab365947002fcd21ab1","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-12-05 00:17:40.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"aa34f03868e3a59e145e6fbdd23b8ff286d589ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 00:57:56.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0e1d09a8e5a4b22bad277a82cc8cd3c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d4ae265e31cc46df887a803ac9d6e1c9 : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":15},{"id":"4ec69c1b27307bb0a6c98ce88da1114fe4b579de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 01:25:33.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02b5080b4fd343a095d4e80b80820231\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cdd8252bb9b4bfcaf7a9d97723cdea9 : SUCCESS in 5m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/31b7659412474984a663d3a8f6b015ec : SUCCESS in 1h 01m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/33fefb7e8da6491391bdee61eade61cb : SUCCESS in 58m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7399d73e6c7644fb986a9962878f067a : SUCCESS in 48m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b45feda7ad8e481c9549498fe32c8515 : SUCCESS in 1h 00m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6fefa9b017764e8ab67e96932bfbb738 : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/534bdd6ee5794ad39f8540f8f5223264 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2152a86081c4e28b7e05eb8c8b772f1 : SUCCESS in 15m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4892ca6f5ef04f05945e36e88c4ba66b : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6b429cf36efe40f58018c5f35b5cef99 : SUCCESS in 17m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d9568967c61046ca89b218436fe2b502 : SUCCESS in 13m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/99d87233b7e342dbaa809658c1c1f0e4 : SUCCESS in 16m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2a127a88878a47b8885e1e4c9b780340 : SUCCESS in 11m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f096bf86339c4c84b88f383c90ca496c : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f2181e375b8d4aa0899e29f6415651bd : SUCCESS in 8m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d35620bdcebb4ae5b6b3f36c2bd965f7 : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e86639015da4eea9d7bfe1b4e2507dc : SUCCESS in 6m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d5656d6d1a904e80ac0ec2781ce20c58 : SUCCESS in 8m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/95ad9c38471a4f31847b9112bc3ba6a2 : SUCCESS in 5m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bf9363769c9487895e5781765da8ab6 : SUCCESS in 1h 02m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e255052f082e43e9ba7f083c701edf7f : SUCCESS in 55m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc6442778528417f9b13869faee8f480 : SUCCESS in 57m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebffd3cb0add4b70ae8ce7c3e0dd8da9 : SUCCESS in 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52e04b53e2ee4e438cbb54f4481ef8a7 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59a6d0c038b840c78afce1b378480029 : SUCCESS in 15m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33e79c41e6714710b5dfd71b7d5cf781 : SUCCESS in 48m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc7f0609c77a4295b8ebcb40c0d47b3e : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"bc1aba75a46811890b8deb00871c4544c5689b5b","revisions":{"cea5dba958294124ab5d7c9573699d985048cbc8":{"kind":"REWORK","_number":1,"created":"2025-11-10 08:01:38.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/1"}}},"commit":{"parents":[{"commit":"da7032473096f6dab750e369b4eb11db593fd32a","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da7032473096f6dab750e369b4eb11db593fd32a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cea5dba958294124ab5d7c9573699d985048cbc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cea5dba958294124ab5d7c9573699d985048cbc8"}]},"branch":"refs/heads/master"},"b32b8d9316e0eced0b742dde80ce17b023a24acc":{"kind":"REWORK","_number":2,"created":"2025-11-10 19:00:47.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/2"}}},"commit":{"parents":[{"commit":"da7032473096f6dab750e369b4eb11db593fd32a","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da7032473096f6dab750e369b4eb11db593fd32a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 19:00:39.000000000","tz":-480},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b32b8d9316e0eced0b742dde80ce17b023a24acc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b32b8d9316e0eced0b742dde80ce17b023a24acc"}]},"branch":"refs/heads/master"},"d78f251c217aa0901f6a1000e3e37b07b776167c":{"kind":"REWORK","_number":3,"created":"2025-11-10 21:01:55.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/3"}}},"commit":{"parents":[{"commit":"da7032473096f6dab750e369b4eb11db593fd32a","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da7032473096f6dab750e369b4eb11db593fd32a"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 21:01:49.000000000","tz":-480},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d78f251c217aa0901f6a1000e3e37b07b776167c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d78f251c217aa0901f6a1000e3e37b07b776167c"}]},"branch":"refs/heads/master"},"d5e1c8678b36fa61f29a556872bee46b9798ac5c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-11-20 18:23:14.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/4"}}},"commit":{"parents":[{"commit":"f5100ad4ef21e22ac4a6eedc22dd48e0984129b5","subject":"Adds labeled cleaving metrics to the sharder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5100ad4ef21e22ac4a6eedc22dd48e0984129b5"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-20 18:18:22.000000000","tz":-480},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5e1c8678b36fa61f29a556872bee46b9798ac5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5e1c8678b36fa61f29a556872bee46b9798ac5c"}]},"branch":"refs/heads/master"},"06c98cfffc653f33998fde1ad996d5262a8cbd32":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-11-20 19:54:42.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/5"}}},"commit":{"parents":[{"commit":"1e8d74a04cc4cc508f9122c652db7bc22bed40e5","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e8d74a04cc4cc508f9122c652db7bc22bed40e5"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-20 19:45:35.000000000","tz":-480},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06c98cfffc653f33998fde1ad996d5262a8cbd32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06c98cfffc653f33998fde1ad996d5262a8cbd32"}]},"branch":"refs/heads/master"},"61855fe4553c5fcd298d3e29152e0820bb3f1e69":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-11-20 20:51:34.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/6"}}},"commit":{"parents":[{"commit":"30a1bf0ba252351e13dc385176ea9b7113d1bc7f","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30a1bf0ba252351e13dc385176ea9b7113d1bc7f"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-20 20:46:08.000000000","tz":-480},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61855fe4553c5fcd298d3e29152e0820bb3f1e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61855fe4553c5fcd298d3e29152e0820bb3f1e69"}]},"branch":"refs/heads/master"},"8301283749d8062fce05cebf2e8aa973e4b05abf":{"kind":"NO_CHANGE","_number":7,"created":"2025-11-20 23:27:54.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/7"}}},"commit":{"parents":[{"commit":"a21528a14bfe4a6a2d9031a71f36341b7d100a44","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a21528a14bfe4a6a2d9031a71f36341b7d100a44"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-20 23:27:34.000000000","tz":-480},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8301283749d8062fce05cebf2e8aa973e4b05abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8301283749d8062fce05cebf2e8aa973e4b05abf"}]},"branch":"refs/heads/master"},"509b92bd2a9a6c60cf19e759e3b59e4b1034fb97":{"kind":"REWORK","_number":8,"created":"2025-11-22 04:24:49.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/8"}}},"commit":{"parents":[{"commit":"bca7826f384a81f5baf61a14559342d4454063ad","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bca7826f384a81f5baf61a14559342d4454063ad"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-22 02:51:25.000000000","tz":-480},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/509b92bd2a9a6c60cf19e759e3b59e4b1034fb97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/509b92bd2a9a6c60cf19e759e3b59e4b1034fb97"}]},"branch":"refs/heads/master"},"4fe30680f9b4e457e75b65772814a1569bc2bd69":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-11-22 04:40:14.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/9"}}},"commit":{"parents":[{"commit":"68dc4bf3bf05738a934d0046a7434c97a4d64311","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68dc4bf3bf05738a934d0046a7434c97a4d64311"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-22 04:40:14.000000000","tz":0},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fe30680f9b4e457e75b65772814a1569bc2bd69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fe30680f9b4e457e75b65772814a1569bc2bd69"}]},"branch":"refs/heads/master","description":"Rebase"},"d5bc37df7213daa2951dbd3cad80f7ddd49d63b4":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-11-22 05:07:57.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/10"}}},"commit":{"parents":[{"commit":"b16a9fc5084c8b519decb27c90e945bb179e4375","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b16a9fc5084c8b519decb27c90e945bb179e4375"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-22 05:02:32.000000000","tz":-480},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5bc37df7213daa2951dbd3cad80f7ddd49d63b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5bc37df7213daa2951dbd3cad80f7ddd49d63b4"}]},"branch":"refs/heads/master"},"985ddbac31cf6d50097f6de47d3499f375804544":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-11-26 19:47:28.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/11"}}},"commit":{"parents":[{"commit":"d5ba3006d16ac284d8cbc54d05e8cff1fca0e95e","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5ba3006d16ac284d8cbc54d05e8cff1fca0e95e"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-26 19:47:28.000000000","tz":0},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/985ddbac31cf6d50097f6de47d3499f375804544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/985ddbac31cf6d50097f6de47d3499f375804544"}]},"branch":"refs/heads/master","description":"Rebase"},"3345a275c841c9b84a60cba6120e11bf78b16fb1":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-11-26 21:20:47.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/12"}}},"commit":{"parents":[{"commit":"b5dc4e36428930da2f63d7071cee80ec08630573","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5dc4e36428930da2f63d7071cee80ec08630573"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-26 21:20:47.000000000","tz":0},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3345a275c841c9b84a60cba6120e11bf78b16fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3345a275c841c9b84a60cba6120e11bf78b16fb1"}]},"branch":"refs/heads/master","description":"Rebase"},"5060f9b1ba2da455eb3e9a672dcf531dd07689ee":{"kind":"NO_CHANGE","_number":13,"created":"2025-11-26 21:40:00.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/13"}}},"commit":{"parents":[{"commit":"b5dc4e36428930da2f63d7071cee80ec08630573","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5dc4e36428930da2f63d7071cee80ec08630573"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-26 21:39:43.000000000","tz":-480},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5060f9b1ba2da455eb3e9a672dcf531dd07689ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5060f9b1ba2da455eb3e9a672dcf531dd07689ee"}]},"branch":"refs/heads/master"},"390a85e6182f774857bccc71317269b301eb57d6":{"kind":"NO_CHANGE","_number":14,"created":"2025-11-26 21:46:29.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/14"}}},"commit":{"parents":[{"commit":"c5417d9a87f3dc1b0c3c30b692a97dd2a5b915e4","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5417d9a87f3dc1b0c3c30b692a97dd2a5b915e4"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-26 21:46:29.000000000","tz":0},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/390a85e6182f774857bccc71317269b301eb57d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/390a85e6182f774857bccc71317269b301eb57d6"}]},"branch":"refs/heads/master","description":"Rebase"},"bc1aba75a46811890b8deb00871c4544c5689b5b":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-12-05 00:17:08.000000000","uploader":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"ref":"refs/changes/12/966512/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/966512/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/966512/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/966512/15"}}},"commit":{"parents":[{"commit":"5ceb8173983b1769484e5f4769cae73b5c7adbfc","subject":"Improves recon stats cleaving context retrieval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ceb8173983b1769484e5f4769cae73b5c7adbfc"}]}],"author":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-11-10 07:52:02.000000000","tz":-480},"committer":{"name":"Christian Ohanaja","email":"cohanaja@nvidia.com","date":"2025-12-05 00:17:08.000000000","tz":0},"subject":"Implements concurrent shard processing via cpool","message":"Implements concurrent shard processing via cpool\n\nThis patch modifies one_shard_cycle in the container sharder to use\ncpool and spawn a child worker per broker path.\n\nperiodic_report_stats is also replaced with a heartbeat function for\nlogs every stats_interval seconds in an implementation similar to the\nobject reconstructor + replicator.\n\ntest_run_forever is also updated to accomodate the new tweaks\n\nChange-Id: Ifaa8ca05a6fb441defd330d33f1cb09a01ff7910\nSigned-off-by: Christian Ohanaja \u003ccohanaja@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc1aba75a46811890b8deb00871c4544c5689b5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc1aba75a46811890b8deb00871c4544c5689b5b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
