)]}'
{"id":"openstack%2Fswift~966917","triplet_id":"openstack%2Fswift~master~I5520eba7b12d6bd3831ddff7935911a0fb61e9f6","project":"openstack/swift","branch":"master","topic":"s3api-proxy-log","attention_set":{"38368":{"account":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"last_update":"2026-01-21 00:12:39.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-01-21 00:12:39.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2026-01-14 02:48:26.000000000","reason":"A robot voted negatively on a label"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-01-21 00:12:39.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-01-21 16:29:27.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I5520eba7b12d6bd3831ddff7935911a0fb61e9f6","subject":"s3api: fix s3api mpu complete log for correct c/o","status":"NEW","created":"2025-11-12 18:55:28.000000000","updated":"2026-02-20 16:32:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"af18d486c395d87c8d7a162963139ecb2caac94e","_number":966917,"virtual_id_number":966917,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-20 16:32:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":-1,"date":"2026-02-20 15:21:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-12 19:36:30.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2025-11-12 19:36:30.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-11-12 21:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-18 18:38:23.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"state":"REVIEWER"},{"updated":"2025-12-01 20:31:21.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2026-01-19 03:23:24.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"2595e03abec51fd9d4ec3ad6113aea39237b9eee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-11-12 18:55:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4064710dee213d2eb8c920b5dc78e2ed5badecf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-11-12 18:55:34.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"3c1249cd4a392a56c12a766e99f5d2e7b9f8702f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-11-12 19:10:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12d832a89757af91d0c9c5929988bbd7e1e1bc10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-11-12 19:12:53.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3f3a2838801a3221bbe779336dbcaec19f4be42b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-11-12 19:31:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"07c1af333885e3b4a9ff3e443c61570beed0e578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-11-12 20:06:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"31aff864bfbdd89433db60072410ffe261db4e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 21:08:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bbc458306c04d9ead26e8e370f64ecb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a8da400010b4577b0203c0a336c7307 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/01fc8d3667a24c2c97287f1606072f21 : SUCCESS in 1h 01m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99b4632970fb4fabad07f705259d10ab : SUCCESS in 58m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0e17df495ad4d689bfb35887d2900a3 : SUCCESS in 42m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fbeddc7375e445795064cab9a769dab : SUCCESS in 37m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0649122f2e4f4343ba46ec1d13ad95ce : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2804d6e4ffa411e8eb61ab152259084 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57631ffa51c24b1b91ca35b2f53d2ef7 : SUCCESS in 19m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9bafbe04ed2a448cbc86cce6bbb5287e : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/364c5edea4d04f31acf0b3d6c662cf15 : SUCCESS in 25m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bfd793cf53614e51b765c89e4e0d98b1 : SUCCESS in 22m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4c8e829c0abd4ebd9603594d8c0ec2c2 : SUCCESS in 28m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f61ee47384994e62b7e5df8d25d0fb50 : SUCCESS in 31m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ac0ed04050594534a6dad72126e9ea12 : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5d7c7a800b3348fa98182124a23728e6 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/656cc795ec43492abdf2f8f94f40783a : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ed6a047d36674569beea758257842a41 : SUCCESS in 6m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9306d68b799c4974ae81886e3f448c81 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c5efb91a4b0447129cf4cb99da928844 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/97a7bf1467db486584d396cf145beec5 : SUCCESS in 38m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0407377518d64dd7aeba80f23a7df43b : SUCCESS in 49m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f053057ccd9488fb7ae8e00fd0495cb : SUCCESS in 35m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64dfe6667dd74048a7d6121f7875ba76 : SUCCESS in 25m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f0bd84f4f424e1b9a983a26b85cb7f2 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/313b3431eb62484ea69866b0489bd714 : SUCCESS in 10m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76e0922e272f4e00be262f1071fffa1c : SUCCESS in 54m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/308a71a07601423182eaec33cc426cf9 : SUCCESS in 33m 34s","accounts_in_message":[],"_revision_number":5},{"id":"bad9971a6a458c940f595a0d07646c704cc8045c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 23:26:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18db05c7abf14d34b4d008277f579be4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e402ba7b29714e1b9c48d4e2e9569f1d : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":5},{"id":"8fdcc23c36e3f4132332fbd134bdf11c1840607d","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-11-18 18:38:23.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e4c66648f00ad0e19910ad7e8c6bf820afccefe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-11-20 21:57:03.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d82f6e8824380ae0ccc9c8387d9e53d749bbc163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 23:04:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bc94c4117f4486a889924b998aaed59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41cbe52eeed84ec4b91185fd1745e3b1 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/c6eb1f8c34ad406daa0816865585e418 : SUCCESS in 40m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e00d177a3ff5431794c64264cf7ed10a : SUCCESS in 56m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0277724a29c4c839740edcb01ebc5fd : SUCCESS in 43m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73dc302fe06d4c32a801130a10adaa3f : SUCCESS in 1h 02m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4bc5c22e60740358c163aa4bc0b32e7 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b3878c5ca2d4d0b8d90408f1ce8c2f7 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29bc975923544129ac0406c8df8c1ec2 : SUCCESS in 21m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/890e8057003840d2b92bdf191f267dc0 : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b422c9274b034baca09c5ee57fbee47e : SUCCESS in 24m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/de2ea85e0bc64b8ebe03190bc85d3058 : SUCCESS in 22m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ed3d0455088f4dbba1b0e65b7a3f9f6b : SUCCESS in 27m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/08c95f9941e2452693e58012a10685eb : SUCCESS in 24m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0275c7959c4b4798aa92ed31ac5767e4 : SUCCESS in 9m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c44e0721e1eb4ea09c68b6bca1703409 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/48e14813388443eeaecc03f6df543963 : SUCCESS in 9m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20bf732a4241455cbfcc46bcb233aff4 : SUCCESS in 6m 23s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8650a85e923e49a2ad01557ed1c457d4 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4cc800a7be06488b972652a468aa1c63 : SUCCESS in 5m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5235324fea2400daa0cdade73b45c32 : SUCCESS in 36m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b92f0a1dd0440d08efddf54a458ead4 : SUCCESS in 24m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc69fa5665914b9d8243f754e0e4f6a7 : SUCCESS in 55m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edc3515c6c414856be5967624c9a4d47 : SUCCESS in 28m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baf6231e9d8a4a108067b38fdeb13fd3 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c048809d399e48ed8c6b2b676522458c : SUCCESS in 19m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17beac6bf89b4435a4b7cf6ea5628dd9 : SUCCESS in 39m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c95f9edf88fe4534a5ab9b1def3a1933 : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":6},{"id":"f90ec2e7fe5a07b56786ad575a4e801184aff53d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 23:36:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c0117a8beea490e8906e37f2967be23\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fc0a374286744e239cd8a4f3c1ce06f8 : SUCCESS in 1h 36m 24s","accounts_in_message":[],"_revision_number":6},{"id":"35c670a1c0f38ac8c2c576efc91c0fcd17621f4c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-12-01 20:31:21.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"eb82ad0e7ecff47f59196a3097fce635a36d3c8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-12-18 19:38:33.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2e6c6c3600448a66c05a6e6d6e967ee01dcfcc60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 20:17:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df55d747c5f0434d9470ba796c8a9f17\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0945b1e33134842816e81e366e72e8c : SUCCESS in 35m 02s","accounts_in_message":[],"_revision_number":7},{"id":"0b08956a61ff6ae491d9de4eb4535b426b67e869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 20:37:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d67150174ec247a3b79f2e57d77f40ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2df8d095f2c46099b68dec1923a3764 : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/8b7b754f43894ff28ff403d07f8fe514 : SUCCESS in 29m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c5ddae54e864d85a7c995a86a937085 : SUCCESS in 51m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a64bd2c063da469fabc1decceaf6c507 : SUCCESS in 54m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/113d78e7fe864c33b426c06d0532c69c : SUCCESS in 33m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f90282cb7ba40d3a7dcb812a314feb0 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0aa2f3b195cb42839cbe7ceba4731e4e : SUCCESS in 10m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e335570f730429d9e37275b00441ce8 : SUCCESS in 23m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12c623b24dff42aa986c4f719ef5d4d1 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/23c5e63f12e64bf3abb715c0c992f6e9 : SUCCESS in 14m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/38158e3dc09441ec96e6b431f2a40803 : SUCCESS in 16m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/898490f27ae14ac6948b13bf5a5f7668 : SUCCESS in 20m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2628d4f89fe542f2a864f23c5a2e5ec5 : SUCCESS in 25m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2946fee74f3e4ce99bb8d12b2db32991 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/940ceb72d7c74d24b756055afa16f47c : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c9fd4551c990431ea30a7aa15e59a7a8 : SUCCESS in 13m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/21738dd331304cd29dc39089b5b66414 : SUCCESS in 6m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03d91058ae864b73ac32346e544b2214 : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0946c0d2f2c74376a06e411281aa208a : SUCCESS in 7m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/52e964ca67b44586b6b265ba462a699b : SUCCESS in 51m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/839b39c723b54a17870da06d3395e46b : SUCCESS in 55m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d282c4caa4bc4d9c86d364376add72f2 : SUCCESS in 49m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c78b0e13870494c949759c87686f331 : SUCCESS in 22m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/625e63786d0a4cb8bb652c16160c685d : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/017e4974717a44c9acdfb0186f7669c3 : SUCCESS in 15m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bc946e45b8446d4a300348929af8b1b : SUCCESS in 29m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b630856861614f758f86d4db0bd637a9 : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":7},{"id":"852432baf12ed59b537b18b79370a72094d0fd6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-01-12 23:07:31.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b45e736714f0ca2fa1e61da976441a1df067147c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 23:45:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74104318b7b145d3a472c1bb796dcf6b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/22162d959ab648de80179bc8725be053 : SUCCESS in 35m 02s","accounts_in_message":[],"_revision_number":8},{"id":"5eb84e9ab15f0d5418d3793c4cf05febb9a6038c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 00:17:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466b991464324a7b8d26304b1c79f1cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f245c9503f9465b81d65349593f332b : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/8cdcc9f43d4a49ddb798f07bc50de7ae : SUCCESS in 59m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fcdec29b34b746e597626f970d0a0df0 : SUCCESS in 49m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b7b3eca59fe4738a1311da212faf6bd : SUCCESS in 30m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32a6cef1f5b64cfead2eb4d93dafa01c : SUCCESS in 36m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b48b33825714923bf700ff0c1f1992d : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22728bff8c0b4d46857217304f548a4f : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60e18b9653f24cfc9f1bfaf230ed5b2a : SUCCESS in 21m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9580b1af52c94bc689070d06a1e0b00e : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/18d704c3710a4ba4b9d2c60538fda9f0 : SUCCESS in 22m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/843c1224e8f5416fa91e4e14579a627b : SUCCESS in 10m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b84a96ac758f4befb7e1fd7302fef5c7 : SUCCESS in 29m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c22df2cbc06d4a80bacfd60ac38cfb92 : SUCCESS in 12m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/67d07cb6c113494a8e16dd8acd7869e2 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4f98a65eb7ef46de8d6ee2de6732accb : SUCCESS in 8m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a87bbbcf7e744abe8868bcb1a3f9d4e5 : SUCCESS in 15m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f000d804dc054d87bfdd211ccc71415d : SUCCESS in 6m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f5a57af3a2e74072a0f848eab281f9bb : SUCCESS in 7m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2881b10534a143abae33b67266f7a8a4 : SUCCESS in 8m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e53bc2826014e0e93a96f4b89dcc6f0 : SUCCESS in 58m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0315bfd1e784341a9866c8225bb95b6 : SUCCESS in 55m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b3c35526896b481b95a6839f56d2dd1b : SUCCESS in 33m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfa0552f2119422d98f3d528a493065e : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12d4c143c0814560a25559aa06e4960e : SUCCESS in 3m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0655129752a14cf6bdadefd2325b1b25 : SUCCESS in 15m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a900f9bce6dc4d4b85dc4b7af5d31de3 : SUCCESS in 58m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90939cd5c2164c5db77f4b4a2727e3a1 : SUCCESS in 45m 42s","accounts_in_message":[],"_revision_number":8},{"id":"846bfaafa2b8376d14cc48505ee93f94caf7797e","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-01-13 00:30:35.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"7bf3704fdef7fc408a5de9c3f8cd60acc011f647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-01-14 01:45:18.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"99733d92cc86d3943639f78a0effcae3045750b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:47:57.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c219bfea3921476fac6308924e07bbe7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4e5ca4ed05ce4ddf94ce12ee072e93c3 : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":9},{"id":"35f003df301cdf36d6ffc21046080e857dc2c3db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:48:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6349410b3ed94bef8a193ba1f939e69c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8113abe50314c3088d186fe1f7dfd85 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/605002cc33044e809e37601eaeb3448e : SUCCESS in 42m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/293dac984c584d92b48aae6a2219fe25 : SUCCESS in 34m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c8b6f07e1b6d4be0bd0b2a2920311ca7 : SUCCESS in 1h 00m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9443e76401f448a3b86a2e91850d7c87 : SUCCESS in 1h 02m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/432195a9ae7941cb823964d88a2f7dbd : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46da25f8c7d343c5ad5879b35c481856 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/956e9e2e62d745748a9c97c83c8c37d3 : SUCCESS in 11m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2334062b467c4b5195954f46e1ecdd22 : SUCCESS in 8m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b182518912d04042a54f24a238ab66f3 : SUCCESS in 17m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/08d065fbd9f141da8f8bc75921533392 : SUCCESS in 15m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9bac179c132540e19b0d0361dba6d543 : SUCCESS in 29m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/537cd629a67f45008e78699b911dd359 : SUCCESS in 32m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9157bc5009584045aba4023d8c51b29e : SUCCESS in 11m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ebc3902c19f0461f8dacf2872aa7c8cd : SUCCESS in 7m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/85237b76d82e494a97c5d193259cb76d : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6065f713710452386cffc9531ecf558 : SUCCESS in 3m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03110b6676584984bc71d23c94e55dd0 : SUCCESS in 8m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0f669b3f2584f3d816db034a0c4c1b2 : SUCCESS in 8m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/12fdf4e9773c42c5b9d6e81d6579fa2e : SUCCESS in 45m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57a96bd0d29f426d9ce65567a8862385 : SUCCESS in 56m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/512b49056ec34a4ba33eb833f319c16b : SUCCESS in 21m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbe78abf5f88448db8b65ac8a67fc2ed : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b36c9065a94091a4ae2f3237a93d06 : SUCCESS in 2m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/53de0259b6c34a21b1a46192254d221d : FAILURE in 18m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/647ee9cd39c947509f3215846a14bef5 : SUCCESS in 28m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95e648cb914c489abe9cee14f9dae25d : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":9},{"id":"22d0fae08b72893fe0d563aa5e02766ab09b19b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-01-17 00:32:21.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"e488c729b8c4a4054c847e6a0a844499404615a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-17 01:40:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b57bde9453e14e3c81520cc4a107da16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd843ddef93f4d2fb77bbf132075f2c7 : SUCCESS in 5m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/1f7506341a8c40e0a0d497ea9f1a87f2 : SUCCESS in 59m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b364ac203384cd283798f24551b10b5 : SUCCESS in 1h 01m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e439933c4ee4236a6858f199d54adfe : SUCCESS in 50m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdef141b643449aca67669513dd7ba6d : SUCCESS in 1h 02m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a16fba6f5ba34b3f970f2b03c58680ac : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c4d5cae03fa4de4aa850247119f762e : SUCCESS in 10m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04cf75cd2b3f4e4c84d6c0f1dbb7cd8f : SUCCESS in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f3a98954440445e19085213aa2a70d8b : FAILURE in 3m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2334ec6412da4daf9fc9f317d3318871 : SUCCESS in 11m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e3e4ead742ce4d89994a6d67c8ff8453 : SUCCESS in 9m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a79045982f104434ab5b4825d7722f90 : SUCCESS in 27m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8b3f91b5bb10499b863ac1f005d9f66a : SUCCESS in 32m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cc79d437355e4604bbb74f22f5f4ae97 : SUCCESS in 12m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e981b4852f2a47ce8c5b5693324d924d : SUCCESS in 14m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9d5dea68203d4010ac2f8f53ea91d387 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/61e465dd0f644a32bdec4ab5b4f8daea : SUCCESS in 6m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da25e7c49ff7435b84761c50abe788a1 : SUCCESS in 15m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b48f1643758147058e4e4b1243047b4a : SUCCESS in 7m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/470fb987be0b4b4bbbaa01c943b2b80a : SUCCESS in 59m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c0715aaf43141f8823c94f84ca3619b : SUCCESS in 55m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4dc5f55494dc4d9990dec060c1544ad4 : SUCCESS in 49m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab92c7a692034ef6866dd80e0466f01c : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd0b1680286494bae9e5f213b455e60 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a653e3f9063a4d94adf99a030e4ea0e2 : SUCCESS in 24m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ad8a32e12594297b93fd885b46a8ace : SUCCESS in 51m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/920ad45485de4444b7959bf668cf13f2 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":10},{"id":"47d4c53994d1699142e6d5e1cf20e8d815f737d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-17 01:43:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69b8c84123c7452c8661847b9a08bff5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb6ffa442cf044e0ba77dbab6efada4f : SUCCESS in 1h 08m 30s","accounts_in_message":[],"_revision_number":10},{"id":"f1e6be81f4bc719cb716e3fcf3efdc39c22f73e2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-19 03:23:24.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a304bb90aec58f585946b2eadc7866baec599081","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-01-21 00:12:39.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"a6e51dea9c81a8553c277f20e9c1f17d48f5af76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 01:26:24.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a74fe527b5f946f9931e2616e1fb6ca7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c19d3a1fb44540bbb6564e13834d33 : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/6a7bfb3d0e1145e3b0da4282ab2b5b03 : SUCCESS in 59m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/566d675346054a1c90a3794a0724a22f : SUCCESS in 1h 04m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3299df334211411280b9787cf54db08e : SUCCESS in 51m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdb70789329f4e03b114370df3b2e8b4 : SUCCESS in 1h 01m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d42824c7d6841988ce01a5c4419fcbf : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9ca5de019d24e53ac89b7b6f5819351 : SUCCESS in 14m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f43de67201a456eb970d62aae69382b : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c918a080276403688fa29531eed665e : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d518d1525e8c40a1bec3d4adfc6865c6 : SUCCESS in 24m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7fe26459b6164558aff10774746b1716 : SUCCESS in 19m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1b7a5784e54f4e8ca2c466bdc4fc6e46 : SUCCESS in 16m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e8ffd5eccb8448fc9341119ca4df29c3 : SUCCESS in 28m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/100ed801829a4bfa8be645dcaec282a7 : SUCCESS in 12m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e869038365614ba9baa24ec16c9ce9ca : SUCCESS in 13m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b2e030e32ada41f0869096f2a2800983 : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c8f26e25f384b2fb6bba8c48d47690c : SUCCESS in 5m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dde0fcd474c04585b3467d5499269557 : SUCCESS in 13m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/44e488a571434a4f933163f8d9d2980c : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f20a545dc074812847bef97b3d69eb5 : SUCCESS in 1h 02m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2449fb32f31466bb942ad37a5e4ad71 : SUCCESS in 55m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/733a97d927594a57bb5623f3b6b6d233 : SUCCESS in 51m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/921a4f8e7e5740059870e39e90552f26 : SUCCESS in 21m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1af3dc78eb942df88336958d1bf0a37 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9463880db8124d0ab29fe35e4a3ed999 : FAILURE in 17m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5600014fb83f47039c8031fedb86522a : SUCCESS in 36m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da9c63d33f8741cc8c2bf5b7a5981486 : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":11},{"id":"61c57ef0881083ca040099f65f5898ea6b2b0b08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 01:33:32.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca3b0db9a1e34d21bc2967518717fee2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aad370aaf60948b484a6652c709cebc2 : SUCCESS in 1h 18m 37s","accounts_in_message":[],"_revision_number":11},{"id":"f04d54689beae4712b64d3a172d916f25511a6eb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-21 16:29:27.000000000","message":"Patch Set 11: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"4f1c64abba24d46f08f6afa88a4edf2d50e786a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2026-02-20 15:21:50.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"65720c1fb23ec303fc342f4decf2d7ea4d89506d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 15:55:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/958e87b6652649eea9db534ec0cecdc5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fa4e4707db434333825182c84d046874 : SUCCESS in 31m 42s","accounts_in_message":[],"_revision_number":12},{"id":"af18d486c395d87c8d7a162963139ecb2caac94e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 16:32:59.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c34d36004ace48f9b07fa6bbb6b5e19a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009b1688845645c48636dc0d6290bf0f : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/12027b18fce849f4a5e67454c8c75a06 : SUCCESS in 46m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/64875a3b02a74453b84f3884a9d5a52a : SUCCESS in 47m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/af97cc64897849a8ac2aed1fa3cefe39 : SUCCESS in 50m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09bb08cc375d40db90edc0edb6fe8749 : SUCCESS in 1h 09m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29dd91a78d5a498ba8991509d7d6fc58 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4064d94a4894da49310b935a83ae3b6 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/227bf35cd249405bbda1b9e0e75d8e77 : SUCCESS in 14m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f7675a5f36744fac830223883cc557fc : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d41b37a7606341b189a01d57423431a2 : SUCCESS in 12m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/00709030c8914cde93e5f668e092bb41 : SUCCESS in 21m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a84ba291bce407895c3e3399acb7920 : SUCCESS in 19m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f8dfffc958e84cd1b24ffe0beb1def4c : SUCCESS in 11m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/79b8188948d94e17ac47a0b1848de1e6 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e292009ee93845578f437ce29bd98177 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1d2fe98377ae46188d9d91f9ea2c9ceb : SUCCESS in 18m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/35422e3aa13e477faf33b64956bde46c : SUCCESS in 8m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d229524723354d83b3aca15050613b45 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6479654b92d54808b61011d4dad8d8d1 : SUCCESS in 7m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea341b116e994414b37ec2e325fe45f3 : SUCCESS in 56m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f4aaae6f91864398a378c405f1b99010 : SUCCESS in 23m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4eca53f0779b4b12ba33f07284f6169d : SUCCESS in 54m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/469a3cc0ac644b6c9dccabb83c070316 : SUCCESS in 26m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/151efc0108a44034bd0d2dd4d6d2fdf0 : SUCCESS in 2m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1381fb2f1ce44ee6ac0da04c397df5ce : SUCCESS in 21m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ba8e76f8adb4a14bd546aef7126d964 : SUCCESS in 31m 22s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"6dd61a8511de727a4c07122a6a964210a4408c11","revisions":{"5c9066e1eb924c7b0015c54341b13ca091666f38":{"kind":"REWORK","_number":1,"created":"2025-11-12 18:55:28.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/1"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:54:53.000000000","tz":-480},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c9066e1eb924c7b0015c54341b13ca091666f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c9066e1eb924c7b0015c54341b13ca091666f38"}]},"branch":"refs/heads/master"},"d4216404da6fa6365ab2f2cac1facd00e4a87506":{"kind":"REWORK","_number":2,"created":"2025-11-12 19:10:49.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/2"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 19:10:42.000000000","tz":-480},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4216404da6fa6365ab2f2cac1facd00e4a87506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4216404da6fa6365ab2f2cac1facd00e4a87506"}]},"branch":"refs/heads/master"},"ab5f24234427856d43b5da36bd54e82e097c3ecc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-11-12 19:12:53.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/3"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"Ashwin Nair","email":"ashnair@nvidia.com","date":"2025-11-12 19:12:53.000000000","tz":0},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that deletes\nparts). This caused sub-requests to log the parent\u0027s\nS3 bucket/object names instead of the actual container/object\nclient was operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab5f24234427856d43b5da36bd54e82e097c3ecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab5f24234427856d43b5da36bd54e82e097c3ecc"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ce29ae52a26c8f3f38e0e92c60833e1d40566b63":{"kind":"REWORK","_number":4,"created":"2025-11-12 19:31:30.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/4"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 19:31:23.000000000","tz":-480},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce29ae52a26c8f3f38e0e92c60833e1d40566b63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce29ae52a26c8f3f38e0e92c60833e1d40566b63"}]},"branch":"refs/heads/master"},"1cce25792c5cbfa0809c528bc91808b45d06d38d":{"kind":"REWORK","_number":5,"created":"2025-11-12 20:06:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/5"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 20:06:23.000000000","tz":-480},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cce25792c5cbfa0809c528bc91808b45d06d38d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cce25792c5cbfa0809c528bc91808b45d06d38d"}]},"branch":"refs/heads/master"},"c1c8a3adda3406d691cf7285ea42c716d229cb35":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-11-20 21:57:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/6"}}},"commit":{"parents":[{"commit":"676e68fc0f563f8971cb97e0b862518fbb08d7c3","subject":"Merge \"trivial: tests: add failure context to flakey assertion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/676e68fc0f563f8971cb97e0b862518fbb08d7c3"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"Ashwin Nair","email":"ashnair@nvidia.com","date":"2025-11-20 21:57:03.000000000","tz":0},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1c8a3adda3406d691cf7285ea42c716d229cb35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1c8a3adda3406d691cf7285ea42c716d229cb35"}]},"branch":"refs/heads/master","description":"Rebase"},"94a4640ca4cd8959360683090bb3f192a1591541":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-12-18 19:38:33.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/7"}}},"commit":{"parents":[{"commit":"0323f9c2ddb5e9ccb12c44ed2a82fd3ef3ea0bf8","subject":"Merge \"Add labeled metrics to s3api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0323f9c2ddb5e9ccb12c44ed2a82fd3ef3ea0bf8"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-12-18 19:38:33.000000000","tz":0},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94a4640ca4cd8959360683090bb3f192a1591541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94a4640ca4cd8959360683090bb3f192a1591541"}]},"branch":"refs/heads/master","description":"Rebase"},"ef3eb6218fd87bab84f37ee1a8e6f842d077561d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-01-12 23:07:31.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/8"}}},"commit":{"parents":[{"commit":"ff79e136704a689f72575277be9c17f205bec8de","subject":"Merge \"docs: Clarify circumstances for using clobber_hardlink_collisions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff79e136704a689f72575277be9c17f205bec8de"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2026-01-12 23:07:31.000000000","tz":0},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef3eb6218fd87bab84f37ee1a8e6f842d077561d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef3eb6218fd87bab84f37ee1a8e6f842d077561d"}]},"branch":"refs/heads/master","description":"Rebase"},"3d0ce3d3926ac31c396bf9d16d4dc3e409e41bce":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-01-14 01:45:18.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/9"}}},"commit":{"parents":[{"commit":"949f0fdac8624c1a40381edd1948e82b3092d350","subject":"tests: assert that X-Timestamp headers are valid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/949f0fdac8624c1a40381edd1948e82b3092d350"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2026-01-14 01:45:18.000000000","tz":0},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d0ce3d3926ac31c396bf9d16d4dc3e409e41bce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d0ce3d3926ac31c396bf9d16d4dc3e409e41bce"}]},"branch":"refs/heads/master","description":"Rebase"},"1d8511995d5f2c1238a3c0e6933d86baecc60bdc":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-01-17 00:32:21.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/10"}}},"commit":{"parents":[{"commit":"72191716d26291e17f46c9e2bd3c0e169a6f3004","subject":"Merge \"Recliam db_dir/*.tmp files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72191716d26291e17f46c9e2bd3c0e169a6f3004"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2026-01-17 00:32:21.000000000","tz":0},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d8511995d5f2c1238a3c0e6933d86baecc60bdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d8511995d5f2c1238a3c0e6933d86baecc60bdc"}]},"branch":"refs/heads/master","description":"Rebase"},"aee1b89a532a02410272517fe35aab8286702dc9":{"kind":"REWORK","_number":11,"created":"2026-01-21 00:12:39.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/11"}}},"commit":{"parents":[{"commit":"72191716d26291e17f46c9e2bd3c0e169a6f3004","subject":"Merge \"Recliam db_dir/*.tmp files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72191716d26291e17f46c9e2bd3c0e169a6f3004"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2026-01-21 00:12:23.000000000","tz":-480},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aee1b89a532a02410272517fe35aab8286702dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aee1b89a532a02410272517fe35aab8286702dc9"}]},"branch":"refs/heads/master"},"6dd61a8511de727a4c07122a6a964210a4408c11":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-02-20 15:21:50.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/17/966917/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/966917/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/966917/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/966917/12"}}},"commit":{"parents":[{"commit":"f15ccf81b4c4c03db3f65b2cc2174367ee6eac5a","subject":"tests: move test_db over to BaseUnitTestCase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f15ccf81b4c4c03db3f65b2cc2174367ee6eac5a"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-11-12 18:52:42.000000000","tz":-480},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2026-02-20 15:21:50.000000000","tz":0},"subject":"s3api: fix s3api mpu complete log for correct c/o","message":"s3api: fix s3api mpu complete log for correct c/o\n\nWhen completing S3 multipart uploads, the proxy access\nlogs show incorrect container and object fields for\nsub-requests (like the DELETE operation that cleans\nup the upload marker).This caused sub-requests to log\nthe parent\u0027s S3 bucket/object names instead of the\nactual Swift container/object they were operating on.\n\nChange-Id: I5520eba7b12d6bd3831ddff7935911a0fb61e9f6\nSigned-off-by: ashnair \u003cashnair@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6dd61a8511de727a4c07122a6a964210a4408c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6dd61a8511de727a4c07122a6a964210a4408c11"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
