)]}'
{"id":"openstack%2Fswift~966980","triplet_id":"openstack%2Fswift~master~I6d1de8caa1322168e5e26682a586d11dc606e19a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"38496":{"account":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"last_update":"2026-09-18 19:28:18.000000000","reason":"Change was submitted"},"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-09-18 19:28:18.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-09-18 17:03:09.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-09-18 19:28:18.000000000","reason":"Change was submitted"},"38767":{"account":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"last_update":"2026-09-18 19:28:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6d1de8caa1322168e5e26682a586d11dc606e19a","subject":"relinker: Add audit step to quarantine stale hash directories","status":"MERGED","created":"2025-11-12 21:43:05.000000000","updated":"2026-09-18 19:29:59.000000000","submitted":"2026-09-18 19:28:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":236,"unresolved_comment_count":20,"has_review_started":true,"submission_id":"966980","meta_rev_id":"639afa5c26f36c316b0c4b351dafb182f7066f17","_number":966980,"virtual_id_number":966980,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 19:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2026-09-18 17:03:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2026-09-18 17:03:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"}],"CC":[{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-12 22:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-19 05:35:56.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"REVIEWER"},{"updated":"2026-05-01 21:35:17.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"REMOVED"},{"updated":"2026-05-05 18:00:36.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"CC"},{"updated":"2026-05-12 17:08:00.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"REVIEWER"},{"updated":"2026-05-13 22:40:06.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"REMOVED"},{"updated":"2026-05-19 07:25:26.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"},{"updated":"2026-05-19 09:44:50.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-05-20 01:00:21.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"REVIEWER"},{"updated":"2026-05-20 02:50:15.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"REMOVED"},{"updated":"2026-05-21 17:23:10.000000000","updated_by":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"reviewer":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"state":"CC"},{"updated":"2026-05-26 13:30:44.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"CC"},{"updated":"2026-05-28 10:49:36.000000000","updated_by":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"reviewer":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"state":"REVIEWER"},{"updated":"2026-07-31 04:32:33.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"d3af7f191e0ee0efa0b0115299565190c2bcf314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-11-12 21:43:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59f69201f218d49a591b0142712387cf3429bc45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 22:55:39.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3b986ac5c744d65b0221137986d5231\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a525fe8e62a4841870567c81db01280 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/41d385ae978248529bc3147298c063b3 : SUCCESS in 31m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0719f84d0f564a75ac638964fda88c1c : SUCCESS in 57m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae0dcf3202af4ceeae67a9389b0b6752 : SUCCESS in 1h 00m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9490bffeca34d40b2aa3fd4166750d2 : SUCCESS in 1h 07m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da73f23beb33497d9c416fb21d44f951 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/748d2f30498f45d7b5f038c195f9588c : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5266f58a25ac47ac9b3915884ec4e69a : SUCCESS in 22m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b41053b2d1324fe1a067e3abcd12aa84 : SUCCESS in 8m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/00b711651651452d92d7ee8e6cbeaf74 : SUCCESS in 25m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4f6e608781fa4b178f87bfafc1cae023 : SUCCESS in 20m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/092dea8d7c6649a88672eae68dc30ed1 : SUCCESS in 26m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b77a298dd202469ea3e353aa7f106525 : SUCCESS in 30m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3b6d6c740b5847608aedb44b7046bd3c : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9f2b561592744bf4a802cab8032d3568 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3e07123c41024365a68f89e43b5b57d3 : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1e0c9e909cfb42aab47296e978344a72 : SUCCESS in 6m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d0c3dfe38aa3451d94db22d2b52e32c9 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/023acac238574740b5d35e0df1960194 : SUCCESS in 5m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b73d55972f04983a6359498d5e7f9c3 : SUCCESS in 36m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/61a2b26b827f45e3bef595671851ff5e : SUCCESS in 26m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/15bb39ea37a44dc6a6cb8cd3d29cc50d : SUCCESS in 32m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e52062043ea347cb862442ec56a53a8c : SUCCESS in 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d56461e6264e48ee9c1faed28b49f4fd : FAILURE in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4c5558ae2d6d4c41ac9050ff6411c854 : SUCCESS in 15m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a5e1331f687424b87f019df162eb62b : SUCCESS in 46m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7556374ec2ec48b4a1172e1db699524a : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ed9e6a32734eeb529296eda18ddb22ada01afffb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 23:55:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/badd4743f7d943999c380e2000b46966\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d68daff1e464a8ca05a3c5f25b28842 : SUCCESS in 1h 05m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8a0f2e08f9b5f3632dcea5d6e70e3bb722343b6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-01-05 16:03:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"25278112c9e7468c09c5bfc88a56e6f5381ca4b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 18:07:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/567b16b301b542b5baf593466ff80465\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23e532be3f4549d6b33fee4130565c67 : SUCCESS in 4m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/3c1ab76389c94d658e2a14aa0778d7c4 : SUCCESS in 32m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c884e3b2d6914b7a8995d3ee5d85c597 : SUCCESS in 59m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c193d87ae1714c5888dcb58f4d04e3d3 : SUCCESS in 39m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1af8a781a95d400cbe06a41272cd3eea : SUCCESS in 44m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ccbecb6843fd4f9f89a3d83429f9a10f : SUCCESS in 10m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8de526edddb046a28a47fee3daf39752 : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3f0bc584716487a849e91743e78b487 : SUCCESS in 19m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/41fd64809b5d42658160d7795499f528 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b1b9f0ef08d4481aeabd6a341f8d626 : SUCCESS in 24m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dfcac504489542f583a46f64fdc27904 : SUCCESS in 16m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ffc3b73fc1fa402e837d969a66d60980 : SUCCESS in 19m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7fe22d601eac4c8e9d416f3ce1e43905 : SUCCESS in 30m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/28e91b869ab64f079665e28fee9e860b : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5a1b71d5b1e943a3bf8facadf5d15856 : SUCCESS in 11m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/058a5519691245779331457263fd3b05 : SUCCESS in 12m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/504d38a2a7104b5b83281d83cbd89728 : SUCCESS in 5m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a6ddfb73f6a4aed80d6ceef0a736a93 : SUCCESS in 12m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8b3fa4cef923482aab27f06cd0e3ced6 : SUCCESS in 7m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3195edbd2d742af99bbc4b4bbc781b4 : SUCCESS in 58m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c44e117b42a4df8a777dcb2c785f45f : SUCCESS in 43m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14c81799fbad47348f5ee017559fc931 : SUCCESS in 39m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b5e9ec2989f403da3d96f1fe3f8e264 : SUCCESS in 27m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f528f0cc2f064ea89781e85644165819 : FAILURE in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8255fe72b17a4c20a9e1cc42f9d06ff1 : SUCCESS in 16m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40485be858844fc79e3693801f648254 : SUCCESS in 38m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/371dea4825a948df85d3b133f3d26cdf : SUCCESS in 58m 34s","accounts_in_message":[],"_revision_number":2},{"id":"c4f515bc19bbab0578f0d06735dbf6cd5f37161b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 19:17:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e59a53fabf6406da9fa0a8b4ec0ea33\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac2c8e57350c44a6845058c66058a5e5 : SUCCESS in 35m 48s","accounts_in_message":[],"_revision_number":2},{"id":"38f575523da7c29b0772e64a72181d6f705ead61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-02-19 05:30:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"92cb63f9202d425848eb452f2cc04ba8dce348cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-02-19 05:35:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"77905ff005523c38ffe48e59803f79925bc3f399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-02-19 05:43:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"07bf5d189ff8d082a9facec1251906036008aedb","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-02-19 05:46:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"afb12840ad1af9b7a38092d698b05bb781d47ec3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 06:13:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60fe318950f74d5eae737d17199daab2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/43b556f8ed7b4dc1a51ae912882ee21c : SUCCESS in 28m 24s","accounts_in_message":[],"_revision_number":5},{"id":"9915535c7da38e8abfa3e32c9b92aeb33215f6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 06:49:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/367722b33e1f475da7ba0dd4444f64a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c7c1dfd1184589bc166933e2c45f07 : SUCCESS in 5m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/6ad08de8b4df4276b691bab9f5e3313a : POST_FAILURE in 34m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/715f1670c6e440cf8a49523c69ceb811 : POST_FAILURE in 58m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75febfa8635941e4bd8925277b16bf06 : SUCCESS in 55m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06d3bc7ec85448cf80cf01dcfbd38215 : SUCCESS in 1h 01m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6d1fe4abea14aad9779739a4c486155 : SUCCESS in 10m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19638a4e03cc46289582abf2ba15c106 : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f26deb54c524a94985ec07db9e1b84b : SUCCESS in 11m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/90b25401119149ec8723f4bab17869fa : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dfefda377b2744cc930f29f645ce9793 : SUCCESS in 22m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3313b1e0531e47358ab14d9e1f07d286 : SUCCESS in 23m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f1154bc496cc47bc8816c1ba94da9bd4 : SUCCESS in 28m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/80c76452229847b4b31c397edcd7d040 : SUCCESS in 12m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/62f0bc114b5e4dbb8dc26a04d1cd0bf8 : SUCCESS in 8m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4e47ee65918742a7ad1a63db28d25f8a : SUCCESS in 16m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e359565c4e5c41e6bd4b55e778671c2d : SUCCESS in 15m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1c9ce18bdc1147dab8c1ec3651fb64c5 : SUCCESS in 5m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e744468705346fa8e649489b5007a1f : SUCCESS in 13m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a2c97250b55426eb477d6cd417776bc : SUCCESS in 5m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f71cfff8fee9422e86e2270d186bc248 : FAILURE in 58m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e79f2fbf01154f91b2301cef56f750aa : SUCCESS in 58m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5925c5cf91d043178e2968566fe8346f : SUCCESS in 55m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c20d6eac8794138bf7e0d22b18e5c75 : SUCCESS in 28m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05b24ce7abb648e9abc5ab4f9fcb571d : FAILURE in 2m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f964d546ccdb4bb796b8568972d2e46c : SUCCESS in 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96c45b4f503448b2954dd314765d6b0d : SUCCESS in 31m 46s","accounts_in_message":[],"_revision_number":5},{"id":"f0f3f74cc60e67bb0b39bba72c657bbd3049d08a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-04-09 18:15:39.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cc92a5903e6d514c4cff2bc6111ced3ae50f25d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-04-09 18:37:41.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"1f44f101d412e63beed9681f7b3aca8758a227b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 19:42:33.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f492eb2d9bf64e52bc0f5f41f51874be\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53f00f0838084503b183e5e18e7612e3 : FAILURE in 33m 06s","accounts_in_message":[],"_revision_number":7},{"id":"f13afac6a3f33d6abe04b2d397f9b6115b9e8236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 19:47:03.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4318527b29904ecda82584357046439c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52bf3071e9784814a45a9400aeb0b2d6 : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/461fdaeb8bb844cb93c13fdbbfa61999 : SUCCESS in 1h 04m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab25ed48a7a247f0bf40e454a2c4f63e : SUCCESS in 1h 05m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2fcdd3d0fd584dccac6e2ff2d423fdc3 : SUCCESS in 57m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8100a83e65e4b579cf61ed8ff7448a2 : SUCCESS in 1h 02m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edbadd45988748ebbd8180d48fdbb317 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5472bd692aea451aa8d72ed714ebe281 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e1c716e8a4e4a069cdff30a2cd29be4 : SUCCESS in 19m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/992185afbbcd49cf88edd2d33a30a8ef : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0b1a30ba1cd41ce8dd4f0d0f9afb5ee : FAILURE in 25m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cbc6b93080f34a3884a88818659517ee : FAILURE in 17m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b37fb0f0ee704796a31cee4a92363590 : FAILURE in 29m 29s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f850d1ca519947a88c2e78ffaaf7f804 : FAILURE in 10m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/65f096c9884b4419867df40282641ebb : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0ca8332ae0544f7dabbb3cfdb945e104 : SUCCESS in 15m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cea627d6bb974fd48eb5b7386f6abc38 : SUCCESS in 14m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d50b8db256c74376ad7786b99e3d2dba : SUCCESS in 5m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eba000d3b38a4fb7b6f05fa46a5c48e0 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/57b25a210bdb43af874d469725a2c004 : SUCCESS in 6m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb78606350ed44b489e5cc45c600bdfc : SUCCESS in 40m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c6076dee4af4ca0a8365b3a3cef9f99 : SUCCESS in 27m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6117380d793f49ed8556c07055cc20c4 : SUCCESS in 27m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc88fa1c612f43fb9c6dfcfbddc8464e : FAILURE in 28m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d5f7b867974d3ba7f9f6aabc39bf2b : FAILURE in 3m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eea8fcdc35ca40469766f627bc16f808 : SUCCESS in 19m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0009fd36c1a4685a117cd1065554d3e : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":7},{"id":"52817ff6fef95d13395b51e76cbc941990e4d65c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-09 19:55:24.000000000","message":"Patch Set 7: Code-Review+1\n\n(11 comments)","accounts_in_message":[],"_revision_number":7},{"id":"70e391383be9e3ff679719823c0873b451107253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-04-15 20:02:54.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"c5566211266647428334efdcdc40c7dfa9a38c84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-04-15 20:09:27.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f3869afb8cc00d7533825de05b3448e81ced7d64","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-04-15 20:25:10.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d0a308ff8963c9a1190fc3c0655d8e6cb5dcb756","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 20:53:20.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a2538586c4d4b0f833bfcb6477e8a71\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c7e4fcb33ccd49918c5fe534ab2685c8 : FAILURE in 39m 25s","accounts_in_message":[],"_revision_number":9},{"id":"665fb5b0d117296144a8734d7a0397543a0c8c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 21:12:41.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86775cd081b24709975b923dc3a73b86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c36e0001cc4c1383de01071783f9b5 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/21393338505b41c4bb16509b8826a9d6 : SUCCESS in 32m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a6087e74a17c42caaf133b7603ae17ae : SUCCESS in 42m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58410ed766b447bcab3f439062ea5024 : SUCCESS in 56m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21319dd877d64c389ef5c69b328ca66b : SUCCESS in 33m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e81d2b5244844e9b38cb2ec0ed9cd66 : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3061628f87cc469ba2bf4cdf31caeb05 : SUCCESS in 21m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6338a5db1dcc4f0ca5ebf7503b5906dc : SUCCESS in 21m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de3f3ee354164ecc8fdea33a62b52b81 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/af6bac8d07a34a4d8a03729b55924f23 : FAILURE in 19m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/afc277ea24ca449fadf18a1ef8076f85 : FAILURE in 17m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/adfae0471dda4971ad8afc0d008ebe29 : FAILURE in 14m 40s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1a4efcf913264c6e899c3f6e95c9c37e : FAILURE in 16m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3c96723b893c4f19bc3d715ef1fc7674 : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ffc2f41e23134168a9c5061e7db2b65a : SUCCESS in 8m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6ae2dc968eb34141b3cc23a10d6208e2 : SUCCESS in 14m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1cd1556fd8c84db5b73e1ac3f8cb7457 : SUCCESS in 6m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/232901089ea5429aaf31b8de57bb1548 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fac35d68eb5642a1bf7ea5d46e19c047 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e8d9bfe0a9a4d2da2d3446f7288abb1 : FAILURE in 57m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c6f5e627f1443b7bb1c3ddefe75087f : SUCCESS in 34m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be3d7f7763bb4d5097e1b36e0f20fc8e : SUCCESS in 49m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aec501a95c849eb8f2cf13bff2f2a89 : FAILURE in 26m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c5189ef8364a2cb6b9864a041352c4 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/609f0f17f1d749eeb545aaad8ae41a5f : SUCCESS in 18m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/670bf640045c4993aa20cf2e0c3b6b8f : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":9},{"id":"bc1dee3bb0ae32336f569b6b1d558553f1c8416c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-04-15 23:13:47.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"b1ee163239f5102dde95ca11c2af414c78db4864","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-04-16 16:34:09.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ea5fd704bedca343d71e593a4e368b5abb4d430a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-04-30 19:58:55.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c7d59681bab6c0b9ff84815ff9811be1410886f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 20:33:51.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/023a7d70902e4eaf9a61b7ba15714a34\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/82c6aa35462f4a1da40c83bcf4e7adbf : FAILURE in 33m 06s","accounts_in_message":[],"_revision_number":11},{"id":"6a0240e1ab19bf4529e4f6886de0da8a6ebfbc25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 21:09:56.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8cb3abd0fee94ac0bbb23d765c25b7ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96af485bb2774fefae9d4fa6090c162b : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/4dc3c8d665c6487886ea8da11053f7ad : SUCCESS in 1h 02m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21624754b1ca46c482288475674577cd : SUCCESS in 42m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8fe9d623ed954c938eb8466ea805dc2c : SUCCESS in 43m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ba656b561e04fe59cfbafb021d70917 : SUCCESS in 1h 05m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a4b7faddd1d44d9b411421927131cce : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67edab038ac242f3a9ccd0e528d562e7 : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da1565d3804e4b1aaf3a7b79c90c55c5 : SUCCESS in 20m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/09e7154f31bc494ea1b03eefa812f44b : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/de51cbacc05e483c84b6955920f7a639 : FAILURE in 10m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/69d1c1f753e147a495ec4eda14f5ec37 : FAILURE in 22m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8544edaded5248ce9375c7cfcf0d94d9 : FAILURE in 11m 33s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ae7cdc00341f4deeab067163d02996f5 : FAILURE in 23m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e6ef056c3fc84086a47666569832bb27 : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c04d1ba0e0624dc3936a1348624486e9 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/37058c912ace4544806a089508d14fe7 : SUCCESS in 19m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/da4136352a234e5e82dd1782b25f4424 : SUCCESS in 5m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aa4e3eb3b6a04fde83ef73dadfc30de8 : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4d3990ad7af44712bf444308a2898fc1 : SUCCESS in 5m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dea38d32aba480cb24b8be44cbc1de1 : FAILURE in 1h 04m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/955ab5da99ec4c1092b7112a185d6c87 : SUCCESS in 22m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3bc9861f92da446c95f7623f1eda4a3a : SUCCESS in 35m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fe6d1fecc4245e5abc140fb98b73e57 : FAILURE in 22m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e595ca69958c4447916320c05523c29c : SUCCESS in 2m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/19f399405c3d485c81a664834fe0d95f : SUCCESS in 19m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/807c41c05e5c42f799b2bfb8e7ec5b7e : SUCCESS in 42m 11s","accounts_in_message":[],"_revision_number":11},{"id":"b6f55c20d15c0575a4966a3f87ab449a9d6abe47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-01 02:49:16.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"4afc99d1905096b19cc0f858ebbfff70b927b90b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 04:01:32.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62166bb87c3947ee8334dcd4d3053d1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27f867f37fa644d4892139970d4fa0cf : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/f7595a2a7c60468fabee7cd518aa67c4 : SUCCESS in 1h 01m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5626676c6120482fadefb62e5293a86f : SUCCESS in 47m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/57e4f6856dc0429baeeb380955d1d8e6 : SUCCESS in 55m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14706573a2de495ca6e772023e67a0e5 : SUCCESS in 51m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b7b0fd9b80f415d879f09db2e3d897b : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d79f53372188474a9485414e6d9ce218 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eec78affb1c84a5f83287ffec55fe8b6 : SUCCESS in 21m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d23dcea56884ac7b5989d972f5e6ae3 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1300da984280430dadc6f7eb4fb518c1 : SUCCESS in 24m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1622d68ee7934272bb2c5f0aa785d7c3 : SUCCESS in 22m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/806c457902ea4028b4cbcd704f4c8af5 : SUCCESS in 30m 03s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ea9e857f337440d3b1b7d2232eb2499d : SUCCESS in 18m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b7879bb61633419eb67f4b4e9b5ecece : SUCCESS in 12m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cdc3a6ecb24b477cb2b010530d379f86 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9982e464e67b4a76a0c99100e569078f : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ff4ce878a579467c804fdeb1ddd8492a : SUCCESS in 6m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f810d753f2b4a4e836a69dee77bba46 : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/867874793483472798dd820bf82a7fc9 : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6135a0f655864f039ba85e7963a23b8c : FAILURE in 1h 02m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff5f4f2b0570462daa7c5526580ed87b : SUCCESS in 38m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b5bcad3d6d942fb8ffec7ddab2372eb : SUCCESS in 50m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e8387771828478f9f2cecfb42ddde5e : SUCCESS in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9970547b15084975ba0ffbf5d14c08b3 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6912eca33d4c438fb810228391709c56 : SUCCESS in 24m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40c960c2b09849e3a9993da747b3b32b : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":12},{"id":"dba8d6ed24e097f111a18d09b4cdcea6a0ca5298","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 04:50:44.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a8a18082bb42048063f9174331a1ef\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea136dceb7a54498bd1b3484590fde07 : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":12},{"id":"9d861f2ed5dfa1077db6cc1cff2265b925da4fed","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-01 21:35:17.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a0e9a8bb0836aa2c570c2c58b623ffcb70d6da9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 22:14:28.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be4fa0a9ceac49a6a5a4b5ea3eba13e1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c42efce130a4446988f09800e1a28fa6 : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":12},{"id":"8f8490ba701ecdd5e1ffe9c43984e3ec7baac651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 22:52:48.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e09496c0d15483a8d8275a43b8fe438\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25bc02596a444489ba80e1b8b9973c34 : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/2279bed8081b46f39e531f03445cc282 : SUCCESS in 1h 08m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3b122558ba64e8597376eac14ddf14b : SUCCESS in 1h 04m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9c4bbffd63d422d84c49aae3a95c9f4 : SUCCESS in 53m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e24e635e299429b92b7a47477d0b2f9 : SUCCESS in 1h 04m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21ee55ac099a4491a4d633f1b5a27d34 : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/babecfb0b3e94e2196b106d10561e314 : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/901ec3c7f0ed4bc3838a70f3debf5d15 : SUCCESS in 17m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9883fff5d5ad40d99e63143f4e28e53f : SUCCESS in 4m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/358bfd1324d44e38b687d4d9e947cd9f : SUCCESS in 19m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1070c5eb728f4dde8bc8dd1d046384c2 : SUCCESS in 19m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/75041c7fde8647749b244d989f4f1a24 : SUCCESS in 13m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a7bc02c0031e479090c1061515c0a3b4 : SUCCESS in 21m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3a4874351b3c4e9183aa7ff328725a64 : SUCCESS in 12m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cfd09e6694904d42b62f0c505905948a : SUCCESS in 15m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/703ec446e1724920a87bb253edb5a66f : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2fa63891c204a7aa700f0cd7cd19c10 : SUCCESS in 4m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a18d895b73b41ac95a8cd5e2295219d : SUCCESS in 10m 41s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3cea0898ae1c470784bfab161b251d5a : SUCCESS in 5m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/827e880a72d0469ab1d0b2d007e309ba : FAILURE in 56m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/285cb8eae99c4675803b6147185802ff : SUCCESS in 33m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/91434729efdb4ef6b32fbb434dfc5c66 : SUCCESS in 27m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57df4c68018b4b3cb26d885873c17737 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7faee561d03f4811a880281e8b466161 : SUCCESS in 2m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f1941ad8d5545b89b77ec876f367f43 : FAILURE in 1h 00m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd2fa7d9b4a240f2a2f7bd744c215ecb : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":12},{"id":"ca0bb70d073faab4735a055e5759de98a87d22e1","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-05 18:00:36.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"80b1b5db3c955a6fca1cd424bb8d96713b0f1b8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 18:33:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f3ab3651bad4210b3899e35df5de5a6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3d4602569b1492c9d6e3d4012c89057 : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":12},{"id":"635c28b10643fbe0f3fb616c10dd2d6523e9a81f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 19:12:24.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebaf53c0d1c442f0a6b764e8133be597\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96a192c713d543b8a7c3a4d9348f90fb : SUCCESS in 6m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/6a49af0586d14e5d84b2e1741820faa4 : SUCCESS in 1h 01m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4838f46288fc46438550ee6e0cc21fa5 : SUCCESS in 41m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c8c817699fd482b8e213e335b326f83 : SUCCESS in 57m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a496f158d4a442ba7a4d900c8d66ad8 : SUCCESS in 42m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e222562054714681bc7f6737d9652e89 : SUCCESS in 10m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0f35291c70a4137a24fc7f9a672acbc : SUCCESS in 14m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/483fdecb0d7145538980de869f2b4918 : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3d8844569dc7413fba788329e665db32 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2220ee46a38446a5959bf8152f57bd3f : SUCCESS in 26m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3420cb8110e24778ae03cbff5f81cc9c : SUCCESS in 13m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0afd064f60ec462faa3d9aafc661b27e : SUCCESS in 24m 14s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3e4048244d5b49efb212c84dd05f02e1 : SUCCESS in 27m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c3b22e0a3d5a41aa9c7e7408e554198a : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0200cd4da035471b9d9b99d85e2f5f0c : SUCCESS in 11m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e7bf2f55e523407aac33694fa60f6954 : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83614dc847d54875b21533252cc830dc : SUCCESS in 4m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/092e4c76e0654bfca088a89133deb4f8 : SUCCESS in 13m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9847ed002f4245d2862df31356eabeb3 : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f36ded346b1f4d50afa536915d9901ea : FAILURE in 1h 03m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0f637c5b73446ddba96989e0272eb45 : SUCCESS in 58m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b9c149efdff2413aab80b791921e0f45 : SUCCESS in 32m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caf90d4e29d6403fa9096d9ced087e32 : SUCCESS in 32m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0342f3c09e442d696afd8609bb7b7a2 : SUCCESS in 2m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/20ebf486cc2c456f96d8717736bdd726 : FAILURE in 37m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c007cc87748f4fd3b74e248d4db54705 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":12},{"id":"fff178b9b476191e801a912e33e64b05f99a0dee","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-05-06 14:47:24.000000000","message":"Patch Set 12: Code-Review-1\n\n(41 comments)","accounts_in_message":[],"_revision_number":12},{"id":"40df00d2598011d48b1feb0c7eca49cc7be20c60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-12 17:08:00.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"f74beb1012862939c06faf6d2079f53ac35f3eee","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-12 17:09:08.000000000","message":"Patch Set 13:\n\n(40 comments)","accounts_in_message":[],"_revision_number":13},{"id":"7800dbc23d4ba15acce377f135cfddd0063a5fb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 17:43:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c83c1abeef4e48beab9c4196855d44\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd5aecacf27344daaf243d4f48c58be9 : SUCCESS in 27m 23s","accounts_in_message":[],"_revision_number":13},{"id":"815f5010f9d9e90ae13230e00e8f04d398e16071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 18:19:51.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70def4c1f6344677ac3afd8ca8d8e71b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf24b315422a4b3db883852927e2d3a8 : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bf561ab2bbb420eb8b0015f151883f3 : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/b7bb01344ea6465a9e9db653a0ec21c9 : SUCCESS in 1h 00m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85042655814a4ef6b4811da936a45626 : SUCCESS in 32m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8dc9592f6ea8478e97e9aaae306c4f42 : SUCCESS in 57m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2976619c8c4f47888b383a8dc27bc02a : SUCCESS in 34m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a0fa656dc1d4e2ea84dd186592dceba : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95b9b309f6f643e6a518741364c5ff25 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9a9764d5f004d2d80b33927f3372591 : SUCCESS in 20m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5bf1e7889298413cbab6ffb3d66b169e : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e46071cf6ab3434bafd18286777a964a : SUCCESS in 11m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/27a8255da7de4f3a956d601ad9555464 : SUCCESS in 10m 08s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/122e68acc5fa414c91fea64b68fc06e6 : SUCCESS in 25m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/49e88b17b925429482e4c9b8473a91b8 : SUCCESS in 24m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/038941fbd3da4745a7c3d3fc4b261394 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3fabca9eec3346e38673a99e653c8d62 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4bd8f36f521e437db19b80480babfdaa : SUCCESS in 8m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e4b34b1461d49e58f4ef7cd8c7b43d7 : SUCCESS in 3m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/18c6e769e1d34d31aac3fc984861bb97 : SUCCESS in 22m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f0d4267ba7694bbebb3c0a927e976aa6 : SUCCESS in 8m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd9ed34b23f54c87aa6b091c2bf98d8d : SUCCESS in 1h 03m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/49ae45d66fb4471f91d06a36f4768afd : SUCCESS in 27m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc2a176633574c2fab08b0a11102c32d : SUCCESS in 22m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/186ca15a92334b17bdfc6fa010ef3bdf : SUCCESS in 24m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a521948c4c04867931e97703e7eecde : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/75eae4cd9e484ce1a0ec5787820df20d : SUCCESS in 20m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9caaa13188cd4cc4aa95b30d2d23b32c : SUCCESS in 51m 27s","accounts_in_message":[],"_revision_number":13},{"id":"0507c251e8dfcdb4f9e7cfa875b8f0c692b0f90d","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-13 22:40:06.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_38767\u003e.","accounts_in_message":[{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"_revision_number":13},{"id":"cd1ecbc1845452a2ea65bb1b743b9adeb2518da5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-13 22:40:23.000000000","message":"Topic bug/relinker-audit-review-comments removed","accounts_in_message":[],"_revision_number":13},{"id":"8a57b2d5bf80adebeda7b8fba114bab485bd9446","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-19 07:25:26.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"ae302eccc90b75a7ac854ce034dbe009775e924a","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-19 09:44:50.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"69f9314c3e908a4cabbac91ae5e567818105e33d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-20 01:00:21.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"be733dcd52ccd95dfc5f0a638795c42efc056fa8","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-20 01:01:32.000000000","message":"Patch Set 13:\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"499335e92f32e9566aa2c4b1a20318bd9baf30e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 02:14:33.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37e2cbe3acf04a5684eed6a31c8efb54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/538acf84591d43e891045b8c69d42547 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f728b116fbfc4096a15c8252589edbe2 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/54f919374d82410482209453a4a45cea : SUCCESS in 53m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1749d29731d49cb836c19958191ff9f : SUCCESS in 1h 08m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86cc436ff41a4cf396e18daaddffe985 : SUCCESS in 54m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53e7434c2a8e445a815d9a04bd6bf3c2 : SUCCESS in 34m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e8a10c939794e618ac3980594ee9db0 : RETRY_LIMIT in 1m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed6356b5d294472b841905d973f0b423 : RETRY_LIMIT in 1m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4de520e0aae74016af77aa8791574bc1 : RETRY_LIMIT in 1m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9abd2173c76a4ee193324bfd9a8cd316 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1322cbc6cfb400da2111f787c022355 : SUCCESS in 11m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ea2b76b0c9a1420c90ecd49e476486a8 : SUCCESS in 21m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/03938eff07914e2b862967349efed2e1 : SUCCESS in 11m 18s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1f3bc03588704044b9ac18ba47e95335 : SUCCESS in 20m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7fc6645ce8d647eb9d1cc3f54cb5d0b7 : SUCCESS in 8m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8190d97571824770acf58b16e317b062 : SUCCESS in 13m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f774c4c3c2684690a6cb6341de4d811f : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9a54da7646b4c47a1079c1b9bbc5d85 : RETRY_LIMIT in 1m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/263774667e4847e391b30a5b72eea42b : RETRY_LIMIT in 1m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e97bbc2e23e3484081116ac085627f2b : RETRY_LIMIT in 1m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c1ca6a62d9347868d964bacc243e2e7 : RETRY_LIMIT in 1m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7edd4f589d4c44a5b9665ab3a6d72f92 : FAILURE in 23m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4b39e12b855c4d7fb0657361989cb277 : FAILURE in 20m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c1de6c624c24e0ebf593a52e2a7ff16 : SUCCESS in 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfb12e30d9124a93a55d8131354705be : SUCCESS in 2m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2268e98087924b45b8ff094d1d5eac4b : RETRY_LIMIT in 1m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de66144bc209439cb6f8427e471dbe72 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":14},{"id":"ea0dbeffcbca09816edf1b8db91b0bc42e8280c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 02:15:47.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a491195c8bd34124857e27e1105e5370\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2eccdfc0cc9146f095887ac12a3aa071 : SUCCESS in 1h 11m 50s","accounts_in_message":[],"_revision_number":14},{"id":"81a529d8d96ad0684fc324f1fe980d7758b4b687","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-20 02:49:41.000000000","message":"Topic bug/relinker-audit-review-comments removed","accounts_in_message":[],"_revision_number":14},{"id":"608a6e6b2088b32acd796dab4f0c240eefd1fc20","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-20 02:50:15.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_38767\u003e.","accounts_in_message":[{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"}],"_revision_number":14},{"id":"7c82ad5426c5aa40373ba3c5a7697dfd3d1abe74","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-05-21 17:23:10.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"381e0eff5d6adedfa7f9142c3f6ed120a24d7350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 18:36:25.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfba621c298645a1bea36d43fc9fe677\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b3106e373c4932b44be5b07054829d : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/736167faa84a417baf0282b4910843d2 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/24ab889a953347e385e131da8bd392ca : SUCCESS in 1h 07m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/530171f5cb194081b260b40769cee0bd : SUCCESS in 1h 07m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ff8e78e7cb944c4b2a759cd2b0c22bb : SUCCESS in 57m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8b92c05a9284a539822537bdcc5720c : SUCCESS in 1h 05m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98eb7582ae0c4a1e820c96f400481aa7 : SUCCESS in 23m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/794b015a4260417b87461804dd4c2541 : SUCCESS in 21m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b03a3caaec4401da8bbdc9480ac4d3f : SUCCESS in 20m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1a6cff59aa244e596ac98d86b7a7c7f : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6aaf35f3f58b4e9eb9117e4ace8a14f0 : SUCCESS in 25m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bb1944afeae244ba80ca629fbd51b5f3 : SUCCESS in 22m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0734213a8a4f4edaab43557489981c67 : SUCCESS in 13m 05s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a764cdeb37a54b9f924b5feedf244f3a : SUCCESS in 24m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1aec6c21d3c948b5a4ef60bd4c464aa8 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/869ada67320d412baf02de897e8b233b : SUCCESS in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5c590391810e44bdb93270f832e1f263 : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/42cc9bd589a244e8a225186f48614d8f : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d67cc0b41e9645b4962327ac641e1596 : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b50593f3ab20494fb3b07872408e7987 : SUCCESS in 8m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4962608411c54925ac75bb9c0e685e9a : SUCCESS in 59m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d16deff329c48c89389da55e2ef99b0 : SUCCESS in 50m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8ac1f07a5dde4a12a2e9ea4bbe075f2c : SUCCESS in 55m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ec185e611d449e2a3037b6ce3c08f2a : SUCCESS in 30m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b777b3f87946b2890ac4a78738b112 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55acbf60f9a14349887885fe921f0950 : SUCCESS in 21m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27d63f2f3b9840f38b64c06bb76db5df : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":14},{"id":"787f30b5c67fa5e0cf710480d53e0fd61ceaaa25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 21:04:40.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/140bb631cd954d6dae2678c73ab401cb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f577320ef3a74faa993aef1863e5b4e5 : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":14},{"id":"359721965b1b9e042b6173754be652c2ae9341af","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-05-26 13:30:44.000000000","message":"Patch Set 14:\n\n(7 comments)","accounts_in_message":[],"_revision_number":14},{"id":"64253b5494337b0b9594b528cef5a27b90d8a5c2","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-27 14:31:53.000000000","message":"Patch Set 14: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"ed4efb621e61a18a58a039e095ef87777d122dc1","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-05-27 14:58:10.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"09c38e484dc7a4b14625df8908f1d1c02136aaee","author":{"_account_id":38496,"name":"Andressa Cabistani","display_name":"Andressa","email":"acabistani@gmail.com","username":"andressadotpy","status":"I\u0027m a Software Engineer at Red Hat and I love Open Source and connect with people! Feel free to DM through IRC, I\u0027ll be delighted to chat"},"date":"2026-05-28 10:49:36.000000000","message":"Patch Set 14: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"fdae17d1d10e41e0867127b21cd8168842617d8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-07-28 23:19:10.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"9889960d3eb324897dc5949f931b44a739cb4dd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 00:11:43.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3af6549803146e9b1b38c3b94fbd0e5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d29084806dd4563832b728ac8052a47 : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":15},{"id":"aa2d0db86fdfc5e60ab691bd5588fc3fa915ec88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 00:27:24.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4425cba2a20b43e5887a1b57d5b62068\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f696230505ff4917b747e220218294e4 : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03bd7c082e7f40399c74505499c90843 : SUCCESS in 4m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/12c52e6c46d94fb2a4341cb28ffc4c26 : SUCCESS in 1h 03m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1bc3d1455c734e55bfcddef1d8c6f7f2 : SUCCESS in 28m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/842de5e468534040b7c44167ff81bdff : SUCCESS in 26m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cb10dd1ed55454d9a2aa94aab4a6c3e : SUCCESS in 59m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfce86fdc00b4134808fe31220d05e54 : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1d14c55b8014076be32d1a3b4ce2ef0 : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34177b256c204ce0bb60bf1d604de602 : SUCCESS in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/17c3cdc77b9c4eb698149559f7c657dd : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eaff884023df40c0986c6e108c3b0cea : SUCCESS in 24m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dbffdff6c66646a4a8d52ef5cfe340b8 : SUCCESS in 8m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0bd60091db25492e8c64d8109a06acc3 : SUCCESS in 13m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/115043a3e8cd4f668034992d59a51691 : SUCCESS in 21m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cfd7593366d345b09ccd3ecd53b95fa8 : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bc2fc2e04c4b4e3fbfbeabb0a3fc5cb1 : SUCCESS in 15m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7292572bef6d4ec99a44dab1dc2be3e7 : SUCCESS in 16m 17s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7c7b7bcfce649c59f95d8df6f6958f5 : SUCCESS in 18m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/18d32d8d4bd244988a3151cc436f9f8d : SUCCESS in 3m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4916daa6ea004ea8a9eee724d1e27378 : FAILURE in 7m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee4308f10ec14b0ba160f8ceb0227f82 : SUCCESS in 5m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b6a8a236c1b4e1a90270580047e53b8 : SUCCESS in 58m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4001549541d54d7390d79a072fa5781c : SUCCESS in 50m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/84f9407e21aa468bb04b16a4ced6e603 : SUCCESS in 53m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d8fdb6d8c894828a0056fc7784d1d51 : SUCCESS in 24m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851c8d56f764432285211d41f8ad9865 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ffde3d520f94d3fb0dc75fc235d56c5 : SUCCESS in 18m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/666d862f31b04358b236d63f2abd505d : SUCCESS in 57m 13s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":15},{"id":"3580dfc3c8f8ed659f6ea77be1fe78e4c81e0a2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-07-30 02:54:01.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"d4595622d317c8bb7f125e87b70d95abb4d386b5","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-07-30 03:06:03.000000000","message":"Patch Set 16:\n\n(7 comments)","accounts_in_message":[],"_revision_number":16},{"id":"6693b06bbb50665aad4b98babbb75bac4e2ac75e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 04:16:24.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8584421aec614224986fe5767f162818\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/068e2b6097e04bcf864a78be1c6b6667 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3edf6e318d5941e0a293a3784c160097 : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/8e3e251f2f0f42f1985c306fd240cb4e : SUCCESS in 1h 04m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca099866c1da44dc90dcd134183bf8a7 : SUCCESS in 1h 00m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/013bb3f033f1484da1815107ba3b631f : SUCCESS in 52m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/530cfa90fe184f7692ed35a903031d54 : SUCCESS in 32m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db44edbaa770473194ecc15c367c88ca : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47b038b4f6c94b4e9353a50b89cedeed : SUCCESS in 11m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b26fd117538411c8910976ef30cb70d : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/338651a826f04972b374819feb4651c0 : SUCCESS in 9m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6f70b23d89e5452a860ba2c315ebca96 : SUCCESS in 25m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f4100ee3666d40bd83021986fe9f7e54 : SUCCESS in 18m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/56fd49edc36e4161b53372e95346b698 : SUCCESS in 24m 36s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/62ca4a4b93ee46dab8b2dd2bb156cde1 : SUCCESS in 19m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7f6dc881e352483e82dfcd1d72d9e701 : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b419e7f480ee4550879a2952b824fb0a : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dac9559e4a674066a95b6b8a9f50f45b : SUCCESS in 15m 23s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/081748bcd43e49db9eb1297e766aa841 : SUCCESS in 10m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a38aa16624fe4b6d878c326485a1cc2b : SUCCESS in 9m 26s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/220ffb68613c42e78998dcbbcffe8112 : FAILURE in 9m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/83c5198c3f21432fb071846162c705e1 : SUCCESS in 11m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdcdc6bcaac547298c578fac51fe6cb1 : SUCCESS in 1h 03m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/996d3483b023418abfbc0b87a149ffdb : SUCCESS in 52m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/af89fc3460984da48f9f3fd54f90dd39 : SUCCESS in 27m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f60e8bd3cc341fdacde517f2eefc3a4 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac27b0e4dd204d2cac99755322d1a89d : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9fb68fad311f416eb8c4920b09c87762 : SUCCESS in 25m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4928f722eff0456085caece9a230b6cc : SUCCESS in 52m 49s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":16},{"id":"c2d9ea6debfc5d98ba5603ec5cf329339d865fd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 06:10:06.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fd25977e91240958b1be5acab3b7d4c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c379b1f1710496c964475db3a8cb1d0 : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":16},{"id":"618da4e3afb8e0d2ca2aaf7df0bbddec292f03a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-07-30 20:36:13.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"ed4dc7a3b3ae6ae7118be9b80d0c9298893508e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 21:47:01.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bda812315c20473eb55eb3af08c716a0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/34eca965941e4ccb846c98a36da5180d : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":17},{"id":"49c2cad5fe8b05ea72e34d857085f38c97d5c62d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 21:49:59.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87c0c0e61b934a8fbe8d53215783b0d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb44273ddcf4b5f9cfe94867637abae : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4622e7b5a1d41c79ccd4f86e2dd94bc : SUCCESS in 5m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/8ffd7b9e4c7940019ccf1becc18120cd : SUCCESS in 34m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/388b73af31a345d08eacbf062afde7e3 : SUCCESS in 1h 05m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3bf78073d754543bf349687544f0cbb : SUCCESS in 53m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8edbea2cc14b46068add0923dacff11d : SUCCESS in 59m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/698ae43749864a39ae889a85beb03d0a : SUCCESS in 14m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4fd9e8faa1da436cab349985350be3a2 : SUCCESS in 10m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de2ab924321e445a9d2fdeb81dde440f : SUCCESS in 19m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d021ef4bca404a0b874f5b15f4a34d91 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/33feb32fea824f48984eb45dec3ec63b : SUCCESS in 11m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f77d6e604de1495a9a6bfd38e9973e8a : SUCCESS in 12m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/eea1ea4580c04a4ca1620c961a60b6ed : SUCCESS in 29m 57s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0ed538b9829e4a7a8cd73c79b9a23aa4 : SUCCESS in 17m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/88e49e31e0934979aed14783de80a376 : SUCCESS in 8m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c6abf251fd9643d39f551a6a9aec89f8 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/10dff96752944b26b1f757810b1af531 : SUCCESS in 15m 31s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/8de593c63461425c8b325069ab987a36 : SUCCESS in 18m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7da1cf7f098a4a508772200b710efa9b : SUCCESS in 4m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/99ea4751c9204431b688a454c75f59da : FAILURE in 7m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6eaf8fe08d514fb5b2968db261a62278 : SUCCESS in 10m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dbc6404db374eeb9a18229486faa167 : SUCCESS in 39m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7490427dc6d241f694d574c78c7c72ab : SUCCESS in 26m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b994b356b6940af8aac63ebae087a34 : SUCCESS in 24m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7120f4e586a4deb8014bbbdd2528f58 : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ff014769e542168808d79923660e20 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f5114cc90601425cb004a3377130484d : SUCCESS in 9m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8edcd75b27034a3b956ae3dd1edc055b : SUCCESS in 57m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":17},{"id":"724cd57f9e47cdb103cc91854987e442d8446ffa","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-07-31 04:32:33.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"9657da57dae52d14c0fb019ea7372a8a66e2a5cf","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-31 14:01:58.000000000","message":"Patch Set 17: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"52a24923ad291f473285732dde734bf86ec8b4ea","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-31 16:15:01.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e7f7644f6785058caac2e64fc3bbd04f036707e5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-08-03 00:53:07.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"d7e0a698e8d61cc7350009ecab40653ff72b7794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-03 21:17:58.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"d9178487d573e2affa092a411be145bd8b3b38cc","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-03 21:20:26.000000000","message":"Patch Set 18:\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"9908fe7e25b158fe9b2c20c13ae68a6851618c47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 22:06:31.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b54a9c82036b4ce8979bb77c9bde333c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95dee8112828422693ed77e72556e2e8 : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":18},{"id":"17e0895ed5d9e219d1b51435105774e797598d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 22:24:45.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f8a44a7ac74d69878209767700e270\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d88e9698cf194237a606a5c0b223574c : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac79174878c346ba8701b7505a96e01b : SUCCESS in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/2a7afb4c41c44c9d981f069a39073af9 : SUCCESS in 28m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44283c59a62d46689d66e6f5c17a5673 : SUCCESS in 58m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7de0bd165de0491eb91419462b08e0c8 : SUCCESS in 34m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f0ba0a0c2f54937a896447794ca1dea : SUCCESS in 1h 04m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5fa63522999d4b12b56f2d6ed7f667ea : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14e95378532c499886c48fc401ae01bb : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2010926c5d7f46ef9f9cbf3c1db49399 : SUCCESS in 23m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6033b1ea0d7c45a9b28f0c61670e6127 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3d736a927a4342cc956054c39b6f197e : SUCCESS in 21m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ed279c4cb6504073a1b05c82bfae47fb : SUCCESS in 21m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7f683b846d1947e39d5b29b01b02613f : SUCCESS in 33m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/40eddda88d984878b6d1ece6e4b79c31 : SUCCESS in 11m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ca8eb09ad2874c81b0a2f619ba0e7497 : SUCCESS in 13m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/257454950043444aaa93c9ab4740b20d : SUCCESS in 13m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/640fdfd1ea74400588a9e9727dbf6b8a : SUCCESS in 15m 21s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9e74f7a81d84d4ba483b77df76349fb : SUCCESS in 16m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/820f0258863b4b1aaeec7b405255bb57 : SUCCESS in 6m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b374494db19440ebb6e30e4fc136fcc4 : FAILURE in 8m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/06b335ae9c514fabbedc1e81221588cc : SUCCESS in 7m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5002cdcef75148fd9b558fae9ce22eb6 : SUCCESS in 1h 00m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9330ebaf0336472dacaba0eb4e7c7245 : SUCCESS in 56m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95ddd92809fe4c0fbe6a4900d4b7e81b : SUCCESS in 56m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e3b0c83df4b4db4946462ead466e3df : SUCCESS in 28m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b69db28bbcf648278052500309995a8d : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92dcfb14e2d44ba7852c15b4c427793b : SUCCESS in 25m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8aaaad64bb024d369641fa127b20ae48 : SUCCESS in 53m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":18},{"id":"bec16ca2a486b8633a9dd03b7c2577afe9a123f9","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-04 13:33:30.000000000","message":"Patch Set 18: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"c852038fac23e431090e15a093ea8309261959e7","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-04 13:34:33.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"bb797a6fcf602a4f50c9c9c19d4d5bbbc4f093b5","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-05 15:59:14.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"add9a4a00cc5e75aeff5ccd5b935d2065a2f2951","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-20 00:28:34.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"4fc4ddd9688afbebd550b0727b0ce463d4ce65c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 01:00:29.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b174b2ccb024470a4da3a0465da2492\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e1f29ae86754f989fbb29f209794f86 : NODE_FAILURE Node(set) request 5d0aa689a01c4329b391d0b1155978b5 failed in 0s","accounts_in_message":[],"_revision_number":19},{"id":"9a985e722c6d58027cd1096d437370249795d016","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-08-20 01:28:57.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"afab630edb668c24d57f26d9f42070b51ccc92fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 01:45:11.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/202165c50b4540acabd88c740532716d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/160fabf785ab46908bd475e89acd7fe1 : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad86eca306b843a39289eaf513fba584 : SUCCESS in 6m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/a101b032606e47708a86082484006dd1 : SUCCESS in 33m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1106aefebf30472587bd761b7655ea15 : SUCCESS in 1h 10m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c26fbd5349ad4095bc8be0fc8cf9dd9b : SUCCESS in 42m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e1f20fe0a094a1da21d44377c3f38bc : SUCCESS in 30m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66dabf47d4df4d12ab3c356f6ca4312d : SUCCESS in 10m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e35cae1dd5504703ab2f848010e79e0c : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6cb044fb72e4e69acf9dc2dc9920f0f : SUCCESS in 21m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d71d03e53583462183b07d4c856b3d8a : SUCCESS in 4m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f979ad7bb87c492b840ac922521d143b : SUCCESS in 25m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b3b1861375e945bb8d31099430fcb121 : SUCCESS in 21m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ba8bc13ec39e4e01b5525e790e2de565 : SUCCESS in 21m 44s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b1e8686b637b42da91c9c90fd550c159 : SUCCESS in 9m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f5ca0106035f4bd583b4d2656e2611cc : SUCCESS in 8m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/00e0616721a44882ae50fc69f0cb1c2f : SUCCESS in 14m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b07ce793d538462490c3b878cb71dc3b : SUCCESS in 18m 48s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e2f55def1d440f5803b97171e45f077 : SUCCESS in 19m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8aa16ac363074012bde3fe71146ca1bf : SUCCESS in 4m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/37da94f5ca8445fe965f819d1572dfd0 : FAILURE in 9m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/077cab2733904390ab41020b9d295e95 : SUCCESS in 8m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0516fcdf7e64197b58d83a901196264 : SUCCESS in 1h 02m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d3750c733fc64f6f974c678e9478fd0c : SUCCESS in 59m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6cd12b951a344c5fbc8b7ee98d69bb94 : SUCCESS in 26m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5505448395949c1bcc2843587c6c040 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5d4490e20654286b34c1fda025e1ecd : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d76f3b78861e497db40b90d1089fa4f5 : SUCCESS in 22m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb2bc4d0e90241fbbb38cbe0e38ecb24 : SUCCESS in 42m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":19},{"id":"b2f4036839baa3130a45633d499272218fa705d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 02:54:03.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ebd32d900ed473083c4c95c5eaac406\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d6e63675fb04241b843e7d2b304a3e9 : SUCCESS in 1h 22m 12s","accounts_in_message":[],"_revision_number":19},{"id":"df460dc5895694a97e9719d196493763e87ebab3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-08-21 05:26:05.000000000","message":"Patch Set 19: Code-Review+1\n\n(29 comments)","accounts_in_message":[],"_revision_number":19},{"id":"7379bad554495b9888fcecc65097429cda868949","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-02 20:32:47.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"ecc7d400bf74f622cf102cac08ea5ca1bb1c2e86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 21:44:04.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab54027b000546e6937c41231dfee6a1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e7310bcbe56849e9a164fa376c36b943 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":20},{"id":"46b9d981160816efa18f46f95dd6fe61567a416f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 01:01:57.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e6ab1d2d9594544820a2d171875d7ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6e42f22f4e4aeeb58817d64f4367d9 : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cb54b3caef2427c8e1b94eb6859bad3 : SUCCESS in 11m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/65e1539028384f75bd5febe1165f731c : SUCCESS in 34m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/edc4851d710a413b8c36159ae371df0c : SUCCESS in 1h 04m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0c65df375ba84ce5a5a4eb9fb2aaa8ce : SUCCESS in 33m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/445c5035e5d4453a821dbbf6877c7515 : SUCCESS in 41m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/faccaa3faa724223a11c85951568af66 : SUCCESS in 22m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b2640c5db5442129a7eae5b5a8f7d91 : SUCCESS in 21m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94426ea499234c2f9fc309177eec6f50 : SUCCESS in 12m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a3700390b45445e820f494db818b01e : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/33a5e1f4d7954911ad9f730f5581d649 : SUCCESS in 21m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4695ddfabeba4ade9b54d9252504d852 : SUCCESS in 22m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d755155a067c4504917cc79524db9700 : SUCCESS in 25m 06s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/49a39d96c7b34121b71c8d172dae441c : SUCCESS in 10m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/75beef3f75b64d639bdadfcaf553fc08 : SUCCESS in 9m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ca2d1953f4fd4d3fac061918984da9b3 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c738c0d927bf43ee958eeb987c8a36f6 : SUCCESS in 13m 59s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/62e39e7476934f8eace8483304ba18a9 : SUCCESS in 10m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/733c9a008db046acb78983ed685e4c8d : SUCCESS in 8m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b11016c435e4015bf9afd25d0f3d584 : FAILURE in 31m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e32b206d0bbe4c27947504cb922562a6 : SUCCESS in 5m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1829209d6274459a3f317b585cb5c69 : SUCCESS in 39m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bdd711e9e6c2482cb3319234a13e1347 : SUCCESS in 41m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0db805c991940f2bc7515cc7fec2f94 : SUCCESS in 34m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/550cef0658404a35ba1536b485c14ed4 : SUCCESS in 31m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa495403fef427bbf36aab12c690434 : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d781f4702db4a6996d18f05a191cf9a : SUCCESS in 10m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3ad1db23e1444a9b65b0ca9d1b980e3 : SUCCESS in 1h 03m 40s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":20},{"id":"3652eb946148d24b320a585e9c03b5b3ec96bd2f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-16 07:52:25.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"fac33d49e3fd9a4b92bd5d89f60720642a4e3203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-17 04:20:52.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"a45eb74b18e87c016b0e98364b8e607acfc9f03f","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-17 04:22:20.000000000","message":"Patch Set 21:\n\n(25 comments)","accounts_in_message":[],"_revision_number":21},{"id":"bbe568ac1f7d6e7ccde3af9989fcdef0683d0f5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 04:53:21.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a661207e7ecf454495e8f67cd4399ee6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f5bb6012eff447f4b8704b7023e66c7a : SUCCESS in 30m 34s","accounts_in_message":[],"_revision_number":21},{"id":"57c95b938d71026119409cb96d8e9b443bd60cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 05:27:38.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcab860b74664f209e0e15e1ce2e65be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99a1ce48e46d4d3280882d287da4c083 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/916d63a85f6b4ba5ac10519fa75c1117 : SUCCESS in 7m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/5d384b26078c4804b7eaf01025d49ef4 : SUCCESS in 1h 01m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7afc4d8d52a747dbbf3a067f48dfd3be : SUCCESS in 45m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd81c26ac6b14c499fef44833ff1a3ec : SUCCESS in 50m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79439ff861d445139ab94769d9faed3e : SUCCESS in 1h 02m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/078b1c2dbdbf4482bcf7765164861ada : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d180edc782a24414b3f480422a9fd205 : SUCCESS in 11m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71a203fab8a9420c8a510372a756c86d : SUCCESS in 23m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/844aba2f211646468bae04fe5a2071fc : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/20ddabe473dc499cb93aa48f551048bc : SUCCESS in 26m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/769327150cc64e64945144177b8a969f : SUCCESS in 11m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/53a68793c09a4120ace2003fef09c3f8 : SUCCESS in 33m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/03960f8dc8d94bf1a30e7473d3ddf992 : SUCCESS in 9m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/973756fda3794fc99e8122b12c2ad0b9 : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/df78b2b6b86242d1a85acfcfc287e36f : SUCCESS in 8m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/028ab9850f5544caa597313299ebe8bb : SUCCESS in 16m 36s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc2722c5d71f435cb176357c1a95e7d3 : SUCCESS in 10m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4385e0c7635146048e0ee4501c06a15d : SUCCESS in 4m 22s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cab9813739394f31b38116cacb2dc873 : SUCCESS in 6m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a4c8e6f6cf8402aaeff7c7b8b9d82f0 : SUCCESS in 41m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e8ace721bf8d4f3893b24a3f91a69f12 : SUCCESS in 56m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4abcf451db8043fd8cb5933fe4b4ad3b : SUCCESS in 55m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39cc038fa3d1453a843c4164ca4cc115 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d0f651e5e1a488b84f5df43ccfea4ed : SUCCESS in 3m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/646514bd11c04cdb843f6db142d08bb5 : SUCCESS in 11m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b7a0e1c7a1e4fc2a26a92ec57b7b854 : SUCCESS in 1h 00m 06s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":21},{"id":"c007458e377dc618d10b3d7e91f676d63cce0117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-17 16:35:48.000000000","message":"Uploaded patch set 22: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"caca452f0c9b9a2b2b80d74de6a66715896209af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 17:08:37.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b05ffbb956e47c9893d4110435694c3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b72e4089848944fabe139b4a1b968c6b : SUCCESS in 30m 37s","accounts_in_message":[],"_revision_number":22},{"id":"8ede7f4f4e01898d7f098aea943eeaeee4cd0e7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 17:50:43.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0430a79d936e4245933c848e360a4505\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da5762779264789823d73260cbfc077 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c40651af7da449f786e29064b5bd6a40 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/1b4ed7a85e3944d08857ede47b74d812 : SUCCESS in 1h 01m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a493e93743354c289dc13b7cc371e86a : SUCCESS in 58m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/735fffad395d4f99923a816396d1ce93 : SUCCESS in 55m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebdc35a35eac4fc7acafd64b312c75f6 : SUCCESS in 28m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73b8ed9745914cd3a31f59e6168df674 : SUCCESS in 22m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb0653a634a64123b16505c512f09ff8 : SUCCESS in 21m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9350332006e4a63ae8b461d047f8e26 : SUCCESS in 24m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0bf94e5f3c114c07b7c9ee3c77424466 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2bbf41a72c4146a0889dd3b48ddba1f2 : SUCCESS in 26m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/76cb9b587c66417086004896eea88685 : SUCCESS in 19m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/993b0a60b5c24a439dc227a9a52bcb5e : SUCCESS in 24m 42s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/63198a94c10840c69bc9b42b2f7cb346 : SUCCESS in 9m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a6b2e8a234494b2185bc1fbdf7fc7ac8 : SUCCESS in 13m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6854d2c9047b40119575319d9817a301 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1a2f08323312440ba62c91e474fbcaac : SUCCESS in 16m 06s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/5be21a00840c4f20b385c96b79014d50 : SUCCESS in 9m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ba8cceab6cc4046b7e3b03014f2ab58 : SUCCESS in 5m 37s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/276b78bfe95a41b7872cd9a2ef1c6383 : SUCCESS in 8m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/607c460c252947e6a08ef980e8dad260 : SUCCESS in 1h 06m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e2ef3c8abda40c9b051440f0af5cd50 : SUCCESS in 23m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf85f4e5c7d94092ad43faa1b07c653e : SUCCESS in 59m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/387937358b7746f09fde01531d373e0c : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812286363799493bafacedccd58ba551 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3335f3b658ba48b0aee29356134f259b : SUCCESS in 18m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16ed165cf284464392528ed94d2bb4e2 : SUCCESS in 57m 32s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":22},{"id":"f068bd3aea842392c408649213ccebb3108d5007","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-09-17 23:40:56.000000000","message":"Patch Set 22: Code-Review+2\n\n(25 comments)","accounts_in_message":[],"_revision_number":22},{"id":"4e2d92be4059dd1041af346c4f76b3b5b4418da7","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-18 16:38:13.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"4833098d7424bbec71110bd8ffc9b16e57ba3dae","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-18 16:43:21.000000000","message":"Patch Set 22: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"fda23103790f618669a11cb0323608f823180f55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-18 17:01:14.000000000","message":"Patch Set 23: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"9176e87bf55969fc685b55c86cd6b30196328221","author":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"date":"2026-09-18 17:01:26.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"fd7f58cb5721d3b934eebd7007669ac2ed26b8ee","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-09-18 17:03:09.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"acc37e629967b2f1f3f666b65a596af306d86bdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:50:44.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8be4c57b76c645e0ad334764182051b8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4ea2c39e716e4ca2b86ee9e4f27c8467 : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":23},{"id":"c95bec0dcab9b825284f3f0c6014119328262000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 18:21:29.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f133d1f107f14ef083408636fc5337d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee2586a338ad45c0a783769363bd5ec3 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfcc7d34dc214a669453e6abe17c587a : SUCCESS in 5m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/cc8540f1d95449caa7ed88d2c0c9fb40 : SUCCESS in 1h 12m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/792967503d144750855db00267751b9b : SUCCESS in 1h 13m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5179060ffc624deeaaa7acfeff40d744 : SUCCESS in 28m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33cb96d095ab45558a6261580cce1329 : SUCCESS in 29m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef202fe00dfa41fea1c60aaf40a21b1f : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd340138bd944e19b2c745271ed90c8a : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cbc72e5d56d490595373b29f19adfea : SUCCESS in 22m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/96983c22bbaf4a32a52935a735f9cf83 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/912c5e5fab6b4497814e8800ff4c660f : SUCCESS in 11m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/da3b3cb4763f40a5b9f5bb72e2545c9d : SUCCESS in 15m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/abc626ac9f3440ffb20e0f1fcd9a6b73 : SUCCESS in 24m 28s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/335da2c48f934f128dacecf82f7ba229 : SUCCESS in 20m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1f945f3a88d841fe9bf61501e10f92df : SUCCESS in 9m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4281d68b4b6e4e89bc019f744b81df2b : SUCCESS in 9m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d65348a4fa434277be589b31a5d66d90 : SUCCESS in 16m 24s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bdc7dbcac5344e2b2b29a891fd8165b : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c04366ea7f5452c8bd1bd49f46c95f6 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b69e0a0c4c3740b6900dc86e3a001ac8 : SUCCESS in 8m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c05d1814d578492bb35263f24f5fbaf5 : SUCCESS in 1h 08m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c508092db109427ca5973b0dddafffb7 : SUCCESS in 57m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e7d5873df7b497f9f417678dabecd49 : SUCCESS in 24m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98a7887aa6034459b5b82cf775323acc : SUCCESS in 28m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c637512561424bd58a11e859e78c5558 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab427a09161a4af0a0fb73bafe86080f : SUCCESS in 18m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7383ee84be5443458f5fd89ab90d3c21 : SUCCESS in 49m 56s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":23},{"id":"28446f5780a5405159d2e06ee0efa116f10ba967","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 18:22:32.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"200ace1cec42dd0a5fd4491ea84080b8d6888a9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:28:16.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ef2b63096cc4710ae5436ce4f5c5387\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6a327b4ad04095bff48f5fa8710323 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/890fc5d8cf254ca79435ec657ed3c29a : SUCCESS in 4m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/b4499431caf54f2085f0684cd8ae8d04 : SUCCESS in 42m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/754c7069a73148c1ac99e226ac87733f : SUCCESS in 1h 00m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6990db7557e04bd1ba2d8fd548bb7e08 : SUCCESS in 34m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49de6f47819e44a59a858d2a45697a9b : SUCCESS in 55m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/da4cf1c7324344d7bbeadb2371826346 : POST_FAILURE in 7m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/56f2ad05a92745049bea46aeb606cb81 : SUCCESS in 10m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a28ea811eb654a518e9c5d6fa9aad9bb : SUCCESS in 17m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/777ef31b6dca48abb16633fe05148374 : SUCCESS in 22m 32s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1431abe6befb413188aa9fa60321a826 : SUCCESS in 20m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c38be0ca89aa44db9d51b5db3e3f6c8e : SUCCESS in 10m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/158912effcc5492dbefed398e900537c : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/433b28c07fdf422585eadb95094f4d28 : SUCCESS in 17m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d60c71c58b36404e96b8116bc9eed787 : SUCCESS in 3m 37s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/42ff5b9574f24e21b1d34a1cf4e8781d : SUCCESS in 5m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f68c6dc6ceec429ebeed5bf6622528ac : SUCCESS in 45m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d3eea4d47a4e425887ec1612a161d732 : SUCCESS in 57m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7821657dd7b243f6b43529dea87e88bb : SUCCESS in 54m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8676e852b454e929716425134e8030c : SUCCESS in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/382f7df3bae542d7b553c7ede3e20fe7 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/559516a579ec40b084e93f12ebcd1c63 : SUCCESS in 17m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d0f3fe2171545c3a3cc3616f4746ead : SUCCESS in 29m 48s","accounts_in_message":[],"_revision_number":23},{"id":"030ab96e9356e91518d47af016ad9e10728a9d5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:28:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":23},{"id":"639afa5c26f36c316b0c4b351dafb182f7066f17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:29:59.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cfeca1d1a29413586c475527eb73ed5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ce86747cc045b791665355b0878d73 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/86b3ce89522d4c028f91e5554364bb34 : FAILURE in 41s (non-voting)","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"7bf3f3d19905e1be2cb1c329c36c58408fc98a3e","revisions":{"275f4070b9eba2c8475094f35551c2c6e69e82f5":{"kind":"REWORK","_number":1,"created":"2025-11-12 21:43:05.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/80/966980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/1"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-10-28 16:25:44.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-11-12 21:42:57.000000000","tz":-360},"subject":"wip: quarantine the misplaced hashes","message":"wip: quarantine the misplaced hashes\n\nTODO: tests\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/275f4070b9eba2c8475094f35551c2c6e69e82f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/275f4070b9eba2c8475094f35551c2c6e69e82f5"}]},"branch":"refs/heads/master"},"714d59901c1d23841ffba1e936ca63779904bb15":{"kind":"REWORK","_number":2,"created":"2026-01-05 16:03:14.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/80/966980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/2"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-10-28 16:25:44.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-12-19 15:55:26.000000000","tz":-360},"subject":"wip: quarantine the misplaced hashes","message":"wip: quarantine the misplaced hashes\n\nTODO: tests\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714d59901c1d23841ffba1e936ca63779904bb15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714d59901c1d23841ffba1e936ca63779904bb15"}]},"branch":"refs/heads/master"},"ee0ebad32c3f361415018eb8f7a9ef55b3e36c55":{"kind":"REWORK","_number":3,"created":"2026-02-19 05:30:18.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/3"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-10-28 16:25:44.000000000","tz":-300},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-02-19 05:28:42.000000000","tz":-480},"subject":"wip: quarantine the misplaced hashes","message":"wip: quarantine the misplaced hashes\n\nTODO: tests\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee0ebad32c3f361415018eb8f7a9ef55b3e36c55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee0ebad32c3f361415018eb8f7a9ef55b3e36c55"}]},"branch":"refs/heads/master"},"e4200a77ef48f107d273c7c2f080f69e4a40fd51":{"kind":"REWORK","_number":4,"created":"2026-02-19 05:35:56.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/4"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-02-19 05:35:50.000000000","tz":-480},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-02-19 05:35:50.000000000","tz":-480},"subject":"wip: quarantine the misplaced hashes","message":"wip: quarantine the misplaced hashes\n\nTODO: tests\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4200a77ef48f107d273c7c2f080f69e4a40fd51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4200a77ef48f107d273c7c2f080f69e4a40fd51"}]},"branch":"refs/heads/master"},"7d9ce3104aa7db412914bd52db0f9430d6ca6ca9":{"kind":"REWORK","_number":5,"created":"2026-02-19 05:43:12.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/5"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-02-19 05:43:07.000000000","tz":-480},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-02-19 05:43:07.000000000","tz":-480},"subject":"wip: quarantine the misplaced hashes","message":"wip: quarantine the misplaced hashes\n\nTODO: tests\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d9ce3104aa7db412914bd52db0f9430d6ca6ca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d9ce3104aa7db412914bd52db0f9430d6ca6ca9"}]},"branch":"refs/heads/master"},"c00fe1c9a481c9ddccbc049219c0b54692b25129":{"kind":"REWORK","_number":6,"created":"2026-04-09 18:15:39.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/6"}}},"commit":{"parents":[{"commit":"694d25bb1a87f9426ac7ad1f3817c51f5bf13a34","subject":"Improved offset support in Timestamp.__invert__","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/694d25bb1a87f9426ac7ad1f3817c51f5bf13a34"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-08 03:05:08.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-09 18:15:31.000000000","tz":-420},"subject":"wip: quarantine the misplaced hashes","message":"wip: quarantine the misplaced hashes\n\nTODO: tests\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c00fe1c9a481c9ddccbc049219c0b54692b25129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c00fe1c9a481c9ddccbc049219c0b54692b25129"}]},"branch":"refs/heads/master"},"806866426dde8c77f72fd85e3ae546f944a6e4e3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-09 18:37:41.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/7"}}},"commit":{"parents":[{"commit":"9d684080daf73f2506655bf1a54c96b8685d7f93","subject":"Merge \"proxy: fix policy index in _get_from_shards response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d684080daf73f2506655bf1a54c96b8685d7f93"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-08 03:05:08.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-09 18:37:27.000000000","tz":-420},"subject":"wip: quarantine the misplaced hashes","message":"wip: quarantine the misplaced hashes\n\nTODO: tests\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/806866426dde8c77f72fd85e3ae546f944a6e4e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/806866426dde8c77f72fd85e3ae546f944a6e4e3"}]},"branch":"refs/heads/master"},"350721c5b30b86f3a40d02d72f2c73a36d2b7516":{"kind":"REWORK","_number":8,"created":"2026-04-15 20:02:54.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/8"}}},"commit":{"parents":[{"commit":"488f8bd8d5a5f86dd699cb4889ca031fde18864d","subject":"relinker: Improve hash filtering logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/488f8bd8d5a5f86dd699cb4889ca031fde18864d"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-15 19:42:45.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-15 20:02:39.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition\nit would have resided in before a series of PPIs. The length of this\nseries is hard-coded today to 1. Hash directories that are misplaced for\nother reasons are logged as warnings and left untouched.\n\nAdditional changes include:\n- Replace the STEP_RELINK/STEP_CLEANUP constants with an enum.\n- Ring-state precondition validation for each step.\n- Only perform the cleanup step on the lower-half of the partition\n  space.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/350721c5b30b86f3a40d02d72f2c73a36d2b7516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/350721c5b30b86f3a40d02d72f2c73a36d2b7516"}]},"branch":"refs/heads/master"},"dbba4ee49f39a9fa079d8b80d7aa94a041143f45":{"kind":"REWORK","_number":9,"created":"2026-04-15 20:09:27.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/9"}}},"commit":{"parents":[{"commit":"488f8bd8d5a5f86dd699cb4889ca031fde18864d","subject":"relinker: Improve hash filtering logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/488f8bd8d5a5f86dd699cb4889ca031fde18864d"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-15 20:09:19.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-15 20:09:19.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition\nit would have resided in before a series of PPIs. The length of this\nseries is hard-coded today to 1. Hash directories that are misplaced for\nother reasons are logged as warnings and left untouched.\n\nAdditional changes include:\n- Replace the STEP_RELINK/STEP_CLEANUP constants with an enum.\n- Ring-state precondition validation for each step.\n- Only perform the cleanup step on the lower-half of the partition\n  space.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbba4ee49f39a9fa079d8b80d7aa94a041143f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbba4ee49f39a9fa079d8b80d7aa94a041143f45"}]},"branch":"refs/heads/master"},"7b6770a647db6cf83f699f377dd01325e18d9b20":{"kind":"REWORK","_number":10,"created":"2026-04-15 23:13:47.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/10"}}},"commit":{"parents":[{"commit":"488f8bd8d5a5f86dd699cb4889ca031fde18864d","subject":"relinker: Improve hash filtering logic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/488f8bd8d5a5f86dd699cb4889ca031fde18864d"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-15 23:13:40.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-15 23:13:40.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition\nit would have resided in before a series of PPIs. The length of this\nseries is hard-coded today to 1. Hash directories that are misplaced for\nother reasons are logged as warnings and left untouched.\n\nAdditional changes include:\n- Replace the STEP_RELINK/STEP_CLEANUP constants with an enum.\n- Ring-state precondition validation for each step.\n- Only perform the cleanup step on the lower-half of the partition\n  space.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b6770a647db6cf83f699f377dd01325e18d9b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b6770a647db6cf83f699f377dd01325e18d9b20"}]},"branch":"refs/heads/master"},"c70fb46a3bbae662326db56719597f24ace1096b":{"kind":"REWORK","_number":11,"created":"2026-04-30 19:58:55.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/11"}}},"commit":{"parents":[{"commit":"453e2356ba92fdefbe9e0886028d8414a1063684","subject":"Merge \"Consolidate swift.common.concurrency import statments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/453e2356ba92fdefbe9e0886028d8414a1063684"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-15 23:13:40.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-04-30 19:31:52.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition\nit would have resided in before a series of PPIs. The length of this\nseries is hard-coded today to 1. Hash directories that are misplaced for\nother reasons are logged as warnings and left untouched.\n\nAdditional changes include:\n- Replace the STEP_RELINK/STEP_CLEANUP constants with an enum.\n- Ring-state precondition validation for each step.\n- Only perform the cleanup step on the lower-half of the partition\n  space.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c70fb46a3bbae662326db56719597f24ace1096b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c70fb46a3bbae662326db56719597f24ace1096b"}]},"branch":"refs/heads/master"},"473e66886d6199d6fe673ce34c7154e80f412b16":{"kind":"REWORK","_number":12,"created":"2026-05-01 02:49:16.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/12"}}},"commit":{"parents":[{"commit":"453e2356ba92fdefbe9e0886028d8414a1063684","subject":"Merge \"Consolidate swift.common.concurrency import statments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/453e2356ba92fdefbe9e0886028d8414a1063684"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-05-01 02:49:11.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-05-01 02:49:11.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition\nit would have resided in before a series of PPIs. The length of this\nseries is hard-coded today to 1. Hash directories that are misplaced for\nother reasons are logged as warnings and left untouched.\n\nAdditional changes include:\n- Replace the STEP_RELINK/STEP_CLEANUP constants with an enum.\n- Ring-state precondition validation for each step.\n- Only perform the cleanup step on the lower-half of the partition\n  space.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/473e66886d6199d6fe673ce34c7154e80f412b16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/473e66886d6199d6fe673ce34c7154e80f412b16"}]},"branch":"refs/heads/master"},"48de28ef95c952347c8f7a8f3f18a05221d6d23b":{"kind":"REWORK","_number":13,"created":"2026-05-12 17:08:00.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/13"}}},"commit":{"parents":[{"commit":"ae89a535aadad6d77c07c12ef4b58eb4470decdb","subject":"object-server: stop using normalize_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae89a535aadad6d77c07c12ef4b58eb4470decdb"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-05-12 17:07:39.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-05-12 17:07:39.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition it\nwould have resided in before a series of PPIs. The default length of\nthis series is 2 but is configurable and overridable via the CLI option:\n`--max-audit-history-quarantine-threshold`. Hash directories that are\nmisplaced for other reasons are logged as warnings and left untouched.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48de28ef95c952347c8f7a8f3f18a05221d6d23b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48de28ef95c952347c8f7a8f3f18a05221d6d23b"}]},"branch":"refs/heads/master"},"49ab5ded587fbab2f0f6ff93f218ca6e76d88cb4":{"kind":"REWORK","_number":14,"created":"2026-05-20 01:00:21.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/14"}}},"commit":{"parents":[{"commit":"ae89a535aadad6d77c07c12ef4b58eb4470decdb","subject":"object-server: stop using normalize_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae89a535aadad6d77c07c12ef4b58eb4470decdb"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-05-20 01:00:13.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-05-20 01:00:13.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition it\nwould have resided in before a series of PPIs. The default length of\nthis series is 2 but is configurable and overridable via the CLI option:\n`--max-audit-history-quarantine-threshold`. Hash directories that are\nmisplaced for other reasons are logged as warnings and left untouched.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ab5ded587fbab2f0f6ff93f218ca6e76d88cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ab5ded587fbab2f0f6ff93f218ca6e76d88cb4"}]},"branch":"refs/heads/master"},"73ec37a6c62687d1471b4fd6dc69db0846649a5e":{"kind":"REWORK","_number":15,"created":"2026-07-28 23:19:10.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/15"}}},"commit":{"parents":[{"commit":"05afa6c87f289bb63a76a0e1f882a2eebe66bd79","subject":"Merge \"s3api: drop native Swift control headers from client requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05afa6c87f289bb63a76a0e1f882a2eebe66bd79"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-05-20 01:00:13.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-07-28 21:26:39.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition it\nwould have resided in before a series of PPIs. The default length of\nthis series is 2 but is configurable and overridable via the CLI option:\n`--max-audit-history-quarantine-threshold`. Hash directories that are\nmisplaced for other reasons are logged as warnings and left untouched.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73ec37a6c62687d1471b4fd6dc69db0846649a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73ec37a6c62687d1471b4fd6dc69db0846649a5e"}]},"branch":"refs/heads/master"},"24f4bebe1637a75ea4ca94dcb5ea2748b650025d":{"kind":"REWORK","_number":16,"created":"2026-07-30 02:54:01.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/16"}}},"commit":{"parents":[{"commit":"fafa244a00afe49ed8abc6cb90fcd1bb883c4485","subject":"diskfile: Support direct quarantining of directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fafa244a00afe49ed8abc6cb90fcd1bb883c4485"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-05-20 01:00:13.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-07-29 23:25:54.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition it\nwould have resided in before a series of PPIs. The default length of\nthis series is 2 but is configurable and overridable via the CLI option:\n`--max-audit-history-quarantine-threshold`. Hash directories that are\nmisplaced for other reasons are logged as warnings and left untouched.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24f4bebe1637a75ea4ca94dcb5ea2748b650025d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24f4bebe1637a75ea4ca94dcb5ea2748b650025d"}]},"branch":"refs/heads/master"},"5b3a81f091b3416ac30f2c5f89c6047fb7a0de59":{"kind":"NO_CHANGE","_number":17,"created":"2026-07-30 20:36:13.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/17"}}},"commit":{"parents":[{"commit":"4288cfdc8972afb8498ab0c94171d8b6776b05ac","subject":"diskfile: Support direct quarantining of directories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4288cfdc8972afb8498ab0c94171d8b6776b05ac"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-05-20 01:00:13.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-07-30 20:36:13.000000000","tz":0},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition it\nwould have resided in before a series of PPIs. The default length of\nthis series is 2 but is configurable and overridable via the CLI option:\n`--max-audit-history-quarantine-threshold`. Hash directories that are\nmisplaced for other reasons are logged as warnings and left untouched.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b3a81f091b3416ac30f2c5f89c6047fb7a0de59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b3a81f091b3416ac30f2c5f89c6047fb7a0de59"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"fafa244a00afe49ed8abc6cb90fcd1bb883c4485","ours":"24f4bebe1637a75ea4ca94dcb5ea2748b650025d","theirs":"4288cfdc8972afb8498ab0c94171d8b6776b05ac","merge_strategy":"recursive","contains_conflicts":false}},"f6cbd18402755ff8c87966cf6011fe26f1df19e3":{"kind":"REWORK","_number":18,"created":"2026-08-03 21:17:58.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/18"}}},"commit":{"parents":[{"commit":"ee2b48494af890a79d4384c59bc6fa5b92ba0d27","subject":"Merge \"diskfile: Support direct quarantining of directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee2b48494af890a79d4384c59bc6fa5b92ba0d27"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-03 21:17:08.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-03 21:17:24.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition it\nwould have resided in before a series of PPIs. The default length of\nthis series is 2 but is configurable and overridable via the CLI option:\n`--max-audit-history-quarantine-threshold`. Hash directories that are\nmisplaced for other reasons are logged as warnings and left untouched.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6cbd18402755ff8c87966cf6011fe26f1df19e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6cbd18402755ff8c87966cf6011fe26f1df19e3"}]},"branch":"refs/heads/master"},"2a9d1644047c6ed06ef59e7bd689ecfc2f774209":{"kind":"REWORK","_number":19,"created":"2026-08-20 00:28:34.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/19"}}},"commit":{"parents":[{"commit":"ee2b48494af890a79d4384c59bc6fa5b92ba0d27","subject":"Merge \"diskfile: Support direct quarantining of directories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee2b48494af890a79d4384c59bc6fa5b92ba0d27"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-20 00:28:04.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-20 00:28:04.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition it\nwould have resided in before a series of PPIs. The default length of\nthis series is 2 but is configurable and overridable via the CLI option:\n`--max-audit-history-quarantine-threshold`. Hash directories that are\nmisplaced for other reasons are logged as warnings and left untouched.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a9d1644047c6ed06ef59e7bd689ecfc2f774209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a9d1644047c6ed06ef59e7bd689ecfc2f774209"}]},"branch":"refs/heads/master"},"ec4e5766e5add8402efb603ce87ca84ecdb170b2":{"kind":"REWORK","_number":20,"created":"2026-09-02 20:32:47.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/20"}}},"commit":{"parents":[{"commit":"4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766","subject":"s3api: add normalize_checksum_algorithm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ff57666f016d4054a46dc2dcaa3ca2ad7ed4766"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-08-20 00:28:04.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-02 20:26:39.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition it\nwould have resided in before a series of PPIs. The default length of\nthis series is 2 but is configurable and overridable via the CLI option:\n`--max-audit-history-quarantine-threshold`. Hash directories that are\nmisplaced for other reasons are logged as warnings and left untouched.\n\nRunning an audit against a replicated storage policy may have uintended\nconsequences: object-replicator may process a mis-homed hash ctory as a\nhandoff before audit gets to quarantining it. Add a guard rail by\nrefusing an unrestricted audit run and require selecting a storage\npolicy using the `--policy` flag. The relinker and cleanup actions\ncontinue to retain their all-policy default behavior.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec4e5766e5add8402efb603ce87ca84ecdb170b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec4e5766e5add8402efb603ce87ca84ecdb170b2"}]},"branch":"refs/heads/master"},"cd2baaad3ec79c760c5df1ad0cebd9ced7d4b59b":{"kind":"REWORK","_number":21,"created":"2026-09-17 04:20:52.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/21"}}},"commit":{"parents":[{"commit":"bd664103928dde998f7b79ec73c572153c8b1b52","subject":"replicator: Continue handoff cleanup after quarantine failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd664103928dde998f7b79ec73c572153c8b1b52"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-17 04:19:14.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-17 04:20:23.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nThe audit step quarantines hashes that were not removed during the\ncleanup phase of a part power increase (PPI) operation. In particular,\nit allows an operator to declare the PPI operation as \"complete\" so that\nthe replicator/reconstructor services can be re-enabled. A pre-condition\nfor the audit step is that no PPI operation is in progress.\n\nThe audit step will quarantine a hash directory if and only if it is\nfound in a PPI-ancestor of its expected partition, i.e. the partition it\nwould have resided in before a series of PPIs. The default length of\nthis series is 2 but is configurable and overridable via the CLI option:\n`--max-audit-history-quarantine-threshold`. Hash directories that are\nmisplaced for other reasons are logged as warnings and left untouched.\n\nRunning an audit against a replicated storage policy may have uintended\nconsequences: object-replicator may process a mis-homed hash ctory as a\nhandoff before audit gets to quarantining it. Add a guard rail by\nrefusing an unrestricted audit run and require selecting a storage\npolicy using the `--policy` flag. The relinker and cleanup actions\ncontinue to retain their all-policy default behavior.\n\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd2baaad3ec79c760c5df1ad0cebd9ced7d4b59b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd2baaad3ec79c760c5df1ad0cebd9ced7d4b59b"}]},"branch":"refs/heads/master"},"0e38b8901576d88c0099e6ef28082014b4e4dd38":{"kind":"NO_CODE_CHANGE","_number":22,"created":"2026-09-17 16:35:48.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/22"}}},"commit":{"parents":[{"commit":"bd664103928dde998f7b79ec73c572153c8b1b52","subject":"replicator: Continue handoff cleanup after quarantine failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd664103928dde998f7b79ec73c572153c8b1b52"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-17 04:19:14.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-17 16:34:04.000000000","tz":-420},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nIn exceptional circumstances, requirements outside the operator\u0027s\ncontrol may force a part power increase (PPI) to be marked finished\nbefore relinker cleanup completes. Once the ring no longer records the\nPPI, hash directories left in the old namespace look like data for a\ncurrent partition. Replication can preserve them as dark data, while EC\nreconstruction fails for lack of enough related fragments.\n\nAdd an audit action that calculates each hash\u0027s expected partition at\nthe current partition power. It quarantines a misplaced hash only when\nthe found partition is an ancestor of the expected partition within a\nbounded number of PPIs; unrelated misplacements are reported and left\nuntouched. The default ancestry limit is two and is configurable.\n\nQuarantine is deliberately conservative. These remnants may be older\nthan `reclaim_age` and the tombstones that once protected against\nresurrection. Relinking them into the current namespace could therefore\nmake deleted data durable again.\n\nA finished ring provides no hint about which policy was affected, so\naudit requires explicit `--policy` selection. Replicated audits also\nrequire service coordination to prevent peers from restoring quarantined\ndirectories; the operational procedure is documented in\n`ring_partpower.rst`.\n\nAssisted-By: Codex:gpt-5.6\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e38b8901576d88c0099e6ef28082014b4e4dd38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e38b8901576d88c0099e6ef28082014b4e4dd38"}]},"branch":"refs/heads/master"},"7bf3f3d19905e1be2cb1c329c36c58408fc98a3e":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2026-09-18 17:01:14.000000000","uploader":{"_account_id":38767,"name":"Wael Halbawi","display_name":"Wael Halbawi","email":"whalbawi@nvidia.com","username":"whalbawi"},"ref":"refs/changes/80/966980/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/966980/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/966980/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/966980/23"}}},"commit":{"parents":[{"commit":"bd664103928dde998f7b79ec73c572153c8b1b52","subject":"replicator: Continue handoff cleanup after quarantine failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd664103928dde998f7b79ec73c572153c8b1b52"}]}],"author":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-17 04:19:14.000000000","tz":-420},"committer":{"name":"Wael Halbawi","email":"whalbawi@nvidia.com","date":"2026-09-18 17:01:14.000000000","tz":0},"subject":"relinker: Add audit step to quarantine stale hash directories","message":"relinker: Add audit step to quarantine stale hash directories\n\nIn exceptional circumstances, requirements outside the operator\u0027s\ncontrol may force a part power increase (PPI) to be marked finished\nbefore relinker cleanup completes. Once the ring no longer records the\nPPI, hash directories left in the old namespace look like data for a\ncurrent partition. Replication can preserve them as dark data, while EC\nreconstruction fails for lack of enough related fragments.\n\nAdd an audit action that calculates each hash\u0027s expected partition at\nthe current partition power. It quarantines a misplaced hash only when\nthe found partition is an ancestor of the expected partition within a\nbounded number of PPIs; unrelated misplacements are reported and left\nuntouched. The default ancestry limit is two and is configurable.\n\nQuarantine is deliberately conservative. These remnants may be older\nthan `reclaim_age` and the tombstones that once protected against\nresurrection. Relinking them into the current namespace could therefore\nmake deleted data durable again.\n\nA finished ring provides no hint about which policy was affected, so\naudit requires explicit `--policy` selection. Replicated audits also\nrequire service coordination to prevent peers from restoring quarantined\ndirectories; the operational procedure is documented in\n`ring_partpower.rst`.\n\nAssisted-By: Codex:gpt-5.6\nCo-Authored-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6d1de8caa1322168e5e26682a586d11dc606e19a\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nSigned-off-by: Wael Halbawi \u003cwhalbawi@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bf3f3d19905e1be2cb1c329c36c58408fc98a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bf3f3d19905e1be2cb1c329c36c58408fc98a3e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
