)]}'
{"id":"openstack%2Fswift~967248","triplet_id":"openstack%2Fswift~feature%2Fmpu~Iad6d0687f35f4d3a2510d04712311fa38af077d0","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-hidden-rows","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-07-22 18:52:52.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-07-22 17:39:50.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Iad6d0687f35f4d3a2510d04712311fa38af077d0","subject":"mpu: store parts in hidden account","status":"MERGED","created":"2025-11-14 21:14:11.000000000","updated":"2026-07-22 18:54:56.000000000","submitted":"2026-07-22 18:52:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"967248-p-mpu-hidden-rows","meta_rev_id":"8dd441250f9d03f61529f3e39b2c8fdc11742567","_number":967248,"virtual_id_number":967248,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-22 18:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2026-07-22 17:39:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2026-07-22 17:39:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-14 22:11:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-21 09:17:38.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"a7b46550e6ce9f35956304cf910d04d76f5f8ec7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-14 21:14:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d775976bcb9c81c7ebdc1de64cf4dcbef492fed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 22:11:35.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c14776d817c54c208eda5dbdb6e5d952\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a06d83afccf749ebb84427b82d4f4f7d : SUCCESS in 6m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbc28bd6af7a47c09271a7f0319ea20f : FAILURE in 11m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c41684aec484563947a2236eae9dcd5 : FAILURE in 11m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cd2c079ee88425e8ac0fb42cbc45f94 : FAILURE in 21m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/159987e4f14f48b8a76faf6edf9dfe35 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e598e0f421ef434884d36ea9825b2eb7 : SUCCESS in 10m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d3e3019415542cdaf3d6e51ea1b6a0d : SUCCESS in 8m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/979c655933d648519631f3d12b93eabb : SUCCESS in 11m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3eb88b3be3044baabe4ac909513f2a7d : SUCCESS in 14m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3dcaafbee7614d22ac99cf305d261b58 : SUCCESS in 10m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/40470971671b487ea1f8aa5959396bd6 : SUCCESS in 10m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b70ef7e3760a4b239495e114a556a40f : SUCCESS in 13m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4143df8747074e4aa6839cc3d03fc72c : SUCCESS in 6m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b20ecffb3a914d9ba87126f7e6a18598 : FAILURE in 18m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/05b643701cc24924853bd0734003f3d5 : SUCCESS in 5m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3579d5f7485e4762b477074301345aaf : FAILURE in 51m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75e2d4e82e9947e08ca66acd907cf366 : SUCCESS in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f5608ccd75416cbe708d673bf55606 : FAILURE in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e43ba110a0d4e89bdf2cae29bc2968b : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c3178476764987bc9e6d14ae3545a95293208f5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-14 22:13:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9f299dc65ce4987847d912345b64862\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cf466750447648718dfd5034903065bf : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b4cbb9ac476ce5aeec2d2716401a3e64102ddf3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-21 11:30:36.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a3da69f73a2976f056467048a5ab3f71139981da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 12:43:56.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9eb55f29e724a04af5475486fa51a15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a26181da4ec49f78493fe91c38d3e06 : SUCCESS in 7m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/291a6dbacec244a181813d8108e66c41 : FAILURE in 21m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/529fdc507aa84a369f5088e5939e11ef : FAILURE in 27m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6280f1868bf440bb68eb96b8e840270 : FAILURE in 18m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e33e27032d244c2ab65015584746a2f4 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/049cee2e6fa342719d242373d2311261 : SUCCESS in 25m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ddc6c239bdcc4f5c968e895db9c55594 : SUCCESS in 12m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/edf2a159b49e49ebbf5ad7bf6fa3a581 : SUCCESS in 31m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4009d82fe27749fba35ac24950ea99c6 : SUCCESS in 18m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e24b04c6939c47c7970cb0068b6ef467 : SUCCESS in 9m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c1ce925ea60f41bf8f5671b77c74be92 : SUCCESS in 20m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/70a460dc63084f4d97592450dc1f3c5c : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/398f45bdef2c42e0a92ddc9803a972e6 : SUCCESS in 6m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/609aaa85908b48aab61c99ac7022dab7 : FAILURE in 13m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fadcdb20fcd843dba3b61cab120fe1b4 : SUCCESS in 6m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/53db88f06f214d30a1cffb8e9cfcd3f4 : FAILURE in 59m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccbc775a387a4b58a4db03f1468f378f : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf626345d4249e9bbf0063dbc70c21c : FAILURE in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc26e0869fe741ccad22dd8723928268 : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"48be4ceab20c5cdfdf8de51c3a3f9ec125213912","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 18:14:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69fc51a89e5140bd9af81b80292885b1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/05263cb128c043c38aa5b7833f81af11 : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f4c18f6c941c92d57fe81574652287ac5229b88d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-21 12:23:59.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"39856004e3cc4061fcad400b2b4c84615c354220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 13:39:06.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a830eea48cf43e78a0f900ae0325d1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd0f8c2aa9e94d3cbeaa9ffe7f35def8 : SUCCESS in 11m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a86283aa2b324df9a7fbab63328ebe9e : FAILURE in 11m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b591c6bdc81144c0bf042d8df43fa566 : FAILURE in 13m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4323fe5679e34c58b3a9a328ac913b64 : FAILURE in 23m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e583318f76440309dff71c44d10a161 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ae49acf4ee084b9fb894d21dad9943c7 : SUCCESS in 26m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6222d5b9e35646e881f412e27d262f09 : SUCCESS in 24m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/67ebbe37f5e74fd596b67e803e89e30e : SUCCESS in 29m 50s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/16f2cb471ced48d180d3ee3ad0756a51 : SUCCESS in 30m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/518632d2006246f88dfc2d66bccbb4dd : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4d6e2758472e4fedbaaf344c8a0788ef : SUCCESS in 9m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/542f4f9dde414030a9408add65675bd7 : SUCCESS in 17m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa94fe3daf1641f2b246e51d4f510cda : SUCCESS in 9m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/926fa3f45026462d8fc42e34b4a103e2 : FAILURE in 16m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac5654a2b6d74e739b7e2e09f30ff5d8 : FAILURE in 10m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ee2d4cb6e614edf8560e08062fa9d25 : FAILURE in 1h 06m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5caf5354fd8d4a84a8555a0e9b759ea0 : SUCCESS in 23m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bdcfd3153dd4cc7b418fad72efcbf8b : FAILURE in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/557772c96b76411db30351a9792e0ca1 : SUCCESS in 26m 43s","accounts_in_message":[],"_revision_number":3},{"id":"094d01e2cbb305469bbe880bcd6fce07cbe90017","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 13:58:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f204153787b044439f3eb5bb09172d2f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0ff614a752446ad9e19d903cc1b7429 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":3},{"id":"79a646a0c39a9c8b7e769777a2be134f1321d8d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-10 11:01:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"fc4e0b6494bd268214fe59e2a04d6a6699ab26b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 12:10:23.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/785c80ec79dd48f589ec7fcceadce618\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5e6b2d1926433d955dd71ff7923813 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e11c56013c0e43e8aad825cd8202803a : SUCCESS in 10m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5516bf4caf9d43d8be9127934427a2e0 : FAILURE in 22m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0390c5db6e24c798d16c7edabc0c0d5 : FAILURE in 23m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00a2a84c81d14e1ab1cd88596d088924 : FAILURE in 25m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd777a6ccc6c4cb6a6301d9792440ae5 : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/79cc54cac1bb41fd90ac46c065fbdbe4 : SUCCESS in 11m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/97066fa4e94246aa80ebbacb8c21ce6b : FAILURE in 3m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/11f4ca1c882646bebae18dbc2aa038d0 : FAILURE in 35m 04s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8fec475c953a4365ac8222e7a258f4e1 : FAILURE in 22m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/547f63ab1de84539af09922bd535ef94 : SUCCESS in 10m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e210b2c74dbb4a58bf988ee3839c7a1e : SUCCESS in 16m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fc3cfb681d0d4c32a800a1c889a83c76 : SUCCESS in 18m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/87ab4fce4f7f4b14a50c96144534974a : RETRY_LIMIT in 5m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7857ff293aaa4330884814ab8a2f5192 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a3dff48d23a41c6b51bf606aecab114 : SUCCESS in 9m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3aba2098a65476d9933ec80dcd66b53 : FAILURE in 1h 01m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf5c2af2f58640f1b702dc84cb7e1895 : SUCCESS in 29m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6f4d179ad14225b42b10edf29b4b7c : FAILURE in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab914a42e47c4d36b4c1c24fb61a7c9e : SUCCESS in 20m 11s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"e98abed2b7afea6c08a7c3bc3631723d1fee3ffc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 15:30:30.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c41fae3d39dd43068e87d566504ed687\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b82f83c997ca4578940330abca8bd3cb : FAILURE in 39m 09s","accounts_in_message":[],"_revision_number":4},{"id":"7aace201abb5319ae9345b38b44b080680d5958c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-15 11:10:01.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f5e55b9fcef0826e120b862631d9760e7c466b8e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-15 11:16:05.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d5f813efde798d682b9ceb78c850e9170ce2c0ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 12:12:38.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0257a43ea9224b51a167ed7cb1e3ac44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/196466de5db94dbbb45979814466e275 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1c3f0a81f6242b8b3c6f2760ccdfeeb : SUCCESS in 3m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0ce82942e2b43ee9d8db7b004645cb4 : FAILURE in 22m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67b8e1334d3d4ee8be8759cdee71ce52 : FAILURE in 22m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b62ce854d6ea409d94c411b363d8e30b : FAILURE in 23m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c7d9af75c294875961804d0a4bab8f8 : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dfff99b444ea458ab3921345d816b7ec : SUCCESS in 12m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c639df6cd6b846aa8d583c6019babaa8 : SUCCESS in 13m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cfa7b4b21196454f8dae0e4cc4f0c929 : SUCCESS in 33m 08s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/3bed72a6c87944bb8e7ab780fe8b299c : SUCCESS in 19m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b5dfe0a922304410919bbf6e3bb2fd57 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ac5186cc4d04445a8dc434e0dd4442c1 : SUCCESS in 17m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f2b3558bab054020a5553f9ac6ba339c : SUCCESS in 10m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c2cbfdf74c24e169e70022dfc7af12f : SUCCESS in 6m 09s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8cb274780fc24bf0a07dcfa3fd7760bc : FAILURE in 4m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e22f051162134d978d2e0e9f82f1e266 : SUCCESS in 10m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f46c123a5864dada14da2cf310c6661 : FAILURE in 42m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/128a9089c7764e33aa4d5c77718ac6ff : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0cc7ebd98c94b298368d76eb424f98b : FAILURE in 2m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7aee2e43c0024f42800ee13befda417e : SUCCESS in 20m 02s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"9f9737c7a88d8800eae656952b747a31f4d683c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 12:47:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15899081bb9d4db08e778c42b8fd5213\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/394fbf4c21dd412fa8b915c06c794d2c : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":5},{"id":"06178dab604d877659c67b71702380f0ebf5b48f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-07-21 09:17:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5e1939b058d0b358a6208b93d674547a435b5a6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-22 14:33:36.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"398c2b0192116f207e3237676426695887a38432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-22 15:17:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cabd45c14b557e39c34b5b01310669711eecf3c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 16:33:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc88da2a9a414ff4a42030542fc5639c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b37e4e0d52843fb8c941537f1ab85f3 : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37351d8c3abf4fa7b76bf07867551ac8 : SUCCESS in 10m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8302aa2dbbb0402eb1396f889ebf8bb4 : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4c6949c37e6495da58e81a2d7951694 : SUCCESS in 24m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5dc669b5f89049a4b9c9868bbc60d65e : SUCCESS in 23m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7630eb9d177241eb82e42afe0b536e88 : SUCCESS in 10m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/477c0464b0ee4efab947d2c5851d2d35 : SUCCESS in 31m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/35f46de4356942c59d949a6232d864b8 : SUCCESS in 19m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2d193c6a24ee479191ebe77ed04dfd45 : SUCCESS in 26m 45s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/58790892591649459dd69c1801d014e3 : SUCCESS in 23m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0d69793bdf49480bb6ef368358cb25c0 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a09f6a66fff74b7b9d48f42c758e7c48 : SUCCESS in 20m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b62994ddacb6439999425602fc45d237 : SUCCESS in 18m 56s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcaeccd452cc4551b0e5a6d88079a982 : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7d4457ae11fd4d16a2f2d33bb67e3778 : SUCCESS in 6m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/79bfd08ab68c4e6a8c51aa225597a101 : FAILURE in 7m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c51b182d31c14bf4a7d615df5ae32d01 : SUCCESS in 8m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0eacbe8f361d44a39e2a24f7c9d10c10 : SUCCESS in 1h 09m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fc41f1b8a7f408cb453af5c7cea4a93 : SUCCESS in 31m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2feeba73efc42eeb4c6f844e42c8b01 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f9963bbe40e4551a6f331e5181666d0 : SUCCESS in 19m 49s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"4d3c40fe681b95fcb318781821c5e566fd78f811","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 16:39:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c10d783943d648c0b0e37db534c95cd6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e5f71c5041bd4f3dbcf3d10cbe12eb89 : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":7},{"id":"fdea44d5d2dc7bf01db285a831814aaeff66275a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-22 17:39:50.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f200898171806d3e59d08eca26140e6c408e6eb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 17:40:07.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"771983a62876430d250b5d02aed69bde6231c27a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 18:52:52.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1bd2a0ecc754bbdb558e47655efce5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ec81ef971c48a99f0cfa2dbcb9ca1b : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/470f82b3984f46eca3123d4afdac4719 : SUCCESS in 8m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a196944b597948ac9d47ce9350e7d893 : SUCCESS in 7m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c894411a7f5b477c832253d1eabbc443 : SUCCESS in 27m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d9942d6b6a264795968d22ec35ce23e1 : SUCCESS in 23m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e8af991cd0dd447fbd8b78f6ad0ea63a : SUCCESS in 32m 58s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/4d0e3b69822e4c25bab37e7fbd994297 : SUCCESS in 23m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/74c4d6573c4b442fa41f1cc90ee14dd2 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8fa77d913b6d4463b902f5d2dd9697dd : SUCCESS in 18m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/29c24d07dde5497f9dd325aa0bb86a6d : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/67f28533d4384c5b8335d31a6197ec43 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/95c7e8f5873646f2912d1d4c0f3a12b1 : SUCCESS in 10m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/93489010d8a448c493465a622e94ce95 : SUCCESS in 1h 03m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f781bbd038a49d982df0b92c912c869 : SUCCESS in 31m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/738b5a9224d34340902a2ee0f9e6cf60 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dfb1723e0b664a83b2bf40f87b91b984 : SUCCESS in 21m 20s","accounts_in_message":[],"_revision_number":7},{"id":"cba0cfba79246f7377b7b1cd6a73208a55056597","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 18:52:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"8dd441250f9d03f61529f3e39b2c8fdc11742567","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-22 18:54:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed1bb822ad5453db573e5d4df9a56d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d62c0c3918c4ee29c5f9322b66fc5eb : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0463d9ca078540df83bd3dfd06475e50 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"40226ecd741e62d3656771f0ecc842c92f74445a","revisions":{"b0fba4a949a2ed4dcd8689dd88a45e17962ce7e4":{"kind":"REWORK","_number":1,"created":"2025-11-14 21:14:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/967248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/967248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/967248/1"}}},"commit":{"parents":[{"commit":"9663d238779568d9cc52ae0bc6bdb8001dc17a8f","subject":"mpu: s3 compat test for mpu head with range","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9663d238779568d9cc52ae0bc6bdb8001dc17a8f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-14 21:05:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-14 21:05:41.000000000","tz":0},"subject":"mpu: store parts in hidden account","message":"mpu: store parts in hidden account\n\nNow that mpu does not rely on SLO we can have the manifest reference\nparts in another hidden account. That avoids the parts container (and\nbytes) showing up in account stats.\n\nChange-Id: Iad6d0687f35f4d3a2510d04712311fa38af077d0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0fba4a949a2ed4dcd8689dd88a45e17962ce7e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0fba4a949a2ed4dcd8689dd88a45e17962ce7e4"}]},"branch":"refs/heads/feature/mpu"},"fe71ca60efa9fef37a6a1677167504bfd6a6799b":{"kind":"NO_CHANGE","_number":2,"created":"2025-11-21 11:30:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/967248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/967248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/967248/2"}}},"commit":{"parents":[{"commit":"e7708da66b413829dbccc750f3938b492444f34a","subject":"WIP: mpu: So long SLO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7708da66b413829dbccc750f3938b492444f34a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-14 21:05:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-18 12:31:59.000000000","tz":0},"subject":"mpu: store parts in hidden account","message":"mpu: store parts in hidden account\n\nNow that mpu does not rely on SLO we can have the manifest reference\nparts in another hidden account. That avoids the parts container (and\nbytes) showing up in account stats.\n\nChange-Id: Iad6d0687f35f4d3a2510d04712311fa38af077d0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe71ca60efa9fef37a6a1677167504bfd6a6799b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe71ca60efa9fef37a6a1677167504bfd6a6799b"}]},"branch":"refs/heads/feature/mpu"},"8457f6dbb6c3a3777d1d949b670fa98b7512b4c6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-21 12:23:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/967248/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/967248/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/967248/3"}}},"commit":{"parents":[{"commit":"d01241fd3cafeef63565d10edef62152a55b34ad","subject":"WIP: mpu: So long SLO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d01241fd3cafeef63565d10edef62152a55b34ad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-14 21:05:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-21 12:05:02.000000000","tz":60},"subject":"mpu: store parts in hidden account","message":"mpu: store parts in hidden account\n\nNow that mpu does not rely on SLO we can have the manifest reference\nparts in another hidden account. That avoids the parts container (and\nbytes) showing up in account stats.\n\nChange-Id: Iad6d0687f35f4d3a2510d04712311fa38af077d0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8457f6dbb6c3a3777d1d949b670fa98b7512b4c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8457f6dbb6c3a3777d1d949b670fa98b7512b4c6"}]},"branch":"refs/heads/feature/mpu"},"79f220a5529627666fc1ff5daa9b6d326ef53e22":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-10 11:01:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/967248/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/967248/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/967248/4"}}},"commit":{"parents":[{"commit":"ca235d08ecadcc5bb46123ea160b25e336ba7630","subject":" mpu: So long SLO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca235d08ecadcc5bb46123ea160b25e336ba7630"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-14 21:05:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 10:50:46.000000000","tz":60},"subject":"mpu: store parts in hidden account","message":"mpu: store parts in hidden account\n\nNow that mpu does not rely on SLO we can have the manifest reference\nparts in another hidden account. That avoids the parts container (and\nbytes) showing up in account stats.\n\nChange-Id: Iad6d0687f35f4d3a2510d04712311fa38af077d0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79f220a5529627666fc1ff5daa9b6d326ef53e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79f220a5529627666fc1ff5daa9b6d326ef53e22"}]},"branch":"refs/heads/feature/mpu"},"2c8a198ba6b02c2abd426d7e2377c1257a04cb29":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-07-15 11:10:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/967248/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/967248/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/967248/5"}}},"commit":{"parents":[{"commit":"d19a42532e68d96bbe38886f8c104d00f228b188","subject":" mpu: So long SLO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d19a42532e68d96bbe38886f8c104d00f228b188"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-14 21:05:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-15 10:34:54.000000000","tz":60},"subject":"mpu: store parts in hidden account","message":"mpu: store parts in hidden account\n\nNow that mpu does not rely on SLO we can have the manifest reference\nparts in another hidden account. That avoids the parts container (and\nbytes) showing up in account stats.\n\nChange-Id: Iad6d0687f35f4d3a2510d04712311fa38af077d0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c8a198ba6b02c2abd426d7e2377c1257a04cb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c8a198ba6b02c2abd426d7e2377c1257a04cb29"}]},"branch":"refs/heads/feature/mpu"},"97bcb509bcaaeb5e62acc1b79b5035396f3d9627":{"kind":"REWORK","_number":6,"created":"2026-07-22 14:33:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/967248/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/967248/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/967248/6"}}},"commit":{"parents":[{"commit":"5e5ab162ac02920d0930d3c3623555dc959a59d8","subject":"mpu: So long SLO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e5ab162ac02920d0930d3c3623555dc959a59d8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-14 21:05:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-22 14:29:51.000000000","tz":60},"subject":"mpu: store parts in hidden account","message":"mpu: store parts in hidden account\n\nNow that mpu does not rely on SLO we can have the manifest reference\nparts in another hidden account. That avoids the parts container (and\nbytes) showing up in account stats.\n\nChange-Id: Iad6d0687f35f4d3a2510d04712311fa38af077d0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97bcb509bcaaeb5e62acc1b79b5035396f3d9627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97bcb509bcaaeb5e62acc1b79b5035396f3d9627"}]},"branch":"refs/heads/feature/mpu"},"40226ecd741e62d3656771f0ecc842c92f74445a":{"kind":"REWORK","_number":7,"created":"2026-07-22 15:17:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/967248/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/967248/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/967248/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/967248/7"}}},"commit":{"parents":[{"commit":"5e5ab162ac02920d0930d3c3623555dc959a59d8","subject":"mpu: So long SLO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e5ab162ac02920d0930d3c3623555dc959a59d8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-14 21:05:41.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-22 14:58:13.000000000","tz":60},"subject":"mpu: store parts in hidden account","message":"mpu: store parts in hidden account\n\nNow that mpu does not rely on SLO we can have the manifest reference\nparts in another hidden account. That avoids the parts container (and\nbytes) showing up in account stats.\n\nChange-Id: Iad6d0687f35f4d3a2510d04712311fa38af077d0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40226ecd741e62d3656771f0ecc842c92f74445a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40226ecd741e62d3656771f0ecc842c92f74445a"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
