)]}'
{"id":"openstack%2Fswift~967396","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ic9655bacd79ed6a6ede0374a25d9c88469bf28e1","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-03-13 11:07:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic9655bacd79ed6a6ede0374a25d9c88469bf28e1","subject":"merge master to feature/mpu","status":"ABANDONED","created":"2025-11-17 16:12:49.000000000","updated":"2026-03-13 11:07:49.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0672f3bb9257ae04ac450fbcd269f3ab5def41a","_number":967396,"virtual_id_number":967396,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-17 17:22:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-17 16:57:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2025-11-17 17:38:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-17 16:56:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-17 16:57:58.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-11-17 17:22:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"454a5f02697d53b962f0c96e08eb2ddcf74b115b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-17 16:12:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e009f1c23513592cb225b76f02abd3537de6b90d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 16:56:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5db37c1bef2543b49d5bff8a46536a72\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8abb15f23d2345588aaa5ce962079605 : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":1},{"id":"b88d6dafa7490427e56ac6eb101e5947844c097d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-17 16:57:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c744027a499fa02e4191299a8f6ea1ba035d1ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 17:22:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65826efcad1a4f36bfed9c5718d72852\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da16ee911ee14a259f437cd3bef2adf3 : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3ee291ad5d0485fb2b41ecb700afef9 : SUCCESS in 6m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d18bfcfd1e54921a659fb4a186dcea5 : SUCCESS in 12m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b0a1e3578e34d38b2ee12af22c56e46 : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e77f69943ca743b998ddf9027250c189 : SUCCESS in 12m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54a04bf0321e47f69f3f806560f1ee48 : FAILURE in 4m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4bb79ec8f49e47ee990df955995c4a3f : SUCCESS in 25m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/301cb423faad4ca796be164426394085 : SUCCESS in 24m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6dbe97a3544242da925bc3e30f36172e : SUCCESS in 13m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d3ad547ed72b49b8a65758243de73e48 : SUCCESS in 21m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/57017e310b1f4d0c8da971b25e09efd0 : SUCCESS in 8m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/649646c2b2c54eb59173999b2465c94a : SUCCESS in 18m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/de224ee8a7fc419d90393ad0e1502335 : SUCCESS in 14m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d27b7d1de42437cb449d7c46b6dd8c2 : SUCCESS in 4m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/920e91ddd845403c8beee1d246fd0b9d : FAILURE in 14m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/81e6eeb27a9e48569bc3f911ed8cf2c5 : SUCCESS in 9m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd50919a2b7b4e27ab5540a641caee6d : FAILURE in 1h 04m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/037b7e1dc74441518de3c625b8899428 : SUCCESS in 35m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a2a80f7f1954b388d4ba8bd4caeb6ab : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee2029b4b54c4193bccd11ac8c1ccc90 : SUCCESS in 19m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ec9a4bf317c36ab3459bc81385457b56fe9987d5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-17 17:38:58.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0672f3bb9257ae04ac450fbcd269f3ab5def41a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-13 11:07:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cec5b69ab2d936da157a463928f3d7ca01bb21b0","revisions":{"cec5b69ab2d936da157a463928f3d7ca01bb21b0":{"kind":"REWORK","_number":1,"created":"2025-11-17 16:12:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/96/967396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/967396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/967396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/967396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/967396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/967396/1"}}},"commit":{"parents":[{"commit":"9e49d2026d0a375c3d4085df2fbaab6ff52775bb","subject":"mpu: use object event history to detect orphans","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e49d2026d0a375c3d4085df2fbaab6ff52775bb"}]},{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-17 16:12:11.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-17 16:12:31.000000000","tz":0},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: Ic9655bacd79ed6a6ede0374a25d9c88469bf28e1\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cec5b69ab2d936da157a463928f3d7ca01bb21b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cec5b69ab2d936da157a463928f3d7ca01bb21b0"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
