)]}'
{"id":"openstack%2Fswift~967400","triplet_id":"openstack%2Fswift~master~I45327cb5a56a87b8f6b9e971232fdce3b3f4bcb8","project":"openstack/swift","branch":"master","topic":"p-annotate-flakey-test","hashtags":[],"change_id":"I45327cb5a56a87b8f6b9e971232fdce3b3f4bcb8","subject":"trivial: tests: add failure context to flakey assertion","status":"MERGED","created":"2025-11-17 16:58:31.000000000","updated":"2025-11-18 12:13:14.000000000","submitted":"2025-11-18 12:12:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967400-p-annotate-flakey-test","meta_rev_id":"7d7af87352ca775a067853fdc46f65a8b63c2d60","_number":967400,"virtual_id_number":967400,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-18 12:12:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-11-18 11:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-11-18 11:07:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-17 17:55:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c191f270ef16851f784e05f47b4d3927a176f6cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-17 16:58:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43cc1bb06c7226a2eab42347ad7c7b04abfb9b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 17:55:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22a6d449445149419b2d66526d875eec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df7e543714fa4b75bf34e3586b1c95c9 : SUCCESS in 9m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e61be37e1d9141faad83fe28b6f273b4 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/480460fa4ee74ae8911f9fbb95c9c590 : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/555860cd9f5d4c7dade73e9b287cc652 : SUCCESS in 20m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/42b4416e20ec45ffa00bfb7885ec3261 : SUCCESS in 6m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3edd4175c6f4541ad525e0f7b07fcba : SUCCESS in 25m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d3ae13b3ba849669ba575051067df76 : SUCCESS in 19m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/43ab0ea6422241b0b1602919c563954f : SUCCESS in 30m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3faf73edd219494fae79e58b341a61d7 : SUCCESS in 28m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2f29876ae9cc4ce3a9a9afee1a122358 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/80ba213b9fb74a77b5b1dd4af7f0ed61 : SUCCESS in 14m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3c27c41567064baba1c62c2b76192c28 : SUCCESS in 15m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06a7831db6db4ce289873981e999f696 : SUCCESS in 52m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2aedcb2a0e564a9cb6a75b346c0eb17e : SUCCESS in 50m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1780d75111604102aa031ef807d68d80 : SUCCESS in 27m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a303f91a0df49d38c5345e6740063ae : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/353601d23b2044fa8d8e408f2421b643 : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"489d5189769be0b9bdf9755a12506218f6962e52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 17:55:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d17e54a0f3884cd08e41071631682267\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/88b92e39bfef4707a79381602c21bcb5 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":1},{"id":"75d582bf86e729ce7ddac7e2eee1df39e89aa8f1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-18 11:07:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04b42594fd59b36a33c7b1f22cc8085ae6899069","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 11:07:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5346bedc16f09abcf028516e77609ff7f4b83759","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 12:12:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6274ffdfa981404fbed5e95455d3057e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f0f3e9f29844628af46ec58d2397cdc : SUCCESS in 5m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d90e433b15ab4009965d76533a5a7544 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0d47af5d896e481c8eaef62907d4f602 : SUCCESS in 24m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a9346880964444e1a0e8fdca8864e0b3 : SUCCESS in 9m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5d878c3e2d0e4a9584fdf1b7c747c60d : SUCCESS in 13m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e5d9d85b41484f12b5b25d75f58f47a4 : SUCCESS in 30m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8744a1984a424f8b95d0d01b06e2f075 : SUCCESS in 9m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c053c1b505df4f87aa10bfb4ae5e2cfb : SUCCESS in 15m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aee0656a64a142b796f54e117dd8afe8 : SUCCESS in 13m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58857b45f3de43caab1470c108461c49 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b41c0ca289a645248191c9db01be7d5d : SUCCESS in 9m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d016e07ef774729900113deb3454018 : SUCCESS in 49m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4272fb36a09a44f787a92e6f7a1107bd : SUCCESS in 23m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78f620f3cdce4e10b25d0464f550d554 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50c54836fd304e85bfa359a7485afc59 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ad4363db9374a4a8d3c0d9dabd269bc : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9186e3e817f419eaece2f25e89cd423e7e50d2a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 12:12:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d7af87352ca775a067853fdc46f65a8b63c2d60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 12:13:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78c2cf98f7b441a2b267d1d55fccc498\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ace0c204f014765944951cae6842677 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4b2010bcb9b641f2933650aaf5ec1e0b : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"affdbf71aaee9e8f9b678592a523c68c8fcddb5a","revisions":{"affdbf71aaee9e8f9b678592a523c68c8fcddb5a":{"kind":"REWORK","_number":1,"created":"2025-11-17 16:58:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/00/967400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/967400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/967400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/967400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/967400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/967400/1"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-17 16:37:45.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-17 16:58:06.000000000","tz":0},"subject":"trivial: tests: add failure context to flakey assertion","message":"trivial: tests: add failure context to flakey assertion\n\nAn object server unit test was seen to fail returning a 409 when a 204\nwas expected. I cannot see the bug, so this patch adds some failure\ncontext to add illumination if/when it fails again.\n\nChange-Id: I45327cb5a56a87b8f6b9e971232fdce3b3f4bcb8\nRelated-Bug: #2131684\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/affdbf71aaee9e8f9b678592a523c68c8fcddb5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/affdbf71aaee9e8f9b678592a523c68c8fcddb5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
