)]}'
{"id":"openstack%2Fswift~967632","triplet_id":"openstack%2Fswift~master~I4c25bac68ecb18ca8860da67ac4c4bc46e14a04e","project":"openstack/swift","branch":"master","topic":"timestamps_nano","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-01-09 07:56:32.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4c25bac68ecb18ca8860da67ac4c4bc46e14a04e","subject":"WIP: Timestamp: Increase resolution to nanoseconds","status":"NEW","created":"2025-11-19 06:05:22.000000000","updated":"2026-02-09 09:57:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"7d1a4f77f250bd035f0dd320ff73762694acf263","_number":967632,"virtual_id_number":967632,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-09 09:45:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-19 07:07:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-19 13:16:23.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"07803f24cbc74ced1af8d40bc18b92d0c55b409a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-19 06:05:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"621fd101d56521b23f5daf302e981942c04585a6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-19 06:06:12.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"dad0c7e90347e9b855286e5b1b5819407c760944","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-19 06:07:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e202270c2815fd1cd22db40cf5a7f0b0dac0be7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-19 06:11:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed2a6f23e208722909d238d1204183f143d5c3da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 07:07:26.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14da7224cc37486dbe5fd40de5694066\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae27aa0b1ed64979a23b9e99562e2ede : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/5f6a0dd332e04bd685150e98544346ce : SUCCESS in 32m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9cc6cb5673ec43cdb4f309ebce51514e : SUCCESS in 56m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1feb5c4a7ab4cdc93e9d1257a286162 : SUCCESS in 53m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/349de74b786343c48f511a109d117dad : SUCCESS in 45m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad74d1286f9643be84f96c96454f0214 : SUCCESS in 10m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/673b734feaa2424aa29af0269fc1f70d : SUCCESS in 10m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5bf4b985d52748cc8f60f7a9e5375d88 : SUCCESS in 22m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a6c8b5cb069d45598f6479c425f41257 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3b83659d1915455b997cb8ea49eae63a : FAILURE in 12m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3fc50a3839bd4be1976b3aaa6f71d943 : FAILURE in 22m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3bde51aa8c834f6b8ace451ede8490d2 : FAILURE in 18m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cb5bdfe94fd547faa0cc450c605ac2bb : FAILURE in 32m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b74013a9ebaa49db9f7b02975983df61 : SUCCESS in 13m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e5003346e2214fc4be71e37d5b7bf79b : SUCCESS in 13m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d519353741114899ae0910982e58567f : SUCCESS in 11m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b559a3845547431789bcdde3a6b01fd6 : SUCCESS in 5m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6f48c7e798994a1a8d09a6eb7f914c46 : SUCCESS in 11m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/39bb1fb5d4a04c058ba53679a5931e33 : SUCCESS in 4m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c791def6458c4a508e29f49c55845352 : FAILURE in 56m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db7baca82a284f6f8ee5bbe30c574414 : SUCCESS in 25m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dcef3731d5b74880a988e4db543f219e : SUCCESS in 34m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49bc5cd4df9244919a1ecb0fe423b763 : FAILURE in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f9e5fcecaa458884495e47a53bd793 : FAILURE in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e62459490fca4938836a09110b6154a5 : SUCCESS in 19m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2fb73c897884cf59c6906e08a5f0ba6 : SUCCESS in 51m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7731407db6ef469db52652e6dc1d2a02 : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f0838dd20d70dfe439f744517375e5d44393cbb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 07:11:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/877e55de6ba34e469e419b3e7ffef05e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e61468c8de3a4ea0910eef674ba76a90 : FAILURE in 1h 03m 54s","accounts_in_message":[],"_revision_number":1},{"id":"61892e34a02e15db8a4594625ec2cbf90dea1f45","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-19 13:16:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5b49ea62c5a6787c843a9bd5ce68c1d1ee5eed1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-21 06:10:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"feb26bc6a8c8e4436c45090d7858662847382a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 07:24:17.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4c2f448aca54313bb305d56a42446fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7e6dc0821b948f59c3d109964c3cb99 : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/6bc760b71c9a4bf89def106e0b0480ca : SUCCESS in 42m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/40b5bb3461dd41959f5e485d5a8a5105 : SUCCESS in 1h 02m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9494a96384e4e0a8a9044a19f34e76f : SUCCESS in 1h 03m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1a12bcceea34ff985385bb90fa8bb4a : SUCCESS in 46m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a0b1c8ddb0146e3b2644591004c9592 : SUCCESS in 12m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cec563497684904a6339aadb97874e8 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7bfeb1541a44a77b26e3972e6c3a34c : SUCCESS in 17m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/90d0b31b01724b8a93c09ab0964fc9f1 : FAILURE in 4m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/138ba51fc4b443e5915d428a6e8076c2 : FAILURE in 11m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fe38918dfdbd4feab6a63da5abc79be7 : FAILURE in 25m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1bdc7a2192564114a2714bdc956b3c91 : FAILURE in 33m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f449215ca9f5460899a393977dffb03e : FAILURE in 13m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6d4c1e5b625348999cd0acc1738c7dfb : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/14bce4ada25b42c5ae0102b9dbc4e68a : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d77ac39394b1464ba30a099f0b5a8b87 : SUCCESS in 10m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7d4a3acc772143a0b02884ffeb6f6f42 : SUCCESS in 3m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2383d9af7ff540ac8ddf5222ca4049b8 : SUCCESS in 11m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b9f34a03328946f49deb6a40a6c2f931 : SUCCESS in 5m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bc70bd56ece4986bc6eb789d37c2d51 : FAILURE in 1h 07m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b368ad3cc8e643c383dfbbe73142c8f1 : SUCCESS in 49m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c042dd23c0841e896db8fa5fb8f2fdb : SUCCESS in 51m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d4b13900b81496dabe7eeeef3e6ca48 : FAILURE in 25m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac09e4ffc5d1429cac50f9007ecd5f23 : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5f68392fae14a48bff887b90368d74a : SUCCESS in 14m 12s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c429593893dd410ba91e1e060b2d1515 : SUCCESS in 42m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c28ae2db3d5489d9c666f313c779997 : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c72be5da303d5cbf5aa793046d245bb6515a6282","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 08:07:26.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d2c9aaf28084d4183e629cecbcb4dd3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/abb910185f8146ea805ce289d77a3fab : FAILURE in 1h 54m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f8e8bc0441f8abfd2aac782c63057b9d8f0bc785","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-21 10:11:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1262502cdca414601e6ddfb386fe3f3790bdf400","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-24 06:26:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"94562eb75bc04733a119295767b02c07fdf52503","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 07:05:22.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74961b2073c64c10a78cf09c21e8f330\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d55d810a6939402a8cc9ab9aa76e0a2f : FAILURE in 36m 39s","accounts_in_message":[],"_revision_number":3},{"id":"fe30d0e7d5de850c1893ebae02af87b329670b75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 07:34:18.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/220fcabd193a470ca035ba5dcd1b7e14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6234e161be7f426fbf6203b58b938f86 : SUCCESS in 5m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/2f4437f77b6d42bda96036e8a113cf55 : SUCCESS in 27m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ba44d8efd5e40db982fc511f8a16ebc : SUCCESS in 56m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a37c3d13974e4987a2fc67ec439ffb0c : SUCCESS in 32m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d65861e000d742868191466da9269bfd : SUCCESS in 1h 02m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a113aed404244b418057b7f01b1d9d59 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a81e07a33ba14952a18ec7ba6abe0768 : SUCCESS in 14m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07e99f8f28c14a2495ae1213c584e0b1 : SUCCESS in 21m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5f60278aee3489a889a8de00760b670 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4b8060dc9c5c490c9accd2d87f47576d : FAILURE in 18m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7b504c879b544ebe90d6b75cf0b9a23f : FAILURE in 21m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d66216f465bc4123b27aab62dacb5bea : FAILURE in 20m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d113da9660b847f995c3d5bc0a50d4b2 : FAILURE in 34m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fc396d70979a4679bb6e660eb00ec657 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5f8469528f114d2d927a242e16557508 : SUCCESS in 11m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b2545c93a49f48de81c4931e09bc0c2f : SUCCESS in 13m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe8255f11b5343979b9a83d43ebf1661 : SUCCESS in 7m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c3282af348414f5882b30fdabfc3081f : SUCCESS in 7m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/657c5b46a5a044f6ba4cb116fe1792b0 : SUCCESS in 5m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab3b09fc75bf4dd986ed6fdb3482d155 : FAILURE in 40m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/32da248d7b3b43e0a927bc0fa7bf70ee : SUCCESS in 31m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/31f5e03c91bb4a78855edfa2be41916d : SUCCESS in 47m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/187bcc308fc449a8b21a4d06300bb3a5 : FAILURE in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354d70896f4549e6ab765ead0ad628f2 : FAILURE in 2m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9189ec7658ef4e36a7ee87b0d1cdf58d : SUCCESS in 17m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dccdc75a58f64980af64cc972ed7f800 : SUCCESS in 55m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75f5b706676444588e513345107c705c : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":3},{"id":"4a94f7abf0fa1fff97b7cbd8a8ceeb9225562a6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-09 06:21:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9f4b4f41a7221546b564edb9e65ea846c3c9a262","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-09 06:22:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f58eb32aa86ad15855dc84bd5edbb8288b8480b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 07:21:41.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1e6ba3373b94675a06fae57eaa9d67a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1389f2c208d549b381d5bf04aeab493a : FAILURE in 37m 15s","accounts_in_message":[],"_revision_number":4},{"id":"631984c755a93ef96699c5db16ee43a55829df5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 07:56:32.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/391b3cd405a343e185c6be2a0ccaa1d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dccc5a93585846b4b3116cc39a6f2c75 : SUCCESS in 7m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/f68cee33c21240ceba712650336bbb18 : SUCCESS in 59m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec8076258a5944a1b21e37eeace9522a : SUCCESS in 43m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7faa62ee58c344f6bedec81b6e56eaf6 : RETRY_LIMIT in 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4c881707b9e437bb1f544424b2cb76e : RETRY_LIMIT in 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4ff40c2cedb4dcabc8dfed8c38f56f3 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a83e7ef240f24a58afe5865d00e0b741 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc4f4f34e015494aba4e579831354288 : SUCCESS in 16m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0a7c35515dea40569f77946f5a3dc63f : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/853f11c83411429e94755e15747a60e3 : FAILURE in 27m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/96d72d13f14c42bba5777bf1e7788844 : FAILURE in 25m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/411a86c248ca4df898956849f2bb795b : FAILURE in 22m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9063a7db2f634a51af80704da89787ea : FAILURE in 12m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/854bf70f635f474db62924d34b581e69 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c532bd348e2d4b3297d64f1c2d1d2733 : SUCCESS in 12m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/96e6e0d996ff42a8ae447235e87030f0 : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dfe7365a0b41488ca5fb81d5784aa284 : SUCCESS in 6m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/31084c0c74ca45e4a242e367d3a6e60c : SUCCESS in 15m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ea4db804cfc4bd4a5f1865b12089c4d : SUCCESS in 9m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed31dd9852e34172bd6c5a32001f27dd : FAILURE in 1h 11m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de63482388d24a9d87991da998c7680d : RETRY_LIMIT in 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d31eedf8e2a4ac8b1084ad4a9901d48 : RETRY_LIMIT in 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2a2d8dc25cb4b61a213973542e79e81 : FAILURE in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc98ed64a594a14a0539fcd69d8c670 : FAILURE in 2m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/75c57b3c16e14610a2fe6556bdba2f65 : SUCCESS in 27m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e41a3118970d4918b6b6a4f6713c18b2 : RETRY_LIMIT in 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0bfa86c2242464cb7cd2c73b390aa99 : RETRY_LIMIT in 29s","accounts_in_message":[],"_revision_number":4},{"id":"01f4409b3e7e44d333d9ee1b88ffb3f0673a2352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-13 05:15:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e2a510cbb0ca5165e24a49826b6f31a80e997901","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 05:53:54.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77cdaf50e82b46a28ec9d8bcfe0eac46\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f65f8cdb0c14f23a1fbee580be8534e : FAILURE in 36m 15s","accounts_in_message":[],"_revision_number":5},{"id":"f406623fc64143ead1e49efddc865319f9042b44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 06:24:50.000000000","message":"Patch Set 5: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0906e6caa991422ebb3b0e39f3918171\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/965d4e324de84316855a988fddf85ca9 : SUCCESS in 7m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/5d52f16579b449278bfc192ddf4280d8 : SUCCESS in 44m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/30a39cfb3280435abeea350412735c00 : SUCCESS in 57m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f6eb5a01a1341e4b0fb3e0fde258325 : SUCCESS in 33m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a5e6420f4be442a845e9f3fb0967cab : SUCCESS in 44m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d54a83aa312443da5d37e6073aa8258 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/732eb32a84c64198ae03086bc5ada38c : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c31c9ce04b7049f6a8b0a41846d41034 : SUCCESS in 12m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1912e2c9e72e43f79f4d26eeee050bb2 : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/14c4258dbd214dce86a19eb89876035d : FAILURE in 25m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0354de7193df432e9f7d6cd3a10a7970 : FAILURE in 9m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fad30f94b75348d78e30b503fb533afd : FAILURE in 29m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/373562ef400049919160ff61ccc67a54 : FAILURE in 30m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/46c5a45bd20146a18cd18ea19c4d4dc6 : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6ff1bbbc592447feaa1209d174f50b61 : SUCCESS in 8m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/555ce78baf3848dab6e1b8a276aeac10 : SUCCESS in 15m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02ff9429dc56402ba9551d55a51b579f : SUCCESS in 4m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f66c261f70e94fc7b8bf45349409947f : SUCCESS in 15m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bd082789d32e4da4a40b51ea9a61146a : SUCCESS in 11m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b198b8deea64be0a87c8edf2034250e : SUCCESS in 49m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8071ae6ffb8c47c5a304ec49bcf3456f : SUCCESS in 23m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/175a2b8e35ae451a974a0c018010454d : SUCCESS in 49m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a6fcf77ff3c4848a9d42280e7a9e4af : FAILURE in 29m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/746f3d87952847669234310d62ef9247 : FAILURE in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf17bfec03864d5ba2e855a176cb957e : SUCCESS in 18m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c47559b3f4b4030ab8b213208215613 : SUCCESS in 36m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25609c06f7384ee9a45b2d7bf99b541c : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":5},{"id":"9f3bae297080fe802b6045e1a4ffc2d200c011fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-09 08:41:00.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"491fe41c232ebbbc7f107649bf8957eb643aae63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 09:45:10.000000000","message":"Patch Set 6: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/210382330a694a5499ae8e96cd68c07c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9105dabdd79e4fba9f0ddc956ad450d0 : SUCCESS in 5m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/1c44f3f551f34aa6b476867b33521121 : FAILURE in 5m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb9f97196a21498eb00bc1e69e97c4b8 : FAILURE in 7m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/037381b9801c429d985a65b1fac68a9c : SUCCESS in 39m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09d245dfea344566b4f80f7f5b91d298 : FAILURE in 58m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/045ab59eff174d28a60b6686baded2a8 : FAILURE in 14m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58983afa39534497b9de6cc73436a280 : FAILURE in 20m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/130de13f4d0843e6970b94c19518a106 : FAILURE in 20m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bdc364a02f01425f937afca3e90f7316 : SUCCESS in 8m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e4637a9bc83745b2a9a0f06fe2c52399 : FAILURE in 25m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f4bc27d4095f46e29327f448b9ade622 : FAILURE in 10m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8448a5fbb9de4fa2b99d20be39c2bd3e : FAILURE in 29m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/78c1818f5e8140aca8fa30beabd4557b : FAILURE in 31m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a74f1918a75d4430b2be0c08befca149 : SUCCESS in 8m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1f3d3b2d9a22411bb407f26b4d90af07 : SUCCESS in 7m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8c9a8be762474bbc9ef37663634b0df8 : SUCCESS in 17m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cfb3fc48f9b24dfca79b878c1880175b : SUCCESS in 4m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e709220342f4d90b43392fdb2329ce6 : SUCCESS in 7m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e10757ffe8d496b95aacf661e70a363 : SUCCESS in 9m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d3ce0032d4b46b7bf96d696c223e42e : SUCCESS in 47m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ddc69781487d42de94feee852838d0fd : SUCCESS in 35m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90fb06d0f28144b3921ba754041e7a8c : SUCCESS in 21m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e3a27b89c6848cf99599892bc0cb2c6 : FAILURE in 25m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fc60ba21faa4a0c953ae58746270ae2 : FAILURE in 2m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c46cf0326ad40538d9b9c93c6989417 : SUCCESS in 14m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42158541c32747f5a90e43b4469fe078 : FAILURE in 49m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/926cc4f6754b4dcb8dfccc105158d236 : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":6},{"id":"7d1a4f77f250bd035f0dd320ff73762694acf263","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 09:57:35.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c97cd180b57046d88ddd77c6d2978a48\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a14a42412c0d49d597a3fe7a99ebce0b : FAILURE in 1h 14m 17s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9e85311db10bdb69bb50a1e611d8a7e0212df6b2","revisions":{"326f6f24149b2302a8a95eee610e3428ea37acb5":{"kind":"REWORK","_number":1,"created":"2025-11-19 06:05:22.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/32/967632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/967632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/967632/1"}}},"commit":{"parents":[{"commit":"676e68fc0f563f8971cb97e0b862518fbb08d7c3","subject":"Merge \"trivial: tests: add failure context to flakey assertion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/676e68fc0f563f8971cb97e0b862518fbb08d7c3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-19 06:00:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-19 06:00:51.000000000","tz":660},"subject":"WIP: Timestamp: Increase resolution to nanoseconds","message":"WIP: Timestamp: Increase resolution to nanoseconds\n\nThe current timestamp resolution is to the deca-microsecond. But\n`time.time_ns()` exists from 3.7 onwards which gives us the ability to\nstore up to nanoseconds.\n\nThis patch changes the raw value and precision of the Timestamp class to\nnanoseconds 1e-9.\n\nBecause not all OS\u0027s support nanosecond resolution, macos, there is a\nmonotonic_ns hack (because maxos supports nanosecond to that) to fake\nit. This is all wrapped in `swift.common.utils.timestamp.time_ns()`.\n\nNOTE: WORK IN PROGRESS\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I4c25bac68ecb18ca8860da67ac4c4bc46e14a04e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/326f6f24149b2302a8a95eee610e3428ea37acb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/326f6f24149b2302a8a95eee610e3428ea37acb5"}]},"branch":"refs/heads/master"},"04ddf888f53170bb36f4f8efdbffbc403d751b85":{"kind":"REWORK","_number":2,"created":"2025-11-21 06:10:22.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/32/967632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/967632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/967632/2"}}},"commit":{"parents":[{"commit":"676e68fc0f563f8971cb97e0b862518fbb08d7c3","subject":"Merge \"trivial: tests: add failure context to flakey assertion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/676e68fc0f563f8971cb97e0b862518fbb08d7c3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-19 06:00:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-21 06:10:00.000000000","tz":660},"subject":"WIP: Timestamp: Increase resolution to nanoseconds","message":"WIP: Timestamp: Increase resolution to nanoseconds\n\nThe current timestamp resolution is to the deca-microsecond. But\n`time.time_ns()` exists from 3.7 onwards which gives us the ability to\nstore up to nanoseconds.\n\nThis patch changes the raw value and precision of the Timestamp class to\nnanoseconds 1e-9.\n\nBecause not all OS\u0027s support nanosecond resolution, macos, there is a\nmonotonic_ns hack (because maxos supports nanosecond to that) to fake\nit. This is all wrapped in `swift.common.utils.timestamp.time_ns()`.\n\nNOTE: WORK IN PROGRESS\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I4c25bac68ecb18ca8860da67ac4c4bc46e14a04e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04ddf888f53170bb36f4f8efdbffbc403d751b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04ddf888f53170bb36f4f8efdbffbc403d751b85"}]},"branch":"refs/heads/master"},"34e77b2de4b8da37c68e67e0a903061584ca48ba":{"kind":"REWORK","_number":3,"created":"2025-11-24 06:26:40.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/32/967632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/967632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/967632/3"}}},"commit":{"parents":[{"commit":"676e68fc0f563f8971cb97e0b862518fbb08d7c3","subject":"Merge \"trivial: tests: add failure context to flakey assertion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/676e68fc0f563f8971cb97e0b862518fbb08d7c3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-19 06:00:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-24 06:25:10.000000000","tz":660},"subject":"WIP: Timestamp: Increase resolution to nanoseconds","message":"WIP: Timestamp: Increase resolution to nanoseconds\n\nThe current timestamp resolution is to the deca-microsecond. But\n`time.time_ns()` exists from 3.7 onwards which gives us the ability to\nstore up to nanoseconds.\n\nThis patch changes the raw value and precision of the Timestamp class to\nnanoseconds 1e-9.\n\nBecause not all OS\u0027s support nanosecond resolution, macos, there is a\nmonotonic_ns hack (because maxos supports nanosecond to that) to fake\nit. This is all wrapped in `swift.common.utils.timestamp.time_ns()`.\n\nCurrent big issue with nano seconds is the encode timestamps, encoding 3\ndeltas that are now number of nano seconds are 4 more digits, which\ncould take up a bunch more offset space... space we might not have :(\n\nNOTE: WORK IN PROGRESS\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I4c25bac68ecb18ca8860da67ac4c4bc46e14a04e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34e77b2de4b8da37c68e67e0a903061584ca48ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34e77b2de4b8da37c68e67e0a903061584ca48ba"}]},"branch":"refs/heads/master"},"19faa17369894f1cc67d490fb4c35b3a903237ec":{"kind":"REWORK","_number":4,"created":"2026-01-09 06:21:37.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/32/967632/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/967632/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/967632/4"}}},"commit":{"parents":[{"commit":"9366678a9e5c5725b67927175d9f56ffbe6a2060","subject":"legacy versioning: use Timestamp to create marker name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9366678a9e5c5725b67927175d9f56ffbe6a2060"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-19 06:00:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-01-09 06:21:24.000000000","tz":660},"subject":"WIP: Timestamp: Increase resolution to nanoseconds","message":"WIP: Timestamp: Increase resolution to nanoseconds\n\nThe current timestamp resolution is to the deca-microsecond. But\n`time.time_ns()` exists from 3.7 onwards which gives us the ability to\nstore up to nanoseconds.\n\nThis patch changes the raw value and precision of the Timestamp class to\nnanoseconds 1e-9.\n\nBecause not all OS\u0027s support nanosecond resolution, macos, there is a\nmonotonic_ns hack (because maxos supports nanosecond to that) to fake\nit. This is all wrapped in `swift.common.utils.timestamp.time_ns()`.\n\nCurrent big issue with nano seconds is the encode timestamps, encoding 3\ndeltas that are now number of nano seconds are 4 more digits, which\ncould take up a bunch more offset space... space we might not have :(\n\nNOTE: WORK IN PROGRESS\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I4c25bac68ecb18ca8860da67ac4c4bc46e14a04e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19faa17369894f1cc67d490fb4c35b3a903237ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19faa17369894f1cc67d490fb4c35b3a903237ec"}]},"branch":"refs/heads/master"},"1aba0a203d94df46d305eccbb0f33da29a32677d":{"kind":"REWORK","_number":5,"created":"2026-01-13 05:15:32.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/32/967632/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/967632/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/967632/5"}}},"commit":{"parents":[{"commit":"9366678a9e5c5725b67927175d9f56ffbe6a2060","subject":"legacy versioning: use Timestamp to create marker name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9366678a9e5c5725b67927175d9f56ffbe6a2060"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-19 06:00:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-01-13 05:15:16.000000000","tz":660},"subject":"WIP: Timestamp: Increase resolution to nanoseconds","message":"WIP: Timestamp: Increase resolution to nanoseconds\n\nThe current timestamp resolution is to the deca-microsecond. But\n`time.time_ns()` exists from 3.7 onwards which gives us the ability to\nstore up to nanoseconds.\n\nThis patch changes the raw value and precision of the Timestamp class to\nnanoseconds 1e-9.\n\nBecause not all OS\u0027s support nanosecond resolution, macos, there is a\nmonotonic_ns hack (because maxos supports nanosecond to that) to fake\nit. This is all wrapped in `swift.common.utils.timestamp.time_ns()`.\n\nCurrent big issue with nano seconds is the encode timestamps, encoding 3\ndeltas that are now number of nano seconds are 4 more digits, which\ncould take up a bunch more offset space... space we might not have :(\n\nNOTE: WORK IN PROGRESS\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I4c25bac68ecb18ca8860da67ac4c4bc46e14a04e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1aba0a203d94df46d305eccbb0f33da29a32677d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1aba0a203d94df46d305eccbb0f33da29a32677d"}]},"branch":"refs/heads/master"},"9e85311db10bdb69bb50a1e611d8a7e0212df6b2":{"kind":"REWORK","_number":6,"created":"2026-02-09 08:41:00.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/32/967632/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/967632/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/967632/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/967632/6"}}},"commit":{"parents":[{"commit":"c7bee623658b464072783096b98adaddfb1005bc","subject":"Merge \"s3api: Support for object expiration time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7bee623658b464072783096b98adaddfb1005bc"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-19 06:00:51.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-09 08:40:48.000000000","tz":660},"subject":"WIP: Timestamp: Increase resolution to nanoseconds","message":"WIP: Timestamp: Increase resolution to nanoseconds\n\nThe current timestamp resolution is to the deca-microsecond. But\n`time.time_ns()` exists from 3.7 onwards which gives us the ability to\nstore up to nanoseconds.\n\nThis patch changes the raw value and precision of the Timestamp class to\nnanoseconds 1e-9.\n\nBecause not all OS\u0027s support nanosecond resolution, macos, there is a\nmonotonic_ns hack (because maxos supports nanosecond to that) to fake\nit. This is all wrapped in `swift.common.utils.timestamp.time_ns()`.\n\nCurrent big issue with nano seconds is the encode timestamps, encoding 3\ndeltas that are now number of nano seconds are 4 more digits, which\ncould take up a bunch more offset space... space we might not have :(\n\nNOTE: WORK IN PROGRESS\n\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\nChange-Id: I4c25bac68ecb18ca8860da67ac4c4bc46e14a04e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e85311db10bdb69bb50a1e611d8a7e0212df6b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e85311db10bdb69bb50a1e611d8a7e0212df6b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
