)]}'
{"id":"openstack%2Fswift~967978","triplet_id":"openstack%2Fswift~feature%2Fmpu~I049c97711261dd5999ea949062fb3e4c3631f869","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-hidden-rows","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-08-05 18:44:02.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-08-10 09:45:33.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"hashtags":[],"change_id":"I049c97711261dd5999ea949062fb3e4c3631f869","subject":"WIP mpu: track parts, manifest and mpu sizes in container db","status":"NEW","created":"2025-11-21 11:30:36.000000000","updated":"2026-08-10 09:45:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"836974bfd2534403736e949ed3edf1fe1f5e8703","_number":967978,"virtual_id_number":967978,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-05 18:44:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-21 12:40:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-21 09:33:45.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"04d163cbe32374433a873512467b177e1a6452b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-21 11:30:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a136ca1872c1405797fcdf94762df731291988b0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-21 11:34:41.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8b51f8f77c10872e01e1752eb243d2c33d83fd4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 12:40:52.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b885401cedc94227ad3a7f6d06665c1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19658a0c839f4452aa7f04ac2c3bfbf7 : SUCCESS in 7m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ce6bb58e0c54f6da17fada4c0e55d70 : FAILURE in 25m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bead110648e4a66b3e4009082a921af : FAILURE in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ba53701bf1b494f92aeed381bb108cd : FAILURE in 23m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49ef6aa21e734206a32a5da6d1c04c76 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/57de78665037491bb81d13ca53792001 : FAILURE in 27m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82ab331090e1400386b574cea0136b9e : FAILURE in 25m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/74d2a5990afc4ba689980e1730b0c29c : FAILURE in 32m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e69307d3081f4a2c9008982a948bfb19 : FAILURE in 26m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ed89c54e6f3c4c2eb3ca9df24971973f : FAILURE in 16m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4e2df50fb4db47a98a7c34144ec9442f : FAILURE in 16m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/68955155b33c4272a861a349a4566c33 : FAILURE in 19m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01e5fca2f6f14b5daccf4bdbafb7f9a2 : FAILURE in 6m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5eb2b0a3faed4922bfcb63cefc4e1787 : FAILURE in 15m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5d86b0da97241c19a32932eb0b2813f : FAILURE in 8m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c528b39ca9c4cbcbde5add7c6835988 : FAILURE in 1h 03m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94e7dad80ce145db8dfbc9d1ca70e7ab : FAILURE in 29m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee408a2c971444294a59dbdac827c89 : FAILURE in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62954b442f3c465e8500de043dce647f : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"05b289a269d6a7902250d8c680a08108f13e82b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 16:47:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c093e0dea66040bfa63241fb10d21ffc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/36ce8fccf6cb4925832bc65340de44e5 : FAILURE in 1h 25m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9153c0e60d4847acfa3105e1f7788b2419e50a35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-21 12:23:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"0b7d0ca67bb1ebaacc8740709079697062a43887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 13:50:03.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28e4b9bbf13d408190bd23975833d532\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db6ccd9b18f341678047ba4bc3424f43 : SUCCESS in 10m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd41ae50fc8d4a42a36d44a8ed1c7980 : FAILURE in 25m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92d78540e53e4d9b873633f7187e5465 : FAILURE in 24m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bb15e30ede84fec9fe668d3f14fd283 : FAILURE in 25m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ade5bf6405142cabfade480ea9dc04d : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6470705255014a8c9a1a0ee658eef4a3 : FAILURE in 28m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f656e40a7198435ca042e8b32c282f36 : FAILURE in 22m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2b5dcc9ee5bd4f2c9926b7076e92c05b : FAILURE in 33m 59s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/e42d2f01e2fa4a83a5b6a7a1a38af52f : FAILURE in 24m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/39c5d4be0b694e828afcb6dbde9d94fe : FAILURE in 18m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ccdc4c0797c84c67976027f97540774b : FAILURE in 18m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9a2cefb868a640da90446af9e4ca295d : FAILURE in 18m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3c2dc5acc4247f7bfd9399add2830da : FAILURE in 7m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6178d94ccb2743af8328eabd447b4bf3 : FAILURE in 18m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/41bb0724634345a8afcb3738a5267b10 : FAILURE in 11m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae1f2278f6294b2b892796972a78faf7 : FAILURE in 1h 06m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef6ed43db8ec49928e958092e73c62b5 : FAILURE in 34m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/619de64d47b3417f9af75548ed4e39f4 : FAILURE in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e98fecb9bc2447fa85a36d08279499e : SUCCESS in 18m 56s","accounts_in_message":[],"_revision_number":2},{"id":"ffb3aa4625256281ee74f203cbb6928930f405a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 14:00:54.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b35c9e7310404fd9a18ff13632ebf168\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42424adfb9e7426984e129d071ec0e49 : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":2},{"id":"66362c3bdec7c2a5d3428375cedef3a6ae934a53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-10 11:01:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"9fcd6feda1e61b43b006afb375a21d4ce7e14e60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 12:06:07.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f868482b58b049c4a7e10f2bc6741561\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10c92be682314abebe9caea0207e175c : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca233d4f501242bfbbf6e97b20f1ceaf : SUCCESS in 8m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf50f7a65b404b6e99296de8e08f4d00 : FAILURE in 18m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a26c5ba64e24831ba384d7089ae68dc : FAILURE in 26m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/964f28f4b8e749918ae858939769cff2 : FAILURE in 26m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b1d6c900a5c49fcac3e5bd5130f9336 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/286e2fcd4e3f4fe5b9f05f6c92d07978 : FAILURE in 29m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2cfcf946b6184e928fc62a3c1462bed1 : FAILURE in 3m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/de417fd978264611be38920b38c58287 : FAILURE in 38m 55s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/241c55c88b284866b7079770290db0ce : FAILURE in 19m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6be84136e61e4e3d98dcdddfe6be5616 : FAILURE in 11m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7ac3c1e947b942e2879527b9b472c268 : FAILURE in 17m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9d2832837e1b42a9aa602edf6ee3b191 : FAILURE in 19m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7f2d1e9864f454088f035ce34c83387 : RETRY_LIMIT in 7m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/89e950e68c9845babb3bde3d063e5029 : FAILURE in 22m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dc641e43e49c4e58b294ef0450c7d5c9 : FAILURE in 9m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b54bfa4549b408c81bc59ec2274512d : FAILURE in 53m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b75124c2176b4fb59ce63e553c783683 : FAILURE in 32m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96e594ae248d4630b6943fe800f4852d : FAILURE in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14ac6648694e47788c082b355b76c17f : SUCCESS in 17m 57s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"ebcdffb6b274e7daeab651458d9e9cfb48e72163","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 16:12:54.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ff011a761a6489fa71f8df864ea66a3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d194201f7aca46e9b1a24040f7dde431 : FAILURE in 1h 16m 40s","accounts_in_message":[],"_revision_number":3},{"id":"2eaedfc26893fb193bf2126cdf33d2eb7343209f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-07-15 11:10:01.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"4aff93acfac9efe607f94a0d68b0b8b510455918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 11:56:37.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85ad58ef1f5b41798dcb8e0477c006ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/814f771f6b8346d896d6ae69b16128a3 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f011ae3cc4f43caaaf43fe7b07163ba : SUCCESS in 7m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/582f6bf56008404196a877fc02f1bdd9 : FAILURE in 20m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f063c9cfdf8b481199d3c1502ffaf571 : FAILURE in 23m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5648fd8ce12f4a2d9b9a11516d3b17f1 : FAILURE in 23m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a979a63563124d4a9c6a7ed0a0939a7f : SUCCESS in 8m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/86129ace981d42c8804977158f090b5b : FAILURE in 30m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7e9fdbec6e7b4739b1309103a00ca001 : FAILURE in 24m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/421ff20f12d4498c87bdf386f1545fd2 : FAILURE in 35m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5874fddc817d4ca08d7aecc1d752b1d1 : FAILURE in 13m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7958844d756b47bb90ef3cf511d3b540 : FAILURE in 11m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8e9cedef41f545259d6b91cff11f5532 : FAILURE in 18m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0900b38cc45b4b4c91e739a51742c345 : FAILURE in 10m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/993e998424f0435cbeb88739b114223a : FAILURE in 3m 44s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1a8ccb7584e433ba40e81f5d00f0aa8 : FAILURE in 4m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0eab34161c74a4095b0279cdae72e1b : FAILURE in 14m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a25c9a75ecb34759acd2134d060d9b4b : FAILURE in 41m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf5b2e7acbcb47fe80a89b2624c517f4 : FAILURE in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/691c147b2a9f4d23bfb2436c187f32d4 : FAILURE in 3m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e02373e8e36449ef8514437598998577 : SUCCESS in 20m 37s\n\nWarning:\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .docker/config.json found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-build-image.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"61b41fd5dcc1e8622f492e62a6fb66add874b25b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 12:58:47.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3476389f7db47c5836ce45169acf2bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/102ae5c849d34523bc7e9f8ea7979c88 : FAILURE in 1h 20m 00s","accounts_in_message":[],"_revision_number":4},{"id":"756f74c2a6f481ad6f5b1582e3db2adc1c91c49e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-07-21 09:33:45.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2d936f32452544905d9142825969970a936e9a83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-05 17:32:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"4ad79843d70f7433664b4ec15d9e1e0ab59a7d22","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-08-05 17:34:35.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"03423551ab046bfeffa067109f7e5d34430ce184","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 18:06:28.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e5047fc42594461bcaa93ae50e699d1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dbd28ea7310046a38a6da3984420caad : FAILURE in 32m 15s","accounts_in_message":[],"_revision_number":5},{"id":"f7320b58953d31665ed6ce8adde8bc7861a660f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 18:44:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16ccbc8479334e839679c85434d4069a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d39ab29fb84e168115e6d1849b9339 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79db0a46c5704802ba5952df3ffcbe1d : SUCCESS in 9m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd86dbb5b53d4105a329cd08f165fe20 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3014781a0f8441cb0a76dcf5efb0337 : SUCCESS in 10m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdc043d0fb1f417ca9952470067fee34 : SUCCESS in 21m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56eedac688e04e7d94d75eea1b639f61 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/26b2920bb46d4ef2bef044b9bfda4099 : FAILURE in 14m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/53f169d51fab45589e4f66ecab3a87c7 : FAILURE in 23m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/053b1b72bfc14af6bef168f439a51691 : FAILURE in 33m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/68049b5297444791a45b097b63129d55 : FAILURE in 20m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/81c186e9bf794a4aa6a6f79ef9a568dc : SUCCESS in 12m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d0da1308608f4af491e63b526924f15e : SUCCESS in 17m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3e198dff14ff427e8ca95b8657c32a03 : SUCCESS in 10m 10s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fceab122f7f463baf007858de47442b : SUCCESS in 18m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f7eb344be8a4e07a374276edbc0bfdd : SUCCESS in 6m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/93e1c9bb652b450b90963e6f43902879 : FAILURE in 8m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0bb42f5bccc414b856d3d7b2fb28534 : SUCCESS in 9m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/93306f2490cc41928cba9a796834a7c0 : SUCCESS in 1h 06m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7394441fde8f47b4a4862bbaea8e2830 : FAILURE in 30m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1124a190459453a8f577282efa1428f : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8ead83e6f224af9b25592883c5d97af : SUCCESS in 19m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"836974bfd2534403736e949ed3edf1fe1f5e8703","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-08-10 09:45:33.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"22595ba352da05fc102d4b6494fda4642a50113d","revisions":{"6d7e656ee33167d3a2dde181c377bdaf2e6ec8e1":{"kind":"REWORK","_number":1,"created":"2025-11-21 11:30:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/967978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/967978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/967978/1"}}},"commit":{"parents":[{"commit":"fe71ca60efa9fef37a6a1677167504bfd6a6799b","subject":"mpu: store parts in hidden account","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe71ca60efa9fef37a6a1677167504bfd6a6799b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-18 15:48:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-19 10:24:41.000000000","tz":0},"subject":"mpu: add mpu_size column to object table","message":"mpu: add mpu_size column to object table\n\nTODO:\nalso add manifest bytes\nprobe test fails because account stats are wrong.\nupdate account with actual bytes inc parts?\n\nTrack mpu size separately from manifest size.\n\nChange-Id: I049c97711261dd5999ea949062fb3e4c3631f869\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d7e656ee33167d3a2dde181c377bdaf2e6ec8e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d7e656ee33167d3a2dde181c377bdaf2e6ec8e1"}]},"branch":"refs/heads/feature/mpu"},"f3ca6dd59908f3d8a352a47c0c03660942ad0a6d":{"kind":"REWORK","_number":2,"created":"2026-04-21 12:23:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/967978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/967978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/967978/2"}}},"commit":{"parents":[{"commit":"8457f6dbb6c3a3777d1d949b670fa98b7512b4c6","subject":"mpu: store parts in hidden account","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8457f6dbb6c3a3777d1d949b670fa98b7512b4c6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-18 15:48:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-21 12:10:02.000000000","tz":60},"subject":"mpu: add mpu_size column to object table","message":"mpu: add mpu_size column to object table\n\nTODO:\nalso add manifest bytes\nprobe test fails because account stats are wrong.\nupdate account with actual bytes inc parts?\n\nTrack mpu size separately from manifest size.\n\nChange-Id: I049c97711261dd5999ea949062fb3e4c3631f869\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3ca6dd59908f3d8a352a47c0c03660942ad0a6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3ca6dd59908f3d8a352a47c0c03660942ad0a6d"}]},"branch":"refs/heads/feature/mpu"},"19912bea1ebc379338d48279047add8e8bce5640":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-10 11:01:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/967978/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/967978/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/967978/3"}}},"commit":{"parents":[{"commit":"79f220a5529627666fc1ff5daa9b6d326ef53e22","subject":"mpu: store parts in hidden account","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79f220a5529627666fc1ff5daa9b6d326ef53e22"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-18 15:48:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-10 10:57:03.000000000","tz":60},"subject":"mpu: add mpu_size column to object table","message":"mpu: add mpu_size column to object table\n\nTODO:\nalso add manifest bytes\nprobe test fails because account stats are wrong.\nupdate account with actual bytes inc parts?\n\nTrack mpu size separately from manifest size.\n\nChange-Id: I049c97711261dd5999ea949062fb3e4c3631f869\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19912bea1ebc379338d48279047add8e8bce5640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19912bea1ebc379338d48279047add8e8bce5640"}]},"branch":"refs/heads/feature/mpu"},"5dc2869626acf7f0423bf674e6648996053de543":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-15 11:10:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/967978/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/967978/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/967978/4"}}},"commit":{"parents":[{"commit":"2c8a198ba6b02c2abd426d7e2377c1257a04cb29","subject":"mpu: store parts in hidden account","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c8a198ba6b02c2abd426d7e2377c1257a04cb29"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-18 15:48:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-07-15 10:34:54.000000000","tz":60},"subject":"mpu: add mpu_size column to object table","message":"mpu: add mpu_size column to object table\n\nTODO:\nalso add manifest bytes\nprobe test fails because account stats are wrong.\nupdate account with actual bytes inc parts?\n\nTrack mpu size separately from manifest size.\n\nChange-Id: I049c97711261dd5999ea949062fb3e4c3631f869\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5dc2869626acf7f0423bf674e6648996053de543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5dc2869626acf7f0423bf674e6648996053de543"}]},"branch":"refs/heads/feature/mpu"},"22595ba352da05fc102d4b6494fda4642a50113d":{"kind":"REWORK","_number":5,"created":"2026-08-05 17:32:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/967978/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/967978/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/967978/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/967978/5"}}},"commit":{"parents":[{"commit":"6a408cc220fbf2d01a75b2a07f3dd6e57b99ecd2","subject":"mpu: improve ContainerBroker listing interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a408cc220fbf2d01a75b2a07f3dd6e57b99ecd2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-18 15:48:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-08-05 17:28:12.000000000","tz":60},"subject":"WIP mpu: track parts, manifest and mpu sizes in container db","message":"WIP mpu: track parts, manifest and mpu sizes in container db\n\nMPU middleware adds headers that are routed to container server with\n\nX-Size: the sum of the size of the mpu parts; this ends up in the\nexisting bytes_used column of the object table.\n\nX-Manifest-Size: the size of the manifest; this ends up in a new\nmanifest_size column of the object table.\n\nThe object insert/delete triggers are updated to maintain the sum of\nbytes_in_manifests and the sum of bytes_in_parts in the policy stat\ntable. This enables calculation of\n\nFailing test in container/test_backend.py likely due to this TODO...\n\nFAILED test/unit/container/test_backend.py::\n    TestContainerBrokerBeforeSystags::\n    test_merge_items_with_manifest_size - AssertionError: 12358 !\u003d 0\n\nTODO: (claude) The migration only adds the manifest_size column to\nobject.  On a migrated DB, policy_stat still lacks\nbytes_in_parts/bytes_in_manifests, the old triggers (compiled at DB\ninit time) never update those columns, and the old container_stat VIEW\ndoesn\u0027t expose them — so get_info() always returns 0 for both fields\nvia the fallback path. The fix needs to also ALTER TABLE policy_stat,\ndrop and recreate both triggers via POLICY_STAT_TRIGGER_SCRIPT, and\ndrop and recreate the VIEW via CONTAINER_STAT_VIEW_SCRIPT\n\nChange-Id: I049c97711261dd5999ea949062fb3e4c3631f869\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22595ba352da05fc102d4b6494fda4642a50113d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22595ba352da05fc102d4b6494fda4642a50113d"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
