)]}'
{"id":"openstack%2Fswift~968538","triplet_id":"openstack%2Fswift~master~I629b49d103a4392833dbdb14224eb20403b09a11","project":"openstack/swift","branch":"master","topic":"p-timestamps-tests","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-18 17:02:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-12-17 03:50:01.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-12-17 04:40:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I629b49d103a4392833dbdb14224eb20403b09a11","subject":"container/test_server: cleanup Timestamp usage","status":"MERGED","created":"2025-11-26 18:24:08.000000000","updated":"2025-12-18 17:02:09.000000000","submitted":"2025-12-17 04:40:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"968538-p-timestamps-tests","meta_rev_id":"5937b47a7ca768f2d097f1c12e686bbdb2d3fcbe","_number":968538,"virtual_id_number":968538,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-17 04:40:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":2,"date":"2025-12-17 03:50:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-12-17 03:50:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 18:57:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-26 19:37:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-15 20:25:22.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-12-16 06:26:20.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"a686145d8ddb450bf0dd556b6436a0d1caa2ce56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-26 18:24:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b562d0966f28555f56150658405010fe9856c67a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 18:57:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7014a1fe4c9d47aa9a9c7f9ecb273cad\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e887754a8dc47dfa2fefa8848ed2caa : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a598272f44cb85b00614aefc2bdb85f6960aa633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 19:37:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43f533bd83404eadaa995dcaf0374dbb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/956dcb74b95942da8e3cb50d67a4d957 : SUCCESS in 5m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/8f960030e2d24fa487c333b61792f3c0 : SUCCESS in 54m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e1600aac45e744c5a5de8f7b5d56b50b : SUCCESS in 1h 00m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/302c7d9e021e4321b68b7a44d12115ab : SUCCESS in 52m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5610cbc4ac74d8db7775f4b1d9e60bc : SUCCESS in 1h 00m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e229ece43d2452fbaf63f727cae6e6f : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbcdb3afd1dd42e2a03cc4b667f7be17 : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a993bea0aa4f4181b5da8b2483adc0bb : SUCCESS in 18m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bb2a3460ddcd4d4ebaa816592c378507 : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/55968d3b260346a5a7d55ded10c61b60 : SUCCESS in 22m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2524e2c2696d423a99581e705dc4aa01 : SUCCESS in 20m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6a28e458cc944d04be60714155d534d2 : SUCCESS in 28m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/20b190b8e2ef4987814447e771a7c8fb : SUCCESS in 31m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c788eb74a8444797a5d56339933fe705 : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a1f3482d610a46d4a6024c99a77f7552 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f9a38f7c5c67467ca0dd55550b4aff30 : SUCCESS in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf14871f271d42a3b3babe1db876ba16 : SUCCESS in 6m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c92853e4a2594a8d9a0dba435c6e4242 : SUCCESS in 8m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d5bb33c1e4b4a7aa32e4054b94d16c6 : SUCCESS in 7m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/93aca2583c7a4d6b855aca7d6e270a93 : SUCCESS in 51m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4282e10501d549e6b9a6d68a9480c647 : SUCCESS in 50m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/551e4e35be5c46fc86cf8c22505528c2 : SUCCESS in 46m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e97e43be84a4e94999e42d71eb1a196 : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eefc0ed7a874c7cb004d222265132e2 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f1ddfb8571de4236b2ce55d1f5dd0066 : SUCCESS in 15m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a778939acff49a0bf10869662bf4cf3 : SUCCESS in 40m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19b1bfbf10cd4727ad394d4aa7181e35 : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":1},{"id":"fff1e2f884ac9d7852ff8ebcf29246898e98236d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-27 18:26:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"db62816045ec7ed675ee39f94566cbf141daf1b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 19:19:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0909b6f372694521b9c4119ed185f2f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e2d6668b4d145b8a3eba7af265cc3af : SUCCESS in 11m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9e3aae2aa924de490ebf1592466490e : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c79d3a63c0e42d5810033e3094d3b37 : SUCCESS in 21m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a3653b957c94386b326058bf401b718 : SUCCESS in 20m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd7465b30e91418b8805a5b40380818d : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/554017090dae4e3eaafa3add806060bc : SUCCESS in 18m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/41afb0f613674229b1823f34d22529c2 : SUCCESS in 20m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/35a49f23051949889b3ad32c8078e052 : SUCCESS in 34m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c1c548b724024c64b097c9e93a6fe472 : SUCCESS in 11m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/909bf5fc1c06476f99f555b7044505ce : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2ab1814fdc76475eb66412f7098cf068 : SUCCESS in 14m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4ed9ed0193914556a2afe0322d2497f4 : SUCCESS in 15m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1604670d76674a678bb2421733606e8b : SUCCESS in 36m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20402f0097744d5fb9ea9ca1d8a4d5ff : SUCCESS in 48m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7243ed051e8748d88cf34ab87dc3cac0 : SUCCESS in 27m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2140a6a012f4703af4e619a90854606 : SUCCESS in 2m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38d79f9833904097af845e0af6ec2dfb : SUCCESS in 17m 38s","accounts_in_message":[],"_revision_number":2},{"id":"82381f6db6724b8e5eb5274c5454184394acf0c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 20:46:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6987eb93be746cc81ed8a69c40571f2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2dea65bc1bc640dea4c943030a3f3f29 : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":2},{"id":"0a68c7736193b6fe7b18dade2d154dcd5ca10d95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-28 18:13:23.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9a30b3c7b52906c79eeb7c16d3c445a8ffd0517d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:10:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6806a526a2dd4e408985a5bd40871737\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab46a9bb14b4dbfa0f9c831e43a03c6 : SUCCESS in 7m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7260d0bf75444fb6911a88f08238f451 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee5cb513375445f08855cb439ab27681 : SUCCESS in 21m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ce1bb74843145ce8c9f32fe0bcf8b8f : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/552f46f9ff7d4ebf888d2e5db8db1376 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a295feced8134fdabc9de9f55fa3a22a : SUCCESS in 25m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d70d789ec72471f8e49a6e68982796c : SUCCESS in 13m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4d218c61552a49e98882bb8f2db151fb : SUCCESS in 22m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2e87ef225dfa4430be2780cd34783289 : SUCCESS in 18m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f28eecaf930747869f77bf994b7fbcab : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8057e04af5da4b37a962d11ca95b2a63 : SUCCESS in 15m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/36ee465e6e634bb58a867917bffa2e35 : SUCCESS in 11m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74730c2a706e421cab0b2b5d9dcc1a44 : SUCCESS in 53m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b18e108171b425cb1862ff66fb427ad : SUCCESS in 36m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5fc2b70a691451eb3edaacd6b7ba754 : SUCCESS in 25m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1ab8933a18d48f9b0d9dddd201f340e : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de899271ceac49ee97528dd3efcf6ed4 : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":3},{"id":"0431c13ecad0617300dbcc25e4435258e8bba6fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:30:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e82cd95926b74adab91f4518b51bfb60\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/287b1fe902654c068af54355f52f3f5f : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":3},{"id":"5384c6e5af9a13a3c2ea5635db1772944ed0640e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-01 15:37:49.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d80ffcc8c5f48681d574751604bd099e5bf0cf37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 16:45:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3de2c91b67cb4866a718ce8bdb0fbbe7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8481fdbe55e44f338e7cedac0e2db235 : SUCCESS in 10m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f8d6d92dae34639b73c31caa6d9b328 : SUCCESS in 11m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5fe17c3b33b04d0a95682aba42fdf48b : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87bc5e4d5a314a9d9911e122c6c8f4bb : SUCCESS in 13m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66197ce00f2e483c91f675ef271e56ae : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b80bd6c65cc748ceb39187c2127adeeb : SUCCESS in 25m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/26b06485f5324e6dbad4e0a01c87d3ef : SUCCESS in 19m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1244988be0ea4bb7bdf165ca97ad7ad0 : SUCCESS in 24m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/569fa79ff7eb4fa9b9b83ea0bb83b381 : SUCCESS in 31m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cd5b776dc4be451e8b17c00a764fcf73 : SUCCESS in 8m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d0f409fccf6f4ba183ab6b55922763dd : SUCCESS in 14m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/184b96ccb3004cbab4e2c60c2fde91bd : SUCCESS in 8m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08a976e3a85e4f9c91cef3ddb3e96119 : SUCCESS in 27m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb25f0bdf0a9494e83c0583d41fdba21 : SUCCESS in 1h 00m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2407407387dc4785bb247eb086453782 : SUCCESS in 28m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead4485ccd05440285fbf9c2d6de669a : SUCCESS in 2m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f118d327b0549248bd0502781f234b7 : SUCCESS in 19m 35s","accounts_in_message":[],"_revision_number":4},{"id":"5dbb074df48d049af0f28fda46f834a7dc0a1c59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 18:50:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/842d2f71c2bd4ba785965b5f76faf2db\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b7b6d1dab1c94be6a65c30b38904a656 : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":4},{"id":"260d01f7297c444f1cadb10b4a394c2063a30b14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-02 18:34:39.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ee80a906bbd58f59c7c2a9d4b78064908fc0ca77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 19:32:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a075c44925eb4c219c6c5175fc6359cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7df06af6b114f39ba491cd68ac4db80 : SUCCESS in 8m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/349408eb39064d6ab701b40706ee37c9 : SUCCESS in 16m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98940d40d4c04867847520751d7b0161 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ee9bdb492524839a4d432ddef887563 : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/428775e81f9f4f83ac87a8b7a0fb46bb : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c0904d81cee04afda19d118f4316b158 : FAILURE in 27m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8d86c908cdad4e0389d20948c3b49bee : FAILURE in 17m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/10a7fda9acd34ffab0738284bf3db3ca : FAILURE in 36m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fee0cc73b57446e2b6387497d59455ce : FAILURE in 28m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/96d89643d1eb45e68186015de1d93c1c : SUCCESS in 8m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/70f7961960ef4bd9a094b0efea941744 : SUCCESS in 14m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5d4bd4a9022947d5a865cfabff78fbfd : SUCCESS in 17m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6fe7a0a6f1cd4f68878b37b509fe6cd1 : SUCCESS in 52m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/229eb28399164fa196e72c3ed072f721 : SUCCESS in 51m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5283b9aeac4b4c51b4efb4cf5d7afc35 : FAILURE in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c698e6756e94111bcb075eab6f95e50 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e11ed6dc87341059a28e6c488408d61 : SUCCESS in 18m 07s","accounts_in_message":[],"_revision_number":5},{"id":"ebf4b8c79cfa9aa209ed624942cd4953e0db1fe3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 22:41:58.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d55180f6bc1346f8be209de08bc67f94\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/55415e02056b48f79184b6fbd181c141 : FAILURE in 41m 24s","accounts_in_message":[],"_revision_number":5},{"id":"08a56542f7a646ad739600a051eb2bf2bbb51eee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-03 05:19:52.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"e34c32d5c80c0329398fa14b97a896b4d49dded1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 06:17:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebf6ef41d42c403b81211fff06f85702\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bc1e85ec1184eb79c7670a7bcf7f3a2 : SUCCESS in 5m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e08b390e82af44d29694886d1a3bb724 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f213761f54844112a67e228a730b105b : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8065cff49cc4494af35427b0088b14c : SUCCESS in 12m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a7c9e73a3d88427f9682433454d983eb : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b31b7f1c180a47d8969cee91c6da405a : SUCCESS in 26m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dc15fcabacbd49289bff76e481c67276 : SUCCESS in 9m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/56f6437a157540da8a2685b1bbd43015 : SUCCESS in 11m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/abc44c8d6da444bf950faf89cb657d6f : SUCCESS in 11m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0f0207dad05d42d0986ff4685a135e46 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aee7a6da479244e19ac7164fab34e564 : SUCCESS in 13m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f160ea017e7747e78018d07de7a8ae1c : SUCCESS in 15m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ae4986501a34c1292eba61738de599c : SUCCESS in 34m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce5331ec130b4fb380519940a71ae5e8 : SUCCESS in 53m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a620d34a448048aebbb1c4c23d6ad6ea : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8708dc38e577444eb36e2c442598d93a : SUCCESS in 2m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6cc95fc0ff1148118f1e1467e39419d6 : SUCCESS in 16m 38s","accounts_in_message":[],"_revision_number":6},{"id":"71f220ffa02106fba2ada57d0cb55a2066a3c5a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 07:07:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8beed229942b48feb16e6b6f06e59b74\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0f94e3acbd6c40d3a39a2f33fc73c653 : SUCCESS in 1h 23m 53s","accounts_in_message":[],"_revision_number":6},{"id":"817ee59f679f2867ce8955aaa3155d7cca414f3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 15:07:41.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"37f82ecf50f3f4655878e66dfd9354023bb7c526","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:46:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9ae02e30b8843a78a4ab5b751abdd43\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8b88a6579ee74462b278db3311f27461 : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":7},{"id":"9dff0e97a242c6de278a159b92f2e61b3f3615cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:47:55.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9710eb8e0414a1c9a7f4f752bc3ec02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7113b1aa1f1d4f37a237858721f05e86 : SUCCESS in 10m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b71c11b695d4f35a8cb50cd15c3badd : SUCCESS in 12m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f4646cf2068419290d874141d54ca38 : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac892d99e98e4267bb57e203c003a54b : SUCCESS in 12m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/31de33d8938e42659d8ebfac61132ea7 : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1526474180054285a11d0a0c74db1d30 : SUCCESS in 10m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/468cda52e17b4cd68ef7b36204d25a45 : SUCCESS in 22m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c9870b87dac644e7aae53f28d7895966 : SUCCESS in 22m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9717b041425a49b1b3e8539b03f946af : SUCCESS in 29m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/337f59b8e04c4bfbb704c359a32a791f : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/063cd4da524d4c74a0ee96676bebf833 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/79ec8996b96240b1806dc0877fde3343 : SUCCESS in 15m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e4d7788dace486fb7dfdc2a7dfb1a1f : SUCCESS in 22m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/93177bcf78774881ab0c89c9359d5389 : SUCCESS in 37m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/959d5dd442ba4dcd93fd701e5e432112 : SUCCESS in 22m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4129c153c22c4a31868d5b1153bd9923 : SUCCESS in 3m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/82dd1ecedbd149fa8e7f6ca58410ad80 : SUCCESS in 19m 46s","accounts_in_message":[],"_revision_number":7},{"id":"3a062e6eff3c72784d90bbff1e67a03b9d0f692d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 16:02:50.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"c7ecd61b3636ec69fefadee2dd6c8c13c72d0c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:42:54.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/235034142dc1487a88f4ea5cded49485\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/301465f2e5bd492a8966ea051a6fef8c : SUCCESS in 5m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1a4984cec0e4d59bcb4e9b068d89314 : SUCCESS in 10m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/637645291a20414b973559d1b9325132 : SUCCESS in 11m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6ae9dec0ec40467ba39b678d8d8cad1b : SUCCESS in 11m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b821435160f04f738e984d62885a2d11 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/818181d322e74df8a28f984cd7f4806c : FAILURE in 11m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d1d9e3bf24c2425c8be10edd46c8b985 : FAILURE in 9m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bb1352fe025c4526bdff6af5b6d94b52 : FAILURE in 11m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/57c20f58a62d4faf8e10992d124d8d97 : FAILURE in 11m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bd635bda722f4a80a07928adee534b4a : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dbd91c1859a84681a24be1e0645c970b : SUCCESS in 15m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/de6eccbec18f448d88e795d2dbd1e866 : SUCCESS in 9m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0240499002f9417d88cd319fbd3aa226 : SUCCESS in 22m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61eb84cbba184b1d9f3cd466282c2dec : SUCCESS in 23m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/192d8f2fe26e4f6894f2599263d2c481 : FAILURE in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e48e9fdbd2741eab4c30f3d5f533070 : SUCCESS in 3m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91964292ef5a44e1ac4128e8b4b951d7 : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":8},{"id":"527e8ea264e3ed6159865e551a59a33e0aa1bada","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 16:53:26.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"6aab3e4fcbe93b8e77a7907dbfb43cda0981ea30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:48:53.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f1c055a67f443cab664705488b417f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9437fa1cce34d59adf52d6f69444438 : SUCCESS in 10m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dee1b471783b4c0cb97eae313cd9a9fd : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59bf71c592b14022a0a42efb460cc7c2 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5cd0209c9d3a4f8eafc89a6fc047b74b : SUCCESS in 10m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7005a46531ee402a964d3098254d2f53 : SUCCESS in 7m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/92fa4c716073469ba2f19525fba78e19 : SUCCESS in 25m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fda3529bc98f465684ac989d0627b28a : SUCCESS in 20m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bd28fdfc79484971916e6e04b9fb5155 : SUCCESS in 28m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/505710e4e69244f3bf6045d06985c6f8 : SUCCESS in 11m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6e159246decb4ceba1ead49b0c2f4142 : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/075b72b756b1450e95107f2092aa341e : SUCCESS in 8m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/15da32390a504f4e8eaee2e2fc07f978 : SUCCESS in 15m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/94759dc0a2ef454581a18b8437bcc8f2 : SUCCESS in 49m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/94b0ed0ba1cb49a7a00fc0efe8ae8eb7 : SUCCESS in 44m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07b1ee5065304fb7a0f32ffc81a92359 : SUCCESS in 25m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/047f27efec2c4bf0a3f79489550227c0 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/245f3dd77e2f426c85be52f55f5b1a08 : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":9},{"id":"30802e59f453442116c62ffbd41c5008ca91be43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:18:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4564162e47947a7962b57a04170a434\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a9fe44d5eb34ea28b877826313c3d80 : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":9},{"id":"d4ee0fab7a8257626b5e340391e3cdbc9d0e29b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-13 12:50:38.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"bcc03ecbe8cda0d8433e831db7891313c6f10904","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 13:19:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c50042ec34407e8189f3ad73eec181\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f960f3a558ec457995f418a852533ed6 : SUCCESS in 26m 53s","accounts_in_message":[],"_revision_number":10},{"id":"de9067491f5428b5897d8e2c114e42465dde8d3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 13:32:59.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02fd342171644dc4a4efef97a0acadeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083e3cf6f4b445d6bb1fbbe6d6349505 : SUCCESS in 4m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/709dc6445cfd423ab8e1b3c0ef4aed50 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f310d0b1c3b5463789d4261686bf73e5 : SUCCESS in 11m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/138d6051d0354fa383145f4f06d1bd7e : SUCCESS in 20m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd965ab7a88e4cbda5693508ddc4c350 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3fff6a4ce90c42d790a1b01b9568fb18 : SUCCESS in 12m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3cf59a04775c40cdb05b5e3faf23bd3d : SUCCESS in 15m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/64653c62a2ef46a29c2e230daff4def7 : SUCCESS in 10m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b288192972d24484876d3c34e567fde0 : SUCCESS in 13m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ecc3bd4cea984b76b5392abe17e47122 : SUCCESS in 13m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bb5c822f03324eaabfe060a698ddd4c7 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f12257b591164dd89c36829102494eca : SUCCESS in 14m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b22caa078534473a842e0790e20a2b77 : SUCCESS in 33m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae774b5f6a0a4448bbdff4c081b4345c : SUCCESS in 33m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7e2dc438fdb4ceaab1e81fe5d70b76a : SUCCESS in 23m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d6ac9f1a744e9f9192f8cf1656056c : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0c0b700e5024124b9907330a4b8c896 : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":10},{"id":"915778af5b11072bd208e57d2b67dbafafcc6c57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-14 18:47:57.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"f482c8d1cbc79abf33875ef7d91ace596ebd2346","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 19:35:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/644b45cffb8941808f34717b0ba1ddf4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e80d822f09c846b0ba5f1ae991804f56 : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":11},{"id":"e9be1be6013766c7661e59e2135b54a2682e881f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 19:43:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b2dd05d82504bc7b9f4e0419f744ee0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd75dd898a864833a37a2424f81d03e7 : SUCCESS in 8m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3262716c2d3f41bdb4d9c070d8072aaf : SUCCESS in 21m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f29ab9dbf8f4893b35e62242794d76f : SUCCESS in 11m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/670adb65a6494465a7b34616891c5fa7 : SUCCESS in 17m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4619d1a1b8484c75971edbba0ad66f0d : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/afea256d10cd497b93da5c500f443f55 : SUCCESS in 17m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b193d346096e4754a12b9f58fb0f00e1 : SUCCESS in 17m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/de4a461a36c04c81b42318ed3886276f : SUCCESS in 27m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7658e20bba174692bde2680b094eefd6 : SUCCESS in 30m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2054ce2ba64b4955b5472c759b89658c : SUCCESS in 14m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b3ee7d56b70d45d68100179db2687d70 : SUCCESS in 13m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2ac0c2689e0444879923749f848b60ea : SUCCESS in 8m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/49425095259249ad8a3c5ba0fb7ffc6a : SUCCESS in 53m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/19f75cbb250e4ff09d35f9d03c183989 : SUCCESS in 33m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c5177f61a7f498d9d6614c5fe3a5070 : SUCCESS in 29m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576cda3563054b7f9f713efa4c3a54b0 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ea9184e51cc41c992df6618a411f965 : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":11},{"id":"fee3ef252395d6fffadaedd0a14f0be1e3fbdaa2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-12-15 20:25:22.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"35aa58cec0046bed4422812ced94681952271bf4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-16 11:44:43.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f01c31d1371e23eadf465d5bf4ed0ffca1c0b2d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-16 15:02:58.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"2edb0aacc3abe88f345a470b115aa4fabf76bdc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 15:32:41.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e4dc70829ff471487f616161962d4e8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c55a4fb03a07430a9686ecbad4718272 : SUCCESS in 27m 27s","accounts_in_message":[],"_revision_number":12},{"id":"71c2262c4514c6a77bb3fff3dea266f21b97f640","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 15:33:51.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b2dbe841a444dceaffb92ec5c3f9fd6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804de6af65144ba7b3786530f3ebd1a0 : SUCCESS in 7m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68b605872fb24435987faccbbc70b11a : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8ec17160ad047eabd288b2b6683670e : SUCCESS in 10m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9abd671f939b4c86aa7d95c871a9e552 : SUCCESS in 10m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/da23b5d6bb2c4541aacb0732394bcc75 : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e9a3013c60464fdda881672f42411a38 : SUCCESS in 20m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a549f99f1c4148a4a48a6559f2b7bd7c : SUCCESS in 21m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ae798d78ebf944128b5080d12a66909e : SUCCESS in 11m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5098c2a2903a4e4fb67e766b973e1ee0 : SUCCESS in 24m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4e8c2e2dfded4684a4f9305dbd6bdafb : SUCCESS in 12m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5f4f2e2d24a742669131e2330e272407 : SUCCESS in 12m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9d9be413ff97456aa056facf81cbf33d : SUCCESS in 15m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ee6505461f9e4c79ac1d49202bd54e8f : SUCCESS in 28m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d3a2558f15db4d2cb3d68641130a5654 : SUCCESS in 28m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8129f18ffe444f23bd181b2c82dd5bf3 : SUCCESS in 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b53cacbae794da9a35f6923f823d806 : SUCCESS in 6m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6966612799174cc3baaf54e8737e2acc : SUCCESS in 20m 32s","accounts_in_message":[],"_revision_number":12},{"id":"41270809a4f6750b52c627bd9c54eb2dc72b5581","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-12-17 03:50:01.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"97a3f6ae5aacd218c5657e5f193544594038a201","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 03:50:29.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"c49d3fe4558b641e21a25e7d8ddde5257235adda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 04:40:50.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c49854bb94fb442c8063a70b4d873651\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b568ee3895473dad8a273aebf5507a : SUCCESS in 7m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/455bd343f9ef4c7b82661b0036e0c90f : SUCCESS in 4m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3ba5fd4eed614368a89f467d401049d5 : SUCCESS in 10m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/54c3246c8bb74b87ac6592260847a670 : SUCCESS in 20m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/388c56e721d143cc81c45639bb0a3676 : SUCCESS in 20m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/027d23a38f554ee0982bfa91a631a36a : SUCCESS in 12m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0d1d821c12a74eb4ab8d5cf43f12c969 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4818349d9f7c4942bcd7868125269360 : SUCCESS in 11m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5f5288f8ee514ab497498247058ed256 : SUCCESS in 8m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7533dc77a9ba41678d89ac74f52f60d7 : SUCCESS in 8m 05s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3eabf08e850d476ead8c3372f0591072 : SUCCESS in 7m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4665e138207e403291838d02728f62a2 : SUCCESS in 31m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb4a3aaa47894d49b14985a8cbf27f32 : SUCCESS in 46m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0987d1cb5ba4694b990f2614589465e : SUCCESS in 23m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58dcf6506b9c4ca3bdd1feaaaa76d317 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f78ba33645204ce5ba45c1624234daf9 : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":12},{"id":"ca019561f205ef3d5effa039cd3898dfe7cadb34","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 04:40:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"89c53629388854ade7ed0f1d0bdda865edeec194","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 04:41:43.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea37347aacd4639bbe6260dbd5d77d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd146358daf34859aae1fb78961af26d : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a3e1c646bdfc44469a6883239cd9eeec : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"5937b47a7ca768f2d097f1c12e686bbdb2d3fcbe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-12-18 17:02:09.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"55eddb9c6e12fbf876ddf76a1aeb365855166394","revisions":{"4dc1e5faeffe6944c47c2fc6c7c97ae46876565d":{"kind":"REWORK","_number":1,"created":"2025-11-26 18:24:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/1"}}},"commit":{"parents":[{"commit":"a1c6e09ba06ba6583e48ce9048a21a352056b094","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1c6e09ba06ba6583e48ce9048a21a352056b094"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:53:39.000000000","tz":0},"subject":"container server: use Timestamp.zero() for default timestamps","message":"container server: use Timestamp.zero() for default timestamps\n\nAlso:\n\n  * Fix some test assertions w.r.t. Timestamp.internal vs\n    Timestamp.normal.\n\n  * Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dc1e5faeffe6944c47c2fc6c7c97ae46876565d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dc1e5faeffe6944c47c2fc6c7c97ae46876565d"}]},"branch":"refs/heads/master"},"7223e96a8849da94ba5e99a42ac64bf1da53facd":{"kind":"REWORK","_number":2,"created":"2025-11-27 18:26:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/2"}}},"commit":{"parents":[{"commit":"32420212f626b928e509e793394d42c55a266fcf","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32420212f626b928e509e793394d42c55a266fcf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-27 15:13:33.000000000","tz":0},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7223e96a8849da94ba5e99a42ac64bf1da53facd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7223e96a8849da94ba5e99a42ac64bf1da53facd"}]},"branch":"refs/heads/master"},"d09919665bed66c46ffd8da2da560795ec88888f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-11-28 18:13:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/3"}}},"commit":{"parents":[{"commit":"c10578a7a0358af65bb970c6ef0cd0e2b658eff8","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c10578a7a0358af65bb970c6ef0cd0e2b658eff8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:13:12.000000000","tz":0},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d09919665bed66c46ffd8da2da560795ec88888f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d09919665bed66c46ffd8da2da560795ec88888f"}]},"branch":"refs/heads/master"},"3d273cb59562689016349a2cdab46aee006c1112":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-12-01 15:37:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/4"}}},"commit":{"parents":[{"commit":"718749744991ed8a74d54526b611194c892790ba","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/718749744991ed8a74d54526b611194c892790ba"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 15:36:52.000000000","tz":0},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d273cb59562689016349a2cdab46aee006c1112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d273cb59562689016349a2cdab46aee006c1112"}]},"branch":"refs/heads/master"},"e536d1681d6f26a77a446db9f764816d4bb1797d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-12-02 18:34:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/5"}}},"commit":{"parents":[{"commit":"5ca6d959388195bb6befcd4dd13a43d97875da91","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ca6d959388195bb6befcd4dd13a43d97875da91"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-02 18:34:21.000000000","tz":0},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e536d1681d6f26a77a446db9f764816d4bb1797d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e536d1681d6f26a77a446db9f764816d4bb1797d"}]},"branch":"refs/heads/master"},"e9f9cf8efd2958372f4e441666f688d4d7bae295":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-12-03 05:19:52.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/38/968538/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/6"}}},"commit":{"parents":[{"commit":"6c48fffb5f5e3aad2a1f9185b8174131ba55fa95","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c48fffb5f5e3aad2a1f9185b8174131ba55fa95"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-03 05:08:59.000000000","tz":660},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9f9cf8efd2958372f4e441666f688d4d7bae295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9f9cf8efd2958372f4e441666f688d4d7bae295"}]},"branch":"refs/heads/master"},"7732815c55865817fd1696f064aa1b2a3731f263":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-12-05 15:07:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/7"}}},"commit":{"parents":[{"commit":"ad6bed6f085b1053e84ae2a26b9aa0aefcb881c9","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad6bed6f085b1053e84ae2a26b9aa0aefcb881c9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 13:02:38.000000000","tz":0},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7732815c55865817fd1696f064aa1b2a3731f263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7732815c55865817fd1696f064aa1b2a3731f263"}]},"branch":"refs/heads/master"},"31834682f65e5715b2e9d0a097a1f776970c24d2":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-12-05 16:02:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/8"}}},"commit":{"parents":[{"commit":"e9f6d3a294d611d5b6baa4b30c4e750e1a03286a","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9f6d3a294d611d5b6baa4b30c4e750e1a03286a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 16:01:19.000000000","tz":0},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31834682f65e5715b2e9d0a097a1f776970c24d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31834682f65e5715b2e9d0a097a1f776970c24d2"}]},"branch":"refs/heads/master"},"45b0d23fdc38c61327e56e584d51ea5acfe1f0b7":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-12-05 16:53:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/9"}}},"commit":{"parents":[{"commit":"1bc64df6eba8ed3b74e3527cb13634c1847c88b6","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bc64df6eba8ed3b74e3527cb13634c1847c88b6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 16:53:07.000000000","tz":0},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45b0d23fdc38c61327e56e584d51ea5acfe1f0b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45b0d23fdc38c61327e56e584d51ea5acfe1f0b7"}]},"branch":"refs/heads/master"},"354c3f0ea5560acba5b762c6eb4845c6164db6ed":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-12-13 12:50:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/10"}}},"commit":{"parents":[{"commit":"de3158da9b80963cdc11746e444b3374c10595a1","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de3158da9b80963cdc11746e444b3374c10595a1"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-13 12:47:40.000000000","tz":-360},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/354c3f0ea5560acba5b762c6eb4845c6164db6ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/354c3f0ea5560acba5b762c6eb4845c6164db6ed"}]},"branch":"refs/heads/master"},"09d73e1f9ca1474354ac3976f204d51c36ce00bd":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-12-14 18:47:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/11"}}},"commit":{"parents":[{"commit":"0ceff0fd5f433209593fa7b4056fa0bdeb2146ce","subject":"test_base.py: tighten up use of Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ceff0fd5f433209593fa7b4056fa0bdeb2146ce"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 17:42:16.000000000","tz":-360},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09d73e1f9ca1474354ac3976f204d51c36ce00bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09d73e1f9ca1474354ac3976f204d51c36ce00bd"}]},"branch":"refs/heads/master"},"55eddb9c6e12fbf876ddf76a1aeb365855166394":{"kind":"REWORK","_number":12,"created":"2025-12-16 15:02:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/968538/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/968538/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/968538/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/968538/12"}}},"commit":{"parents":[{"commit":"1698fc32489572787af45cab2bb9b8774ab1fb90","subject":"OldestAsyncPendingTracker accepts Timestamp.internal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1698fc32489572787af45cab2bb9b8774ab1fb90"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-26 15:49:08.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-16 11:29:34.000000000","tz":0},"subject":"container/test_server: cleanup Timestamp usage","message":"container/test_server: cleanup Timestamp usage\n\n* Fix some test assertions w.r.t. Timestamp.internal vs\n  Timestamp.normal.\n\n* Clarify Timestamp format usage in tests.\n\nChange-Id: I629b49d103a4392833dbdb14224eb20403b09a11\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55eddb9c6e12fbf876ddf76a1aeb365855166394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55eddb9c6e12fbf876ddf76a1aeb365855166394"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
