)]}'
{"id":"openstack%2Fswift~968740","triplet_id":"openstack%2Fswift~master~Idcf28c29d52a52830ff12be9c9b1235fd8d23738","project":"openstack/swift","branch":"master","topic":"p-timestamp-v2","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-04-21 07:06:19.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-04-20 18:43:30.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-04-21 08:10:53.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-04-21 08:10:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idcf28c29d52a52830ff12be9c9b1235fd8d23738","subject":"timestamps: add NormalTimestamp for ShardRanges","status":"MERGED","created":"2025-11-27 18:27:35.000000000","updated":"2026-04-21 08:12:30.000000000","submitted":"2026-04-21 08:10:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":143,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"968740-p-timestamp-v2","meta_rev_id":"e0a95d5796dac66ba0064310965fa4ba29edd386","_number":968740,"virtual_id_number":968740,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 08:10:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2026-04-21 07:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2026-04-20 18:43:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2026-04-21 07:06:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-27 19:38:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-22 04:01:31.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2026-02-02 05:51:04.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2026-02-19 06:22:12.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2026-03-11 22:46:42.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"f7134940313cbbd665461259577897a681e6e0c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-27 18:27:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6ebb733966e5f0db52278ffd5f77298f3e747c6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-27 18:38:58.000000000","message":"Patch Set 1: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"eaa8ceb62f34ac77a6ecdc40d8f30871665322c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 19:38:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a2e5bba53b1400ea482cd0847d503aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2488cc9ba444427b7a9369443c3cab5 : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/80ef7be793b142ffaeb4557c4225b085 : SUCCESS in 58m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d7e6da4cba5f4e968abdd25a33a3270a : SUCCESS in 57m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/807d82a4b4cb49ed8daf912529c5eb85 : SUCCESS in 53m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/950306ccf37d468d8ad2075889d68c43 : SUCCESS in 59m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/487a9de65ef046f996abf0540c6b27d5 : SUCCESS in 22m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d501797804f64868b22add10731b16fc : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f099b951a764a8a99d4b1846b59b323 : SUCCESS in 15m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/597bce8b6cb64b1bb175f391c3f562ee : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad1a4f50a5354dc7a3f86e137472a7fa : FAILURE in 29m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2dd40b2b432646e9abcceacd8bbdc403 : FAILURE in 21m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/debdb740278f432781166bc4c1e45a6a : FAILURE in 17m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6b2ec3984a9b4f249de7a7cd2c51e8ef : FAILURE in 31m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4dd8cc31e25348a0991d6f3a49212f98 : SUCCESS in 8m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fa51e710891d4c5ba3b9b47b64fdd1f1 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/04215e8b136049d8a012114c0f50d26f : SUCCESS in 15m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a795b617d6d4aaab151d113d736c305 : SUCCESS in 6m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4192297dce40432d9e5a791c68400743 : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/711dcb171e1342a482c9d72f952c578a : SUCCESS in 4m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f80f640ea1b4022a0eff4610e1b3b93 : SUCCESS in 55m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/775f67e4562849a2b2e7cafceeedc053 : SUCCESS in 55m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee19b3cda86f4777a1146fb9aeddcd2e : SUCCESS in 55m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdd557a5a7b04984983dd48b6a596ac1 : FAILURE in 21m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4ddafc2a2b94a1cab14cb3fa7467589 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/239ddf229fa34a3ba9c5d6e8fb8a52f7 : SUCCESS in 15m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06593a3b63644f95871420291981274f : SUCCESS in 50m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8061be872784386936777e3e59f1245 : SUCCESS in 1h 02m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ddd9cf735e37c3d33bb32b862e24f5a65876d2d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 23:32:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb88ed85ddb34032b5e0412342897426\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80f505f79fd941e994c4989cd46b3aff : FAILURE in 1h 27m 35s","accounts_in_message":[],"_revision_number":1},{"id":"619359f09c6e067b847cde2c788d2a192a0820b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-28 10:34:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7a488452980798566ecf8837cd943b555989f76d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 11:48:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c26f10a587084cb7b1f955342fd19a35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aaad96a090144fd8ebab5525120b757 : SUCCESS in 5m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/39a57eab5f404599948cae68390ef0ef : SUCCESS in 58m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/33313a4ed34842bc965966158e839652 : SUCCESS in 1h 00m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/66dcc22cb4a04401b86628ca2f5bc2da : SUCCESS in 30m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d963361224034dd5bbe51cfe8fe841ea : SUCCESS in 1h 07m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4b75a4561584a62bc632ebdf46f4ba6 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1305edf7fc804466bd76a43f468031a4 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e9d7edac0944e28b222d4afd1b78a46 : SUCCESS in 18m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7333a73d5f02463b83f6cd1cc19b3b9e : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/804d2ccb13314dfbb4cf2c547d24ed34 : SUCCESS in 25m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8bc87d7c6b24e6d886161de1630b794 : SUCCESS in 21m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d3c3df6cae4a431ebc4bc9cb0d6c37d6 : SUCCESS in 18m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/00c901b704424ba1868e50dbbe64000d : SUCCESS in 23m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e220d428ad614079ae7c15ecfe9e1ef7 : SUCCESS in 8m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7677b4bc1221496eaa892a426db440d5 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7394f146e05745f7b77dc16b6f9b95c8 : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e38a69e61f304a81942171b7794843c3 : SUCCESS in 6m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/836fbbd05cca4c60b227f0280fb21ecc : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/106200f7753e4fc5b1f7915314ca9a71 : SUCCESS in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/45e386dff2c540798cfd7ca1bb51b764 : SUCCESS in 34m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbb3d16a41a64071ad78abdd2e72fb81 : SUCCESS in 34m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0a20c3057c54d1e9582637c899e2b25 : SUCCESS in 35m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f90cc74eceb3492d995a125b4aa851d3 : SUCCESS in 28m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ef9b1018eea464f8175cb29d8d2fc64 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/66af645513d045d992138d376016c880 : SUCCESS in 19m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8aa84090dad5428dadd8f0f7e340f160 : SUCCESS in 50m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc15cf0176234bb59f741ecf8c7f5870 : SUCCESS in 55m 29s","accounts_in_message":[],"_revision_number":2},{"id":"1aebc74a0da240aa961de1844df5354d7019e8e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 12:21:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/741b27851edb45a98fb1b5f10fcfb708\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8536a0fbb3af480baf35ec2e0f8436ff : SUCCESS in 37m 17s","accounts_in_message":[],"_revision_number":2},{"id":"3961eb4189bf32356d20ec6b64f56b60fc1361ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-14 18:47:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f560f773bfe8bd61be3aed7bea06ca98212d3ec6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 19:16:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a867945d6a4a6285e202ed2e9624b2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/72c30aff0aa94817a31283eb7864d944 : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":3},{"id":"d1c959fbf8787fa9b5c0f83b08ac6a169bd32928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 19:56:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ee30d176254b66bf0b4b7d03f77403\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a735d5690ba4a15be97c53e3c5e9b2a : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/2d7b0e0aa5dd41389e94e8209e241def : SUCCESS in 55m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0dfbdb9048164a5a89e4d53d978ac43a : SUCCESS in 1h 00m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8cedf1b83607457a9f965c1fab374ef3 : SUCCESS in 53m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b37492c66d9d4df88a57fe82a5b99366 : SUCCESS in 55m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c081f2da060847dc8d53590b6cbb987e : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d67da7a98f24ffa8e93136e290b0582 : SUCCESS in 21m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ebcfb16efa344d68425599302c00eeb : SUCCESS in 15m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e32ca30fc3d47079e97bb94996634b2 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/201a4d01893344c79db9278044df8f9e : SUCCESS in 28m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f44c16cdeca439bb823a7ee45f3af66 : SUCCESS in 21m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c0a431a4a0de454eb102bd97260e9cf6 : SUCCESS in 28m 03s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5204c6fb8fbe45ee8b6518b0642cfc00 : SUCCESS in 30m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/df1a971b43f843ddb1e1c5c2ea2cb79d : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1c2356e571444d93a65dd6dc5c11704f : SUCCESS in 14m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a2b66260a6ab449c8c0ff278136cf29f : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20158d7a8f6f479dbe76731702ca38ff : SUCCESS in 6m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/224bc662109c43c7b582603e8791c82a : SUCCESS in 15m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e49c00bc3e24a90874e3aa4352bccc1 : SUCCESS in 7m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb36b361fc6e4cce9b54fd7a70323069 : SUCCESS in 50m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76f8f356cff84822973f685c6fcbb575 : SUCCESS in 21m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ceff43b18a04daba411ada25b9248d9 : SUCCESS in 49m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce513ae6167941f1a05fa0ad63e28ac7 : SUCCESS in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9008b9a463d4c14ab10d5a925471b72 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4a42b6a1f1c4e84b7a34a8e2dc7c142 : SUCCESS in 18m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee1f4fbb84364143aab5f70f0f18028c : SUCCESS in 41m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/023833a3f09b49bbbd795c1c591e18ca : SUCCESS in 54m 47s","accounts_in_message":[],"_revision_number":3},{"id":"e39f0a85f59dd411212513afca1b62a6a15a74b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-19 16:25:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"36cf4eae5b2d807f01e761cdc209cba3c0e5867a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 17:26:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cae8e8628c9e4d71beeb0f290c4a66d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7f057c319194242adc4a71f9d47797e : SUCCESS in 5m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/261f382b4e3041d7a88614ad6856cd3c : SUCCESS in 32m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/050419d2d1154fffb52e86cc6e98e4fe : SUCCESS in 33m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/44214ef2a493483dbc593bb54e8d0af7 : SUCCESS in 35m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c1ea8be45b54c10ad9654c9d2035348 : SUCCESS in 44m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93cbe59d23414deb9d1c67450e6ef34a : SUCCESS in 10m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65741e1e7218436ea9b1a9b413d93920 : SUCCESS in 16m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db7f9bca288e41f2a224fa03b969deec : SUCCESS in 17m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/31767e8288024b529fae0be52383b368 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/897a7a168e514dd4bbf9bffe3c4d24e1 : SUCCESS in 11m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4f125f8c8d2c4dd9a284c01dd25598f0 : SUCCESS in 9m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b618170a36474d23a5776bba637298db : SUCCESS in 19m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7594f9ca5d8244e79b762d0ac5477b2b : SUCCESS in 29m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3e808686778a411b8165f0b2ebd247d0 : SUCCESS in 12m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/512b13509c2d472fb7ba1a81087223f6 : SUCCESS in 12m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6facb628713644ad9041b6b3220a2ad2 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/abf7893b9e324deeaba5f3c253696443 : SUCCESS in 4m 04s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c5e34c5cd22f458292cfc78bf09f16a7 : SUCCESS in 8m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dfad905d34254d10900e20bccaf337e6 : SUCCESS in 7m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d1cd5c396114b7391905f31aee705ea : SUCCESS in 56m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1862f68bff349469e93b8dd6f2c09f9 : SUCCESS in 54m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce348a522e954672960ae12283b58e73 : SUCCESS in 22m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a71a2a518c64b44a11ad374ef34faea : SUCCESS in 20m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77debe305767483ca0e149575cd2e33c : FAILURE in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c9bc6790dff847419c9873d9261f44b5 : SUCCESS in 10m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3ef554b002040339b591ad19b30a3ca : SUCCESS in 32m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c495add04fd24ca6b04556de1843660d : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":4},{"id":"3ceb5ca57a594ea8c3b22b663b7768dd4c31f3e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 17:46:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/764a38049c1c4f55a75639610a25db67\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5644f89665074e6b889209c271f426c7 : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":4},{"id":"2f9ad91ae23fa4eb0b7fce9e2db1841e698c52bf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-19 18:20:39.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b1194a9c5f9754053d0941738e4f54ce3add924c","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-12-22 04:01:31.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2691c650cce4f39b62da0f580beb2741a173bf17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-22 20:22:46.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ab80ae8384172c063df0673f11425a034fc1ca81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-22 21:01:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cc17beaa09545e4901806f49506c7dc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ea9df0f92aa4cda828b4b34274e590a : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":5},{"id":"2819007f3eaee676e366170b18ee1500e19a097c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-22 21:39:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d8ef9b272e34fccbcd0f851c9322964\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f242fdc34bad460eb08aacafc38d7001 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/793a6a5af78a4992b61240e5b5875ed5 : SUCCESS in 42m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/02be184ac24946ca95888b00671871e8 : SUCCESS in 1h 10m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d82d5716dc7e45728384d45a45c8cd37 : SUCCESS in 52m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6ebe589a7ef415f99eebc9cd2e0e659 : SUCCESS in 1h 06m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39d00f36c9e74b74a48a2c7aeb0e561e : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/178978076f444bfab83707df4fc35f96 : SUCCESS in 12m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96bd4b64a17d435487095c2ed6c1474f : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3ed38a5aba64aee9cea3c761ad5afb0 : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5b0813560769486586a0af53c961344a : SUCCESS in 24m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/16aafc2be5bc4070a1032cc92a97c1ae : SUCCESS in 17m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/daf429942d4c4f9e8640ef32fcc16827 : SUCCESS in 13m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6643ac577921420d9bdfb83eff7acfb9 : SUCCESS in 13m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0f4fdd94c36848209cefcbb341943810 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/13bc1aff22944c8d88651dba1809e04e : SUCCESS in 12m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/79a889cad62e41738b7785c087dad877 : SUCCESS in 8m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/443941bc1ea24d898c5b76a7015108fc : SUCCESS in 6m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/18f7cd160b7f4c1b9d10ddf8a7099306 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/49c2903f3b86458a8e00f65215d03407 : SUCCESS in 15m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/24c29f4a30d646aaa8d96c93d348dbe8 : SUCCESS in 37m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e6bf775322df46fb917e6d892ad320bf : SUCCESS in 56m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f1f5cd2bd4c4442bfe88de28634b8ae : SUCCESS in 21m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f57a30fc4f7645c49ba589a9f925224b : SUCCESS in 22m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849961d2fed74513be906d07a3e40319 : SUCCESS in 2m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/681bbddea7ee4003816aeb1037d54afc : SUCCESS in 18m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e115ba466be477996abbeeb6269196c : SUCCESS in 51m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee10d715fe654a44a21c0faebb7477da : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":5},{"id":"8b66979c2fde91de26ff0ce633811ef228046d51","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-12-23 05:53:53.000000000","message":"Patch Set 5: Code-Review-1\n\n(12 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4eef62170af6cb936776da6e2a730d967e924a7f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-23 12:37:13.000000000","message":"Patch Set 5:\n\n(9 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ef6c325c66b7edbba9076e1795edfb2d4969cd39","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-23 18:27:48.000000000","message":"Patch Set 5: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1cc2655a1fbec542fb9a474e1de66cd595b4f507","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-23 19:13:26.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"e1a784da8cb7e1145f4f78f09ef1b9d2e901fc30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 20:12:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0e68bfcd9a46e8b3225c2c398ba3b8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/922dcb9212c44e45b4d9047fd107947e : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":6},{"id":"dd13af27c80cfe1c2740b963c95d4c9cdf5f7de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 20:34:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fca78283caf4490cb9bdc9f117feb894\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad04e61a8d494c68a3f966ca1cae107e : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/d3ab25da70ef4602bca06fd81a40cb8d : SUCCESS in 57m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/96c9fc28ee204a939ad82ecdfc64c308 : SUCCESS in 1h 00m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9ab65ed0f014e7c8e16ceff89b4f6b9 : SUCCESS in 52m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92be9009f78545a9bb11ff36cd297688 : SUCCESS in 1h 06m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31d26fa281654c4991decab891265959 : RETRY_LIMIT in 1m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e676378aebc948ebaeb0160f98cf6f34 : RETRY_LIMIT in 1m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5151690fe539462aafd01fa55900610a : RETRY_LIMIT in 1m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51e7c254ed9a4ac18950cbfa2219fdbc : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f7e44200544747ff941867dfdb27de3d : SUCCESS in 22m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dd62ce1cfc24434f813859bd72c1d35a : SUCCESS in 8m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cedeb85726b84df69de799591797572f : SUCCESS in 21m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b8e8750f8619467d8f3efd8dfa22dc74 : SUCCESS in 21m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b7635e5659534d989452bdf6e19e4efb : SUCCESS in 12m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/193020b43c1c4538af235728ac4da5e6 : SUCCESS in 14m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d1eaeb3e138249068fe306ea9e7cbdcb : SUCCESS in 12m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02beca9eba7e49c0a4057499038fbb97 : RETRY_LIMIT in 1m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd4bb3be87b34d34832de59e0dae87d2 : RETRY_LIMIT in 1m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f90e81a80214183a2a5f2972f4f6aa3 : RETRY_LIMIT in 1m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d983a3253c54b768061f32618e78473 : RETRY_LIMIT in 1m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39ceca6e235441bb93477011d1a23bd3 : SUCCESS in 21m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/053f359f1ba74b3d8dd14fb37bc455f0 : SUCCESS in 35m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b08a463d0ff746b9a1b23d2fc8c89697 : SUCCESS in 20m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86755f6452d14fb097613339ffc59824 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aaf96b4e144e423e85889cf6b720b9f9 : RETRY_LIMIT in 1m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ba50db1eb544ea3a2aa20400c477d50 : SUCCESS in 46m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6aa954476924e14b2350bf40cf0b898 : SUCCESS in 41m 45s","accounts_in_message":[],"_revision_number":6},{"id":"2de8f5af1493ab3f5a4d3a76d8972ebc38393635","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-24 12:51:08.000000000","message":"Patch Set 6: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0b93062356dbb09082f5f7af20a780baf6e7ddb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 13:23:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ee4ecd3e32c4b07a72a64d5c269616d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b69691bc10ee465088980a17b68fac96 : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":6},{"id":"bafe38ffca0f817615035a223de2cd4bbcf0dc14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 14:14:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22cbd262d45a448eb0d5ca56390b1af6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/915fe1e2356d4e9296bf941dc7c36c2b : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/1162f516bcaa436090f07dc239f3c9a1 : SUCCESS in 1h 06m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c31bf12020a64085b8425ececfa21632 : SUCCESS in 1h 03m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/652b5b7d23984bdbb0ba947c76c07801 : SUCCESS in 53m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebd1f89008c44d5f99351993ec2d518c : SUCCESS in 58m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/553c799301ef4b4a9cab1dfa6dc929e9 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/695769238e6040ae818bb9a0d899c8c4 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e7cc3e6009e40849b3f6125bd496ffb : SUCCESS in 18m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/318108adda484cd684c38e008e739a77 : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0457c434e1964f159f9b7eec15be6071 : SUCCESS in 19m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7502be3d1af44d20b3fa6e2319fcdcc6 : SUCCESS in 21m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bdeeff4583fc4a6da8a4db1a7da4861a : SUCCESS in 28m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b20ed7f6df3545eba3e8ba30a0fb08fc : SUCCESS in 30m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f9d9ef9bb70c4125a5fab030d0f91c87 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c6be0d703fc347e1a9eff548b7474f24 : SUCCESS in 14m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/18032a6126594785ab13447f4c4db30d : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1d34d957dc84e74ae47c23fb996317f : SUCCESS in 5m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb5f1d4d6a92479da67357d2cfc74587 : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9af5fde2ac1e4a5c9409f01b9662655a : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/309064e2d7274a5bad5b7100b53ddb2c : SUCCESS in 42m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8cb4667798d340dbb5464098bdc23585 : SUCCESS in 47m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1df0d47c7fd14bba97ac7288a5f201de : SUCCESS in 33m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17bad3edb0dd4497b0515f8000e7420b : SUCCESS in 27m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d48c028e6ae43518a2abe61b7270558 : SUCCESS in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71dc2cc60c5d420e8a311fdb6c017235 : SUCCESS in 18m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c274dfde19943ee99cb133ba733a47c : SUCCESS in 45m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d66de8cccb3f45e399ba15e37755570b : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":6},{"id":"95616f299ae5eac126f60c23c703cce984c69d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-02 15:19:52.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"18c67b062e1d1e02175cb406def9f7f079eca265","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 16:18:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e0959efe0054b48ba0b6e183593b5ff\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/972b4ad5014b4426aca16f4fbdd4c750 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":7},{"id":"7150c72bf15cf6a085aacfa0df17e2c411c0b9da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 16:26:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/258308ab1e4c4190a0c215989136184e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6dacd662c0e43638426a01212052b14 : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/edeb3ed635d944fe8dee5731ba646534 : SUCCESS in 56m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c22dff9b45c144f0bfdbd32444bcdb22 : SUCCESS in 1h 00m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c6e248096d60449e8539fa20a96efb98 : SUCCESS in 56m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/436a46fdc3be4c99a6b06048128c5777 : SUCCESS in 35m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/682f9a0aa3fd4de98e7995a7b4376554 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/310ff9c2db234ddbb6829b29a3a72a28 : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f610745e945d4160b74818d1fba928ba : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a24ea2a94764ad38ca6c6a5aabe3cef : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/212ddeb6195e489b9a88dad0fbf5cb39 : SUCCESS in 21m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/845d55e322d3438bb5952aa0c9eb4f9b : SUCCESS in 22m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9d6cf8980fed4b35898d59d70f163c20 : SUCCESS in 19m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cd2a202f132b45dcaa4d62ad2ed52082 : SUCCESS in 18m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f513ad7eb30f46089d889e862ec7da71 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/58bfa1489af54b84a2fe77586d68c0ee : SUCCESS in 14m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f2fb657d980543588bece18d185796dd : SUCCESS in 16m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/46bd98ea0fa14a888973e098fce08645 : SUCCESS in 6m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c69ca14056aa4d64afe942c2cc9d9358 : SUCCESS in 9m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/aa355be8b5754902911926cf275785f5 : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/81561f6d1559414daec88e7b9a6447ab : SUCCESS in 56m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f208d80767cb404f9f1aeafcb31bd450 : SUCCESS in 41m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14b4daaf423a407b9b319d415b767d37 : SUCCESS in 49m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69b9335d1bf140f6862d659c2e496e53 : SUCCESS in 28m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36512ab54dec4df08f00b6ff99f40f2b : FAILURE in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/679004c764974e69a26e6c67d40f6cf5 : SUCCESS in 18m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e7601ddd3064e1595471b3e6793e4a2 : SUCCESS in 52m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2aa7fe0cff704ef4bb1683f5eb2fbaee : SUCCESS in 47m 27s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/versioned_writes/object_versioning.py","accounts_in_message":[],"_revision_number":7},{"id":"0c57dce5dbdd95d6886f162c5e23a24203ae2b0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-02 18:27:12.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"846b19e83d82b973048dc6fac1cd6d208749572e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 19:00:32.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95ccb457f8674b98b2e6e2dc9a8f952e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/76f4dc5f0d5f4b9ba991d5c6a2c9c510 : FAILURE in 31m 21s","accounts_in_message":[],"_revision_number":8},{"id":"315512192b4cc11c6ad3dfb7f5c310082463101c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 19:36:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f9686caa3104a218564589a36a4606f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d88e4aa88f544a0b6be75ab9bf19830 : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/641e1fdf222f43e4a79bafa488d4bd30 : SUCCESS in 42m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a126cd7ef0ac45fcb0ee8e5d7a3382c0 : SUCCESS in 41m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/70443c5da36e4a0c990ce027a0179dc9 : SUCCESS in 28m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dabb97e02cd54082bdc3329437fe9b56 : SUCCESS in 1h 04m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e44e9fc5a884e41b18b05b9cc98cce0 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e872e25eb344f16a170b22a37cdbfe0 : SUCCESS in 14m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a04823af2034a8799133bd929297515 : SUCCESS in 20m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e103b25ca3c445fb74ac38d16d33d9c : FAILURE in 4m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0ce9df3602144efca1fef4484427b08d : FAILURE in 26m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2859e8f800d6466fa6d35f5b1e25dbc8 : FAILURE in 15m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2abe3528800e4d3096e86f14002496dc : FAILURE in 29m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d18124daf50345a69ce98c41d4d630c8 : FAILURE in 24m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d13ce42e48404110835514d8d1928e6c : SUCCESS in 8m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5b826fa3368946d5ade76fbd7c6f6b28 : SUCCESS in 11m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/df41774f6a4044b6943b68e49595386b : SUCCESS in 15m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1fb53eb99a984b41891d09ee8e41e43d : SUCCESS in 6m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/397026235fef4a5c8ca0fd840109f307 : SUCCESS in 13m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e897f7d84a054c09ba5bae8af136309a : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e17bbacc01dd4171b4a98c0d385d277d : SUCCESS in 55m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9115e5c4fe7a4a0c82f493ac90600b60 : SUCCESS in 49m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e61f1335aa54ed9b8d4aa14a74391cd : SUCCESS in 43m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eba6e1dc9e7d4547bf66200769f12499 : FAILURE in 20m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15724b3ed248492a998b203feaac4dcf : SUCCESS in 2m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5ffbb10c88d435784756aba909498f4 : SUCCESS in 9m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/784e70581ce6459e9a3472ba27ec9d37 : SUCCESS in 34m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db848c146eb5412494f6382c73ed2559 : SUCCESS in 57m 10s","accounts_in_message":[],"_revision_number":8},{"id":"c3cb2f0bd459666610e4daac12b81c232288ca04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-05 10:38:39.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"dc42e5bb0bbedb976e6581f679d97bc1780ce100","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 11:18:05.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0be6af235f541cbb35c592decafaa9d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/72e51828e0224da3a6eb1368123b72f8 : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":9},{"id":"c40335a53f0377495dfe23054a728c262a795409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 12:03:42.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a840179c7944cc4ab6db55c953df0a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc5ea4b726a418b9e81ed13d3525e71 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/f0e1ef1c9812460aa145c7a9818937c9 : SUCCESS in 57m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6101e0bbf8f14a0f9c786a65dcc8eddb : SUCCESS in 1h 02m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ffe026d65d54069a6c8c71d6e4361de : SUCCESS in 55m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e899c9871e5a4c548d8768b69921dfa8 : SUCCESS in 59m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74cadf30bf2744ceaea983dc9a075899 : SUCCESS in 22m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/728d8abc49684ea185a166fb56856f7b : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb798b96a65a414bbc8f2869d992a6e0 : SUCCESS in 11m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b06fa6f692d34d79a84174f1e5bc9efe : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1cda868e6c4645bf93200c73bc5d05d2 : SUCCESS in 21m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/407532b317ac45269828b42d85852f0f : SUCCESS in 24m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6663dbf384ef4b20b87a641149f15965 : SUCCESS in 10m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b963de0d956c4a6c880f0193d0e48740 : SUCCESS in 30m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d6c8e11345c947ae9665a9b9e3109aae : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5b511e4f34904ca5a627647d02bd074e : SUCCESS in 14m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/77c263dab8b14db9beb9582b8817f800 : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7af14d7cceda4240a0f41a1fdf00fc04 : SUCCESS in 6m 13s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/29b349144c7f408db4332297b13561fc : SUCCESS in 13m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c68044bbd95e4f00aa51b9c3d043c3c1 : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d93df931489b4a6fbd989b89d794aedb : SUCCESS in 34m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26068fd0bd024e988c91a15c2809ab78 : SUCCESS in 21m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/442b99e4215649538b43906af0504a8d : SUCCESS in 35m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9af0ad79a4be453988ec7028221ab907 : SUCCESS in 21m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21b13edb4bed453991ec8eac0fd9568e : FAILURE in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a36339e328e4550b4b48a9f6e6f9f1a : SUCCESS in 9m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a08b68a917af4570b54ea352e521aa22 : SUCCESS in 36m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b177ef8cb3d481e9ed723adfabe05e9 : SUCCESS in 30m 22s\n\nWarning:\n  Comments left for invalid file test/unit/common/middleware/test_object_versioning.py","accounts_in_message":[],"_revision_number":9},{"id":"6412e6413b88a14c96459a6c46b2cc12b20df845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-05 16:17:50.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"6c28a2c84dd1153dbb6ee09af766597b07ee7dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 18:35:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/169d7b94e70741a28c1eb9116a93e8b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058fdb2da697491f8f441265c1301828 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/4443a3d69b1d42708623659adf814043 : SUCCESS in 48m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1432875363914462a0975138baa58e63 : SUCCESS in 56m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/560933de7d074456a448cbad6245abb1 : SUCCESS in 53m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/164634767bd44aac86c3ec358bdaade7 : SUCCESS in 1h 04m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e1021a8183d4e5cac2d25fede7527ea : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6004bbe6a764c31b38c6fa0b21f5ca4 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a83f95a964344d0691bae457794ad2ed : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d54c021fb83340089cef9c72a6cf1d4b : SUCCESS in 8m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ffa2a5557c6c49d1a47f634d917e7b83 : SUCCESS in 18m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/31ab9ec252754ab1bd1d5adc16281a4b : SUCCESS in 21m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f940497f9e744d9fb10f5c4751e7f273 : SUCCESS in 30m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/eb3eb9756f484671aacfb5241a53aa57 : SUCCESS in 27m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/75a393b79f954fe682b9a35461b1ae08 : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9040ea2a80434b39a75dd3cc0e97f350 : SUCCESS in 13m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d69b90e464224aa592a908c12d8e37e3 : SUCCESS in 15m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/644f168c36274e7185d881b85f98125b : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c2683493ab614df0bc7001742de42e1f : SUCCESS in 13m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c934e4832f9240c3bc532218c4e46c54 : SUCCESS in 7m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4a3644093384e4fb52af7b0a3ffe539 : SUCCESS in 55m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91a96aa8d5c84bf99d91b684aea739cd : SUCCESS in 49m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cca72eb814de44dbb13454925ffb9a3f : SUCCESS in 49m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0ceb3e4523a4d29b87b0208fcadab6b : SUCCESS in 20m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffa9a8280484a13a0819a9e2da770cc : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b8489c92340f4715987f438ee044280b : SUCCESS in 17m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acfba30624a8476e91d34816cf55500d : SUCCESS in 44m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cdcfa0cfd02420ba90dad080bcf6da7 : SUCCESS in 1h 48m 28s","accounts_in_message":[],"_revision_number":10},{"id":"fce9ae8694bd13fafee3dfb26674275933d4d058","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 19:34:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/860f1ab6e93e4d6cb776ccc7eb001f16\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2028129aab524ae4bf7b642f0f6b3cb9 : SUCCESS in 50m 39s","accounts_in_message":[],"_revision_number":10},{"id":"10ce1e20b5f9cff3991e4c9de856485e7be31a31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-06 17:43:16.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"5ef966832e6a985dddaa5108f07f998c69dbc4a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 17:59:17.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfc5d45525fa460a8ee116e853547be2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/09c1e37ace544101b37fa99dea9dde62 : FAILURE in 8m 08s","accounts_in_message":[],"_revision_number":11},{"id":"6852bc293c8a67c7ce37ab39e3907a859e17c197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-06 18:04:45.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":12},{"id":"530222cb20b97636782c38bb404b09311ac5bd3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 19:01:14.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a07d0dd8ae8f46f2a95b96f0da4b7f17\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/21237364b34740e291af4448c6ea6c74 : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":12},{"id":"252b4dda8f45de50f8017c04789dd8621fb478f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 19:14:52.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42a3826c2d424355ab56b6dfec416bfd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af1398f1d2d644aca2aa92354c841b00 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/104a243c708b4295ad561deedf16707a : SUCCESS in 47m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea05bf11ebcd4682afa5c47a46c6e82e : SUCCESS in 1h 03m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb8a8e23268644ea87d5fee05abc8edc : SUCCESS in 51m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbee30db53f04a989bb733a318a4c257 : SUCCESS in 45m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fa6938979c84d069c2665cda3aba4a3 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dea49d47e1104ff9a4e892987a9430e3 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27f88a05169141b0963541de73871c15 : SUCCESS in 12m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0c7b9587934b4f0cbd1a2173c6120b26 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7d4cb95684fb423da2bb341b32f9567f : SUCCESS in 23m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d8453f2623449da9628ebe2c1c9d3e8 : SUCCESS in 20m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cce2a27560cf49d489a9ae9bfd11b374 : SUCCESS in 20m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c19ca1c3587d467ba500332191866de6 : SUCCESS in 31m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/037538e1c3324f23a43cd1512e02d2f3 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/30a0d8494ae040a294e394330bca887d : SUCCESS in 14m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/325a6d71015c40aaa76c84972aaaa36b : SUCCESS in 8m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4e254312b5a3416da17ddfb1dd468bc7 : SUCCESS in 6m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f0ed63f824e4419b8509a244355fc231 : SUCCESS in 8m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e8b9f1cd8db547599212ff7a6ba5cefa : SUCCESS in 4m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3883db45f7242c6a2761eb108ec267b : SUCCESS in 36m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f016ec2bd6914c52b0f5e04aa6ea2da7 : SUCCESS in 51m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e22bf6685b2c424b831be7d2cd874741 : SUCCESS in 51m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed8d85d37be04e81a900a4c6ce2ec79d : SUCCESS in 25m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a518328c932d4b76802b96f759da532f : SUCCESS in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ad1d643c4d840e38d6fdd6f15f4fc70 : SUCCESS in 19m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/930fee8669e2404d88fa1997f4e0c51e : SUCCESS in 32m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a806f93963ed46b1aa19affc5777a44f : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":12},{"id":"8502212cd4096d03a20ef6846cde2c4fa65976cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 12:12:42.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"2a4088cdb8e7666c3b5f376ec832c7c70ef13915","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 12:59:22.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3663327893d5481a8b29f06e66987f5f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/055c489367e24daeaf18a4326c754a1a : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":13},{"id":"90f9aef6be61367f64dd9f3e524e9d5264dc3155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 13:32:49.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c25d168cd8e845498bddfe1bb599d00f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40808d1638af43fe84445976af37043f : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/40981824d437419295511c17bf53949e : SUCCESS in 57m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aa3f07d2657c4fec8411ab0803e16b82 : SUCCESS in 1h 01m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/16a41d36d18243b1a5b7e9f7184297a5 : SUCCESS in 35m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b19366a552e645c1a4b6bfc9b6f20a6c : SUCCESS in 1h 06m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c11218d9e1284275b8a8f17c2f9750a9 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cda410fbd54a4106a86a751fdf9577a8 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8ad50a962cd481cae45a7476f6a0d23 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b607b91b0634d47972eaea06c8f8a5f : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b3cba0970fd7409aa1a4cd20e27967bf : SUCCESS in 26m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4f9d563b23b044bdace38fb5422a422e : SUCCESS in 21m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f24ec44c703f42b992a421dcf6c951fd : SUCCESS in 17m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9e0ae9003950435e8b2e689468388631 : SUCCESS in 11m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fdcb969c16da418b993cb48120fc8ca1 : SUCCESS in 12m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5fda1bda7e8a4814baff5344fd6aaf55 : SUCCESS in 12m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/781e39f6b8b7496b86d9a6bc4ee43b7e : SUCCESS in 15m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/66901312361047e1b1bdcbccc93be3e1 : SUCCESS in 6m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e30fadd1c6f448ff8cd01e5f234cb085 : SUCCESS in 7m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d239f94b1244daba84c4db848d55f93 : SUCCESS in 4m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/27dac4e1689740378ea9a237bfb548af : SUCCESS in 50m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/59ed4f547f1f4461bd1e3d3d71226069 : SUCCESS in 21m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8001ad57c4ec4bdfafcd2f7e50cf05a5 : SUCCESS in 56m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bcbb5af310d45c6b1c7be724d7158d7 : SUCCESS in 29m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9372b3e274a94d6fbfe8db04c7702167 : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/972870344be047e489b0f13b8a152862 : SUCCESS in 25m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6ff66a2c7094873b7a7a357b740be96 : SUCCESS in 53m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e11b110373324598bc739b1250601256 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":13},{"id":"17a6e402925135e2f70063fe61d9aeab98cc4690","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 18:16:52.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"ca620d33becb9c085e2752d090a5ddd2266b1d2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 19:01:48.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78fe3dacd66d4d30a92a619db640955a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d9e040d48cf04cf997b0750154d9e537 : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":14},{"id":"10c8228ab74fab104256faf64e3867054fac59b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 20:13:11.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b7fa6d558644b8ab33ee586e39b59b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d025234a569406aa714910e5997c316 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/83445179921d4406a8a209e7d51ab3c0 : SUCCESS in 57m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1f204b7b957d486db67decde8101fd41 : SUCCESS in 56m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cae0e2e7e473460ea3fcaa631dea8d80 : SUCCESS in 1h 45m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/721d221574514d1f9b67bd1279635c4a : SUCCESS in 1h 04m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d6b73a509cf4e8087dca39792e591d9 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c53ba7af1c974c5591520a57015098c6 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e2574b3db064a30a87598784092db8a : SUCCESS in 21m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f29ba2144ee402898c0e437c88c20cc : SUCCESS in 8m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9dd3efddc63e4b188f8de38475e18bed : SUCCESS in 24m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d99b44a02474fd1b84abb8fdb765534 : SUCCESS in 20m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2866301b2e274b7584b108b11007af6a : SUCCESS in 35m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c351fdd50c254f708df9e6de64728dd3 : SUCCESS in 29m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4ded9fcefb6c443581518d957d90eba9 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1c1fe6cf46d544038bf2487f1c244385 : SUCCESS in 15m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2b13e29e8ed349ef8a8ab055f0e2536d : SUCCESS in 14m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d810f1585c314db492d58ef3640c7322 : SUCCESS in 7m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a186e2844af4f8da4d62bf34b2ece09 : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a8508393dc634b24a8273e60e1ac0e19 : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/afc43d5ab28a4c4ab736eb8f2bbd6104 : SUCCESS in 1h 00m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3a6bed0b85944947ab93606ee6508b49 : SUCCESS in 1h 00m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30f9aacf9f05425b996316a4c7a703a9 : SUCCESS in 51m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/892c1ff070b84518a1e6ab25c2feb8bc : SUCCESS in 28m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae0e049bd9b41f5a44b7159cafbe5f7 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/af8efca380c04ce9b0b7ab852303ade2 : SUCCESS in 18m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dd6b6b66575422e9e5bcc3846553d65 : SUCCESS in 55m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51188c5e696b4eda8873957a6402bcb0 : SUCCESS in 48m 17s","accounts_in_message":[],"_revision_number":14},{"id":"e01915d170ba9b6200aabd25168dbb0b21441add","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-13 14:18:58.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"eea2a3f347a8bcd35144bf213b0ef37d88d1de1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:20:25.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a780269c27a40ffb8dfa66e37b60e69\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a08fa7d1c1484049a8b1fde54d8e12e7 : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":15},{"id":"55d54194b5bab183b9360e12df3de1274d41a7d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:29:45.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e9830b8a873431aa7a215a6f153442a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3a4c8f14df24da4938e191e0d9056c7 : SUCCESS in 6m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/4514c1a7e81b4e5aaf284b4f689e0fe7 : SUCCESS in 59m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4e3bf5c862d84982bf919f7a259797a7 : SUCCESS in 35m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a0f9ed22fc64dd596a581129d6293b4 : SUCCESS in 37m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53316af8ef3c4f82b9312d74472fbc65 : SUCCESS in 1h 02m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb16cd071eef43c88cf640314f3f2410 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0de147e0336342349803032f251bb9b2 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eacddda09fe44a0888293359cd5b8aa0 : SUCCESS in 13m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/05f9e0d4e2834ccb8fb265b693017680 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aca20ccc6aba4146b16e5d37377d6cd7 : SUCCESS in 13m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fee0a6d35e324e24895328bfccfbe578 : SUCCESS in 22m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e8b8ece300014c9cbee6a065b4b991f7 : SUCCESS in 28m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9747e183d7f04ea587b26b4ce1dc35c9 : SUCCESS in 31m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9a23e501d3484c96bcea6bc069b9997e : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8a3a1453f48d46ad9cda959144572aa6 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/310af8df36b7486ab83ee32dffab91e4 : SUCCESS in 14m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8309f205c40b47548a6ddfaf67b23a39 : SUCCESS in 6m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6a2bc59b6b14b62a2bdcfb32c62a907 : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/69485afc189441a5903fabf820999984 : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8618783b900d44ce8f34b368ef95076f : SUCCESS in 55m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0def5b884fb429393c035dc0619bd9b : SUCCESS in 45m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aa342b2423a14383a8416b49e874c73f : SUCCESS in 43m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62e2fe78402e46de9994a72d75aa3430 : SUCCESS in 27m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e280058bc3884bc1b5cfe8ff61ecfefd : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/079fdb9bb34e443bb64a0bb9005e0c0a : SUCCESS in 18m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e2223b6be784ed59249fb064fb121cc : SUCCESS in 47m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4d4a7c83d674897b10e4a6f607edbb9 : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":15},{"id":"eaa397546dbe266af20cc942f69b06747c7d4c08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-14 14:51:16.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"2b079afaa65e64db1fa1323d527e20cb92c49b8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 16:12:27.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb0f33ad226840e38cbf953ef9547a57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd5f33f646d48baae13cd62d5921198 : SUCCESS in 10m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/5a3bfe41ecfd4001982d6cbaf37b89d1 : SUCCESS in 1h 08m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0df7ddbd4aea415c82070d9cf0a27e0e : SUCCESS in 1h 09m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da2145ebf26d44d88756107ccc1f3c74 : SUCCESS in 39m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a5fce0b9ef64394a131ecbc1b4cf91c : SUCCESS in 58m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/996b258a510e4c48acff1a75027965a2 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1c37cf78a524b41a8bdb0c111b22fc6 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a2b95352e01486584728a964bef3406 : SUCCESS in 20m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47b6b4a7494e4e61a2cc14ffc009c542 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/de8d5314712d45098340e0e851e6df48 : SUCCESS in 11m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/16ad6722671b41abb3910aa6afa8e721 : SUCCESS in 23m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9523f86120074f77a56614b6dd27680e : SUCCESS in 36m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ee0f655d3a424120ac993515db9146cd : SUCCESS in 15m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c54adf93049240a2b8948a220d744304 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0fe1991b659d48e4a90042bda2f7c457 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/426102fa70454b688d140cedb6d805f3 : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ced6279a2a7e44dfb42572caa489ab2c : SUCCESS in 4m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bc329a34dd6a45d8ad9ec15b5739eb42 : SUCCESS in 8m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a787c34f0bd4ce4b5bb48cae422c8bb : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3b273168a294e809e7aa0a30dae6fc8 : SUCCESS in 56m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de8b013df65b4c65851caa551fd27aca : SUCCESS in 29m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8bdb84e17115466484932a8fc5e3b197 : SUCCESS in 28m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1caed430dfc477e92e7e64a43283fbb : SUCCESS in 28m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d644213d941f4613864221776d8d6210 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f562391157e4ced99c784bd38d187f7 : SUCCESS in 9m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4f35a4a1c1d4e77b3eec6ba830f4c5e : SUCCESS in 51m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68b57fc1d08b4ba5a6ab28a09f90b053 : SUCCESS in 31m 16s","accounts_in_message":[],"_revision_number":16},{"id":"058ba5919c3556d320c03a655c91d40eda9aff3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 16:26:26.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb92a66608dc4e1399f29527cf9b128b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c8e7334b6a834eb3a4a8c7dfa71276ca : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":16},{"id":"07ae394940ce3fec6d3c0d28a8d540ef805a9b03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-16 17:43:41.000000000","message":"Uploaded patch set 17.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"4e1baa2c0a3bc74b12c28812f0c61fe43a2a279e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 18:36:39.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83bb6c17558b407e975ad24520039bfe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0ddc10db040c4968b264489a0c547e2f : SUCCESS in 29m 32s","accounts_in_message":[],"_revision_number":17},{"id":"ef19520bbdc1ddcec32c02341343b5a15592452b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 18:58:32.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/628500f30c11498e8b6abf41b26ef219\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05419ff6741840ec8b405ae51616357a : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/4dc75bb6848a4db99a9ac81be28fe5ba : SUCCESS in 55m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ce52da783f844491a9bf70384de8f4e9 : SUCCESS in 1h 02m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c2a55a00e07c43d091867e61b4a745f4 : SUCCESS in 47m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/deb7d075e502424285ec5cdddd358c2b : SUCCESS in 1h 00m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f7b5eabb716495b84ca0ea327d35959 : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81240d3f12f14c0ea9c29c161e86c7ef : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae887feef7184a8f83b143d8a6b980a4 : SUCCESS in 21m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c1eca689c17c4cbfa4d8dbd1dcd56f85 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a159e1f4f66e41b79db3f7851ad2de8f : SUCCESS in 25m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8c07bed392fc434fb34cdfdbfeeb4cd4 : SUCCESS in 21m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a2daa536b3c742aaa5d141bf6241e561 : SUCCESS in 29m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a316fe6835084903800e3fc83df7bd97 : SUCCESS in 14m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5d9018f9551247aa85f5e27a77735b6b : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/88be13f998c64b59b4e8894d8c19aa9f : SUCCESS in 15m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3e917c95acf1450cb0ac57969c883065 : SUCCESS in 10m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5f3ef49cc4044ba8464e1cb0e884bae : SUCCESS in 6m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea0a748b05054811ae757b90fda8d39a : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/434c742de3564d27908c129c8a0d30fe : SUCCESS in 5m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a8e8749d62944409df438b01bbbf30c : SUCCESS in 56m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc928e0824c241d7984c06530fb4b931 : SUCCESS in 24m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e942b7569a7e4894ae72770029ee2fd8 : SUCCESS in 51m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a8d5cdaa63a4a27989b2f7bf619fa68 : SUCCESS in 25m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f80b54e943d3474cb055d9baff9b194f : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4fb34dd48858470cb810351ffdc70e7d : SUCCESS in 19m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9c9eedd9ea14d3ba8246c95c5c530ad : SUCCESS in 52m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e580c61c70b34af7bcd444eda672ebfb : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":17},{"id":"42108f2152615a1e76b4237aafbd87299f5dfbe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 18:25:13.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"6796ae9936241e910dec19b6795ad5a56d1c547d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 19:51:59.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27cbd7080bb747238bae0dba18072d4d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a4ad7f739d454dabb04d02a2a22b41 : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/486893f7acf8418b97a4d07353302f3e : SUCCESS in 1h 01m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bfe636194386481b8fe3b9a5117e18a7 : SUCCESS in 58m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db4473a1938743faa0d7f69d5b62680c : SUCCESS in 49m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6269040c25c443d3bf0f1ddbc36dbcdf : SUCCESS in 1h 05m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/751f13fcd6ff40c4b71141abfaf31102 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93f8db2383b14dd486889ca2400a682b : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de6b1ee3c9fe4597b8e1e6440bc1e197 : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5bb5a7fbc4134c86b8fffd345049f2b6 : SUCCESS in 6m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aa9e05655d4a4254a754130dac4638e1 : FAILURE in 25m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/491dd5c25d7e48ac8ed97779cd50a4f7 : FAILURE in 20m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7d8ba51ec4064c6782f50cd9739d8ccb : FAILURE in 21m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/62827982c98446878f60b70251942380 : FAILURE in 31m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a78f331aad6b4d309b2fcd91c7a57566 : SUCCESS in 13m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/da649c9ab6b14f10b3ec5ad490b5f450 : SUCCESS in 9m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c32eb36acc1c4d90a6a656a50a077284 : SUCCESS in 9m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e31c538d959f45f7be7f9a82ba4a2fd4 : SUCCESS in 6m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/819175200fb547f7999d9d2b1c6f43a3 : SUCCESS in 14m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0b43c29e7c564be98948d88bd9e23f48 : SUCCESS in 8m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8fe6b3b7a97466a8e3f8cd69935d6de : SUCCESS in 39m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2e126e2c71548589a63ec0a8542d0cd : SUCCESS in 23m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c396a0c060c42aa9af29ffe1d52a970 : SUCCESS in 51m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a02979f47d34b5ba6f3b972b36ebc6a : FAILURE in 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/280012b86cb54a2da29f8dff6f25f8b0 : FAILURE in 5m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/052b3f2307d24ed385ac23851e73d265 : SUCCESS in 17m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c1ad5997fb9415782d14e7332e8b1c3 : SUCCESS in 33m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5191d6beb88c4f579848bc4ce583b052 : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":18},{"id":"0a11bbc70757dceacfe407ab4cea63929b5cba37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 21:17:01.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52b420726b40441a977165bd3ed8bdb8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e55db7f761f146b681bd6f9ce14245ee : FAILURE in 58m 38s","accounts_in_message":[],"_revision_number":18},{"id":"224840c8f0ff4c209da7d4fe1f322a2b3c6deb40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-20 17:05:12.000000000","message":"Uploaded patch set 19.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"afcbb28b977e6a91621e496ac2fd0432a8fe326d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:07:07.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5b23e94844c4e02bd35c4c1390fe91f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/808f0b1deded4c3facd7bcb2558cde2c : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":19},{"id":"f01a9ad5294cf70371a07b8c531d51be6e0f0fbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:24:33.000000000","message":"Patch Set 19: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68fddb4761434004a2291747ff93c0b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/231c6abafa8f4a37b60baaa66d4ddcd7 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/c4900c9e9fd946ff8f9e8df17aced635 : SUCCESS in 1h 04m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92449671e3da4d969fa3241f19745b0b : SUCCESS in 1h 05m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/60c7344e64be488f81a4198d1a5a90b4 : SUCCESS in 30m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52c77944a4484a8ead2af2c0b4d00d66 : FAILURE in 51m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19f11404b3944e17944739a8ba5f659b : SUCCESS in 12m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbad91c7c66e4fa88aa235057339f030 : SUCCESS in 12m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1477517eee147648916a891b135b011 : SUCCESS in 21m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b944884227a04bc79be9b006b8333e99 : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2b321c50bac44f6094539a47ccdb94bc : SUCCESS in 24m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/25b6a8ef98714deda6163684a5dd3a42 : SUCCESS in 8m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/95316f728f534059bf2d0ffea07f840b : SUCCESS in 17m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5b460eb4804640b2803eb4e795063306 : SUCCESS in 30m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ce7af707a84d4ceb8e29865fd19e2ca5 : SUCCESS in 14m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/84091943559b4283a81659bd3a5d2515 : SUCCESS in 14m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cd1f6affa11241e58a7520988c65e2ea : SUCCESS in 10m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9b95439fe18c487fb5e5598d3b9c8ab6 : SUCCESS in 8m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52b3c0f474f1485e99d6fac4bf38c105 : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/44dc55edcecf4228ab33229abe7db2f0 : SUCCESS in 8m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/49a60d6892164ac5873f5057f9e47dec : SUCCESS in 55m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc8d135fb57c4f378770067c339a0f2b : SUCCESS in 47m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/403cc756a08e4e7a9b74f09c6a2dd81e : SUCCESS in 54m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bec4b44a5c0c41c4a81d7c7608fccbfe : SUCCESS in 29m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75f4181dc71948c3adcf63b95deb5418 : FAILURE in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f0f8b4582ea49db99a94530f2e63490 : SUCCESS in 16m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37ba9f07ccc642f7a65b8a3a18453023 : SUCCESS in 46m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45ca36af42744dddb59ef038458bda97 : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":19},{"id":"b2d7e8bedc7c4d144be487e262e4ad142e8841cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-21 10:11:20.000000000","message":"Uploaded patch set 20.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"091129e87b5078e230962622a3a9a47e419431df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:21:22.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66705124210049afb8856756eec782c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/721ac8507c72459b8160c15e2f20c4ac : SUCCESS in 6m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/c49313def4744a3f8b9bc1c985ff388d : SUCCESS in 36m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2c62cd48380940628d02e7a5663bd467 : SUCCESS in 34m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20c3dd24e8a643a886c057f27f029649 : SUCCESS in 35m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/570ef70e960e4a12b730ebedcda439a2 : SUCCESS in 37m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d77b695d19484bb29de6d43ed9b42288 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a56b73181804e7cb85ce0b9f97d4ca8 : SUCCESS in 13m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6247a28485a479e94049b0bd8d4b4c4 : SUCCESS in 11m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/159a5ce101e64b858cdefada230eddde : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1edcd38869e34dab944ff648d0ada092 : SUCCESS in 13m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a0bbc501ad6e466c97d5a1faa45166d9 : SUCCESS in 11m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0bd5acf0494243e1a3bd84c739e5e727 : SUCCESS in 26m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1ecafd6d36fd4e6e84d0bd7ef6a969ac : SUCCESS in 22m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0b44a9151f0240db88635e2212494641 : SUCCESS in 9m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b7ee8f92393e4010b2b6ed99e157bb3e : SUCCESS in 10m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2479da9559024282b55983ab303406ff : SUCCESS in 15m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/080e9e2912e94724ac55e4205c3997f2 : SUCCESS in 7m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/190206f1925b4c258fa656ea7455c1e0 : SUCCESS in 15m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0188a3445284e20b034d3eab897c697 : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddd3b8cd04c140dcb8c05ac0b82e540d : SUCCESS in 1h 00m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3229f54d70e74691973dd7c1acdcb894 : SUCCESS in 22m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d0483fbd63f4c8e85334b9c648f8a2b : SUCCESS in 23m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/379453e82ec943ca9fd335761628367d : SUCCESS in 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/577ac5c891c44f1d999edaec54e1d23c : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac3d51f08e2f4056b7b16ee186589115 : SUCCESS in 12m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ceae8b2f3cac42f08f3aa140c259cb7f : SUCCESS in 35m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/651b4a0983de4ef6a16c7058d0feaa4c : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":20},{"id":"766b144590c001a79570e0e8c3f3b839069c2f33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:38:42.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b965650fa9d14759bc2fa7c786dbc61a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/75a250c519dd4999a94e37f3c1960c77 : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":20},{"id":"47b1f03b8a4d78a9de3d2523bd109cd3bc1e9c4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-29 03:56:58.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"052089d37e802f192942d16b6d9d753a8ca4cfa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 05:12:51.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c84b0908827443cbb161f7c1f5a5bc06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e811fe3cc5451eaf19959d4b296689 : SUCCESS in 9m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/457079d857e543baa2fb4f167fc9e7d8 : SUCCESS in 1h 06m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cb5a3b5afd2f4c9ab8a34cb8410e0291 : SUCCESS in 1h 03m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8e167bfc21a64596888840ec93d89415 : SUCCESS in 49m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7181762ae32741d5b27c8cbf17d0cbd4 : SUCCESS in 48m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/364cad7e98ff41fa90af6ba9c8c55c30 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74d55fe6af064398be25e6922953266a : SUCCESS in 13m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16390806ad8c41f1bbb3a00146ef8a31 : SUCCESS in 19m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d926ab14f2854e6695d3cace5792bc41 : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cb03df242a414f408ff9c4122e8beee0 : SUCCESS in 22m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ae8ac455882449b68ef355824a8fb565 : SUCCESS in 16m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/697c0933982b4144af040ccc54aa8b30 : SUCCESS in 35m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ea8434cc3cfa499aa1edc45802bfc891 : SUCCESS in 18m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/17aca007fdc74d0ba1aa22355fd54581 : SUCCESS in 10m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7737efdd472549e2831e903dff42b9f2 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6d0e85b490134e1995a8d2682b379d50 : SUCCESS in 19m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/990bf4a1c404430d9f23fae869c5fb7a : SUCCESS in 6m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff79721c00df439bba8dff409fc1646d : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/30dab254b6b040f7aefdce728eac94f6 : SUCCESS in 4m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a81f47591f3e4da2983c90be5aa47236 : SUCCESS in 1h 02m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/572c5c1d117b4843877edde9730e2cb2 : SUCCESS in 54m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33667ab5b7934b9390c7b963494e53ec : SUCCESS in 47m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6762207fe766484eae300a7bd4b64e04 : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eea01a246ef4c0dbdc86872aa3345fd : SUCCESS in 2m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65d88dcd61e5460f9e64cda0747c3731 : SUCCESS in 17m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc2dbb4fcd0e4cda8c5a713b39dd7411 : SUCCESS in 52m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f19e8ef30769410098e7cff74a4954d1 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":21},{"id":"c7eae459f3a726e127925d1c1455dfaec95cb072","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 15:55:46.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b981f550e3b4220841233f626752fb6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d3d816024b444d28a56f98ebfb53218 : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":21},{"id":"a5472c6890c1c279417ce2e40bdc04e9286335cc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-02 05:51:04.000000000","message":"Patch Set 21:\n\n(4 comments)","accounts_in_message":[],"_revision_number":21},{"id":"b1f6423916fc6c6b1e98409ba3263f36e32e56fe","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-03 15:28:48.000000000","message":"Patch Set 21: Code-Review-2\n\n(6 comments)","accounts_in_message":[],"_revision_number":21},{"id":"6494f891b0773d6b5de32c416a959e85295a730a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-10 18:55:59.000000000","message":"Uploaded patch set 22.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"f0d66b7fbf004e2bd5f07816ce4efd024a7f9fe9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 19:36:41.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8534d6a338c44f48cba8166c26a1349\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d5201f84821245a283ab6939df7fc3e4 : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":22},{"id":"7c9bc5d883a844c35a5de6cfab49d7d8406cb698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 19:57:33.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26c5162910444b8abbab0ddbc537df69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c1f08bebdcb49b39a2762d4596baf65 : SUCCESS in 5m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/affaa2123f2547fa9f66bdc7bedd775f : FAILURE in 5m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8e3b13b3800e4b7180d16a9b3a3e1f97 : FAILURE in 10m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/652dfbbf84ee472c991a07ed857c6686 : SUCCESS in 50m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/529115e1b5c14a79969cb018f67b19ce : FAILURE in 36m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5943b19f64614934b6f5d4d0403d3312 : SUCCESS in 12m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d3b36f7f57041a4a4efc5b02604c331 : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77db1b9af9374e32a76a3252e7ab72e9 : SUCCESS in 16m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2dea0415a7dd4d2c92a0fad305f60726 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/13f68437314b4809a508d361b24d1d6a : SUCCESS in 11m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9d085867186947d3bb5f4e644958e46f : SUCCESS in 17m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/851fd0e8edee4112b3dba161c6b24bf5 : SUCCESS in 28m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/aecb97a5f43b436885932bdadf7c20d7 : SUCCESS in 29m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/067cdf64ad1149fc8d970aeb384ce3d2 : SUCCESS in 11m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/432874da270f4dce9b710ca1753484f3 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4bcbe740e28c41489bb3dd9c81656aa3 : SUCCESS in 12m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03d3c7de8514425ba703de6682e80b3f : SUCCESS in 4m 16s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bdda29b50d994bddb4ec023aef7cca2e : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/76346a162155471fbd34d2e4abe9d555 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0505bc19f1794a05afa7ff8372432c4d : SUCCESS in 46m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc3057c92585418eb776797fddb668e7 : SUCCESS in 56m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e863bea521c4fb08b9c3dcf922bb001 : SUCCESS in 25m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dd72fc23a994d609e784b48ec209970 : SUCCESS in 27m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c1abdf4ad646759decb4c5afe4dfea : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80d2380166b04bb4a0662ed36190f897 : SUCCESS in 18m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97039c17e7a74fbbae7f0de913a466b7 : FAILURE in 33m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a45d7fe4f2a4571919f3fb8a76273c1 : SUCCESS in 29m 19s","accounts_in_message":[],"_revision_number":22},{"id":"c1d043c54f10b7bfa388406f78f6c18ae03c92c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-11 19:07:43.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"001328265246633678cf12a08ba4c542aaaba981","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 19:48:52.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aaad179b6a2410badf56165d5d2d25d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7191a550dcc947ba954068a0612f4ab6 : SUCCESS in 39m 19s","accounts_in_message":[],"_revision_number":23},{"id":"692756338f42069335b95550ae882b0f5f8258cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 20:15:57.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98665f5aaf2e46b68fa368e8535834ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77a9fc77c4dd4eb7b5b33dfd52ad58ae : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/501eaa98adbd4533bde94d335b09c445 : SUCCESS in 34m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/acdf036902af42dbb8c90cf77730b6ee : SUCCESS in 44m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/76af24ddffe34347b1ef68919be7dec5 : SUCCESS in 43m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb9d8f06d1444ac2b29d99d61ff71494 : SUCCESS in 38m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cbbed505389430587be848e400b3f29 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b7f4f0b8876488aac29b2345d58bc88 : SUCCESS in 21m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/283f8216ee2b47e5867126529aa5f4a7 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6da7b8eb8d4e4a2fa926c37fdd9bd387 : SUCCESS in 10m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ea56315c6ee84f3d97286e589999fa14 : SUCCESS in 17m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f0cfa50bcba7472ba227afe155179539 : SUCCESS in 21m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a51dcf3164f64624a182d38c05da5e36 : SUCCESS in 37m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8f3e0913add94e84a45ea09f5851da01 : SUCCESS in 11m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9b8fcb9cf36a4c19a5706543780ccb6d : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a1924ed4095843c6a984fcdf8b104b4f : SUCCESS in 14m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c54a33d3aabe4505849d425ee602173a : SUCCESS in 15m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17caa0d008eb4415b87160b8dfabf44f : SUCCESS in 6m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d4547811f2e547ab9ce607499a60f5bd : SUCCESS in 16m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/075588f07f864eceb420fc4d2335d4bc : SUCCESS in 9m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7c5894ed5db4446b007d6a208a82ebe : SUCCESS in 1h 01m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/caf17d9790da48c0881746ac896a1a30 : SUCCESS in 34m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/16194efaf87e4df7bfcf45ca63a5d0da : SUCCESS in 21m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c47645c706340918bef5f547f049abf : SUCCESS in 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc8d1f1bdeae44ed9ddf3a7f6392dfff : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7fbe2030fb142fc90ad2612d99ee6c7 : FAILURE in 33m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38c84fb27ce940d3857051d30d23146e : FAILURE in 52m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ffc2603b9614cbe9fe881f251743038 : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":23},{"id":"98ce7480b09a037d071bc3bc0bf769af5e6dfce3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-12 19:09:11.000000000","message":"Uploaded patch set 24.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"3f523d00a2f88a1d32fbc29aa90351bd61761c66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 20:35:40.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5db0bf703ccc4bb08b95169977938521\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50d85e78587a49fcbf4ea0fc3d7a7963 : SUCCESS in 5m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/ef31796faf01439eb83fc639e67f312e : SUCCESS in 35m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cb97545b7abc4ced8ea92fa31ff32acf : SUCCESS in 29m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e6aa3f5ef65e4019b285bd871ae5617b : SUCCESS in 31m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df1b4e653e0a4621bf09f422aa71a9ea : SUCCESS in 32m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a80d690268e49038564fb2a3d185225 : SUCCESS in 11m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ddeb128da5c49a28ac44f4176c342e7 : SUCCESS in 11m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08fc00b99a9542359c9ea56c34603bd3 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e220e57141994967bb33f507dee9b938 : SUCCESS in 8m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/adb0dcfef1f3408388bdbec87e6f6e34 : SUCCESS in 11m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b23d3d71578c4c8eb67867469955973b : SUCCESS in 22m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c188ccbf6e29477087ff5df25ec36aef : SUCCESS in 29m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3b208789c3514cc5836615d02edf9fc3 : SUCCESS in 30m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/31b38faa8ca24ccf95703a6155828502 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d97c86773f6048cdb2613716a7d1ad79 : SUCCESS in 15m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0c9e155ae6a044cbb25d6294fda1f2f3 : SUCCESS in 14m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0e9aa32505f34b869ca21e3c543dff26 : SUCCESS in 8m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/536bc543478b4c058c862357e028dcd2 : SUCCESS in 16m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2874643af70448efbd98379e295123ef : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/759a4a7533bf411187e00aa1679e9c60 : SUCCESS in 56m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e770f08ef8f24010b8e73190e56aa8fc : SUCCESS in 46m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c26acf28d2ba40319cd2da92850393c3 : SUCCESS in 22m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9af75e62f0504847aae1449d7ad0b807 : SUCCESS in 22m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d9e9a31f6649a28d88919223aac90c : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e95c3df277ed4d4d8b3b61263c3d1a48 : SUCCESS in 10m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b07f4d805b7d4631b1fd24395ca26903 : FAILURE in 35m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f894f43887347d58a1a6f8a4991aa29 : SUCCESS in 30m 06s","accounts_in_message":[],"_revision_number":24},{"id":"9aac530671850f1d76e257aedc7d6e374f43ec1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 00:42:32.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aa56225a5304008860ef86974cf48f8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67d9baadb9eb44c6b6a1baaa71da4ef6 : SUCCESS in 44m 03s","accounts_in_message":[],"_revision_number":24},{"id":"572d340c845c40e8642037d5f840dce46721d3a2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-13 06:56:37.000000000","message":"Patch Set 24:\n\n(6 comments)","accounts_in_message":[],"_revision_number":24},{"id":"2b502d8617e24424a4ac2bb164dce3e4cc7f1625","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-13 10:40:39.000000000","message":"Patch Set 24: Code-Review-2\n\n(5 comments)","accounts_in_message":[],"_revision_number":24},{"id":"be073ffd3e17252bbc0d051fb396ab09490f5df1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-13 15:48:12.000000000","message":"Uploaded patch set 25.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"8c52914928ded2e95cfa5d02408405cd92995b04","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-13 15:48:34.000000000","message":"Patch Set 24: Code-Review-2\n\n(5 comments)","accounts_in_message":[],"_revision_number":24},{"id":"1a2cf9e227b16718d81af5a3aa04c77754652c0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 16:24:28.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0710a0f20fab4500b8209c3d5a1233d6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/153b1e16343a420393dd869c6c28dfb5 : SUCCESS in 34m 10s","accounts_in_message":[],"_revision_number":25},{"id":"feec348c6be889ea34d612fbf0e2ae3ab8a9a8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 17:01:28.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aaf7923a571d4641b33cb50a4bbf1f67\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9acfc17fdc4c4d099d6534dbe70a4c4d : SUCCESS in 4m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/929259bf40d4468ca648ae10a21dcd19 : SUCCESS in 57m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4eb519a2c902425fa3c346648a24ed77 : SUCCESS in 28m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0fc50a512f664a798812661fd2cf73de : SUCCESS in 57m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f049b7a6b18245c1a3b3baf0aa6e51d4 : SUCCESS in 1h 03m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa4e0f4a87ab4b128d0f90633e447b38 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25db7c607cba4a39badcd99ba2a23838 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/30f6ef40cb9e4847a940d9b3845b607c : SUCCESS in 18m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9f69629d0fda4d429c66276b46e724d1 : SUCCESS in 7m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a82707ab17464d13ab59798afd613e49 : SUCCESS in 19m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/350619833b3c4f22bc66cea31e1bc52c : SUCCESS in 20m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/36de2e34888245deab56ff0b366760ed : SUCCESS in 11m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ce408c3754af4462afe1240dc33b2376 : SUCCESS in 26m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/47f3ecc3889a4882b6c413a40840ae43 : SUCCESS in 7m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0899cffe3b7842be83a17645f1461f5e : SUCCESS in 12m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b1909d75838e4183a9d1ee165f4318b7 : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bc78632eac24474cb37a1683ad6a2fbb : SUCCESS in 3m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ffeb27e38f2a4754bb71964589de9eb9 : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d9cb94315b64f13ab88ef9be39e49eb : SUCCESS in 9m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce9a8fe22d41455192c9b45f1c211fb2 : SUCCESS in 33m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df8ad562b744454b9571c296e1d86bd1 : SUCCESS in 56m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6e05f20bfe014ba198813eb5b250e668 : SUCCESS in 51m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/292396e0dfdf43c99a354e8169014a78 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40709f6afdcb4f5baf3fc5875fe6b059 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd7c25b067c9441b90ee2387ae3e1178 : SUCCESS in 18m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d2657e978874c4994de4fbc3390812e : FAILURE in 42m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8042732c1154e0fac64f806aa3080b1 : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":25},{"id":"f029ce90ccdc9cd05885ce5aeddd6afce797b106","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-16 17:42:54.000000000","message":"Uploaded patch set 26.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"8805b808ab19c3a3f24cadeab23f1c496a46434c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 18:43:42.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/619b45734cf8466ba924900de9e140a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5d971b690c943848c5d86b5bc9dfbc6 : SUCCESS in 10m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/a627f27cb6fe4d31b563d8c239ea17ab : SUCCESS in 33m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd1907db40154202acf12b670d875f5e : SUCCESS in 34m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b21c5d0b5a224692b4bad06d631c209f : SUCCESS in 52m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cc0dd2680454c61924b0e3f1ed50689 : SUCCESS in 36m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1271ec88bd81462cb2d8ba28541a4c7d : SUCCESS in 11m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84259ae5c0ca4a37aa550209877eab75 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b12629a0ecee48f4ad701efea9437b29 : SUCCESS in 21m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/715bdbb5b51b47a482eabf3a577ef566 : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/272d3b97739c4f44aee41181a977a0df : SUCCESS in 19m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a718f791df1647309eb174b3027ef1c4 : SUCCESS in 19m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7ba2484ca86144b4880e608b70505bbf : SUCCESS in 11m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f5243afed79c4caa8d6338ab5d2f6f6f : SUCCESS in 24m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2087a3bb9a0642a393e0caf85511e88a : SUCCESS in 8m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/faaf9900075d4f2cb75aa44d68ea175d : SUCCESS in 15m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/80a74e28129a46d18cada4740eb02dc7 : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/32d0a2a1c67f45638dfae99b71dea6b6 : SUCCESS in 8m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c15a38cdadb345ab925d1ecb5a8608dc : SUCCESS in 15m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/efd99c6446cb4c64928323ded6a2bed3 : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4edcfec32b54e6fb7ca28efccbae2b7 : SUCCESS in 54m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a42659183c84597b22080d0f922194e : SUCCESS in 35m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71e4e6ba2e1c44b1bc86d8cf5aa99c5f : SUCCESS in 38m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9df50de07cf343978257aee596c8bdb4 : SUCCESS in 22m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f6dd091db3448128bd9724fc7061cd7 : SUCCESS in 2m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f1e972edc924410800520b64eb8162e : SUCCESS in 16m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49c219fc6f324113a0663d57be812429 : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":26},{"id":"b1852a7a4ac9a63a02948d62269b2abaf148bcde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 20:40:01.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13445c9d27ce4110982a7aa9cedf60f7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f077e0681c4140878b27b5ae880504e4 : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":26},{"id":"f2713af9e0b3d6bfd3684c2a296791b75d63ddb8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-18 12:04:09.000000000","message":"Patch Set 26: Code-Review-2\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"07a1c2c68055ebd7dcec0c4449bce9763453a723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-18 17:13:51.000000000","message":"Uploaded patch set 27.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"b3187ca798d6929231bc7994067d70b1b02ffe3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 18:25:56.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06841d337a604bf2be49267ecaa02d90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14027aa60a814379817635b09c09839c : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/bb0123722f4949d595952074d39d20b5 : SUCCESS in 1h 01m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2fee6d9af9ee45cabdeacb7bdf89d208 : SUCCESS in 1h 01m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d47ddfa9b5614af4b02b3491f4a443ff : SUCCESS in 52m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cae775d591b447e8900aecb7ac020c53 : SUCCESS in 36m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21e8f1ec19fa4e308b0a8566f2961d97 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/200b8fcc294f47a9b23bcc25a9cf8122 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c190f65a1f242cf8ff9c112470de6b3 : SUCCESS in 11m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/328de57173e14a3da9f52f1f1b1693df : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8082a5ce455145a783cb2ef1a52a87c6 : SUCCESS in 24m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ef70169e7a14d56aa5f4c998541e7aa : SUCCESS in 24m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1eebe6eaea544580b980e167863cf55f : SUCCESS in 20m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ffd7c67ac462493fb080b9e090391fa9 : SUCCESS in 11m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3e312084af414ec2ba5a1edd54e7911f : SUCCESS in 12m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a50939d0c69b4706ad5599d3099ef8f6 : SUCCESS in 13m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0f2a0e9842a641ca896530e9ea0de5fb : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/87d1a64a33de4dcbbb0d3d897d026d7f : SUCCESS in 6m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a806b22dcfbc4b8ba503bb28562cbe97 : SUCCESS in 8m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fb3f5b60f0f545aabfcac1105e6f7ad2 : SUCCESS in 8m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e33f08c4e1ac4ba0a160b8a3909a6473 : SUCCESS in 58m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a88537214a940a48dc7d58908372bb2 : SUCCESS in 58m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/421041b67c5f4944bdf29f9527b41bd4 : SUCCESS in 23m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf2b28d2a1db47798e1d1a92ac6b7b42 : SUCCESS in 31m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fcd06765a424225bacedd47bbe1e8cf : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/95593b33c4364ea59657e8a3c6ed5394 : SUCCESS in 27m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35cc44d411564eb6b32fcf1bad831f37 : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":27},{"id":"eff789ac42fc4a016040cc3ce803f6dc53f28fab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 22:45:39.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d6dac650ff24424a4683d27dad04277\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e1ae061a1ef4173a89462794614032f : SUCCESS in 47m 44s","accounts_in_message":[],"_revision_number":27},{"id":"8b74aaadcdf7621f0196b2e88621eb027e29f980","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-19 06:22:12.000000000","message":"Patch Set 27: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":27},{"id":"6a559e4bdf8bd0be99295b2961c9af8f1e02c13c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-19 13:13:14.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"6316ff266ca82d5a72425c79c698e5be972ea0a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 14:06:01.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/441cd1835f584f96b0a9c7f1672e048f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a108498c518485c86821d2d357105e5 : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":28},{"id":"af2a1bee6eeb6777895e2958587b3e85be8dac77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 14:34:56.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b91d32755dda4f20a02945551d5d8647\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70864c6685f64d99b9e8f95bddca7481 : SUCCESS in 5m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/9d30294a3b154fab93b96d8cc0026c58 : POST_FAILURE in 1h 09m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9bda0d83d865462d851de1c3f728d30f : POST_FAILURE in 36m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f61182e4b9e24ef489f6a24fb9b4d05d : SUCCESS in 35m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5311dad865e64c40bdd35270bec70cdd : SUCCESS in 50m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f269dd228f3486ba8b53bc4924f3a88 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2db7cbeba6a6439281d87a501a2deb1b : SUCCESS in 12m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59ef0cf4977d41d98a7181b0a2c72504 : SUCCESS in 16m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8961b4cc064a4be5a3d14bd6e7afef69 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fcf1e11fcaf64af5bfba045f81102ee5 : SUCCESS in 11m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32a82b2e224c4b6286ace15fe9745e60 : SUCCESS in 20m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3347882225ba4d738f1122317c745a4e : SUCCESS in 14m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/24026a79ed8e49cfb277fc273931aae1 : SUCCESS in 13m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4390984c2ca140b395fe9e2ddbdb4f54 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6a855c43b11b49688f028676af7fa17c : SUCCESS in 15m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/11c1736d839449369c4a2099902ced74 : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d7af13899afd4eceaacb382e88f94fc0 : SUCCESS in 4m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d0a1e8a2acfe44b7b29bf9bea28c8bf7 : SUCCESS in 13m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/66f31ef01d6e44ccaed8777ae35c5986 : SUCCESS in 8m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/36d44b5f2eea490c98fffbac0c626d4d : SUCCESS in 36m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5d8fa95d3f3424fb9cb78d1c5ce0553 : SUCCESS in 28m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b09226371e2840f08c2aca2e7d99eb8a : SUCCESS in 55m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e279bea5c0247bfa43f595d29a105b6 : SUCCESS in 26m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa5c2f5bc5e34ef58a217ff9a8b4fab3 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f567014153924a2fa97a8187c1ff1967 : SUCCESS in 18m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38137164e04147fc8bc7f1d7b62f1d20 : SUCCESS in 44m 52s","accounts_in_message":[],"_revision_number":28},{"id":"1202426517992e300293be3a5a391e866c1cb7ff","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-19 14:46:16.000000000","message":"Patch Set 28: Code-Review-2\n\n(11 comments)","accounts_in_message":[],"_revision_number":28},{"id":"1d0136ce850a32dec8dffef9b61d9c1a6858cbe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-19 15:35:08.000000000","message":"Uploaded patch set 29.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"234facf6b72c6cd3275ee9cf088ea75ef9c152f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 16:14:11.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d1e13c122204c5b9e9611a25c51cd89\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e59761640c5244099079913004116caa : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":29},{"id":"26adfdfc0c5abf1eb2a6a930bcce35f8f010c220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 16:50:21.000000000","message":"Patch Set 29: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a070af391484805a4f34a3fb2c389c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f84704bf70f4cbe9f1fd75756770f10 : SUCCESS in 5m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/962c71d673014b9eb0a54b3c8ab310d7 : SUCCESS in 1h 03m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d62974cdc81440ecacfc4eccff96854b : SUCCESS in 30m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f734115b526f4095abea6beaa0b02c32 : SUCCESS in 53m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12485258568943839064091797cd2c17 : SUCCESS in 56m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73c4b38893c24de0bdc9cc6f464d500c : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/689647d1f5864c479d88355ebde45c17 : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5faf463e709e4dcf87ec3dd3f69d6362 : SUCCESS in 10m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a79aa54272c84535bfca2ce151eab098 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/951b161e79db498eab13b60845df92f7 : SUCCESS in 10m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2ba5f6a78294bcb930954b1724499ce : SUCCESS in 10m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/758b2cd599594308a81b7a00acbb7703 : SUCCESS in 29m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/32692f200507444e8a66acfa29e9d260 : SUCCESS in 25m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/94687059c4e94853b83b254fff864d01 : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/383c2a4fe00d472a8a93a99cc6a164e0 : SUCCESS in 13m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4bbf25e05197469b9b9e436fca508bed : SUCCESS in 8m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f38b783ae51a4375b406711087faef84 : SUCCESS in 6m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c9d8cfe22204427a9ffc9873690f8352 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6aff14ad4900418faf8c37b7c6d63b76 : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d41bdbf6540640a2ba748d8a5b05ab09 : SUCCESS in 47m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4740e49e89c94bae8e7758b79987c338 : SUCCESS in 56m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0d778ea18014107825443d0ad6e444f : SUCCESS in 41m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/917c0d796a7c4027bb4f3e7f8dd0f4b9 : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bfc90fcf02644aebc1147d7652b34ec : FAILURE in 3m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b94b77519228461590e1db2df17c2005 : SUCCESS in 16m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74fdd6a70faa48d283bfdc515ee45a7b : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":29},{"id":"89ac85832bd90c21cbf39b0c94e9f629e9520848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-19 17:44:50.000000000","message":"Uploaded patch set 30.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"f4984a741435bd3fc15b07b22e0ebaf76961f315","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 18:20:51.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9108ab58088a4be0b5d1135a8b5c6ba8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8cc43890485c40c39acec143bba2e31f : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":30},{"id":"e8878a86b211c867c6a8bc5763e028cffb87467b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 18:53:47.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f13729e29d4467cb6017b84c165bfa9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e24118eb7154a8cb6c78b8a4dbf2905 : SUCCESS in 6m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/d255c0c0994a4929a9bd0058fd7d611a : SUCCESS in 1h 00m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1785ae4eb167491ab95ad33dc6e1fce1 : SUCCESS in 1h 05m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d55a5dc4b182474c863c6beb85ece3ca : SUCCESS in 51m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31fbe0709b344536a02c528058d7aefd : SUCCESS in 57m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/465f4a5282984fdd87e88a9a243d9f1a : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51623a5079a0439a84093ce861ea3e18 : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b62c72a40224656bd325bc1d2df6cc7 : SUCCESS in 20m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/89205650df31442da75cff6c1a528739 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fc5486e0d32d4f98af5924ac89ef6285 : SUCCESS in 18m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9a1cedebe7e141e1b3ef8b8f162873e7 : SUCCESS in 21m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/348158a6eb84455bbbc716058f4c94e2 : SUCCESS in 31m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c87d674d1b11493c8ef00c85ab3f3bbe : SUCCESS in 21m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/299cc04db7d844c9b5628a02335e9fda : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/37917da91d1c485db8e8efde22a46bcd : SUCCESS in 14m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4a6dc8eaacee48b2ae2380be8fea01f2 : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/253cddf7819e418ca859cb699d600502 : SUCCESS in 7m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a5d631b58c404fda937e9fd44c7196ce : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ed3a4193e7044c2a91ed42da61d29c53 : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a191ce0142be40fda87e99499c3405fa : SUCCESS in 54m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eeade1a409cc4c6faa296736509d51be : SUCCESS in 50m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cfebd66e637244a6a6319beffd926e1e : SUCCESS in 51m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc59518f8bb244fb9f4d53ce491db3be : SUCCESS in 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb1491ccc4d4a91bcb42e34b03aaccb : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/365f37a05bc34c6fb73b44b6d54c8b02 : SUCCESS in 20m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1d6df0dcf1a49518528dd869fe7d7c4 : SUCCESS in 42m 14s","accounts_in_message":[],"_revision_number":30},{"id":"550430898fb1bb1f643c040cc3878e47d7a64601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-23 03:45:31.000000000","message":"Uploaded patch set 31.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"eb1063372cdcdd12c2ef0274397e92bb2470d126","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 04:20:19.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77c38773221a44399db187ccbb7d8c10\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0802e8482be74cefa52b8a650861bda0 : SUCCESS in 32m 59s","accounts_in_message":[],"_revision_number":31},{"id":"6b903b1d1aeb6dc46986354a1eeb83ddc9b3e341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 04:50:02.000000000","message":"Patch Set 31: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da71a18097164ab09e151a530a1cc741\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edd7d15ecf704ca5825084819f713dcb : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/1f5e6922652345fbb09a79d687897402 : SUCCESS in 48m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/feb42e32dc7d4963a49426a9727c0c5e : SUCCESS in 47m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ace6b0f589794560aa3dc163addc3911 : SUCCESS in 51m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbee8d449c3b4254a0934757fe27dabb : SUCCESS in 50m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d568d316c1542e0890ea85882544f5a : SUCCESS in 17m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e28c9d8dfca641479b428d240fa17c15 : SUCCESS in 21m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01645e807a434422b869d99e1c8c69c7 : SUCCESS in 17m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3de3ede63ea34fa09cf7f50abcc8d9b5 : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c86bf13efc14d369f9b183e08011c1d : SUCCESS in 12m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4804be8af52c4339bf40ab4d0843ebdb : SUCCESS in 8m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b8847df60e8045c98e959d3deb18ed9e : SUCCESS in 30m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dc223ab6717f49e1937d89216515d576 : SUCCESS in 11m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b3c54d2d52b94329a4b51e981d9bb743 : SUCCESS in 12m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e4e274d958ab4d3787722f73f4c825d7 : SUCCESS in 7m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3fd54610263e4f42ae28ec2e84f02d9f : SUCCESS in 8m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4de3d91be1bf4d95b3908415a2182139 : SUCCESS in 8m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bfc9b612ce9542c392a4bf603701edbe : SUCCESS in 13m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/be035ab5a1e64d8c9e7285e040f09cbe : SUCCESS in 7m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/90fe0632d9a24c1e920ea3755321a33f : SUCCESS in 49m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/54d605d4550f402fb0933dddb4ce1cc2 : SUCCESS in 33m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e108a12354fb41a28e80372e99c722e3 : SUCCESS in 25m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bcd7d55cf7f4a0cbdcd580ed5f666c3 : SUCCESS in 22m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/949fa1af0c0a4885ae59888fe91ac69f : FAILURE in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5a0d4604ecc497aa069aa183556689c : SUCCESS in 15m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a8b2cfb650f4a368726afe835afa261 : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":31},{"id":"f7e28ec7a31da3234e159392dcdbb817e58efe20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-23 19:06:36.000000000","message":"Uploaded patch set 32.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"b6eccea4304d253563f9a2010a085eca635082c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 19:44:08.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eda1a931363141c59ff03ebbffe3fb81\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d5bb0a7a45f543e2a28ab06a85af6ca0 : SUCCESS in 30m 01s","accounts_in_message":[],"_revision_number":32},{"id":"727b5b33fd0116f97403fffe62619b132b8eb1f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 20:06:06.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f816d2f26d416aa308969c440e356d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46708f4d685e444ba5de4e727f987f16 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/aff95267f77343bb88fcc48775a880c9 : SUCCESS in 44m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eee2fde0ca98435e8914f9df6288bb90 : SUCCESS in 48m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4fb01bf7766a40b087ad970ff2b43405 : SUCCESS in 43m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f71956f6d2fc40638b8d7908dc28b227 : SUCCESS in 33m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc0c48069ece4f53a619767f03bca10f : SUCCESS in 14m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a8767d099574bf49f07bb70a80df0e6 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ea40426585e49e89473d0a275daaf79 : SUCCESS in 10m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8fcb3e8beb8d444a96b61df1f28324d7 : SUCCESS in 4m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/893beaafc1dd4f93be24b8fbc85699e4 : SUCCESS in 10m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2e2eda300f7e49888c48a54a6d0ddce4 : SUCCESS in 8m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/94a1cb6eb9c64cc4b63782a4d4e30d01 : SUCCESS in 18m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b200705e3877485792e61028563dedb6 : SUCCESS in 30m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5716a9b7511b45bb8927b2dae586e1e4 : SUCCESS in 8m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4d2948f5cd734e50a1981da3712b457c : SUCCESS in 15m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4d7d7c1c47ba4032a29455ee1c78f683 : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/951196d205904eaa8bd7b6ea46dfc18c : SUCCESS in 3m 37s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d10a1939f4194ae2acef3ad467d013b7 : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/df071f5618c34d1f87ef0654877d05b4 : SUCCESS in 4m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4f1a683cc3d4e0782bc60f117205f49 : SUCCESS in 34m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e8ebafb3db46483c8bee749927184bce : SUCCESS in 55m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51d6b575b5f44a21962354e48bfb3d4b : SUCCESS in 49m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f27e8307db9e433ba71ce90d2960b220 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5faec7f64a5c478abdee7bbc84bad89c : SUCCESS in 2m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a7adc462629847cda278713588cf6d0c : SUCCESS in 10m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d74ffdc0492146339df8d237dd7911c0 : SUCCESS in 28m 29s","accounts_in_message":[],"_revision_number":32},{"id":"519b5b2fbe16cd624e5d6d50a0e36bc58ee61511","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-24 18:04:17.000000000","message":"Uploaded patch set 33.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"6f1f817a5f51686ef53a4dcaeeace3f121440253","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 18:47:14.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c459d0924834e579fc5118666d2c900\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a74b0d5d30fa4c23802fe5e782f4bf02 : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":33},{"id":"4ad56a3f276f3451134a699dd718910520105615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 19:16:43.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/076a6e1a2825478fb08db84ce02d3366\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1d1d624666c4371b01640a46f433d13 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/6c533050d09c49fc9d28dc21b510ff96 : SUCCESS in 31m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25e08ab9956a4cc895769f7466654d02 : SUCCESS in 51m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fff08786bce74a9fa082efe3015f1926 : SUCCESS in 30m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09b5172f9af949ba872426f2e151741a : SUCCESS in 33m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9890ff03c013436595873dc8e5f2b39f : SUCCESS in 10m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0ff60fc73194a0bb7d9e61980bf0a1c : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/008d3a7a1e6749cdba5b9a07a9aa0a22 : SUCCESS in 10m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/321bab4ae8b34a28837cd52d661a06b0 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e4e0f18d060647d18fdf2122e9dfcb58 : SUCCESS in 22m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5a8804115d2c41bd902dd960360b9704 : SUCCESS in 22m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d6b9275bbb5a483580172a20010d44d6 : SUCCESS in 28m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0a7e4ec53c0d4044873a9996550c3b05 : SUCCESS in 24m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/114a9ca03a7d4073bdc3f1d21d331604 : SUCCESS in 8m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a1b2f0c36b9e4509aed4e81e71de8033 : SUCCESS in 8m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a2dcb70512d64a30928573b26f912eb2 : SUCCESS in 10m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a059539975a9469dbcf6a759e16a691e : SUCCESS in 4m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f3a6915fcf34a3b9feed109892e4ded : SUCCESS in 12m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6656e2d8aa07452d8d3562f8477878b3 : SUCCESS in 8m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a7f567c18bb4381a16e050fcc8b8456 : SUCCESS in 57m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3413e31402504b65bed7fbcf7810d89c : SUCCESS in 23m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/64808a231a37428d883d25dfe7943959 : SUCCESS in 32m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c5f40fb9a834b5089df3d17b6f4a705 : SUCCESS in 24m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/063632d7f8a043f789f0e96949b64713 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ed26b0adf1f4f1ba56811957f5793db : SUCCESS in 16m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cf28714df3641e8a3e5efaefdb72be3 : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":33},{"id":"5135c846c08d20aa34f1e6dafac1e0e86f6be69d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-26 17:52:51.000000000","message":"Uploaded patch set 34.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"e3bbfd83d964c254d91a6ba76e1bff0b8c1a47b2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-26 18:07:11.000000000","message":"Patch Set 34: Code-Review-2\n\n(2 comments)","accounts_in_message":[],"_revision_number":34},{"id":"3f45bcc99928eb03cd1298d3d16f2a74fdf3d5da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 18:34:24.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d1785f4ccb74d02b199a7da61d65510\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c98e03416ab4a2eb3678c70239d9539 : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":34},{"id":"d8d642e55f57c393b442ae7d663d18f7d9c4af0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-26 18:48:23.000000000","message":"Uploaded patch set 35.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n","accounts_in_message":[],"_revision_number":35},{"id":"0da391026eabad469f20db5f77e1e4d48e2bfc74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 19:46:41.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77e5b5b0232c4bf894a54d07422aab54\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1043209fa566436096da93a9ba47295f : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":35},{"id":"dfeb12ab42668673d5830b5249fe21caf9cd660b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 19:58:26.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89f7ebfeaa2c4a0eb4e73d1e78e98081\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe69ce8092254d8b9d9d2148117f582a : SUCCESS in 4m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/2756d6f7015b472ab34f12c152955fe1 : SUCCESS in 32m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/08eeb46c392441439237020c1b2fb94c : SUCCESS in 1h 02m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e40ce8ceaf9c4433a4830c9057358a1d : SUCCESS in 30m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/070e8121400e487f8c6bebb650fffd29 : SUCCESS in 1h 00m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db935e14a8de4f39a06b3587890ba443 : SUCCESS in 10m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32742aae676242969f749466d74866f1 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f509887866c41159bbc2fd6c318fab3 : SUCCESS in 12m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/052880e67b054c0087daec031d4a5482 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/375f8f25e03548298204d7059eb02e53 : SUCCESS in 22m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fa06bbe0a9cb4b8a92b43bac8b95d951 : SUCCESS in 8m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f2e5cbdf7ca04d27881ecffbd7862d4b : SUCCESS in 25m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/700e53478e484c7582c2320d927d1d17 : SUCCESS in 11m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4ab86410fdfb4e8ba65405b75eff1054 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6e43caa0d5b34bb8ad45f0964e0f6d9a : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a88ac5af6d604aac8ed08d047cbdb51e : SUCCESS in 15m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b400aa40abce4b9eb83202e331386296 : SUCCESS in 7m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e325c02603a9466688293015827528eb : SUCCESS in 15m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4d254d88c3664dd0a43b2e52e105378e : SUCCESS in 8m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9cec5263c83497aa97f76d3125996cb : SUCCESS in 49m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb6ce190a892484a82c275e3a981436a : SUCCESS in 39m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72e8ea73f9b44b079f28602b5ec4cb16 : SUCCESS in 1h 02m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd7c790af72244f6b0c55e0a709723df : POST_FAILURE in 30m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e034b6fd75bd4ab084eb7ef7f1977d64 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12d5e400e521430d8210c4cf443fabf6 : SUCCESS in 21m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77eb02d6a2ac4db1942a22415338c038 : SUCCESS in 30m 58s","accounts_in_message":[],"_revision_number":35},{"id":"1d2f884ba88848d2bf6ba028e936c15f359c1266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-27 14:50:10.000000000","message":"Uploaded patch set 36.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"48e1bb9bbcbb3b67fb1c6f7fbb233f46c6434272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 16:12:19.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5bd1f30b3b74c0baf288e353814f64f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/701328bd73214047a3691cc3b41fc0f7 : SUCCESS in 5m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/d3b5e1b3aae44fdb9d93376eb3de8ad3 : SUCCESS in 32m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3947df48d036485a96dfb79326322e2e : SUCCESS in 32m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/adf2c5ad6ef34258975479071f052b11 : SUCCESS in 46m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54c8bfab86974cca90907bf73052c4e1 : SUCCESS in 1h 00m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a7ab44e2151451c9aced68ea1edc322 : SUCCESS in 13m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f5a861edc4040069c6b6366ee48582f : SUCCESS in 22m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6e26807acd843948069721a1e6fb07a : SUCCESS in 13m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a489cf81269847299fd24afca9605957 : SUCCESS in 7m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/baeda09048704cc3b6c6d83bc79fa540 : SUCCESS in 27m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/adafb66280a14b129363ed0b92d17ed6 : SUCCESS in 24m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/17bb3af52bd44effa4338d6d9c2d6f97 : SUCCESS in 30m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/49bde26a3adb4146ac2facf8834986c9 : SUCCESS in 30m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5ef310f64d884b57a3aebb3f0be3cd9b : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fee7bd338256421e967528812f2649a4 : SUCCESS in 14m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4001fa10ff3a4742841f7aa273934b1f : SUCCESS in 15m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b06e17fe8724f37af67ae9c61bafe60 : SUCCESS in 4m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/570b319f87064541a79fc01cc5dbf76d : SUCCESS in 11m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1248c3028b29463b8bb9e87846355c25 : SUCCESS in 5m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0d8d25ae4ee4abd8b1876642edc697c : SUCCESS in 53m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a7511cfcd4344b12b3276eb52071eff2 : SUCCESS in 57m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc10405b46be460ca4d8d18ce5fc4a57 : SUCCESS in 53m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7ea5a85562f473bbef9df031fcc49cf : SUCCESS in 28m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e4ce4cbcbee4ab28bc418d40acd04b6 : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3fb9689859a944e682e84e4f8d1549c3 : SUCCESS in 20m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72ad5958ad6243f0bc1fc0fa0dc4c51c : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":36},{"id":"0a918fabce31002951e150e46657802e05a65c76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 16:13:57.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668ae2c157c0459b97a48b50b1ea56dc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c497e5408da44f948bb454874fa2019e : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":36},{"id":"d5b70cb8c3fde6f0115e4bab9a0d7e51c280fb0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-02 14:49:05.000000000","message":"Uploaded patch set 37.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":37},{"id":"26bf274f3c90ddadf1dca40da1184440071aa388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 16:14:21.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d7e3f6824034139989d433a47876cf9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76008a9592b5475a8dedd91aa4c0832c : SUCCESS in 6m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/c3da5f7696b54483bfbfd977d62842a0 : SUCCESS in 1h 02m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9ebcc6678c44e1583f9325e70d9376b : SUCCESS in 32m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc24d9bf5dcc4164b9708d04b66e4274 : SUCCESS in 33m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/add7848c10804a4d80614b8f9cfcfc39 : SUCCESS in 33m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c477bbe4f9b9440b83018f9a9c318148 : SUCCESS in 10m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c404c31d43744c92a4138d1ba35584da : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1df2425a4c92409cabe8591722b883f1 : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c604a549495b42ec8790881d76b9dcea : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/53033b3d2ef64a29ad5115e0bdf108f5 : SUCCESS in 14m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/37f4d5461f2845a38ad882712a430c1d : SUCCESS in 16m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9820ce1aa9f24af4a11394880e734fe7 : SUCCESS in 15m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/46c8a0d94db84a69a8438b2c86d7424d : SUCCESS in 12m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b3ebe02d85df4d369f0aac9ea3fcbe32 : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/06277a41a8894ca99a1a4c71fc2b480d : SUCCESS in 15m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/03bfc2b1de35489b8df65fe8503997fe : SUCCESS in 18m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0609b8d38e934661898141828a4adec2 : SUCCESS in 7m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5bd6fe8891a144c48250bcfb8e5a3108 : SUCCESS in 16m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fcd2d2a821e74257a91a48bfdfe16666 : SUCCESS in 6m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ce43db4a2e84d48b64ed275883c951d : SUCCESS in 40m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1bb30473a1184284a07f0f01e087b722 : SUCCESS in 32m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53af2fc55afd4c848314b1d492c195a8 : SUCCESS in 27m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd3c597892f74620bae84190d5d0c948 : SUCCESS in 28m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0354809a980f4d0587f28faf1491416e : SUCCESS in 2m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/170aed85e8ff4700988af64e574fad49 : SUCCESS in 18m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc688f0d499f447fb10b5490554a9bbb : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":37},{"id":"1a6e80757476af569a8bced393bd8dcb3affbfc8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-02 16:19:44.000000000","message":"Patch Set 37: Code-Review-2\n\n(3 comments)","accounts_in_message":[],"_revision_number":37},{"id":"94fb59492755a24162502664e874d6bec281d084","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 19:40:30.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6280982c465b4b1ab4efa34eaa0f6780\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32b303d74e56470a87e5e0309f2f98b4 : SUCCESS in 26m 50s","accounts_in_message":[],"_revision_number":37},{"id":"43237360d292199489bcf2e02af9e3cbbfba2425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-03 20:59:26.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":38},{"id":"92e9fcf7ef76d3afc0f5fc5aebff960c57c114e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 21:17:51.000000000","message":"Patch Set 38:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a3d19c94a824b89b14e28ee28949ecd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f0ef3a5535a4d6a81bfd7c5cc1fe3bf : NODE_FAILURE Node(set) request ed3be07cb29148779c7dcce0b962b00c failed in 0s","accounts_in_message":[],"_revision_number":38},{"id":"5ff3a628076ec18ff2e5374ee0829828c47e5789","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 22:16:30.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff9c7e0fa9614f82a6cbc320ebc98b42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c614ca19892045e8b89d8af47d532296 : SUCCESS in 9m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/97153eb177e9404194e905b5a7c0e5b8 : SUCCESS in 1h 02m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec1c7044c9404c51a4a2aa0827a09976 : SUCCESS in 1h 05m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aace13d0bffd4231ae13381230fb7489 : SUCCESS in 56m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/268e8afffd0f4d6d8461f6eb4606fe49 : SUCCESS in 1h 04m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/739dd13f17ed4dac8a0a5d8066739513 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9a95fbdb5bf4c1b92129689504cc206 : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/457d6293ea6348aebf7492d60d7a6de5 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b18ad4360a4a47fe8e7d314ff9d840e1 : SUCCESS in 8m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2e96fdc66cc8464c8170b142fa69968d : SUCCESS in 25m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3439e0abaefa46cb8e468a8f13d2aa31 : SUCCESS in 21m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/96cb7110db934931b2a56e7dede627de : SUCCESS in 34m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a3750a1ca37642b8a2e4d438940e1f2f : SUCCESS in 31m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3bc37bce923c49298bccc471c9539351 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bc14ffc8ad80425e94148e163405cf5d : SUCCESS in 16m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5a20dda4bc25454a82ad01830b9fee38 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2fab78fa6dbc419398ba6bb367c9ea71 : SUCCESS in 6m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e518ad98978d4b66a1e56a4dbb6b4307 : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/192528ad3fa142dfa271795341dfb2fd : SUCCESS in 9m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a00812412c40449986249327ae04071d : SUCCESS in 59m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8638cffb1b6647f8a0765228bab588c4 : SUCCESS in 42m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9955522033634f708cb9241685649d8d : SUCCESS in 50m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1951cf4c7b68498688acb51305438dea : SUCCESS in 31m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f9ba07c575f4387915af53e566134c2 : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24441712d0a74b409875628240b49d63 : SUCCESS in 18m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c89e8e1e11964120a6c8505cc8cfb38b : SUCCESS in 1h 10m 15s","accounts_in_message":[],"_revision_number":38},{"id":"19ef084fb95228d6c14938d4f5d3c3b4b426219d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-06 15:48:21.000000000","message":"Uploaded patch set 39: Patch Set 38 was rebased.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":39},{"id":"6380aff841eded6acbc35dffcee165944c5b1db4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 16:52:47.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2379cd6794af45fe84503eece805cec5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07245b476b434ca6ad443379246f8589 : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":39},{"id":"947e684aae225b211569e5eab3cd63769eb56a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 16:59:30.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82273afc33df4f2aba146e3db4e99726\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4116b0ee9c314cddaf8fc64f3660808a : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/b9ac4675869f4971a455e6a4c4410875 : SUCCESS in 1h 00m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f952e8f7422f4f47ae95d758800acf1c : SUCCESS in 58m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f1f93eb675041da98dc5130d5ec6171 : SUCCESS in 56m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/059baa50b46d40feb15335dc5ea798ff : SUCCESS in 1h 02m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74246d704f5141f8b28bdea14486b0d2 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9227e71aa3c345ef99e1961553de1e34 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/21088863dca7465eaf1f74c7a54e2978 : SUCCESS in 11m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7147ab8fa70142a8a39cc33dc4c86726 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b0e41e9217d4ff1ab61070ef24056b4 : SUCCESS in 11m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0000f5f2525a4659a0add8f2db323964 : SUCCESS in 21m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9c3affe2dc084e3b8ba052f790d95a7c : SUCCESS in 12m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8c4e65db979c449094b385f1a799d30b : SUCCESS in 22m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/56736178c0c64021856a18ecee7eb6ef : SUCCESS in 8m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/50d15832ae7d4f738ea57949aa888705 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c6d208ed236a4375acd5f2e2a65b6329 : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7c987a07802c44559182606a13d43c66 : SUCCESS in 6m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bc7a6b719e1649c3b1b7d46e6a882b69 : SUCCESS in 7m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/782e4ecaca554f39b7b1cfca63497784 : SUCCESS in 9m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/595fac53e1c94010b1b4e80e88d94cb6 : SUCCESS in 1h 00m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d9a220127794bd5a2d39a9095dd62de : FAILURE in 9m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b64effe629554cce8e916bafa13f4ca2 : FAILURE in 19m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d3472f043a94578b291d0135af4fc03 : SUCCESS in 31m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e565abe930420cafa878249ea1f6e9 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cbd415adfbf43f6881b2d7ece827c27 : SUCCESS in 20m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62e475a748fa4f34892a0b10f589349e : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":39},{"id":"6b127ce0315295374b1d250730fa342903265fd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-09 13:12:34.000000000","message":"Uploaded patch set 40.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"02aa89216fbcc747ea1ae5071b6640cbd19318fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:02:39.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/534a65715e6e49f193870188bbf6e6ff\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1933adc0b7d94222a9cf2c7e65e399d3 : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":40},{"id":"e05c6b306d22b389dc8f3688d9317dc0cfd0afa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 14:38:51.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19211c138ff24cf688e8088438c92a52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7afd83c121a34049aae97191503efa68 : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/25b6cf656bf54f3b8d0e836ba07b8a73 : SUCCESS in 1h 05m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c676501b95c44287be1c8ffef41cba8c : SUCCESS in 1h 08m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb725c2a9a284077bf6654bcc1b1cb0a : SUCCESS in 45m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c5d09873b7a40f98848848e3a9343e8 : SUCCESS in 1h 10m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffb8bffb9981450182de701fc0dcf83b : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ee481c2998c4054aba6aa47ae4c6f10 : SUCCESS in 11m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13846c61017b4991993e1b34af10791b : SUCCESS in 14m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b7b2ea21cac455d92dc7124f094ae9b : FAILURE in 5m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4cf22dbbd256468b99cef4d6a02df81a : SUCCESS in 15m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ca3901ce54744058ea9aaf06cc1492c : SUCCESS in 20m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b8400817f7d5439e93a02ad3c2bb2621 : SUCCESS in 25m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fea8b82605ef4a3b9a614472abf5687a : SUCCESS in 19m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e7f3a6574e2941139a983411a46405d6 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ab65ed56150748e89c754735f230c9d5 : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9c542cde4c2a494a806f804e79f03437 : SUCCESS in 11m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8485688802814f8783b5ce41cf4f3b8f : SUCCESS in 8m 57s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/808ecc5802774014b1e4d52c26ef9905 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/17f4d37d167e4dcb895e5ef133014b77 : SUCCESS in 6m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c03da838b8f41b7bbc91f469c2851be : SUCCESS in 56m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d5aa4ab4011f4adeac9aa4900733046f : SUCCESS in 56m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6879f901b4164d518acec876401af323 : SUCCESS in 1h 04m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75bfcbe7f41e4422b16ed06a4a14180d : SUCCESS in 29m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513ae783156f4f7a83272d4bb7e6ed64 : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d2001f6cb4840a8b992552dace363c4 : SUCCESS in 19m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c20c23d0e3ec46b6b759293112d7b66f : SUCCESS in 1h 05m 25s","accounts_in_message":[],"_revision_number":40},{"id":"9db2848c98870841ef0a302e8657e3f9e698a667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-10 16:53:30.000000000","message":"Uploaded patch set 41.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"b336455a8546d044f5a020fae1216d86adab347b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 17:40:17.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d2261bd26d4407baad0a09eb01c281b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e669693b5f694653860f2276586ee9e9 : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":41},{"id":"078468fd5a5d3b0d2e3ec5a64a019b89e25c6cf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 18:12:47.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c66acea3145e4737a598cd5b9190fcab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e87f509a274c68b4e49ec49aace04a : SUCCESS in 5m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/edb4c066d81f4b56bf323a6bdbdff7be : SUCCESS in 58m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/667e2ac05ed44d71be3d18594d82fd24 : SUCCESS in 33m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0864f6410fdd4464a970bb7757a72e74 : SUCCESS in 31m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b51de673a384566b8814e080aaff5f3 : SUCCESS in 1h 03m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c1e334a933e4e6ca6f6e17a7e7ee816 : FAILURE in 5m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe68e08d8e60405899a6017f0337cd80 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e8250e2153f4850968461fb54c52aa2 : SUCCESS in 21m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c951d9e59a94abbb623cc16f585ad9a : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c0762c0ad9c42ba8a6903370da076fb : SUCCESS in 11m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a9bb3ee2a33d4fb1a0a2d3ca1d940300 : SUCCESS in 10m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/046513f27c7a4fdd86f056bf3be98e09 : SUCCESS in 30m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6c41e0763f50483dbdeda493767ce303 : FAILURE in 5m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b3c876ccd5ad4cc0977977fd813c27d2 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6f4881a2517b4abe982de9c7703504f8 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/738158333557483fa3958bd5ecb48f01 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2391ab0f965449ed938b499af707c1dd : SUCCESS in 6m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45891c8786d442aaa52771190a5125ee : SUCCESS in 15m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d9f295c8ed24540beccae474e75ae03 : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/725a2fa30af343919fcd598b8387b584 : SUCCESS in 55m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58b9497c7f5749cdb3f1719e8c67a0cc : SUCCESS in 52m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/662c66d944c14ea9b37093ec6c5ee85c : SUCCESS in 52m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10a5458b3c1e4e7a815682fc363dbbf9 : SUCCESS in 21m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c351482290f44979b794e046bb289dc3 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/98c74094180a4718b46a688cdcb937a5 : SUCCESS in 19m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2a197ce582d47029388c31d248d0e23 : SUCCESS in 1h 10m 18s","accounts_in_message":[],"_revision_number":41},{"id":"4ab308355ac576df7584ce8123580de091df9dd3","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-03-11 04:33:42.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"943860aa1fddf3b7aa4a7cd775b48a6e0cf43f83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 05:29:45.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/457f7cabc80c485596d30a168c66d34d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e37ed16cc080489488039c42de312d75 : SUCCESS in 41m 46s","accounts_in_message":[],"_revision_number":41},{"id":"45cfc958adb53cc4bbec51028bd91bd745ef88d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 06:00:21.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42649c6c71f144439ae418510a87a24c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5537ecaae504cd2be6f369102c82abe : SUCCESS in 5m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/df9748418f914ceb8a2f7b6ec16374b6 : SUCCESS in 30m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1013db321c2a452ebd71373da00dd445 : SUCCESS in 28m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4779c306c5944bdb2afc95c1e07f0a7 : SUCCESS in 1h 09m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17c9e45a459b426386bbe126541738ef : SUCCESS in 1h 01m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03e91104a23948ca9bf34189995f870f : SUCCESS in 11m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/147b7349422245d6819b7de034e5d4db : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c04885a853424609b370e07c75f485c4 : SUCCESS in 18m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/39d2efcadf034640ad0be25d5f3ffba9 : FAILURE in 3m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a936fa4375f4f0cb3b8221ca5d07821 : SUCCESS in 24m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ead454e0c51d46d7847985de972f3d78 : SUCCESS in 8m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9f9f07d740e74989833bf425979aab31 : SUCCESS in 12m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/350c40c0a1ad40a09a76a0c2066f4278 : SUCCESS in 29m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/eda0e7aa229c4bff99794f4baf239347 : SUCCESS in 12m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2834657b5ce849f6a107ed6f037875a9 : SUCCESS in 15m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/812e31dccf194539af5b07486ae826d1 : SUCCESS in 13m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d59660e467854694be9628c513a9d420 : SUCCESS in 5m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da39502d3f004861902184f9d4f07907 : SUCCESS in 13m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8b0c312106c8499fa98791133cf160fd : SUCCESS in 7m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f313e0893a614d70bd9ff6988ef639aa : SUCCESS in 47m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5b14600878c4311b2e13aa308860fba : SUCCESS in 21m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e92a0fa401ee48baaa31ccf1c61ce6f0 : SUCCESS in 34m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12c28fa5d7424984ba731cca8011ba35 : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c39d8a11d1a4489bc165dc3b36c3ca3 : SUCCESS in 2m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0acd8d6b4cb944f384144292633c4f09 : SUCCESS in 18m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68963a60b69f42068f443ad171b4b633 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":41},{"id":"83ba7f3d242a9e8612aab1c2549ada7eb5149402","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-11 22:46:42.000000000","message":"Patch Set 41: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":41},{"id":"96c762f39250e489a10d1e005cc66c0a43225890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-16 18:48:01.000000000","message":"Uploaded patch set 42.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"58fbc3c2ad1ad44e76aa6d3da81ff4dbae6f6ea3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 19:59:27.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a6627a5ed9040049d07309a6f456710\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/073aa459474a4e72913faf8952d41d80 : SUCCESS in 1h 05m 49s","accounts_in_message":[],"_revision_number":42},{"id":"c34cdf4df92d962d3517f802e44ad7ef7222696c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 20:13:13.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/196a70d0ebc645a5953df82b0f931510\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c56c076606744cbab176632d2ee7ac07 : SUCCESS in 5m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/5d2e57a1063540a2865464293e674964 : SUCCESS in 38m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/814e671e8bb241a8a699936920f69ef5 : SUCCESS in 1h 01m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1bba250da7a14bcebb95df3fe9073d68 : SUCCESS in 1h 00m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9472edc0d2334e25b211ee532791ec2d : SUCCESS in 1h 12m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7344e9c78fbd4437aa42eaf779d393e4 : SUCCESS in 12m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8094ac1112af4c569a241be5e0d85824 : SUCCESS in 11m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7b875a7b84a4cff91ee6031ac5fa50f : SUCCESS in 21m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67857ce313a749b2a100f523a9f3230c : FAILURE in 5m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8c5f76b85b6a4d7f95e9f0a728bc339a : SUCCESS in 10m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fcea26200f6a43d09046eaefd576811e : SUCCESS in 11m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c42141338e394fc08590e91523fb456c : SUCCESS in 38m 15s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7eb0c5eec3874e3bb80f3afd533d4e14 : SUCCESS in 9m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ac854d7e696544a992830b07b2bfbee9 : SUCCESS in 8m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3f9fb8d7cd5d4b7fa97d4546e5455861 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/da86fc73c9b449e0854c718bc7a5fbd5 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/694e38550eb346178c7390ddde3519ba : SUCCESS in 5m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9caa2497c477472a94f393d6b06682eb : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/69a3da6fda3c49aaad158562fbd5bf53 : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a30b1f901284f8183d139a0326b17dc : SUCCESS in 1h 03m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24c68870ebb64ddd82016fa90b369a94 : SUCCESS in 55m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e890f7e6c904c169034a9908b2e8abe : SUCCESS in 53m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/616db2e51cf14ae8adc8497024c2f018 : SUCCESS in 21m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f6812293112463ebe7ee64a35146726 : SUCCESS in 3m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1051dd188186490095c31ad9a070e9ff : SUCCESS in 19m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/290a4625f4f345ad99bf727b19663230 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":42},{"id":"e5c23935a8a693a8c49d3b9b4f026f39a10cdab9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-17 15:34:31.000000000","message":"Uploaded patch set 43.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"85fed4dc5112142ede245d8ce71a0a055ece6306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 16:41:33.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bca3fbec44674c34b1ed55f90e0ca427\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fde5a7810f54084ba661c4a1229f52b : SUCCESS in 11m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/18f334addcf9464e9244d7a7a7862f31 : SUCCESS in 1h 01m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7702972145ec4f7ab017065177a1af14 : SUCCESS in 46m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/daec8ae768b549e1b78c3095ef1e8401 : SUCCESS in 37m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b888a1a6a598444081786f26c8387633 : SUCCESS in 56m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b205e3269d1749a3874d324b51c2d61c : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99b781094a054d1d9450fcd796f49409 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fe3ba2ae6684450b4244cad69cccc89 : SUCCESS in 18m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b142b89c2674c2aaa708d327d8ea8e0 : SUCCESS in 8m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f7f9d5dbbb824ac8a81f42f59d364c9b : SUCCESS in 21m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ffd24fb4a0e6436fa166bea38f4dfa2e : SUCCESS in 22m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c50bb870ff2f40ed83761f7bd7a8bed3 : SUCCESS in 32m 56s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/03ab191790854238832b8ce914670ca7 : SUCCESS in 33m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cc596d10784a4b989b07a0b84a89a4bb : SUCCESS in 10m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4dda273ad1f84e9ca63710da4f845994 : SUCCESS in 11m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ff9864ac46fe4d30bee79370bdfcf6fa : SUCCESS in 10m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac3d64e3c9b94745be84d20ab5d0758c : SUCCESS in 7m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/619e6126efc641c78cbf65304b749c4a : SUCCESS in 15m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/09f9fd0618c24bcf9dd55dc9643ff8d9 : SUCCESS in 6m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfe80210a87c45849c5b74c3fced7a1a : SUCCESS in 40m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/817e6ef32ef846a68a445d4ffba96eb5 : SUCCESS in 25m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/41066e15c04d489594dd80c60018cdbb : SUCCESS in 54m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc0610cbc514474a91c2d23a35436ca0 : SUCCESS in 32m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80a6971ecf9147609d962c7fd7bf3b38 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/454bf537aaf747758425348539990c06 : SUCCESS in 18m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c449f83d3fb74453b10baab53559ec98 : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":43},{"id":"567f0dfb48f144c0baa36ec17a34b19ae264659d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 17:33:51.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea7fff8d8bf5431b8a690637251fd0ac\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ea7d197c814475094322bfcbf00b793 : SUCCESS in 32m 11s","accounts_in_message":[],"_revision_number":43},{"id":"8cd672231e76da221d61a021cdbe634ed1ddfe3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-18 14:50:56.000000000","message":"Uploaded patch set 44.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"2bf339078f7b21c493a1a614054f3110d7f01a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:16:55.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67ef9bda35d040e68f5a0c119c0079d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d68cbaa92067459485626526c28b825d : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/e7901e54e0ef460eae698b7eee378ab2 : SUCCESS in 1h 01m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f0fe659db9841859f9ce952068fbde3 : SUCCESS in 44m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/050414f6ab5f47dea6da004ed895264e : SUCCESS in 55m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c530376d44f47d7a9f63c55620021cd : SUCCESS in 1h 08m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/306247979db84f39bb4428400d72c836 : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4b1c4bef7bc480f9a3fa2cc8e436933 : SUCCESS in 12m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4709086a282b46d69ece3272ddbfacae : SUCCESS in 10m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0e7a88ed3664ed7b5b646adcf468bbb : SUCCESS in 8m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4b6c9663152d44bc97d9291c0126e3f0 : SUCCESS in 29m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2d711034cf93402782089e29a74dd2dc : SUCCESS in 24m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a140e73dc3a24307a55923f0be05aeff : SUCCESS in 31m 29s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ca7b75bb7b794244acadcab9ddc83eeb : SUCCESS in 15m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e821dcf256f84f6583313aa7234052af : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d6d7535cb62d4aaea74ca0fc5d451a00 : SUCCESS in 8m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eacd706b6960481582fd6b0b35e4f06d : SUCCESS in 16m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f892650e87b743a3bd606b1b8149596d : SUCCESS in 6m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/47e1df9d3cd34dfc8f14907bbfb004b4 : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b7623132b68847c5a9d7415c7c9c69fb : SUCCESS in 13m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c205d2db32c48369ed707575c7d624f : SUCCESS in 57m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d373a3835ac64a6d9cf8a4cbcfb09cc9 : SUCCESS in 50m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cf78b420029644d6888e1d93832c1e04 : SUCCESS in 1h 02m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0983794c30d419385e7a72af861bafc : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f099faa5910f46a2a19f38870b9d3f68 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4508d255dbc540ef9bd76a9ad337219a : SUCCESS in 28m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c51cadb4001e48eea43938e134f1a390 : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":44},{"id":"c7d453084aaf7f41671c2b637f6909238eca6235","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:00:51.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f74a97e8036e41ea869f37e8f2835953\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/18b9075cfaad4247bc72f042e8f14e80 : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":44},{"id":"dc5ebe62a063e1d63c59a93c3e3903f710c76811","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-20 02:51:19.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"7b548427722a0ee5a0f2401c2a4e613f7e2bf123","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-24 17:58:55.000000000","message":"Uploaded patch set 45.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"fe92048ef16bc20ddc48a5da18ed3b35ada16d16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 19:25:45.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8793ee3ca44146a79ba33a329559f192\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79847ff5bfc54695b8d5e64817b676e2 : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/bc1a9d3537c849b88d28cf4a0006770c : SUCCESS in 1h 09m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/88fd8606913648d6b2bca01d73c18ef8 : SUCCESS in 1h 02m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef4ea9ea006441048984c478dcdce921 : SUCCESS in 57m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae7247586de447cfa12ba3bd5b35b447 : SUCCESS in 34m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ea54ea9e1534131ad1dc81f65c54c8d : SUCCESS in 11m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b5c88750fad479c8aa269a61b0ace5f : SUCCESS in 21m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/239d8fd95cd9486f8918440bf8e60f31 : SUCCESS in 11m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bdeb499068944e9a946a7bac21cb87ad : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f64911212fc4492939fd05a4e158a06 : SUCCESS in 11m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e2df10924d0e4d88a6b44a3931be3cde : SUCCESS in 22m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b5b4306ce391400890a5525650af749e : SUCCESS in 26m 52s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/aca3e7d6c9084d4db247b358a1e62fcc : SUCCESS in 20m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e7dda85e2d594b8db5ad204a633ef17e : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/39bb1577ea1e4ae180f753e552fa2a37 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4dfacca558fe4fe3ab6fde4235db718b : SUCCESS in 15m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1df2fd56a97249e0b36befe15fefeff4 : SUCCESS in 8m 57s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45ee3963871d42d89ddeefbb306661ba : SUCCESS in 15m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/106fec48faea4939aeb769c36ae1071b : SUCCESS in 8m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8628077e353d4b8dbcd05def09f05da6 : SUCCESS in 50m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed8eb9bbf5c842e58b143b0a29bde392 : SUCCESS in 22m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95e6403e0b094e228d9d34db9b43aee3 : SUCCESS in 53m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a42d2955c6a491a89f0ab597acae6f3 : SUCCESS in 26m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/579a2a0dbc17492abe13737ad78fa149 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30234d4c45174978a02edee61fd898c3 : SUCCESS in 10m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff31b080a49343e998416372b01e7187 : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":45},{"id":"60a79189e4237cf9b271dcadd7c834a7b388c91a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 19:36:21.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dab970193150427896393d8c1e091084\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/050fe34a83f24fa6b367da6477f6bf84 : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":45},{"id":"605211df8290c9c852bd7f6405ca73d882a3d417","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-25 16:11:48.000000000","message":"Patch Set 45: -Code-Review","accounts_in_message":[],"_revision_number":45},{"id":"f0abf2e95e641302976f9b920e77d286f24b2ef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-25 16:30:13.000000000","message":"Uploaded patch set 46: Patch Set 45 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":46},{"id":"9fc0eb4651179ade216038bdc32a077f3e0c7347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:46:48.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cda2e07044d41b181b222fec171e1b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa7df8f344547d6b7e3faf339d26749 : SUCCESS in 11m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/be9c9aed497643e5bbd37e52e03572f4 : SUCCESS in 1h 06m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9767102e304c46a9a330c771f02113ff : SUCCESS in 1h 02m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a3b285948de44df586f4b47bd1fae652 : SUCCESS in 56m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f19c42caf2c49c48f75152d8395c760 : SUCCESS in 35m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bc0e9bf097c456197cdf95478be6780 : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be822b89c5494e7b92f20d84a26bcef5 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e17e681cf4774d5284491482c2d1203b : SUCCESS in 23m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66ecd601dcd342e7a8d6293564816cae : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/383274aaa3354c03bc138a0695bfeba3 : SUCCESS in 24m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/86b8d5af3c1a4183814f62d35fcf1b93 : SUCCESS in 27m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/855a0afad430419c9f9f07c1139f5e00 : SUCCESS in 31m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/157e099ea1e0491eb1c65129bd9ede60 : SUCCESS in 20m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/96edadc3908a413eaf2b0e4ad0256385 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5c242f8984804fd6bd9c7761f9a480b3 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8defff1bfacf4d4382bda89735cb4d4f : SUCCESS in 9m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f0f7f355368c4239bb2c692aa8ba71b1 : SUCCESS in 9m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/705026053815427982c87313e929678f : SUCCESS in 16m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1a460e21f5704291927a555f076788a7 : SUCCESS in 9m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5a285fa719748d5845b644f4b927814 : SUCCESS in 1h 00m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67999ab58bb944baa5909ed6fe464893 : SUCCESS in 51m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4a90216d0f947d3bf6440a30a9bf6db : SUCCESS in 24m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a9fb2a31f9948a4abeb43c4dabf22ea : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a461ee3f8f47a4b5a50d109a8750b0 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8c15936088504681b8ca473abaf64a50 : SUCCESS in 19m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5c45f49fd5e42aca6034fe1e54e2d07 : SUCCESS in 55m 38s","accounts_in_message":[],"_revision_number":46},{"id":"0dfa210b6814928565e7ad1a1e967da18fe8e830","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 19:16:39.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99432292bbf8413fb62f0ffaea57f9f4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a82621b64504b36b990b69e0f2598c2 : SUCCESS in 38m 08s","accounts_in_message":[],"_revision_number":46},{"id":"fd13b640c0332451254a276d246b6b3ea27b512b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-31 14:54:16.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":47},{"id":"f0b2c277fe87174f617b7db8037c3a37e5458920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:06:10.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/300ec456991c4958b487c3a4d6dc2bb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41848bafb46d46e28d89cd6fc51f985b : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/38441ac2fd904c2b95380d8822f78e21 : SUCCESS in 1h 01m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b15e21e5aee47fc8a7faa3c35bdadb3 : SUCCESS in 1h 02m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/efcc3d688c314673a430e843fafc3636 : SUCCESS in 33m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d331f632a354f75b6378a621560cd18 : SUCCESS in 54m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab812af5ac604e0885c82fe2fcfb27b9 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/940ada81b23247d3b1ee274288705ef7 : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3dd6b82399704ce6aa23543f383a1ca7 : SUCCESS in 21m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/638cb600ccb14dada4d4960a4e55ac37 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2e5ad76683ca44feb3329656cb3fdc9d : SUCCESS in 28m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dee747f5d2f6408288f0fab29b5584e1 : SUCCESS in 15m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d641dfdceb5e4231834919a960a08312 : SUCCESS in 12m 47s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/8d30639204db4ec7ae666557067a4d3d : SUCCESS in 20m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2f47f0d79ebb4073a4966137ecfed5af : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cfbaef2605a84193b692fb0a4da6a8b3 : SUCCESS in 10m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/593b03d57d7b418eb144940162a69f70 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4484809baff04e1a81a388eeb7fe9fe4 : SUCCESS in 6m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/14fa836afcb6477280a021d5c56c2ed6 : SUCCESS in 13m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9d0c98f8d5004a75bd301e412911b962 : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e2f6c615b084b34ae68f3df6086015f : SUCCESS in 54m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2f6d1aa6cdf4654b99fcb42fd676f69 : SUCCESS in 56m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/111b9656e3914879a757d2119bb64e26 : SUCCESS in 53m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd1143a4b929497791dfc9f28a31fafc : SUCCESS in 30m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaf0a83f5fd447748498264071d371e2 : SUCCESS in 3m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f30beb5097441b58eebbf723b3f3b4d : SUCCESS in 17m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa884b0bd2b24558838372a3ac50934c : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":47},{"id":"b32abebf5f78d0afe02306d004c086e31697e049","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:10:26.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c67ab30eccf9414caf6f1d9b2e537a5e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9021ccd7a79a4ffab5fc5328b39b8b74 : SUCCESS in 31m 29s","accounts_in_message":[],"_revision_number":47},{"id":"b36e5b02e769711018f02ead919b4266da5c20cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-09 09:21:56.000000000","message":"Uploaded patch set 48: Patch Set 47 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":48},{"id":"bd256710677213b3cdc51604eb0bb737d50bda85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:35:40.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc6fb4265a49465bba99eb3f52e8c557\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c919159e852a4e99922456138245becc : SUCCESS in 4m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/f2548337df474a97b18cae2d163a2075 : SUCCESS in 1h 07m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/87b4ad42503843db88f1f2c76a26b12c : SUCCESS in 33m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1ce7288819e4c11985070cd1cbaae14 : SUCCESS in 37m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9fd6597486f4147a05cf094138b560a : SUCCESS in 1h 09m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/efd66ec95dbd4bc7a3be3565ef0ab85c : SUCCESS in 10m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1d69c6d81204330a49da548276142dd : SUCCESS in 22m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31e3c7cb270e43e39cae15939114922b : SUCCESS in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01522aca3a774c52ad884d9b82f654ec : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/14534ef5489b4ba4b124ce47dc446806 : SUCCESS in 15m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3a8bb7e799f945c0a47367948f8c9542 : SUCCESS in 21m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3eb33cf2f92a406aba3b0b509b4147a1 : SUCCESS in 37m 38s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/eb2c9609dbdf4dc4a6e3e53d4fce546c : SUCCESS in 19m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b2cbc23b39c041f097f18a578d1f15c2 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cdac6ba4e8374a1eba002f1064acb654 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/310659a4dd294ae8bceb90756598fbc5 : SUCCESS in 8m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e172b41e412a4243887111f0c9ce57f7 : SUCCESS in 5m 04s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0396cb6604324daeb079d6f2cf3d1dd5 : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/42f4aafb6c37473f9f180b3cb4a2d311 : SUCCESS in 8m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6fb8d435d594a1ca02281b883301d05 : SUCCESS in 42m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e7e57b717d2f44cc8411cc6bf15aced2 : SUCCESS in 57m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e62c51cb109d4677aeca1637fcaead8b : SUCCESS in 23m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0173cf83a88d418f991b8b2313773293 : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec80d8cd7fe94b109626300d554e07aa : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0814049e830b4b35ba4bca68910cfdc6 : SUCCESS in 18m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea1e6a43eabd4f4e9150b4c6dfd00586 : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":48},{"id":"f2862b3584aa7aec902334dbd81c9f5d605ee433","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:57:47.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/426be680de0d46de922a4d544e043cdb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/306918fa0be64174b8653bea90631d0b : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":48},{"id":"f48936d48919e0c678b9a5fee1ad804dcaf72e68","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-04-13 02:59:41.000000000","message":"Patch Set 48: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":48},{"id":"e886afeaae2f822f38c6ed577d82f70378c6574c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-13 09:47:13.000000000","message":"Uploaded patch set 49.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":49},{"id":"40f773454f4ab9254f2a881f7479ee869424a58c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:39:27.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef7fe969dd424d818c6994736b401cfe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0664e61ffc62441986853a8d53587403 : SUCCESS in 11m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/ace91b9a275d4d648ff9957b2148b59b : SUCCESS in 1h 10m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/93affe5bffc64eb390ab1ad9baa301bd : SUCCESS in 1h 03m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/180261b49f7b429b963e6595b9f71657 : SUCCESS in 52m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00db204e179e42a2ab2a259319647d78 : SUCCESS in 1h 08m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c126c7f16c35413385b2a4ab108e9336 : SUCCESS in 11m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c49a8b357340475d97ab611cdc44e3ca : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ace8d3e565be4c2e8f86b0c24b6ecda6 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/98ad13b6becd465cb0fb31b6c7ab0a75 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/da2f1045f9a6499095c4cc66f09b996a : SUCCESS in 24m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/81ecc4dee5d34a9a9858c6446b128d54 : SUCCESS in 22m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8cb47aa481cd4aeaa478b1b37da3b046 : SUCCESS in 22m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f19afcf255ae4fd1b5ee1745fe01d6ab : SUCCESS in 24m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bc2d009acd814195932728bb699fd7d8 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7f01d6d8862749a9b99f6ad5431f26a2 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fd82bc623eca4176a62841cc876136ce : SUCCESS in 13m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/842b23211dd84d2aad17997c0f65ea2f : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/574b036082fa480a9bdcc040ce80e461 : SUCCESS in 11m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/14d075e7328d4bf386c30c8223f56356 : SUCCESS in 9m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3e01de937524bd0b595b428e96ff091 : SUCCESS in 1h 00m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f69829da5c0e42afb35bfaf9778c3877 : SUCCESS in 24m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/42144b7e86334e9f965967050d2bcc30 : SUCCESS in 54m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea08b4fe9d204ec79d13ce822ca8d3da : SUCCESS in 27m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67bd201a83bc4669a6465941492114b3 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35e6c3edaff243b69e99e19000fa8c80 : SUCCESS in 17m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2477657c0c51400e9e430e0bad1ea31b : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":49},{"id":"b38d947ede0bf47d353717e86eaae105d9ef6f02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 15:33:23.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66449b85a1184f17a9c32cef7f5437ab\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b0b2e05539724d609bcb6ab2ea0cd68d : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":49},{"id":"2f9781a2d6843bcef0622b1042a1db0ab10c98f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-14 15:09:20.000000000","message":"Uploaded patch set 50.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":50},{"id":"35426086fd99ac4b60da967942c9929ca395c972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 16:22:41.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be5ea0833f9b4857886d2e9a28015932\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/978db52eda4044fca4fbd551bad8785b : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/f8e52baa271346e4b023f6bc92ec6655 : SUCCESS in 1h 00m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0423bb1736d340e0bf613cb76b918fa0 : SUCCESS in 47m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0156c2788b4e480b96d59498eb9e65f1 : SUCCESS in 55m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bffba95afc1b41059b49627c86bb140f : SUCCESS in 1h 04m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a267c525be024c9fbe54905e472ba34c : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e97ff7c422740bba6fd137e993789ec : SUCCESS in 20m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8184b8f182b84ab38ed7dc6ac1e1a330 : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/55954d47c159447f8cca73cbd853455e : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c53989dbb0a146a9ace1052b31b6901f : SUCCESS in 14m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e5443f4d5ff34d59baa76e8335c202eb : SUCCESS in 22m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f93675a55fe141dfa86670a2781747f7 : SUCCESS in 23m 21s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9bc4125d6824484294fef6f8a201e854 : SUCCESS in 19m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/36364bb7d8144509ade2259a93b8e60e : SUCCESS in 9m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7e91c8905ece4febbf792ecbec3357a1 : SUCCESS in 9m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/707a5c8cdded4e3fb5cc78c09c665e17 : SUCCESS in 10m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64059684b4a04243b94dcec14f1b2f7f : SUCCESS in 7m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/de23cfae2dd54a169c1d8e0f4e4e27b1 : SUCCESS in 17m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/71f73327457246dbaffcde4eae3f4f5a : SUCCESS in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/41dd62def25b4f96b381fba265515560 : SUCCESS in 54m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e8d36d13d7c84c37b5c0b5fa251923ed : SUCCESS in 28m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/426b774fc0304b4bade585a423da3b02 : SUCCESS in 25m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b31b087d76744d1b210cb1f0af81296 : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/622c74cc88454dfe82f33d6f699d1659 : SUCCESS in 3m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/857ad4fa3e464877bda44a93d7d4df91 : SUCCESS in 21m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d447447e995743cfa4ff4a5d1242caf6 : SUCCESS in 31m 49s","accounts_in_message":[],"_revision_number":50},{"id":"8d8e5afaeacda53fc33c393c177ad0b9f450d829","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 16:50:17.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/343c42ac51d04f388af8d08559a5406d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a236a1bf2324bb49bf755ceae44b583 : SUCCESS in 40m 00s","accounts_in_message":[],"_revision_number":50},{"id":"64fba0d90b3e7f669a54fcaa4c6e5e2628b0aa90","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-15 05:39:11.000000000","message":"Patch Set 50: Code-Review+2\n\n(8 comments)","accounts_in_message":[],"_revision_number":50},{"id":"ff0598f89ebbfb8df994132f6792ccf4575a785b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-15 14:23:58.000000000","message":"Patch Set 50:\n\n(18 comments)","accounts_in_message":[],"_revision_number":50},{"id":"fdb5e37840b69bfb890e62c205465fe2ef00d108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-15 14:24:12.000000000","message":"Uploaded patch set 51.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":51},{"id":"ff8ded29106fdb72782ba2dd0da2da2ec3bf3ba5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-15 14:31:52.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"a91d192397ffde54582c8275afe5d443378fec0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:26:11.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd31e1810ed4ef6bbf39f2ae599e2fd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fb84bd419f947bca90805a263493fe5 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":52},{"id":"bf92ed264f198db043f660f5258402ce281fa0a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 16:03:31.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8745f69088094103b0e56be529356ca5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce38c3745ca46d99c4bcd9210de10ab : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/dcd51d0bbb35403695364a4916be0b4a : SUCCESS in 1h 08m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74742a38a6ed4ca69241d86a7c0cb7bd : SUCCESS in 1h 19m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/49df8cf045754bd6b87c52243efcab81 : SUCCESS in 52m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/136cfaf9fe0a4bfc9d48fde8dbde087e : SUCCESS in 1h 03m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c52b79a965bd4af9ae8d11184a728311 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c59bd336c1ef46c5bc5525cc37f10d2e : SUCCESS in 20m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9847e8df47c4a17a1202a8a9ee69460 : SUCCESS in 20m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53337f550f9040068e7ffb6b3bade92e : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/29549087f55a4b9ca5d17963f116ba9e : SUCCESS in 29m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3643fc3f88774a43b4740bdde1b92eb9 : SUCCESS in 22m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/785f69070918465cbfdcd82f48e96403 : SUCCESS in 33m 30s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/12d08445e425482fa7c6ea3d627b7d3a : SUCCESS in 23m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d0c2e5ead47046348ec3df0398882043 : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5b51ad29e8ae431daaaa395c1614663c : SUCCESS in 15m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6ef06769cdda4fc18cb85c256d24f95f : SUCCESS in 17m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f1f265d828a34ab3959fd92e5c333987 : SUCCESS in 6m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/40b971433b1740f2b5d6e853e6558bae : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/79d0945ffd2541fe97c53a217391d066 : SUCCESS in 9m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/95930623d00244d2bcf293be66c477f6 : SUCCESS in 36m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d5cbdbb7f05940eea355482a9b11df07 : SUCCESS in 40m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b01753ae2afa41808f3baf5309601aa9 : SUCCESS in 54m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e427a19a386c4a438bba7f8151509b23 : SUCCESS in 31m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e31d26ed02c485ab8ad1f7326db4ff1 : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/881cd82a1a0249d1b44ff7a9fd105a78 : FAILURE in 55m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0ff9ed6672c440b8d4bf732785d28be : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":52},{"id":"26342b50f5cb0d1ecb2e88d3ff2963a9fd6a7ff2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-15 22:58:05.000000000","message":"Patch Set 52: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":52},{"id":"88fcf6f7147d071c490bdd698c87b5f1edd7d1a2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-16 05:03:42.000000000","message":"Patch Set 52:\n\nrecheck","accounts_in_message":[],"_revision_number":52},{"id":"b80dc16e94aa9d2282573072796598e734440eec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 05:55:56.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/216a8a4f9f014df1a090c3613de72587\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/71334a77f92f47c5adc3aa4181cf0145 : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":52},{"id":"e689162822a8a4db3ab21d4ccab0dc0ed3a9f726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 06:14:21.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38aa8fe64b1a4f00b7592d826020ceb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f1c5ccda6b34bedb8dc4f470799a7f5 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/4c5feab5f6c74bd290abe2ec8831413c : SUCCESS in 49m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7ad4d5a632a42c987b99bcc601352e8 : SUCCESS in 33m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b90b19cb23114486ae18862224587e55 : SUCCESS in 49m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00d2b402f4a244809edcee0f6d540729 : SUCCESS in 1h 04m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8befdfd6b114e5db70010d31b3432e1 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/090e6034e0864033b253b3798fcf242b : SUCCESS in 16m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b71b898348743deb544fc92f5aa18f9 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/17b5998f65d34c6a90c909dfc64c2487 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/65d203f2036f4bd687d5943b32ac6738 : SUCCESS in 24m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b131bcf663d34a7f9543103761b67404 : SUCCESS in 19m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dbd97ec24d9245a6b62d37f4c9586074 : SUCCESS in 29m 43s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/a117d353a1a446d78d6ee245e7bedc52 : SUCCESS in 20m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f6a44ee3395c492ead7259a8813cb283 : SUCCESS in 8m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c0db90499362497d8bddb409e6fbca35 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d24c202647674df0b98d935c8dfe02b8 : SUCCESS in 16m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf41819cab9d4d29b49d7feed3b0e21a : SUCCESS in 6m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6620dc714a0d4f84838729f522b8f30a : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7c3fbc1e28d748f5974248ba1697179d : SUCCESS in 8m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/72e118debaa3458a984c456c5d837d94 : SUCCESS in 1h 03m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afbd71484ecf451ab35e2ad2dd963dc6 : SUCCESS in 40m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/38d6b5c7d12642ddbec52c5fc5da1d39 : SUCCESS in 53m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e7d9bc532b44a9a901552cc17146818 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d892898eb9f45f6b63b99a1f8924621 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c049541ca6574d0dac549e1b11c8d23b : SUCCESS in 9m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1b80f3d250240beba84dc69cd11913d : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":52},{"id":"bf837843df5c9b581d833b3a385dbea0bd74c1a0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-16 09:31:52.000000000","message":"Patch Set 52:\n\n(1 comment)","accounts_in_message":[],"_revision_number":52},{"id":"457b910a21146715e806eb29265e4bdc9bddcf37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-16 15:49:05.000000000","message":"Uploaded patch set 53.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":53},{"id":"077616cb97e83fd33ce8b169199d7e898253580c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-16 15:56:44.000000000","message":"Patch Set 53:\n\n(2 comments)","accounts_in_message":[],"_revision_number":53},{"id":"6e00f75d6bc1fbdd82ceade20d3c9cbfc48069dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-16 15:58:13.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"935563fe869134fbc83a950b2c241584cc4719bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:58:40.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e405711c4e0a40238b2898f2dc8e7ba3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b12d344e25427daff71f4a1ced3aac : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/126db114f88f4429a84afdacc23c5c60 : SUCCESS in 31m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73943d65c3ef4d6382c18df767e674f7 : SUCCESS in 31m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/06ff604405d7406ca1cd2d2894dae7f8 : SUCCESS in 45m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c31fa2ef0e204c38b3dac6541ec1cc25 : SUCCESS in 38m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd7f0e42323f451693d1ee3001d0c5cb : POST_FAILURE in 12m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7676301a7e6458082b672743dea4f37 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd2e735c684c48d5821556f5de309226 : SUCCESS in 19m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9d0dffabc7c24423ac9461f7bba7efd4 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/db0d37dabacb40edbec7a9079182ed55 : SUCCESS in 11m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/648f7524c3ba4db59a6569d4b3839eb9 : SUCCESS in 19m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2c522978533f41058c2eda5ccf058adc : SUCCESS in 22m 26s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ed3057e44fb44197a10f5d0796416df0 : SUCCESS in 11m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9fcf368c0c224227aacd1660ba260c82 : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/53ff66c175de4db3ae46738b3326c8ae : SUCCESS in 8m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6e006f0a0cbe4ee5a44c5bdade9fe97e : SUCCESS in 9m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4229a882a736498dade6a36a2a8a7b0b : SUCCESS in 6m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0082f09641f54b9b9386cf61ca31d32b : SUCCESS in 10m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1f744e3537284f90b35ec98457e892e9 : SUCCESS in 9m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f8796c81de34d5dab007d4e63a6bc93 : SUCCESS in 57m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e74028dbfa547b09be1e26c02868315 : SUCCESS in 21m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3fc0e16e2d6a4b7e9d9b1a9911ed59b7 : SUCCESS in 50m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53149a1e4b0944abb549016f461a770b : SUCCESS in 27m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f8aa4e40bd3478c8045aa904425afb6 : SUCCESS in 3m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/72cdbb6c6993492fb1a49e625ea1b29f : SUCCESS in 20m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc3a34737d2348c2bca6a4980dab5b5f : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":54},{"id":"8c9ee91758fde19b794d7e00a1c45ab601973ab1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 17:27:43.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1106178d60d49fe8f7f3224a438bbc9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8012c4a529a04c348014d8260400a312 : SUCCESS in 37m 55s","accounts_in_message":[],"_revision_number":54},{"id":"37bc60e79489953e9ba47e2267406af4798cb835","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-04-17 00:27:26.000000000","message":"Patch Set 54: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":54},{"id":"9c8423858953192777b7b9e1269ff8c67798b2ec","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-20 11:40:10.000000000","message":"Patch Set 54:\n\n(5 comments)","accounts_in_message":[],"_revision_number":54},{"id":"4334f948ce7bdcd0585eeac6703c79de7464dd53","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-04-20 18:43:30.000000000","message":"Patch Set 54: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":54},{"id":"8df668cb5b03540c0799557488ef406925c17b4f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-04-21 07:06:19.000000000","message":"Patch Set 54: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":54},{"id":"a29180b39569fa0724d43e5378ec778d6db78ae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 07:07:40.000000000","message":"Patch Set 54: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":54},{"id":"80837024c59dbcd73ffcd1dbc556aa98e8c5a11d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:10:53.000000000","message":"Patch Set 54: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f7b7a93d234d34ae01a1f7d6082122\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6127c501c51e42769f9a01ff7c87d29a : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/bde44635ef464796bf52fdf5f1ff268b : SUCCESS in 58m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2abc8f87d9db492fa80c32c0fd90b850 : SUCCESS in 35m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4c7bb2f849f448981630337b6aaba9b : SUCCESS in 57m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57730410c4874ea4a15e7e16f347ae70 : SUCCESS in 57m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b655deedbdf84c95bf748a2b6be838b9 : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bc51bf27855b4af88deba08abb32ed54 : SUCCESS in 17m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d3c1be58027744919f5ec6c54e44140d : SUCCESS in 21m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fa8fe9d1a77d43e294b1f13a448d6e7c : SUCCESS in 19m 51s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/41f770bf45944d6da1e09fe46df6e4bf : SUCCESS in 16m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bd75dcc2f22d4944b1e8e69f70755a3b : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8514bb935847430daf62e9b1a1c6db19 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ca503caeab6448ecad51e08fecfdab03 : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/312d865367774cffbbe0e4b527d2a561 : SUCCESS in 5m 32s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3765c45134664e41963dc8382f359ed8 : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3a74be2338f410e810429cc81df68a4 : SUCCESS in 47m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0f126835cea4d68bd080e09f1c1ef53 : SUCCESS in 49m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ffbc8f55f11745c393293a0e7dc8cf62 : SUCCESS in 53m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6a51bc70e4247418784c7b3f1c5eb8f : SUCCESS in 25m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad12d05e522f4f30af3dd0eaaae8535a : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b16ea1c070e84d9681d4c69cd268ae1c : SUCCESS in 19m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/613ff3705e0c4e16ae62ddd9a01cbd66 : SUCCESS in 44m 23s","accounts_in_message":[],"_revision_number":54},{"id":"a5008f8e4b80fd48041d2178fde55b18638652d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:10:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":54},{"id":"e0a95d5796dac66ba0064310965fa4ba29edd386","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:12:30.000000000","message":"Patch Set 54:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0279da590ec4bba8dca4378b873c6f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d73bb8d318148f69437726ed8eedeb7 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0be74ba7032946b596bd7e9394546c02 : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":54}],"current_revision_number":54,"current_revision":"cf5df8fac80ff354ee8e6369dbf18a5e1fd75343","revisions":{"c815f735c1ed95ca41559dec6b8797b5cc4ebd98":{"kind":"REWORK","_number":1,"created":"2025-11-27 18:27:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/1"}}},"commit":{"parents":[{"commit":"760829a75a6598aab5a1865b20ac9e2b988355b8","subject":"account: Use Timestamps better in test_backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/760829a75a6598aab5a1865b20ac9e2b988355b8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-27 14:35:12.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-27 18:27:25.000000000","tz":0},"subject":"WIP: timestamps: add SimpleTimestamp for ShardRanges","message":"WIP: timestamps: add SimpleTimestamp for ShardRanges\n\nTODO:\n* I\u0027m not totally happy with the class heirarchy yet.\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp.\n\nChange-Id: Id9bb653b9c997cd1b8a67d86c736e71626ce8a86\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\ntest sharder mods\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c815f735c1ed95ca41559dec6b8797b5cc4ebd98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c815f735c1ed95ca41559dec6b8797b5cc4ebd98"}]},"branch":"refs/heads/master"},"4037fcab7015395a3bf57dec6069af07179c9957":{"kind":"REWORK","_number":2,"created":"2025-11-28 10:34:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/2"}}},"commit":{"parents":[{"commit":"760829a75a6598aab5a1865b20ac9e2b988355b8","subject":"account: Use Timestamps better in test_backend unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/760829a75a6598aab5a1865b20ac9e2b988355b8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-27 14:35:12.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 10:32:21.000000000","tz":0},"subject":"WIP: timestamps: add SimpleTimestamp for ShardRanges","message":"WIP: timestamps: add SimpleTimestamp for ShardRanges\n\nTODO:\n* I\u0027m not totally happy with the class heirarchy yet.\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\nShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4037fcab7015395a3bf57dec6069af07179c9957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4037fcab7015395a3bf57dec6069af07179c9957"}]},"branch":"refs/heads/master"},"bbf7ad5a3856e29c3d05127b1f2901578bdb9b96":{"kind":"REWORK","_number":3,"created":"2025-12-14 18:47:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/3"}}},"commit":{"parents":[{"commit":"61d084626d422321900f57252ea566a7c1ca8434","subject":"OldestAsyncPendingTracker accepts Timestamp.internal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61d084626d422321900f57252ea566a7c1ca8434"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 18:46:56.000000000","tz":-360},"subject":"WIP: timestamps: add SimpleTimestamp for ShardRanges","message":"WIP: timestamps: add SimpleTimestamp for ShardRanges\n\nTODO:\n* I\u0027m not totally happy with the class heirarchy yet.\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbf7ad5a3856e29c3d05127b1f2901578bdb9b96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bbf7ad5a3856e29c3d05127b1f2901578bdb9b96"}]},"branch":"refs/heads/master"},"de2ad1c2885dc0548755ab16cbe4f20b8422b432":{"kind":"REWORK","_number":4,"created":"2025-12-19 16:25:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/4"}}},"commit":{"parents":[{"commit":"8bf6e090f4753f7e9d1df1ef2bea46aad9c6e088","subject":"Merge \"test_diskfile.py: fix failing test if O_TMPFILE unsupported\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bf6e090f4753f7e9d1df1ef2bea46aad9c6e088"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-19 16:14:31.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de2ad1c2885dc0548755ab16cbe4f20b8422b432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de2ad1c2885dc0548755ab16cbe4f20b8422b432"}]},"branch":"refs/heads/master"},"4cfae210350dcb0bf82ca2f3b4dcc23c6c3e5e9d":{"kind":"REWORK","_number":5,"created":"2025-12-22 20:22:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/5"}}},"commit":{"parents":[{"commit":"bb4196971880a84f35c7898d40ac7af9dfa0d747","subject":"Tests: Mock Timestamp in object versioning unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb4196971880a84f35c7898d40ac7af9dfa0d747"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-22 20:22:28.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cfae210350dcb0bf82ca2f3b4dcc23c6c3e5e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cfae210350dcb0bf82ca2f3b4dcc23c6c3e5e9d"}]},"branch":"refs/heads/master"},"894b05ed2a9597f2fa27dba0f0d34904f52a824e":{"kind":"REWORK","_number":6,"created":"2025-12-23 19:13:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/6"}}},"commit":{"parents":[{"commit":"9366678a9e5c5725b67927175d9f56ffbe6a2060","subject":"legacy versioning: use Timestamp to create marker name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9366678a9e5c5725b67927175d9f56ffbe6a2060"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-23 19:13:16.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/894b05ed2a9597f2fa27dba0f0d34904f52a824e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/894b05ed2a9597f2fa27dba0f0d34904f52a824e"}]},"branch":"refs/heads/master"},"444ddad4895f7cfc11e5160d113d86f300f15b0b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-01-02 15:19:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/7"}}},"commit":{"parents":[{"commit":"35882b7b06af9ecc8543d9cbdba1ac7e4c958c9e","subject":"versioned_writes: use new parse_date_header helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35882b7b06af9ecc8543d9cbdba1ac7e4c958c9e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-02 15:18:57.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/444ddad4895f7cfc11e5160d113d86f300f15b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/444ddad4895f7cfc11e5160d113d86f300f15b0b"}]},"branch":"refs/heads/master"},"36bd7966b54f9b7c4408910c3ff91d6bd8a9b9b4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-01-02 18:27:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/8"}}},"commit":{"parents":[{"commit":"377c518bbab549232abd0c5884671d665581f804","subject":"versioned_writes: use new parse_date_header helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/377c518bbab549232abd0c5884671d665581f804"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-02 18:26:59.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36bd7966b54f9b7c4408910c3ff91d6bd8a9b9b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36bd7966b54f9b7c4408910c3ff91d6bd8a9b9b4"}]},"branch":"refs/heads/master"},"617fc726df7cb8caf54a631bc7f07595bd4b7480":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-01-05 10:38:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/9"}}},"commit":{"parents":[{"commit":"b81842b8c05b280013144faa866c7e743f67cd50","subject":"versioned_writes: use new parse_date_header helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b81842b8c05b280013144faa866c7e743f67cd50"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 10:37:27.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/617fc726df7cb8caf54a631bc7f07595bd4b7480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/617fc726df7cb8caf54a631bc7f07595bd4b7480"}]},"branch":"refs/heads/master"},"bb5ffae9d3d5dc3fe10950d1034bd80e5356b0cf":{"kind":"REWORK","_number":10,"created":"2026-01-05 16:17:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/10"}}},"commit":{"parents":[{"commit":"97ec6333c417dec196c79787596a5a28885638d4","subject":"tests: relocate timestamp tests to test_timestamp.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97ec6333c417dec196c79787596a5a28885638d4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 14:08:52.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb5ffae9d3d5dc3fe10950d1034bd80e5356b0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb5ffae9d3d5dc3fe10950d1034bd80e5356b0cf"}]},"branch":"refs/heads/master"},"ec99e8bcca8068b6a9bf17583d15baffb41c3586":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-01-06 17:43:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/11"}}},"commit":{"parents":[{"commit":"33ef46ad5631191cd270436a8f953e7f5c9bd23b","subject":"tests: add assertions re. Timestamp headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33ef46ad5631191cd270436a8f953e7f5c9bd23b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-06 15:18:50.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec99e8bcca8068b6a9bf17583d15baffb41c3586"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec99e8bcca8068b6a9bf17583d15baffb41c3586"}]},"branch":"refs/heads/master"},"2a6341107c076ec290431db7c9e05f89d2c4720c":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-01-06 18:04:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/12"}}},"commit":{"parents":[{"commit":"c2506aea01407f4274afff5b229eb7b025d32c38","subject":"tests: add assertions re. Timestamp headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2506aea01407f4274afff5b229eb7b025d32c38"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-06 18:00:26.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a6341107c076ec290431db7c9e05f89d2c4720c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a6341107c076ec290431db7c9e05f89d2c4720c"}]},"branch":"refs/heads/master"},"9381ddbab810802cbb475fd0d5537aa29041ff46":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-01-07 12:12:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/13"}}},"commit":{"parents":[{"commit":"3ba803a2085b98454adc6f0f3306ec97a7548cfa","subject":"test_sync.py: fix timestamp handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ba803a2085b98454adc6f0f3306ec97a7548cfa"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 12:03:29.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9381ddbab810802cbb475fd0d5537aa29041ff46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9381ddbab810802cbb475fd0d5537aa29041ff46"}]},"branch":"refs/heads/master"},"806504fe1fbe18e316967337af92887ca952dfc5":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-01-07 18:16:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/14"}}},"commit":{"parents":[{"commit":"44c8c44770618013b8d3679c758271477a07e392","subject":"test_backend.py: refactor TestUpdateNewItemFromExisting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44c8c44770618013b8d3679c758271477a07e392"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 14:53:54.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/806504fe1fbe18e316967337af92887ca952dfc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/806504fe1fbe18e316967337af92887ca952dfc5"}]},"branch":"refs/heads/master"},"1631460545016036746736f712efe843d1c7c59c":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-01-13 14:18:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/15"}}},"commit":{"parents":[{"commit":"1971983dece9f889ccb4dfd5dde628727390f074","subject":"test_backend.py: refactor TestUpdateNewItemFromExisting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1971983dece9f889ccb4dfd5dde628727390f074"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 14:18:10.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1631460545016036746736f712efe843d1c7c59c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1631460545016036746736f712efe843d1c7c59c"}]},"branch":"refs/heads/master"},"8aaf3c9852460ed7d76fe748abb4893f99a1d83c":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-01-14 14:51:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/16"}}},"commit":{"parents":[{"commit":"de7c88432bcfefbccb6ae161998b375a6898082e","subject":"s3api: do not set x-timestamp header","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de7c88432bcfefbccb6ae161998b375a6898082e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-14 14:23:43.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8aaf3c9852460ed7d76fe748abb4893f99a1d83c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8aaf3c9852460ed7d76fe748abb4893f99a1d83c"}]},"branch":"refs/heads/master"},"5bd6cb569705f1a8aee4796fd679ff299b7dff27":{"kind":"REWORK","_number":17,"created":"2026-01-16 17:43:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/17"}}},"commit":{"parents":[{"commit":"5cf424fba2da3d9cd46eb7c6daad42d8b9230aaf","subject":"proxy app: assign timestamp in a single place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cf424fba2da3d9cd46eb7c6daad42d8b9230aaf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-16 17:42:38.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bd6cb569705f1a8aee4796fd679ff299b7dff27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bd6cb569705f1a8aee4796fd679ff299b7dff27"}]},"branch":"refs/heads/master"},"aeaa648c52e00bf1a357db50ad1ad83a3d86eefc":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-01-19 18:25:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/18"}}},"commit":{"parents":[{"commit":"c3f778dda203b38b5b6084bd550809f72fca2cdb","subject":"direct_client.py: proxy app: assign timestamp in a single place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3f778dda203b38b5b6084bd550809f72fca2cdb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-19 17:20:43.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aeaa648c52e00bf1a357db50ad1ad83a3d86eefc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aeaa648c52e00bf1a357db50ad1ad83a3d86eefc"}]},"branch":"refs/heads/master"},"5a7cb15186f0dd60bbb79623cfd69622f1c5da28":{"kind":"REWORK","_number":19,"created":"2026-01-20 17:05:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/19"}}},"commit":{"parents":[{"commit":"ad6b5a5ff21be405879be9116528b35e4fde7b47","subject":"direct_client.py: proxy app: assign timestamp in a single place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad6b5a5ff21be405879be9116528b35e4fde7b47"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-20 15:10:26.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a7cb15186f0dd60bbb79623cfd69622f1c5da28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a7cb15186f0dd60bbb79623cfd69622f1c5da28"}]},"branch":"refs/heads/master"},"092d5a5a5658c746d967493b81740e869ea77bf2":{"kind":"REWORK","_number":20,"created":"2026-01-21 10:11:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/20"}}},"commit":{"parents":[{"commit":"ad6b5a5ff21be405879be9116528b35e4fde7b47","subject":"direct_client.py: proxy app: assign timestamp in a single place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad6b5a5ff21be405879be9116528b35e4fde7b47"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-21 10:09:55.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092d5a5a5658c746d967493b81740e869ea77bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092d5a5a5658c746d967493b81740e869ea77bf2"}]},"branch":"refs/heads/master"},"c12de992b0bdf36e3fc2ca3fb6b828fc0f2cdd69":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-01-29 03:56:58.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/40/968740/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/21"}}},"commit":{"parents":[{"commit":"7bf728b6198400aa2057bf42d34f434e6dc71407","subject":"direct_client.py: proxy app: assign timestamp in a single place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bf728b6198400aa2057bf42d34f434e6dc71407"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-01-29 03:55:28.000000000","tz":660},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c12de992b0bdf36e3fc2ca3fb6b828fc0f2cdd69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c12de992b0bdf36e3fc2ca3fb6b828fc0f2cdd69"}]},"branch":"refs/heads/master"},"7171594f9218e4259f1b93f1309ad9d440cbaa29":{"kind":"REWORK","_number":22,"created":"2026-02-10 18:55:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/22"}}},"commit":{"parents":[{"commit":"7bf728b6198400aa2057bf42d34f434e6dc71407","subject":"direct_client.py: proxy app: assign timestamp in a single place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bf728b6198400aa2057bf42d34f434e6dc71407"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-10 11:46:08.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nSimpleTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7171594f9218e4259f1b93f1309ad9d440cbaa29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7171594f9218e4259f1b93f1309ad9d440cbaa29"}]},"branch":"refs/heads/master"},"7a4aee28a52741128bd1268b15effe003fc6c153":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2026-02-11 19:07:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/23"}}},"commit":{"parents":[{"commit":"c7bee623658b464072783096b98adaddfb1005bc","subject":"Merge \"s3api: Support for object expiration time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7bee623658b464072783096b98adaddfb1005bc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-11 16:46:52.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nSimpleTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a4aee28a52741128bd1268b15effe003fc6c153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a4aee28a52741128bd1268b15effe003fc6c153"}]},"branch":"refs/heads/master"},"68014f782523fc463f314435c730fba2c2ac0b76":{"kind":"REWORK","_number":24,"created":"2026-02-12 19:09:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/24"}}},"commit":{"parents":[{"commit":"c7bee623658b464072783096b98adaddfb1005bc","subject":"Merge \"s3api: Support for object expiration time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7bee623658b464072783096b98adaddfb1005bc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-12 17:02:16.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nSimpleTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68014f782523fc463f314435c730fba2c2ac0b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68014f782523fc463f314435c730fba2c2ac0b76"}]},"branch":"refs/heads/master"},"458e63b686ce0204e40dafa7487011c539bcf45d":{"kind":"REWORK","_number":25,"created":"2026-02-13 15:48:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/25"}}},"commit":{"parents":[{"commit":"c7bee623658b464072783096b98adaddfb1005bc","subject":"Merge \"s3api: Support for object expiration time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7bee623658b464072783096b98adaddfb1005bc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-13 15:26:52.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nSimpleTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTo avoid confusion, SimpleTimestamp only has a \u0027normal\u0027 form of string\nrepresentation.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/458e63b686ce0204e40dafa7487011c539bcf45d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/458e63b686ce0204e40dafa7487011c539bcf45d"}]},"branch":"refs/heads/master"},"bd39a2dee98c79794a586c2259099442d9f2b9c5":{"kind":"REWORK","_number":26,"created":"2026-02-16 17:42:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/26"}}},"commit":{"parents":[{"commit":"c7bee623658b464072783096b98adaddfb1005bc","subject":"Merge \"s3api: Support for object expiration time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7bee623658b464072783096b98adaddfb1005bc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-16 16:39:02.000000000","tz":0},"subject":"timestamps: add SimpleTimestamp for ShardRanges","message":"timestamps: add SimpleTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a SimpleTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be SimpleTimestamps.\n\nSimpleTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTo avoid confusion, SimpleTimestamp only has a \u0027normal\u0027 form of string\nrepresentation.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd39a2dee98c79794a586c2259099442d9f2b9c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd39a2dee98c79794a586c2259099442d9f2b9c5"}]},"branch":"refs/heads/master"},"2df6b46445869d551a226c9071597575ec6f1170":{"kind":"REWORK","_number":27,"created":"2026-02-18 17:13:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/27"}}},"commit":{"parents":[{"commit":"2c980ac94682e8baed89c7f3c3b5af824514e8c3","subject":"Merge \"Improve relinker test assertions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c980ac94682e8baed89c7f3c3b5af824514e8c3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-18 14:41:46.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTo avoid confusion, NormalTimestamp only has a \u0027normal\u0027 form of string\nrepresentation.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2df6b46445869d551a226c9071597575ec6f1170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2df6b46445869d551a226c9071597575ec6f1170"}]},"branch":"refs/heads/master"},"4494e43b26e8a374b66ecaa80897ce9c630514d9":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2026-02-19 13:13:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/28"}}},"commit":{"parents":[{"commit":"810478eed645cc896511fd20245a9fe43bc0a276","subject":"tests: Stop using normalize_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/810478eed645cc896511fd20245a9fe43bc0a276"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-19 12:29:06.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTo avoid confusion, NormalTimestamp only has a \u0027normal\u0027 form of string\nrepresentation.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4494e43b26e8a374b66ecaa80897ce9c630514d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4494e43b26e8a374b66ecaa80897ce9c630514d9"}]},"branch":"refs/heads/master"},"6941e0c22708f5a73e900146bf9bcd033a748334":{"kind":"REWORK","_number":29,"created":"2026-02-19 15:35:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/29"}}},"commit":{"parents":[{"commit":"810478eed645cc896511fd20245a9fe43bc0a276","subject":"tests: Stop using normalize_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/810478eed645cc896511fd20245a9fe43bc0a276"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-19 14:50:50.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTo avoid confusion, NormalTimestamp only has a \u0027normal\u0027 form of string\nrepresentation.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6941e0c22708f5a73e900146bf9bcd033a748334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6941e0c22708f5a73e900146bf9bcd033a748334"}]},"branch":"refs/heads/master"},"b190af8eb99b196201c8cb98bfae9119f276d887":{"kind":"REWORK","_number":30,"created":"2026-02-19 17:44:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/30"}}},"commit":{"parents":[{"commit":"810478eed645cc896511fd20245a9fe43bc0a276","subject":"tests: Stop using normalize_timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/810478eed645cc896511fd20245a9fe43bc0a276"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-19 17:44:07.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTo avoid confusion, NormalTimestamp only has a \u0027normal\u0027 form of string\nrepresentation.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b190af8eb99b196201c8cb98bfae9119f276d887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b190af8eb99b196201c8cb98bfae9119f276d887"}]},"branch":"refs/heads/master"},"b90a92f31a53ffd6b29ff01d9120ac7064f529b5":{"kind":"REWORK","_number":31,"created":"2026-02-23 03:45:31.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/40/968740/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/31"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-23 03:37:39.000000000","tz":660},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTo avoid confusion, NormalTimestamp only has a \u0027normal\u0027 form of string\nrepresentation.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b90a92f31a53ffd6b29ff01d9120ac7064f529b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b90a92f31a53ffd6b29ff01d9120ac7064f529b5"}]},"branch":"refs/heads/master"},"75f75c2b3cd2548bb765bcef6c2497a874e719dd":{"kind":"REWORK","_number":32,"created":"2026-02-23 19:06:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/32"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-23 11:43:41.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTo avoid confusion, NormalTimestamp only has a \u0027normal\u0027 form of string\nrepresentation.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75f75c2b3cd2548bb765bcef6c2497a874e719dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75f75c2b3cd2548bb765bcef6c2497a874e719dd"}]},"branch":"refs/heads/master"},"03bc65a6273824041654ca95b35481b42a1e2e4b":{"kind":"REWORK","_number":33,"created":"2026-02-24 18:04:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/33"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-24 17:44:44.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03bc65a6273824041654ca95b35481b42a1e2e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03bc65a6273824041654ca95b35481b42a1e2e4b"}]},"branch":"refs/heads/master"},"c8343131b0d830f5be20c6a6ae21267ecfa93d7d":{"kind":"REWORK","_number":34,"created":"2026-02-26 17:52:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/34"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-26 16:38:25.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8343131b0d830f5be20c6a6ae21267ecfa93d7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8343131b0d830f5be20c6a6ae21267ecfa93d7d"}]},"branch":"refs/heads/master"},"0bdde42c52bb7499db343b3dbff021b629b1340f":{"kind":"REWORK","_number":35,"created":"2026-02-26 18:48:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/35"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-26 18:40:00.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bdde42c52bb7499db343b3dbff021b629b1340f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bdde42c52bb7499db343b3dbff021b629b1340f"}]},"branch":"refs/heads/master"},"2c928272ee3cf4fe70b51c97b0ea0cc92441cb44":{"kind":"REWORK","_number":36,"created":"2026-02-27 14:50:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/36"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-27 13:04:26.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example when\ntimestamps are being compared against wall time.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c928272ee3cf4fe70b51c97b0ea0cc92441cb44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c928272ee3cf4fe70b51c97b0ea0cc92441cb44"}]},"branch":"refs/heads/master"},"bea356d47e1e96aacc5b2ac33ed820c137f15bb3":{"kind":"REWORK","_number":37,"created":"2026-03-02 14:49:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/37"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-02 12:55:43.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example when\ntimestamps are being compared against wall time.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bea356d47e1e96aacc5b2ac33ed820c137f15bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bea356d47e1e96aacc5b2ac33ed820c137f15bb3"}]},"branch":"refs/heads/master"},"6600749df64533239163d135d30172b0c1a7bd63":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2026-03-03 20:59:26.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/40/968740/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/38"}}},"commit":{"parents":[{"commit":"886790163c0663421460f7a7497a367c9a524cb9","subject":"obj-reconstructor: use Timestamp.internal when sending out fragment requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/886790163c0663421460f7a7497a367c9a524cb9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-03 20:56:09.000000000","tz":-480},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example when\ntimestamps are being compared against wall time.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6600749df64533239163d135d30172b0c1a7bd63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6600749df64533239163d135d30172b0c1a7bd63"}]},"branch":"refs/heads/master"},"94cde7d9949fad6224b6674b1486bd6304bb0f57":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2026-03-06 15:48:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/39"}}},"commit":{"parents":[{"commit":"1f488b2842eb27c9b1caad44caffb18f8f95a228","subject":"obj-reconstructor: use Timestamp.internal in frag prefs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f488b2842eb27c9b1caad44caffb18f8f95a228"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-06 15:48:10.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example when\ntimestamps are being compared against wall time.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94cde7d9949fad6224b6674b1486bd6304bb0f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94cde7d9949fad6224b6674b1486bd6304bb0f57"}]},"branch":"refs/heads/master"},"1899ac3702bbfa724c9fccd6b06a19627f91ad51":{"kind":"REWORK","_number":40,"created":"2026-03-09 13:12:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/40"}}},"commit":{"parents":[{"commit":"1f488b2842eb27c9b1caad44caffb18f8f95a228","subject":"obj-reconstructor: use Timestamp.internal in frag prefs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f488b2842eb27c9b1caad44caffb18f8f95a228"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-09 13:06:30.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1899ac3702bbfa724c9fccd6b06a19627f91ad51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1899ac3702bbfa724c9fccd6b06a19627f91ad51"}]},"branch":"refs/heads/master"},"26432faab511caaa3a7b4ad12c13ea69e17d7b0a":{"kind":"REWORK","_number":41,"created":"2026-03-10 16:53:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/41"}}},"commit":{"parents":[{"commit":"1f488b2842eb27c9b1caad44caffb18f8f95a228","subject":"obj-reconstructor: use Timestamp.internal in frag prefs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f488b2842eb27c9b1caad44caffb18f8f95a228"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-10 16:52:26.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26432faab511caaa3a7b4ad12c13ea69e17d7b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26432faab511caaa3a7b4ad12c13ea69e17d7b0a"}]},"branch":"refs/heads/master"},"ae06117bd612bf96e288c7ad9d9a1eced9f4e63c":{"kind":"REWORK","_number":42,"created":"2026-03-16 18:48:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/42"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-16 13:04:21.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae06117bd612bf96e288c7ad9d9a1eced9f4e63c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae06117bd612bf96e288c7ad9d9a1eced9f4e63c"}]},"branch":"refs/heads/master"},"9444127fc8e799d3db1d4c6ee042c965ea2a1262":{"kind":"REWORK","_number":43,"created":"2026-03-17 15:34:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/43"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-17 15:24:01.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9444127fc8e799d3db1d4c6ee042c965ea2a1262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9444127fc8e799d3db1d4c6ee042c965ea2a1262"}]},"branch":"refs/heads/master"},"297da3cb38579e724d2fde5f165ad23b25615bf6":{"kind":"REWORK","_number":44,"created":"2026-03-18 14:50:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/44"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-18 14:50:35.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/297da3cb38579e724d2fde5f165ad23b25615bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/297da3cb38579e724d2fde5f165ad23b25615bf6"}]},"branch":"refs/heads/master"},"ae0ef5d59cc824b2ac05c717a79aff61f86999de":{"kind":"REWORK","_number":45,"created":"2026-03-24 17:58:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/45"}}},"commit":{"parents":[{"commit":"7814c0cb5620527724695ac6b1f48d27572a3767","subject":"timestamps: support offsets with Timestamp.__invert__","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7814c0cb5620527724695ac6b1f48d27572a3767"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-24 17:39:56.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae0ef5d59cc824b2ac05c717a79aff61f86999de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae0ef5d59cc824b2ac05c717a79aff61f86999de"}]},"branch":"refs/heads/master"},"49dc0711dac95caefd9bb857440eaf1effa2384b":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2026-03-25 16:30:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/46"}}},"commit":{"parents":[{"commit":"92ca1b8df7cc6521206a72b73d457254d9517725","subject":"timestamps: support offsets with Timestamp.__invert__","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92ca1b8df7cc6521206a72b73d457254d9517725"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-25 16:24:45.000000000","tz":0},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49dc0711dac95caefd9bb857440eaf1effa2384b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49dc0711dac95caefd9bb857440eaf1effa2384b"}]},"branch":"refs/heads/master"},"1e818bdd5f757de1411e685f5d2baf04018d0259":{"kind":"REWORK","_number":47,"created":"2026-03-31 14:54:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/47"}}},"commit":{"parents":[{"commit":"694d25bb1a87f9426ac7ad1f3817c51f5bf13a34","subject":"Improved offset support in Timestamp.__invert__","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/694d25bb1a87f9426ac7ad1f3817c51f5bf13a34"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-31 14:53:54.000000000","tz":60},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e818bdd5f757de1411e685f5d2baf04018d0259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e818bdd5f757de1411e685f5d2baf04018d0259"}]},"branch":"refs/heads/master"},"fa792e5ad5eec34d69c624136d995c5f9ff689b1":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2026-04-09 09:21:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/48"}}},"commit":{"parents":[{"commit":"e0b5aa3a26c53cfbf9969bda9b8f9541f9252bb3","subject":"Merge \"tests_multi_get: fix timestamp formats\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0b5aa3a26c53cfbf9969bda9b8f9541f9252bb3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-09 09:11:51.000000000","tz":60},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa792e5ad5eec34d69c624136d995c5f9ff689b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa792e5ad5eec34d69c624136d995c5f9ff689b1"}]},"branch":"refs/heads/master"},"5c8af87ca882ef822fe675f5a2cead1180a210b3":{"kind":"REWORK","_number":49,"created":"2026-04-13 09:47:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/49"}}},"commit":{"parents":[{"commit":"76fa89cc2823212bb1de195943648fec9edebcc2","subject":"Merge \"versioning: fix listing gap when version_marker\u003dnull\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76fa89cc2823212bb1de195943648fec9edebcc2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-13 09:46:51.000000000","tz":60},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c8af87ca882ef822fe675f5a2cead1180a210b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c8af87ca882ef822fe675f5a2cead1180a210b3"}]},"branch":"refs/heads/master"},"2710b1ff9fc434b41f2d8427e864451bd2c064d7":{"kind":"REWORK","_number":50,"created":"2026-04-14 15:09:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/50"}}},"commit":{"parents":[{"commit":"7829fb90b630b2f80e3fff1b18fcd5f9ca8677d2","subject":"s3api: give MPU Clock Skew 503 a reason","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7829fb90b630b2f80e3fff1b18fcd5f9ca8677d2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-14 15:08:56.000000000","tz":60},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2710b1ff9fc434b41f2d8427e864451bd2c064d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2710b1ff9fc434b41f2d8427e864451bd2c064d7"}]},"branch":"refs/heads/master"},"e4ec23afbfee9a57fd0dfce427d2867d694f480b":{"kind":"REWORK","_number":51,"created":"2026-04-15 14:24:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/51"}}},"commit":{"parents":[{"commit":"7829fb90b630b2f80e3fff1b18fcd5f9ca8677d2","subject":"s3api: give MPU Clock Skew 503 a reason","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7829fb90b630b2f80e3fff1b18fcd5f9ca8677d2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-15 14:23:19.000000000","tz":60},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4ec23afbfee9a57fd0dfce427d2867d694f480b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4ec23afbfee9a57fd0dfce427d2867d694f480b"}]},"branch":"refs/heads/master"},"0253a6bb44f68d8a9cf50763db7bb9baddfd8f97":{"kind":"REWORK","_number":52,"created":"2026-04-15 14:31:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/52"}}},"commit":{"parents":[{"commit":"7829fb90b630b2f80e3fff1b18fcd5f9ca8677d2","subject":"s3api: give MPU Clock Skew 503 a reason","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7829fb90b630b2f80e3fff1b18fcd5f9ca8677d2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-15 14:30:07.000000000","tz":60},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0253a6bb44f68d8a9cf50763db7bb9baddfd8f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0253a6bb44f68d8a9cf50763db7bb9baddfd8f97"}]},"branch":"refs/heads/master"},"7b5c965238ddf7a43f69684074ea3a98e53e7efe":{"kind":"REWORK","_number":53,"created":"2026-04-16 15:49:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/53"}}},"commit":{"parents":[{"commit":"7829fb90b630b2f80e3fff1b18fcd5f9ca8677d2","subject":"s3api: give MPU Clock Skew 503 a reason","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7829fb90b630b2f80e3fff1b18fcd5f9ca8677d2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-16 15:48:39.000000000","tz":60},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b5c965238ddf7a43f69684074ea3a98e53e7efe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b5c965238ddf7a43f69684074ea3a98e53e7efe"}]},"branch":"refs/heads/master"},"cf5df8fac80ff354ee8e6369dbf18a5e1fd75343":{"kind":"REWORK","_number":54,"created":"2026-04-16 15:58:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/968740/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/968740/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/968740/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/968740/54"}}},"commit":{"parents":[{"commit":"7829fb90b630b2f80e3fff1b18fcd5f9ca8677d2","subject":"s3api: give MPU Clock Skew 503 a reason","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7829fb90b630b2f80e3fff1b18fcd5f9ca8677d2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 14:54:26.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-16 15:57:15.000000000","tz":60},"subject":"timestamps: add NormalTimestamp for ShardRanges","message":"timestamps: add NormalTimestamp for ShardRanges\n\nShardRange timestamps, in particular the epoch, must never have an\noffset part. This patch introduces a NormalTimestamp class that does\nnot support offsets, to be used when setting ShardRange timestamps.\n\nThe intent is to be very explicit about the different types of\ntimestamp:\n\n ShardRange timestamps should always be NormalTimestamps.\n\nNormalTimestamp also replaces the use of Timestamp in other places\nwhere an offset is never expected or needed, for example:\n\n* when timestamps are being compared against wall time.\n\n* when autocreating DB\u0027s.\n\n* for container DB status_changed_at timestamps.\n\nTimestamp.normal has the same form as NormalTimestamp.normal.\n\nTimestamp.normalized() returns a NormalTimestamp with just the float\npart of the Timestamp.\n\nChange-Id: Idcf28c29d52a52830ff12be9c9b1235fd8d23738\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf5df8fac80ff354ee8e6369dbf18a5e1fd75343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf5df8fac80ff354ee8e6369dbf18a5e1fd75343"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
