)]}'
{"id":"openstack%2Fswift~968852","triplet_id":"openstack%2Fswift~master~Id85e2f3decb386e5f865491357bc98750d287af9","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-01 11:55:26.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-12-01 13:42:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id85e2f3decb386e5f865491357bc98750d287af9","subject":"test_reconstructor.py: make hashes assertions less brittle","status":"MERGED","created":"2025-11-28 18:13:23.000000000","updated":"2025-12-01 13:43:20.000000000","submitted":"2025-12-01 13:42:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968852-p-timestamp-collision","meta_rev_id":"e9876c61e87a01f027c112160a47246b1834c62f","_number":968852,"virtual_id_number":968852,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-01 13:42:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-12-01 11:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-01 11:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-12-01 11:55:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 19:10:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-01 03:25:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"7ffc9403623e65d017cd72bbc6e3a55426f068e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-28 18:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe96aef0033799a696465c80f0869f5f48f46962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:10:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/773298fd03d6441b89d787dd9b97c253\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24fbaaf23779431cbbf154b8649215bc : SUCCESS in 10m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afb0013ef7204f5ab03e0c1482c9e709 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/200d5245b17a490f9ace1b4f0d1ed072 : SUCCESS in 23m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/baf91be38af9422eb734b8fff099531b : SUCCESS in 20m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f214f0068699411ba572d5965c4b5308 : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b9a92ee188245f6a6e823fa50e7d3fe : SUCCESS in 24m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0af3ca1c153741d0be0d3dc5d18efac1 : SUCCESS in 8m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1dd77ab0503d40349da0806f2b4b4764 : SUCCESS in 18m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/60cf7048ac374db6be6a6abf210d1fdb : SUCCESS in 30m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/991d888e47b148eea7b8a0488bea5727 : SUCCESS in 12m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4a9797123dd6487288ac2a4d5e355d49 : SUCCESS in 13m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d96902f0ec6d4b30907d39cbd73cd298 : SUCCESS in 8m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77c6681c477b47b6b72b76df372fc6ec : SUCCESS in 49m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/57e4e662fb72424e80c2dcc74c40352e : SUCCESS in 34m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a315d3092aa743c7a6aef705846710a2 : SUCCESS in 26m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c6950e42314a1f90fe23f8b362b69d : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/008b31ad44b8483b957b3b1bc1080aa0 : SUCCESS in 18m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7e7f01c0b79ed79104bc37099be77754003475d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:31:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cb18c1e68f64332899063e99f1c2dad\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1cca157014ec41859795743f13c8c66d : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":1},{"id":"149df010dd7bdfd23bc4fb836db1605c3f9ccf98","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-12-01 03:25:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f87ed260a503362e1bac8d7821e9d9ccb073756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-01 11:52:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae8ea868afcbee98d2c11a71e59f654e48b51ae3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-01 11:55:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e6f0a090bbea7f81e888c63ae11822188134d98e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 12:47:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb9677a21294c72bd7d968e8d03763b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eeae7367301430ca275877c7bdb46cb : SUCCESS in 9m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6748461bb3df4330b52dad328628ed24 : SUCCESS in 16m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc4a0628cb6c4069a22fec92ee0695b8 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a0328967c5e448399131eb99676d620 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d81152a265fb4df7a5b8319a4fc6d35a : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/df35a1ed39b0443ea4f183f895b0f6eb : SUCCESS in 24m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9cc6427934684c37a8fe30da1b1880f0 : SUCCESS in 21m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5eed2d8a2cac4ffdb12e7392369ccbd9 : SUCCESS in 27m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2644e73cb0374f7096e89b36b1d87fa4 : SUCCESS in 22m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3623341f6d214cdeafb62d935214888c : SUCCESS in 13m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/05a25c2529dc4087be2421dc09a0ca9f : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/110d45cc74794b0d9627fc60bfc5f781 : SUCCESS in 8m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/12d0f3d630604837a142cffceb8512fd : SUCCESS in 48m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/177e18fdfc8c43379bdea3bafb1b1560 : SUCCESS in 54m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ae566c0c39d4182b48776e5dcf93a93 : SUCCESS in 22m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9058586b05e04d56ab86232bfcc16dad : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93cff55d7efe4a3282371b10f6c586cb : SUCCESS in 19m 16s","accounts_in_message":[],"_revision_number":2},{"id":"f695045c054de0f760c1a7b83acc7a4acb5b73b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 12:47:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a3fd50cec1721acd16d818b590553c011d2adf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 12:52:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25c6f6b4276c4f1ca142260b4ba7e597\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dacbe6cba9824d36b59b49d9d7502527 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8f1929ff7ba6d414cac73668d54d1b8ec4925d26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 13:42:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a2069e17a7c445a9bbcc9af1dd13ccc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89938ecfccb4945b50a59fdf08b283f : SUCCESS in 9m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/74181156d7464fdf9a21cc16337929b0 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6be51d8fd9f1416ba2ff96b6a571aff5 : SUCCESS in 25m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6f11fc13de4a46cda43c6d0ea1f5d5b8 : SUCCESS in 18m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6ac80a425fdb49fc913c3c1d72e21b47 : SUCCESS in 26m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c6f446906f7d45b6b3f32939326ba7ab : SUCCESS in 31m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5fdfb5c4e4954e4d86139c580ad9867e : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1d748be5edb34fbb988f172888484e28 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/69bb09f46d7744519e9b32ad68434b6a : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7be90d094ea4e438ec9a0279de2b413 : SUCCESS in 3m 56s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/25257cbb84394d28ad4f0b92c75a2394 : SUCCESS in 8m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6d948853a77241518b7e57b816e95cc5 : SUCCESS in 22m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b6a652b8c25447297afb496d5375c88 : SUCCESS in 49m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00c2ac67e39748b2b9e2721e3b74351b : SUCCESS in 28m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c832d127f040fc814553862ae78ca3 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/604902e071d643d792c246d10e00dfea : SUCCESS in 19m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8fa29f48a96861ab97e1bc246128e28195544031","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 13:42:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e9876c61e87a01f027c112160a47246b1834c62f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 13:43:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12d4c8dd65cd44d28d63c754840ba86f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c8c986a87cc4f25af08c0c46092edd1 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/23fc52745b82442abb56102c0685727b : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"800cf321140c276950cc94c01eb13a10a4022fcf","revisions":{"41ffb4e37fcf251d18cfcb5aec9db8c08b649f2c":{"kind":"REWORK","_number":1,"created":"2025-11-28 18:13:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/968852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/968852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/968852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/968852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/968852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/968852/1"}}},"commit":{"parents":[{"commit":"755d9d41e9868f428afcdace628b6437bcbdc224","subject":"tests: container/test_backend.py cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/755d9d41e9868f428afcdace628b6437bcbdc224"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 17:03:25.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:13:06.000000000","tz":0},"subject":"test_reconstructor.py: make hashes assertions less brittle","message":"test_reconstructor.py: make hashes assertions less brittle\n\nReplace the hard-coded expected suffix hash values with values\ncalculated using the timestamp. That way, if the timestamp ever\nchanges then the test will continue to pass.\n\nChange-Id: Id85e2f3decb386e5f865491357bc98750d287af9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41ffb4e37fcf251d18cfcb5aec9db8c08b649f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41ffb4e37fcf251d18cfcb5aec9db8c08b649f2c"}]},"branch":"refs/heads/master"},"800cf321140c276950cc94c01eb13a10a4022fcf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-01 11:52:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/968852/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/968852/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/968852/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/968852/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/968852/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/968852/2"}}},"commit":{"parents":[{"commit":"d31b201d6b6bc4fbb34a8bf183b0965d95d29fcb","subject":"test_db.py: clean up usage of Timestamps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d31b201d6b6bc4fbb34a8bf183b0965d95d29fcb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 17:03:25.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 11:52:24.000000000","tz":0},"subject":"test_reconstructor.py: make hashes assertions less brittle","message":"test_reconstructor.py: make hashes assertions less brittle\n\nReplace the hard-coded expected suffix hash values with values\ncalculated using the timestamp. That way, if the timestamp ever\nchanges then the test will continue to pass.\n\nChange-Id: Id85e2f3decb386e5f865491357bc98750d287af9\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/800cf321140c276950cc94c01eb13a10a4022fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/800cf321140c276950cc94c01eb13a10a4022fcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
