)]}'
{"id":"openstack%2Fswift~968853","triplet_id":"openstack%2Fswift~master~I8b84fa258ae0071cb171c4b5dcdb86b01980c33a","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-01 11:55:46.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-12-01 13:48:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b84fa258ae0071cb171c4b5dcdb86b01980c33a","subject":"account/test_utils: cleanup Timestamp usage","status":"MERGED","created":"2025-11-28 18:13:23.000000000","updated":"2025-12-01 13:49:48.000000000","submitted":"2025-12-01 13:48:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968853-p-timestamp-collision","meta_rev_id":"66652e998dc03d601861ad1b8172ef6d2d681f5d","_number":968853,"virtual_id_number":968853,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-01 13:48:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-12-01 11:55:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-01 11:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-12-01 11:55:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 19:08:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-01 03:37:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"675882115191f3f6c1f4ed0b4143de331f4bcfa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-28 18:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdbc59138c695cddb925cbf6ac8b43c80093fb99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:08:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ee717500c8c43369063cf5a3206d2b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d24a404a004dd5b7654b46e8ea4a3e : SUCCESS in 9m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5afcb791d4444073a1e3460b1ec2afb5 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef7bc8dda7364f3d8f848ddd33ed4830 : SUCCESS in 23m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e267b6ee1d6449cf9c1a9b282cd0a880 : SUCCESS in 19m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/853d6e7b35094644ac7c5653584f6d2c : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/267f66627fa942c78f783f6f4d7529ab : SUCCESS in 18m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c9340ad496594cd9ad0ba43ed533e556 : SUCCESS in 13m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e1d2a921c6724fafb617d2a2254339ed : SUCCESS in 30m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e029d240455146199a3a97309903b3d9 : SUCCESS in 31m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/645dfe3df14d4dc88a57aadcc59e60a6 : SUCCESS in 11m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d2b952ed50cd42a48a382f71a97f86d6 : SUCCESS in 14m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9e92e348cb6b43a5bdc878b081720f90 : SUCCESS in 8m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/21bd17a580814a7a896019ff4650740e : SUCCESS in 48m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2761307c5364bfb9b5c8f4ebd5481a5 : SUCCESS in 39m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7940f0b9524e46f0b8d5800b321a45c1 : SUCCESS in 28m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7414a04028aa488eb0c9056a0ae161fe : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd84cc0c7a4542d1b91050df30b28295 : SUCCESS in 17m 14s","accounts_in_message":[],"_revision_number":1},{"id":"34f3a096941eb9fb70b4b643799c4a5c1fb68658","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 20:23:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f679d45d8747e2acf3c9545ff29a71\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4242b0dd3e74a9a8db56abeb076a74c : SUCCESS in 47m 57s","accounts_in_message":[],"_revision_number":1},{"id":"88264aa9bc7bc10eda54a4a8e176c482d1eeede0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-12-01 03:37:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f916311cdd497026f196b65b9ca9cd591cd3e71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-01 11:52:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"972acf80d9ff0438f7b5d287859bf8e6f3c60ddc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-01 11:55:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"54de995347c23029b252b64d3503620f7179b68a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 12:29:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e0b338b9fed490c9619ea663206d11b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c9332692c754136978017fdbda759fe : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":2},{"id":"170f534a48ef53ecb039c17b924396bf15757f55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 12:42:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43eb33f7ed0046d6b415f56ac0f57bf0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/992e2d73cead479a893fecb0d9415691 : SUCCESS in 10m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1832699591ba45ba84195e40a06721af : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/256fba15063046e8b0a3d7e7832ba6fd : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4da169e4b234ea080daf48003d38180 : SUCCESS in 20m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dfd1367fec7440ed919a2bf6330f2051 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/69183ba7011e49f28eb6fb88c7b5e040 : SUCCESS in 22m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6c1d49bc957433181efb586af75fd4f : SUCCESS in 18m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5d67861c3d8d486b8edb86629cf0080c : SUCCESS in 33m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a5db392b600e482fba025159191ae2ba : SUCCESS in 27m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b600ddb75ddc4d86bf4db41cd6783e29 : SUCCESS in 13m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cf1fc3579bf44ac48aba13acbb0898ff : SUCCESS in 13m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/82f3a89cc7f643038a2364bade301560 : SUCCESS in 15m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ac5a552834e43b4ad464f2f950f8fbb : SUCCESS in 43m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45e556a6ec9a4c1cb65c1309ba8bc4c1 : SUCCESS in 44m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e058c033ba44ac7a9d19446548c08ae : SUCCESS in 25m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb1544c6812432ea7283d1a1ed3c6aa : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0e10fa6f6f94f32a0cd46244748fbb5 : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"e3c425f700212e1284f615ae36e4f8c6af4c1656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 12:47:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3a6a904bff4f9b9a777a923095cce0e5d8f2e529","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 13:48:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dfa202befb24af1a6a3fcc0886bbb97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be422b329dd64cd39635729c07196301 : SUCCESS in 11m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5588bba28c104b89bcb0ecb0c9a3b6d8 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c2bccc5f05bb465695c4729a3000edf6 : SUCCESS in 22m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/885eb6fbc15647c2b6518b417c3d5c16 : SUCCESS in 18m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dd08523f0de14480821c5e23bb543281 : SUCCESS in 23m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2d3b80b293c74c4baa02484974dec9e6 : SUCCESS in 24m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0ed5b2226a5f4e4f92be3c79f24c7243 : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6dd0a6e276654470bfeb9f2e9cc86cc8 : SUCCESS in 9m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e4f4cdf117f3462986bdaa5dd84e2628 : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/08f3cc56770f42af86a6aae48673c290 : SUCCESS in 8m 31s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/02e31525dd8d42fc955f052d1e131868 : SUCCESS in 8m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f90c5d48c5b24300a52911f9d605d046 : SUCCESS in 50m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ffd890b83dbe493ba0b3e0bdcec7a94b : SUCCESS in 53m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/413889de9f72420098f1054627db3a82 : SUCCESS in 28m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65bc3489a95a41beb12a1abcf941fb97 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5817900f5a064971aad05902ff26b943 : SUCCESS in 17m 27s","accounts_in_message":[],"_revision_number":2},{"id":"03f1281889f4fcb022260bf413de3186f40de110","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 13:48:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"66652e998dc03d601861ad1b8172ef6d2d681f5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 13:49:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6cef7cc56784e99911643e6065f1b38\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0331a1f09b9c47779cab3107019e2814 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/549a9f89aa034853ada85ec24b45975b : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5658355bf5a7d888c8682a7925f5d5fa4a5ee3c3","revisions":{"16269f4984b4acd97d4766e9b0e9c524c0e4b12c":{"kind":"REWORK","_number":1,"created":"2025-11-28 18:13:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/53/968853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/968853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/968853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/968853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/968853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/968853/1"}}},"commit":{"parents":[{"commit":"18421c1b92e6e41bc3ae7c09b2f544b1c551a167","subject":"test_diskfile.py: fix Timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18421c1b92e6e41bc3ae7c09b2f544b1c551a167"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:07:49.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:13:12.000000000","tz":0},"subject":"account/test_utils: cleanup Timestamp usage","message":"account/test_utils: cleanup Timestamp usage\n\nChange-Id: I8b84fa258ae0071cb171c4b5dcdb86b01980c33a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16269f4984b4acd97d4766e9b0e9c524c0e4b12c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16269f4984b4acd97d4766e9b0e9c524c0e4b12c"}]},"branch":"refs/heads/master"},"5658355bf5a7d888c8682a7925f5d5fa4a5ee3c3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-01 11:52:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/53/968853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/968853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/968853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/968853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/968853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/968853/2"}}},"commit":{"parents":[{"commit":"18b861bd6ce48e45251d709ba190a32392ec9954","subject":"obj test_expirer.py and test_auditor.py: fix timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18b861bd6ce48e45251d709ba190a32392ec9954"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:07:49.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 11:52:24.000000000","tz":0},"subject":"account/test_utils: cleanup Timestamp usage","message":"account/test_utils: cleanup Timestamp usage\n\nChange-Id: I8b84fa258ae0071cb171c4b5dcdb86b01980c33a\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5658355bf5a7d888c8682a7925f5d5fa4a5ee3c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5658355bf5a7d888c8682a7925f5d5fa4a5ee3c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
