)]}'
{"id":"openstack%2Fswift~968855","triplet_id":"openstack%2Fswift~master~Ibd5ec2fe65cdc454c46ecb5a490e864763f503b7","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-01 19:49:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd5ec2fe65cdc454c46ecb5a490e864763f503b7","subject":"test_constraints.py: cleanup Timestamp usage","status":"MERGED","created":"2025-11-28 18:13:23.000000000","updated":"2025-12-01 19:50:25.000000000","submitted":"2025-12-01 19:49:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968855-p-timestamp-collision","meta_rev_id":"1179bf61756ab96b4f14c57e13cd1d2bb518bb65","_number":968855,"virtual_id_number":968855,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-01 19:49:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-01 18:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-01 18:12:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 18:51:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-28 19:16:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-01 18:12:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"146831574e2d3e51b23e245cbeda71b90f446eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-28 18:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f010b6f61b828ce396b617f0cce413dcd084f445","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 18:51:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/548fa13daad642bd9de813414c724521\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac48fbeebb7945f884b28a8f0270c72a : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8da888642bf44575d0a0010a15fb238ee1d29138","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:16:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b755a427af84b3b850eadc0f1945905\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5c39fbc7ee4c75ab5618969f956392 : SUCCESS in 11m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51629c8472af407cbfc4cb9b8d29a597 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/783a66a87eb04a3ea072c095ccaeef1d : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3743c33bebe44e549083e809f214b9e5 : SUCCESS in 24m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/41bfe9d2533445baad5def97285302cd : SUCCESS in 8m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dece997476f440e99faecaefba4d9b55 : SUCCESS in 14m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5ac23cb78ae348128deee4f0ae566a54 : SUCCESS in 8m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b85b855472234ad593f7eb54d3d7b91c : SUCCESS in 24m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/77c1517a8ae54330bf1ec91ae0ce51eb : SUCCESS in 29m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/80cf708b3a594ccaaae69b37edad4c41 : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0736a03fa91e4bafbc207668110c2e8b : SUCCESS in 15m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/755176d5bbd54b5b9fb89657a1605986 : SUCCESS in 17m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8dec76e4432a425d9ff380c3004e1c56 : SUCCESS in 43m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a28230d93b54a4fb8a1c24d3f64f3a1 : SUCCESS in 56m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1912de9fabf947a096b3d3663bbef670 : SUCCESS in 27m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/300f16fa66bb41d385c1c35825dd3fbf : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/af4da02f0f0440bfa02953d126244f6a : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f9a148806c334f468503b0075bee08f542f8b781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-01 15:37:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"36aba8d9d8c039fca06a96fec0bef925b77e5a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 16:30:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d693cd311cb14d17ae15edb19594b17c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e498eef46184868811b5ba28d85d6d3 : SUCCESS in 10m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8168adf81361497c80212fb0a9b6632e : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51fc9b3ff3b14c9dac6b0b7ef86a54ea : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/438b1e5a7c0247e49141fc9d445e11a5 : SUCCESS in 21m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f9fdf8bc588439a8607c87997876535 : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ab777d04ac5f4ddf8507bb847a0d4e49 : SUCCESS in 25m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0803eb8b1b9d44928e2f86a8cc813a03 : SUCCESS in 20m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/111f4aac67c54e20b00bcf0f2759b438 : SUCCESS in 27m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1152398efe3e4167a842f92d07c42d3b : SUCCESS in 22m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6e40c604a4ce44928ed7ad6bacc66773 : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1301f24849f74d8b87fe8d686786538d : SUCCESS in 14m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a61bebdb52b04718b3aabf27c9072190 : SUCCESS in 9m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1142480768e74656ac6e4e8a50e4969d : SUCCESS in 48m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc312c41ffce4853a68490265eed8f59 : SUCCESS in 26m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97ee27ab8eb3455caacdebaabf2a1ecb : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fead4a93b2db4c03968ce0d8fbb380f4 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f872a13c2e04d7eb5e5c4b7c887f255 : SUCCESS in 15m 37s","accounts_in_message":[],"_revision_number":2},{"id":"89329aab8f434a91f7e43bb62ffb0b4441ee303e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-12-01 18:12:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b4d63a2e28d437cc89c69d7d2ae31a6ad3c5a371","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 18:12:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dfedd343e0b6d482df34f6c18befb549c42e5bca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:22:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d2487285ac54192a3ed812fc3bbd7c9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f4776d7063ee4598a54f59723d5d06a5 : SUCCESS in 1h 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6acd85cedeaf79cbc4f8aaef24737705741c24b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:49:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94b322bee0ba46c68fbc97462eacea2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734969b947664217a50a432a01f546d2 : SUCCESS in 6m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c6c818574b774a48bee2b6fc8cd82c02 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d589142899c443bb8fa66082e556cc99 : SUCCESS in 29m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1417e5e2c40b41b99206617d612e7f1c : SUCCESS in 10m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/108f326811eb4402aaed5a74801ecc1e : SUCCESS in 28m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6630f4ced9864726906a5c4144b0a584 : SUCCESS in 30m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/15cf2abc6e6a4bd095be3814d3894607 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c6c035d1f8b5405aa124afda09a7097e : SUCCESS in 15m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e9df64bb4a0940a791705b8b11c5483b : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/284b98f89dd64a068fa1bad720f3fdfe : SUCCESS in 7m 01s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f2a6aab0a5644dc8a7c9c83e50710675 : SUCCESS in 6m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5fea78dd2c1541dc8c697a5efdd1a49b : SUCCESS in 57m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ae8bfb301c246c9920693f692a8ecd2 : SUCCESS in 35m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1e73714a3c74032b6b6f2882f5bb96b : SUCCESS in 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d1a1836d69451cb5942800f6de4414 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/549bccbedf554efcb51173585bc55521 : SUCCESS in 11m 32s","accounts_in_message":[],"_revision_number":2},{"id":"8e1962827aadaa51f4277240ed04318bd08e0787","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:49:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1179bf61756ab96b4f14c57e13cd1d2bb518bb65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:50:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c6a0b0ed5344a92a54741c5fd2bb637\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebb065d3c0f54904a89ac318e971f1ed : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b9e8d440ef9243e99ec9e646f123f582 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a70e33e73a361a6622c093d025165e855ec6c100","revisions":{"101b643a2e3486539aa49cd3410d84da39e38503":{"kind":"REWORK","_number":1,"created":"2025-11-28 18:13:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/968855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/968855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/968855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/968855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/968855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/968855/1"}}},"commit":{"parents":[{"commit":"a8ffc8165705a130af7be03517429fef2dcac0cc","subject":"test_symlink.py: cleanup Timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8ffc8165705a130af7be03517429fef2dcac0cc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:08:51.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:13:12.000000000","tz":0},"subject":"test_constraints.py: cleanup Timestamp usage","message":"test_constraints.py: cleanup Timestamp usage\n\nChange-Id: Ibd5ec2fe65cdc454c46ecb5a490e864763f503b7\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/101b643a2e3486539aa49cd3410d84da39e38503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/101b643a2e3486539aa49cd3410d84da39e38503"}]},"branch":"refs/heads/master"},"a70e33e73a361a6622c093d025165e855ec6c100":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-01 15:37:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/968855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/968855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/968855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/968855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/968855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/968855/2"}}},"commit":{"parents":[{"commit":"91dd65d36b88629aea4230279c03f6dd679d1edd","subject":"test_symlink.py: cleanup Timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91dd65d36b88629aea4230279c03f6dd679d1edd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:08:51.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 15:36:51.000000000","tz":0},"subject":"test_constraints.py: cleanup Timestamp usage","message":"test_constraints.py: cleanup Timestamp usage\n\nChange-Id: Ibd5ec2fe65cdc454c46ecb5a490e864763f503b7\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a70e33e73a361a6622c093d025165e855ec6c100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a70e33e73a361a6622c093d025165e855ec6c100"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
