)]}'
{"id":"openstack%2Fswift~968856","triplet_id":"openstack%2Fswift~master~I750ac1ab2a75d1a49c87c6e1704ebe733fb31ec2","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-12-12 18:56:24.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-05 13:14:37.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-12-12 18:56:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I750ac1ab2a75d1a49c87c6e1704ebe733fb31ec2","subject":"slo: use Timestamp.zero() as default in RespAttrs","status":"MERGED","created":"2025-11-28 18:13:23.000000000","updated":"2025-12-12 18:57:35.000000000","submitted":"2025-12-12 18:56:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968856-p-timestamp-collision","meta_rev_id":"06dbe41ea678332f05b9d3146d9555ddb99cddf2","_number":968856,"virtual_id_number":968856,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-12 18:56:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-12-05 16:53:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2025-12-12 17:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-05 16:53:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2025-12-12 17:31:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 19:33:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-01 20:46:36.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-12-05 01:17:06.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"570607ebd48239f23bf13bc92513b398c0f53ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-28 18:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff93a3eb1c83e4276dbe70bcd6d50e0b73c14175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:33:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4b21f73f774fb9be1a1516f6c774f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b6b9f88c264b3b8a752203a31f3458 : SUCCESS in 5m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/b23c70498f7a402b8e789ac64a1ba0a2 : SUCCESS in 59m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8afa3a259b1d42ccaf88af80a032ecc8 : SUCCESS in 1h 08m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f24331d7a28c44f9a319bd63ea99a44f : SUCCESS in 39m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b442d7c911d34d08ab0e5f0e9e4a966f : SUCCESS in 1h 09m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5d55f1762f04bdc9cb344928a7a82d1 : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0dfccffc09564130bbc63b3c3f352a2c : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10d23a1f626c4baa809232e34733039d : SUCCESS in 21m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a1a90164329f481bb8ce1753488ec686 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/06bf562e4deb4c4ea3314ea138b6c4a2 : SUCCESS in 25m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/064d591b648e4fd98ce76b029115a217 : SUCCESS in 8m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/db1e0568ed5946dfbafdaadd1f172326 : SUCCESS in 28m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0aa842c6244a47ebb01ee29e23b42a96 : SUCCESS in 11m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2a513e9587aa4d60a64649ba978357fd : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c2f8097def7c436abe7428e6f47dfb06 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/965fe41f739b45dba549736f0a1b4475 : SUCCESS in 15m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f05b075b35be43698ccee0cf41f6383d : SUCCESS in 9m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6fbe794fcbaa433fa22654d0bb417021 : SUCCESS in 15m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/59eec10bcb0a434e840e3d4f309a1507 : SUCCESS in 8m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d235e9b60dc4753a0aeb51a9039d5c2 : SUCCESS in 56m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f934a88bec24499bab5d74a80f2f216c : SUCCESS in 53m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c71484000a82447b9ef311c42838a14b : SUCCESS in 48m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/533f6c9abf1a4ee39198e03b80e14d97 : SUCCESS in 27m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232630b8146e40359d535b39bc2ee8e4 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0c8e13cdd494796984f0b6040b3b55f : SUCCESS in 19m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae2d6d6d98d84593923cea380143000d : SUCCESS in 38m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6af94cf81ded43c7aa71e613c485ee5f : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":1},{"id":"bafdce1077f1c847d233859a55b6969b2385749e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 21:40:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/538502ab61d84b31bddd342d0843c4b4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/87217ba8b99d41448be2020ed5050525 : SUCCESS in 1h 44m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7c200292068bda2c166b74bc42f1d09f108cd278","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-01 15:37:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5cd0939ddc6d32349067c019a7ae4af96e29ca7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 16:46:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dcc7f29fa354b1898d0ad0bc127514b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb4cc05ffe84f438c8f764304cce45c : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/276a84f7cd114e5090c52b025e9293e5 : SUCCESS in 1h 02m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37eecd8b51e844c19eb3d4f349c31dfe : SUCCESS in 44m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/abfb18142740473a875bf48eb35e45a5 : SUCCESS in 42m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/501d67b7f3a4453aaac1de944c154634 : SUCCESS in 59m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f5edc278d4f4afa91e796e1a8a72811 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00f779991a0e4835a826f8f3cd700323 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5c90b8544ea04ba995f61d161aca0682 : SUCCESS in 17m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53af877dd85c420aa7641f07b2c42dc5 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dcb87a483251478094531f9f08a24065 : SUCCESS in 26m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cc69365350024a7e9f7b6745a75480eb : SUCCESS in 11m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9e73574b4e904c648cccac068a36606b : SUCCESS in 14m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/198ee04a03154723aa60c672b27b785d : SUCCESS in 11m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1644434470874b9a9432dbdd0457b78d : SUCCESS in 10m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/219619418727450a9f5f53844f946024 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8167639073334ad28a557a1db3396b19 : SUCCESS in 10m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2ab52d671cb1425f93a4f6347e86900e : SUCCESS in 3m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e68b670bc59143eaa5dc687d9268a57b : SUCCESS in 14m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f9ac786dd744dc38d4f051b3718ebd7 : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a13e07b9b2074eec99c84c17e4b45150 : SUCCESS in 51m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/665e8720fcae4394bfb39f94c38ca3b2 : SUCCESS in 22m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d56a09aadf1646bca3d0b2534fd195f5 : SUCCESS in 35m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b143eb45e6e44738bb6574aa879f61b4 : SUCCESS in 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26eee8b0531c49f09f51b5939b8ae2ab : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f991d571dfb3478eb97e6d3daf780207 : SUCCESS in 17m 14s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a132735e885941b6840f85f299a6c2a9 : SUCCESS in 45m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62c41c903fb94978abeda9601c31a6e6 : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":2},{"id":"15e88b9d6b1a8d6e7bc06b4655a00becba031d35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:24:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82a1662e4c19460a919db857b54005f5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f1da3279528243c288887725cf709aae : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":2},{"id":"256bd67093b592555d93201628bde12c034c4385","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-12-01 20:46:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"860b828d3dd3c2e6304e05d5affef68c7933dd1d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-02 12:31:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dcca8cb53095d2087fa62ac41ca72e3d7e9fc8f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-02 18:34:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a4c93cc99420b2c9420a6bec5d659dd9e26b7605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 19:44:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33f09086fe6045e1989c918c61484688\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b72add29ddd47bbb9503475699bf9da : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/4767aef6de9b465cbd4839077f7f5c3c : SUCCESS in 57m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e8a7093a2d2345c3995de0b6b41dea78 : SUCCESS in 1h 01m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2ebb553f86654a49bcc98089a53e16d3 : SUCCESS in 35m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bb2620c2c1c426d883100edc89cc6d0 : SUCCESS in 1h 04m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b12360986b44d72bbe4495056b182ef : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0565c253c37b494a8c276d3e4ef197be : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33989b977824428dbfbaca7c34e2eefa : SUCCESS in 11m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/65547c4bca3447c2b5bc86ab5e08ea28 : SUCCESS in 7m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/60507d86ca4a425b9911b3552708f776 : FAILURE in 24m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b0dc6a09a9674f468e3c8ff01e9928a0 : FAILURE in 16m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c781af0cd42e42c681f1bd70efa8f6df : FAILURE in 37m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/448146f6a9e64e728b1ffc0d60c01d11 : FAILURE in 12m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/56ffb84eef294b03a3d76ddc52f4fdbe : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d55026b6974d448d9673b856888fd94a : SUCCESS in 15m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5fb367ff05b849699aaf3eb29592d5c1 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f4e8e1a05f448c1bc8ab16cabb2b3ae : SUCCESS in 6m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6b889c3c81b465fbb3764ab1de50462 : FAILURE in 19m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef2c41f7c6aa4515bbb54b7d39ed1164 : SUCCESS in 7m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e34ee64de16647d4b1dd772d98bd00db : SUCCESS in 50m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/134dae7d2b744c6bb3d38ff70553621b : SUCCESS in 37m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/adb6d8e65e204f6e853f68fbfe63e46b : SUCCESS in 48m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b68e4c2f0bc4fc3a51f3513f36d004f : FAILURE in 28m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e21b1ce73147218b53ea442c12f28b : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6bb22c6a0dcc40418aacb3ab64ce26cc : SUCCESS in 25m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15757425a3944630a72da23b60c989e6 : SUCCESS in 48m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d27b9fde9a24714904d93476454e551 : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":3},{"id":"fa45bb3ce35b2fa4d57b7e7beec4e6d2c80b8405","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 21:07:21.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e44e143e67034884855480810c9af06c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57099a30b158490d9b2675571cb846ab : FAILURE in 33m 49s","accounts_in_message":[],"_revision_number":3},{"id":"eecbef80f743c652143d5394010820fbbd2e166e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-03 05:19:52.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e50d211a250e88a03a3e50dc0d5f5b84dd192dc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 06:19:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26becb64b2e24be58966fe54dcda4501\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5e0d27eb6838418d95d2b257001c05de : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":4},{"id":"96d88c564b40fb86ad5a22988be8bff628983690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 06:32:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbe2caccd5ee41368c26e7c49dae7d68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/719078f23f2c47b1a38e7793886caf50 : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/f536f9cb74e14063967ea59fcd395d65 : SUCCESS in 57m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/13af234d3ae942a59fd18c6fa3f47bb4 : SUCCESS in 1h 03m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b4cb9540f8cb4515aab073f230fb00c6 : SUCCESS in 48m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0aaef3b14b644cfebbbf3801f25d7442 : SUCCESS in 1h 02m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0bd3e65dae742018c10bb915d37d2a4 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03daeccd8558436b97fcff47cfce1cd9 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d00a7948e85a4945945f658417f91a3a : SUCCESS in 21m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e83daf6d9733492b8a9c204a03df7f6c : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/487c5e5840534b1887a2d72c7c2f0344 : SUCCESS in 25m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/67c6abc12a5541acbd18540df906308d : SUCCESS in 9m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6c022b257f0d4a46ab7221136d2814e9 : SUCCESS in 28m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/37604ea4205949dbbaf1742a4a6da1d3 : SUCCESS in 12m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/02fc6f950d644bbaa3e2c0e0ad15e1de : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/141dc69802dd4a1d9382eff9d5d5af41 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/153747d24ced4683bf2a456c4c6a83cf : SUCCESS in 13m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0aa928d688574a76aea26d3cf475f889 : SUCCESS in 6m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/015947fb1d2a4822b4df4d2b84436a3c : FAILURE in 9m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c81ad1f7cabf439184967185b3c47748 : SUCCESS in 5m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/839e9452d5de4b2db1727d1d55eac1aa : SUCCESS in 36m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d40f932c39a4083b00ca069775e8da9 : SUCCESS in 52m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aa1647527e814b32a2b3519cc1d0db6e : SUCCESS in 54m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6be5a973f66e445e894d0e1f56b80fdb : SUCCESS in 27m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b72245deb6b4deeb1e12459a30bbed5 : SUCCESS in 2m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6ac85924ee7343e5935aabb4bca71a7d : SUCCESS in 24m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/527eaa7f090a4cc799af6d5e8aaf5bf6 : SUCCESS in 38m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c649f730c1444b1fa00ba82b9d328387 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":4},{"id":"091a5571b4e480a132e2d41dea6bcd5c4af3a962","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-05 01:17:06.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9bb349d89349af58301ce54b8810dd1c3a665f58","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 13:14:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7e2522cdff8deb719f19a33e8d05e239ebfb49ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 15:07:41.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cf5c8e3e2cdd728a3145e7d2c9eff3b26e895ff9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 16:02:50.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"060592459ea89c355e84f779b4447a39e80527fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:50:53.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/930c8d14844a441ab0038472701de959\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d638a32bf940477197244ab481e90378 : FAILURE in 36m 22s","accounts_in_message":[],"_revision_number":6},{"id":"a14ad45268e928bf17f03e13d6ee48d6ee821e58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 16:53:26.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"0d0252ed4a40669456effa12a8cf5c43b82e4b5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:41:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/430599870e4a40659d0280a1f6f45fbb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b18e6b3f989d426ca6aac01f1b61e50d : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":7},{"id":"fafd97534dd2efbbba01ed956f8fcf4013dc2639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:06:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ada86926014436b96af4533665c8aa8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f13d3f201924dd0aa930c50e4ebd377 : SUCCESS in 11m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/04bbc3a18c96451eb69342ba96f34bb9 : SUCCESS in 1h 00m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bcca8cf9a038411ba94f5710af852647 : SUCCESS in 59m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0e35f31be0af475d9cf68f89326389d9 : SUCCESS in 49m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8fa2966276140d3a93ab5d48814a23a : SUCCESS in 56m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a139b8bf57d94cf5b15d6dcfa2dd2879 : SUCCESS in 24m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b9776b854df486696cb576afa46fc15 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c72a3555cab9411eb6c24e46f46b0c4a : SUCCESS in 20m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2f1bf1b1bb9a43f29afc8fb1fbc1c4ac : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1f62e6180d2b45c588333d00aaee4000 : SUCCESS in 20m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b9440e16ddc242edb1a0b9a495423762 : SUCCESS in 9m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/438b1e78d26d4cc798c583d392c4dda9 : SUCCESS in 20m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/207c9600ba4246fbbbddf164dd49d599 : SUCCESS in 29m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5c3ea5c5d8584ca3846896e14ff98d58 : SUCCESS in 12m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/57ee4168a59544a0b82a976ff3859689 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b13e861da38e43408a80d11c61c4163f : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7253091eef5a42dbb0a0a1756ab91c01 : SUCCESS in 6m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0dbd067541ad4bd9abe3e6df3424bf5b : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f9b4394f8f0f45b68cdee93171a81429 : SUCCESS in 12m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0031783072144238b065fbcab0d8e8e : SUCCESS in 47m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/84bc407ac6cf4bc7940a91fa182ebcad : SUCCESS in 56m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea65a21de3bb46f5907f0e76433b030c : SUCCESS in 55m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3afe248a6a084f5ea735fed9c44c6075 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1569eff9e49d4ee8b71a4f5fb59f8c72 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/32690951baab4323bde21e0ca44c7273 : SUCCESS in 20m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66575231217c4ed0861374dc6e270f66 : SUCCESS in 54m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96be226c35414366836402d49e027996 : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":7},{"id":"3641a99c5ac2453ed7a9876652b982afc3fad398","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-12 17:31:24.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"97937a46eac581fbacc7e802c22d36be6b0f5165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 17:32:22.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"54d78984db8d46291ad176d4ef18d6276b025380","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:56:24.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61eaaf59c7a24b3287f9732a7d74279d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a531a30fa9481f9d7e6b02ce1850b3 : SUCCESS in 5m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/0cc3a5f6b36f4e899ed4c4ef97dba395 : SUCCESS in 1h 00m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ce39e6c4cd8c45e1ba62fda5ea7b9b9b : SUCCESS in 1h 07m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/73fff90b3bdc4adbbf3208f881af53b7 : SUCCESS in 50m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28300a6ff81d4e5abc526b79059ce504 : SUCCESS in 1h 01m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/aa2f90ac9d6c416199f085463ff7f5db : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c7a353c5b2f74e589e768a5a56123c30 : SUCCESS in 25m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f33d1b80795147f2bd3061f5be290d59 : SUCCESS in 16m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d71ac0acdb0448598e2c05f1f745cec2 : SUCCESS in 17m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4b6182edc2b34ca28f3ae4428d2aaf1a : SUCCESS in 24m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2a8ae2caf21945d193cd4594e783a3d1 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f0eff19809a94c0abaa9433bcb711cec : SUCCESS in 15m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/600ede2e7dd94334922ffb1b0ef6a953 : SUCCESS in 15m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b8d4b0c06ad848b1b6041a3d01159281 : SUCCESS in 7m 01s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6a92c99eab3b4419aef1239e8238c04e : SUCCESS in 7m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfd83b140fbe4361a51b0d04d49a7e8c : SUCCESS in 49m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/29b0ba7e5ad446a4b74ebc713b82385a : SUCCESS in 52m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e98a88cc56c48c99e8e3aa744bfdbdf : SUCCESS in 25m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aed2ac0b9f1b4df8841630f21caaf8a3 : SUCCESS in 27m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b64dd1dc7d0749eca49297e8f31276ed : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30ee182a70e44725a6ba61446d228ebc : SUCCESS in 17m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c2385ef7cdd463baf0aad320cef1b57 : SUCCESS in 57m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93fc6defa4a241ce82da4de247b3c6a2 : SUCCESS in 31m 51s","accounts_in_message":[],"_revision_number":7},{"id":"df58a24e8aff52866ca28d6d42ed44654acdac0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:56:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"06dbe41ea678332f05b9d3146d9555ddb99cddf2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:57:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f2f020bd1946cb84b543c0346eaafa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54f970836b1247df8e38bf63f27a7b84 : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/58f47f0f91aa451ab453283314da42b2 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5df4e758f35d5453bdbf73e3591139d339d157e5","revisions":{"102e137ebecf6dc76d8f4f509a3b679a3e05d9fe":{"kind":"REWORK","_number":1,"created":"2025-11-28 18:13:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/968856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/968856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/968856/1"}}},"commit":{"parents":[{"commit":"41e6e3a6d74143434ef67708d6c4bf35e33decc6","subject":"Add a Timestamp.zero() method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41e6e3a6d74143434ef67708d6c4bf35e33decc6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:06:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:13:12.000000000","tz":0},"subject":"slo: use Timestamp.zero() as default in RespAttrs","message":"slo: use Timestamp.zero() as default in RespAttrs\n\nAlso make test assertions for async delete at times less\nbrittle.\n\nChange-Id: I750ac1ab2a75d1a49c87c6e1704ebe733fb31ec2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/102e137ebecf6dc76d8f4f509a3b679a3e05d9fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/102e137ebecf6dc76d8f4f509a3b679a3e05d9fe"}]},"branch":"refs/heads/master"},"ff5a4f3d8ce7701189e9d4a76a698e5e5f544ba4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-01 15:37:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/968856/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/968856/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/968856/2"}}},"commit":{"parents":[{"commit":"cb480a5bcc19ce3edb63f5e34a42a4915bcd61de","subject":"Add a Timestamp.zero() method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb480a5bcc19ce3edb63f5e34a42a4915bcd61de"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:06:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 15:36:52.000000000","tz":0},"subject":"slo: use Timestamp.zero() as default in RespAttrs","message":"slo: use Timestamp.zero() as default in RespAttrs\n\nAlso make test assertions for async delete at times less\nbrittle.\n\nChange-Id: I750ac1ab2a75d1a49c87c6e1704ebe733fb31ec2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff5a4f3d8ce7701189e9d4a76a698e5e5f544ba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff5a4f3d8ce7701189e9d4a76a698e5e5f544ba4"}]},"branch":"refs/heads/master"},"0ae92b2bfbc728134962d28c35242d3e1ff28e59":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-02 18:34:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/968856/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/968856/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/968856/3"}}},"commit":{"parents":[{"commit":"08a4215345439bf227d381f8a12192970494ad6f","subject":"Add a Timestamp.zero() method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08a4215345439bf227d381f8a12192970494ad6f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:06:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-02 16:46:10.000000000","tz":0},"subject":"slo: use Timestamp.zero() as default in RespAttrs","message":"slo: use Timestamp.zero() as default in RespAttrs\n\nAlso make test assertions for async delete at times less\nbrittle.\n\nChange-Id: I750ac1ab2a75d1a49c87c6e1704ebe733fb31ec2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ae92b2bfbc728134962d28c35242d3e1ff28e59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ae92b2bfbc728134962d28c35242d3e1ff28e59"}]},"branch":"refs/heads/master"},"3ab9fc470a858595a72451893036786394550c5d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-12-03 05:19:52.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/56/968856/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/968856/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/968856/4"}}},"commit":{"parents":[{"commit":"5f3ed0834467717085fedc5df82a43f308403f85","subject":"sq: Fix failing tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f3ed0834467717085fedc5df82a43f308403f85"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:06:05.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-03 05:08:59.000000000","tz":660},"subject":"slo: use Timestamp.zero() as default in RespAttrs","message":"slo: use Timestamp.zero() as default in RespAttrs\n\nAlso make test assertions for async delete at times less\nbrittle.\n\nChange-Id: I750ac1ab2a75d1a49c87c6e1704ebe733fb31ec2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ab9fc470a858595a72451893036786394550c5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ab9fc470a858595a72451893036786394550c5d"}]},"branch":"refs/heads/master"},"13c5856147b9ec598cbec40b4716dcef49bf71f7":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-12-05 15:07:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/968856/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/968856/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/968856/5"}}},"commit":{"parents":[{"commit":"67696bb4cc95bc3d11388b83e91faffc5d6bb874","subject":"Add a Timestamp.zero() method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67696bb4cc95bc3d11388b83e91faffc5d6bb874"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:06:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 13:01:14.000000000","tz":0},"subject":"slo: use Timestamp.zero() as default in RespAttrs","message":"slo: use Timestamp.zero() as default in RespAttrs\n\nAlso make test assertions for async delete at times less\nbrittle.\n\nChange-Id: I750ac1ab2a75d1a49c87c6e1704ebe733fb31ec2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13c5856147b9ec598cbec40b4716dcef49bf71f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13c5856147b9ec598cbec40b4716dcef49bf71f7"}]},"branch":"refs/heads/master"},"4f2deaf26aa5c3ac7a51515d905fbd9fc75c591b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-12-05 16:02:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/968856/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/968856/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/968856/6"}}},"commit":{"parents":[{"commit":"0bbcd6c2cec7ae8b472453d917ed1db85723a091","subject":"Add a Timestamp.zero() method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bbcd6c2cec7ae8b472453d917ed1db85723a091"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:06:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 16:01:19.000000000","tz":0},"subject":"slo: use Timestamp.zero() as default in RespAttrs","message":"slo: use Timestamp.zero() as default in RespAttrs\n\nAlso make test assertions for async delete at times less\nbrittle.\n\nChange-Id: I750ac1ab2a75d1a49c87c6e1704ebe733fb31ec2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f2deaf26aa5c3ac7a51515d905fbd9fc75c591b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f2deaf26aa5c3ac7a51515d905fbd9fc75c591b"}]},"branch":"refs/heads/master"},"5df4e758f35d5453bdbf73e3591139d339d157e5":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-12-05 16:53:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/56/968856/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/968856/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/968856/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/968856/7"}}},"commit":{"parents":[{"commit":"db578c1077d25aa28ef704df94caa7de6cf782bb","subject":"Add a Timestamp.zero() method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db578c1077d25aa28ef704df94caa7de6cf782bb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:06:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 16:53:06.000000000","tz":0},"subject":"slo: use Timestamp.zero() as default in RespAttrs","message":"slo: use Timestamp.zero() as default in RespAttrs\n\nAlso make test assertions for async delete at times less\nbrittle.\n\nChange-Id: I750ac1ab2a75d1a49c87c6e1704ebe733fb31ec2\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5df4e758f35d5453bdbf73e3591139d339d157e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5df4e758f35d5453bdbf73e3591139d339d157e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
