)]}'
{"id":"openstack%2Fswift~968857","triplet_id":"openstack%2Fswift~master~I3281c2f446d26b31c7531a79c09d9d9111ba9b1b","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-12-23 18:45:23.000000000","reason":"Change was abandoned"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-23 18:45:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3281c2f446d26b31c7531a79c09d9d9111ba9b1b","subject":"test_diskfile.py: make allowances for v2 timestamp","status":"ABANDONED","created":"2025-11-28 18:13:23.000000000","updated":"2025-12-23 18:45:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"666c9eb37c03e1b5013929e784cd71062650e1f9","_number":968857,"virtual_id_number":968857,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-04 07:05:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 19:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17385698b026799dc00db4941a311f74106bc169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-28 18:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11d7d4c67a2aa988bb2d4fb1a3ac6d28f6c4a053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:17:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9832526c8e8c4053aa352c36d49c7c68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3934211ccb94fffa4181fe289ed1a35 : SUCCESS in 10m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/848622f90d7848e1a993da00a877997e : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50f9701b63234e0fa048fd2e88053d71 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e38fc004d0744ed0b75dd2c8abdcc815 : SUCCESS in 21m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f627293ce1d447f2bbfdeb058ec064b2 : SUCCESS in 7m 12s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/84b7a51b82a34e2dafbbdae3bbcc0890 : FAILURE in 29m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d4e3e68035894fa4b5458ee1a82b573b : FAILURE in 15m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a68d27948735496bb7fb4dd7ea17278f : FAILURE in 42m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/43280993d4b3423081ed2ef263d9b922 : FAILURE in 20m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9fdcfe0d0ef34c1da3579f7a4384a42c : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2050e3d4f7874088b6a1d381b0e30705 : SUCCESS in 12m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e0b37b8e54854117945c2031b55fbe3b : SUCCESS in 9m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9af3f23cb2b47108cd1196d8c759867 : SUCCESS in 54m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4900735e352444f09f26b441e334ae00 : SUCCESS in 22m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/079552fe3a7e4c37b9aafd060880e364 : FAILURE in 32m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c951156bdb94bebbf11ff4e2e822718 : SUCCESS in 3m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b801da1eea48474d84643f739b6ce811 : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7e331d15e1b9c7927e9b0de24959b4ab59bc2afb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:37:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7aa7de5c28242f08afd909da894130b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aba6181559374f4b838be3ab33f7b2c1 : FAILURE in 1h 19m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3d448475f578be50e7d21ff9c28bb78e7af191e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-02 23:57:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"607dd841e821d4ec3fca7f47baea91280adc79a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 01:04:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b25cb304e0554f97993fb57ca482e3fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e8d8203aa0e473ca33e43746aff57b4 : SUCCESS in 9m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/beebd98a183546cb965ceee2d8529e05 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b394a53e50d44789b4a5d0f0163f2f38 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9b92ac70a97417eb1fa66518f7eb30a : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9ed2f873ac742129973ef263c765571 : FAILURE in 4m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c36f3b54fe764616ab480b700eecf097 : FAILURE in 30m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2a34b8a1e6ef43c294ed5d9fe491bd28 : FAILURE in 16m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9cc0e7d5b62a4324a23334ddae0bb462 : FAILURE in 34m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ad2ff4a22dec48eca42339e6ce2d8280 : FAILURE in 23m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/50b754c9ca4f4d1ba4368521b9378616 : SUCCESS in 8m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bfa512cca2a14b1b864ed2f323114dea : SUCCESS in 12m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a1633654d3e645aaae2154967ea6bc51 : SUCCESS in 8m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ef969551fc342d38b3f5504db7b8479 : SUCCESS in 55m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d2683c442a44a5e83a280c6c5e988f8 : SUCCESS in 55m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/242584444aaf4e45ac19ab963246c2c2 : FAILURE in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17498b906260444fb140899c0b5a54e6 : FAILURE in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/425357cdd4d046c6a0db3a51aa1aae5b : SUCCESS in 19m 37s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_ssync_sender.py","accounts_in_message":[],"_revision_number":2},{"id":"b5ddeaac80f8f5108a66dc7255606e4fbefe3921","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 02:16:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17c9a1cc0f1b4a0ba3b6673d0d2f5125\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2dac3a6d105a4d60ae4a028529d6a43c : FAILURE in 36m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3b5824e3ae4d8297fa6cefaaf47d59aeab6e4f86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-03 05:19:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"606614649f28811af931990449f94733bee333c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 06:03:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/737952673a724c7e80ea2142458c9fb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4e22d1d0bb942c1b9a20c67bd5a1a9b : SUCCESS in 10m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3e92603f8a6412dac1f6b260ab2807e : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/667f6cd2d74f4077aced13f384c2bd4d : SUCCESS in 14m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bd4420752544e4db1c84f7dcd06a656 : SUCCESS in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c364ffda1a164e46b36a5443b7f6d38f : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/43a83fce7f6d448db6153663189f04d1 : FAILURE in 12m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a43340a88a064882a7d439d6bffd9ca9 : FAILURE in 25m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/42ad8ea01ff645f7a07d46cecd6c9a89 : FAILURE in 33m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e4ddd1e647044dc3b875be837b626443 : FAILURE in 35m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0b3e0db34f4d4cd4803cfaf0db94e89c : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5e9660fe25694b128a307966fe73a2cb : SUCCESS in 8m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e15101776ad4485fadd88bd6293083dc : SUCCESS in 17m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/72c92ed440a645d89e078ab37aecd74e : SUCCESS in 36m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b343e51d82ce4d2c9308f2e38c38b99e : SUCCESS in 22m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92fb85b2c8da4a12ae83cfc787ad21ab : FAILURE in 32m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bf23ccf21a4478fbbf92203e5573853 : FAILURE in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e10e312b509449a782c32398a60ddcad : SUCCESS in 18m 06s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_ssync_sender.py","accounts_in_message":[],"_revision_number":3},{"id":"c58f6668ae66ca9e2e2dc415a7bb17494bc33762","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-03 07:55:22.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"43edffcda18a6b75b67e16276f7a486d6e2dd70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 08:51:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec9db4d5dbd14fbfaa3d9b7c3449a141\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816492e997a04d699a5684cbc702195f : SUCCESS in 10m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/116dcb3ab0c94448a093808eff4bc305 : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31611c5e02fe4bd388c929716211e098 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a21db35bffdf4d029efa99833be893c2 : SUCCESS in 21m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/adf6c2a5e09d456e884fd675ae96eb55 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf35c4d9f5644f41b92c3b2c3aec36ac : FAILURE in 28m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ec49ccf13011413caf961847d62fa192 : FAILURE in 24m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/30c47ad3db074159a24b862e7784ba18 : FAILURE in 21m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fa8078aa42c040969314c198278c86ed : FAILURE in 26m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/674b1a13cf3e4c919932c3c0646b0a79 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bd098a9e790a4726b735c77d7c7f0dcb : SUCCESS in 15m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b9fb2391723443c694b2291c2efbe13b : SUCCESS in 15m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/875d05d5625347a498820eca2bd0a8d6 : SUCCESS in 48m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8773c68731e64679af3eedc97e583f10 : SUCCESS in 47m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ba8a187f8d9471fb9bf26cfcd7fea52 : FAILURE in 35m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a724bf86ed8e4d028ff8bce4ec28e24d : FAILURE in 3m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1bd558a452404056b13994375239e389 : SUCCESS in 25m 01s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_ssync_sender.py","accounts_in_message":[],"_revision_number":4},{"id":"a2877305169d64d60d2413198bf099cccee9b383","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 11:29:47.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c8dc6229e0940909ca3bcc186651068\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/28f5a4a69a834cb8a8adbf8b6bbe2e4a : FAILURE in 38m 46s","accounts_in_message":[],"_revision_number":4},{"id":"7f4678d4b1c7b8f1b56f23c01554a08b94a66e7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-04 06:07:12.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7660db5878ee0b9287b3d4ef0c899dd8135dcfb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 06:51:44.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e77887af5c04bb08f2f6d41485c7e9d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f7a50ae7f5bf482d9ae38b17678624d3 : FAILURE in 40m 30s","accounts_in_message":[],"_revision_number":5},{"id":"b28d850cb94d2c1d5435f5a39a197ed4c7a60165","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 07:05:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89a19b8fece34a92a73b3bb46c7414de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5323707bc5724cf3aa74704b3a2d36dd : SUCCESS in 10m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/accd916798854e07bbc9a7c7df11fab4 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e2afdeff32b490ca0df7357df748bab : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55f99f570ca548ddb6cb9dc3c66d57e7 : SUCCESS in 22m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/03a90af7680345a3b5f2a58696d67ccb : SUCCESS in 7m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4a9f1dc986e5410e8400275d8ec1220e : FAILURE in 26m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/81676ceb971b4827ba22d40b70279fd4 : FAILURE in 24m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4e03f773f61b44e18cd830cb3bd2eb1d : FAILURE in 38m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7890ad65a798410b9b99e4ef80f195d2 : FAILURE in 31m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8fd942782b0b4241af17a0b1d7ec0c5d : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/61e873af1115436cb76f986fd82b1a62 : SUCCESS in 13m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bfa78304b293419ea71e31d25ac03041 : SUCCESS in 15m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea65884ea2104e3b8a47e2321eea482b : SUCCESS in 51m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f5b69dee66d4163b299d0fd7f20f995 : SUCCESS in 51m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75527e1a53d64bce9c743aa704cb0462 : FAILURE in 30m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d790b52345b148e7a544c19a769f0ac0 : FAILURE in 2m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88b9e171a87243439ccde65bdffaabbc : SUCCESS in 18m 40s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_ssync_sender.py","accounts_in_message":[],"_revision_number":5},{"id":"666c9eb37c03e1b5013929e784cd71062650e1f9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-23 18:45:23.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/967738","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f1b785424cfad788a586952f303fddfc944ce785","revisions":{"12766c3c0978b059080050c84e29d5d056a04ff9":{"kind":"REWORK","_number":1,"created":"2025-11-28 18:13:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/968857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/968857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/968857/1"}}},"commit":{"parents":[{"commit":"53c445cd38b7dd5f350d482c2d6b877df0595c02","subject":"test_sharder.py: add test_cleave_root_legacy_epoch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53c445cd38b7dd5f350d482c2d6b877df0595c02"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 14:41:39.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 18:13:13.000000000","tz":0},"subject":"test_diskfile.py: make allowances for v2 timestamp","message":"test_diskfile.py: make allowances for v2 timestamp\n\nOn disk filename encoding cannot represent the hex part of a\ncontent-type timestamp when it differs from the meta timestamp. As a\nresult, the hex part precision is lost.\n\nChange-Id: I3281c2f446d26b31c7531a79c09d9d9111ba9b1b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12766c3c0978b059080050c84e29d5d056a04ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12766c3c0978b059080050c84e29d5d056a04ff9"}]},"branch":"refs/heads/master"},"b71f3ad495ce80042873b217c13b01fa115087ea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-02 23:57:02.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/57/968857/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/968857/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/968857/2"}}},"commit":{"parents":[{"commit":"6e3c456f7c43f8bcbeb241a69246a7dc83d7b40e","subject":"test_sharder.py: add test_cleave_root_legacy_epoch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e3c456f7c43f8bcbeb241a69246a7dc83d7b40e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 14:41:39.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 23:56:40.000000000","tz":660},"subject":"test_diskfile.py: make allowances for v2 timestamp","message":"test_diskfile.py: make allowances for v2 timestamp\n\nOn disk filename encoding cannot represent the hex part of a\ncontent-type timestamp when it differs from the meta timestamp. As a\nresult, the hex part precision is lost.\n\nChange-Id: I3281c2f446d26b31c7531a79c09d9d9111ba9b1b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b71f3ad495ce80042873b217c13b01fa115087ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b71f3ad495ce80042873b217c13b01fa115087ea"}]},"branch":"refs/heads/master"},"86b25c10e4c048743d899cc756423b23f574a85d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-03 05:19:52.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/57/968857/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/968857/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/968857/3"}}},"commit":{"parents":[{"commit":"d1c6e1ad423b861a382eea5711d36c214c8e5623","subject":"test_sharder.py: add test_cleave_root_legacy_epoch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1c6e1ad423b861a382eea5711d36c214c8e5623"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 14:41:39.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-03 05:19:39.000000000","tz":660},"subject":"test_diskfile.py: make allowances for v2 timestamp","message":"test_diskfile.py: make allowances for v2 timestamp\n\nOn disk filename encoding cannot represent the hex part of a\ncontent-type timestamp when it differs from the meta timestamp. As a\nresult, the hex part precision is lost.\n\nChange-Id: I3281c2f446d26b31c7531a79c09d9d9111ba9b1b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86b25c10e4c048743d899cc756423b23f574a85d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86b25c10e4c048743d899cc756423b23f574a85d"}]},"branch":"refs/heads/master"},"d7be1c8b5c24ef2c47e0a991a2164da74e770f6d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-12-03 07:55:22.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/57/968857/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/968857/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/968857/4"}}},"commit":{"parents":[{"commit":"f01040ab487def6f8300d030f2fd41ad1c30316c","subject":"test_sharder.py: add test_cleave_root_legacy_epoch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f01040ab487def6f8300d030f2fd41ad1c30316c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 14:41:39.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-03 07:55:03.000000000","tz":660},"subject":"test_diskfile.py: make allowances for v2 timestamp","message":"test_diskfile.py: make allowances for v2 timestamp\n\nOn disk filename encoding cannot represent the hex part of a\ncontent-type timestamp when it differs from the meta timestamp. As a\nresult, the hex part precision is lost.\n\nChange-Id: I3281c2f446d26b31c7531a79c09d9d9111ba9b1b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7be1c8b5c24ef2c47e0a991a2164da74e770f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7be1c8b5c24ef2c47e0a991a2164da74e770f6d"}]},"branch":"refs/heads/master"},"f1b785424cfad788a586952f303fddfc944ce785":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-12-04 06:07:12.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/57/968857/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/968857/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/968857/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/968857/5"}}},"commit":{"parents":[{"commit":"e74d9101d9c45f35591d7c610718acf444c9d59f","subject":"test_sharder.py: add test_cleave_root_legacy_epoch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e74d9101d9c45f35591d7c610718acf444c9d59f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-28 14:41:39.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-04 06:05:48.000000000","tz":660},"subject":"test_diskfile.py: make allowances for v2 timestamp","message":"test_diskfile.py: make allowances for v2 timestamp\n\nOn disk filename encoding cannot represent the hex part of a\ncontent-type timestamp when it differs from the meta timestamp. As a\nresult, the hex part precision is lost.\n\nChange-Id: I3281c2f446d26b31c7531a79c09d9d9111ba9b1b\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1b785424cfad788a586952f303fddfc944ce785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1b785424cfad788a586952f303fddfc944ce785"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
