)]}'
{"id":"openstack%2Fswift~969104","triplet_id":"openstack%2Fswift~master~I836cf1e1bdb48aab59b4f3d618f5e4f9320d80b5","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-01 19:49:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I836cf1e1bdb48aab59b4f3d618f5e4f9320d80b5","subject":"account/test_reaper.py: cleanup Timestamp usage","status":"MERGED","created":"2025-12-01 15:37:49.000000000","updated":"2025-12-02 12:25:22.000000000","submitted":"2025-12-01 19:49:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969104-p-timestamp-collision","meta_rev_id":"ac49394903da450f09ec13cd07b53a750bf305bb","_number":969104,"virtual_id_number":969104,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-01 19:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-01 18:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-01 18:28:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 16:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-01 18:28:00.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"47a3ec5fb60a4c3a25587088d7715bd67b25a920","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-01 15:37:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b0d0cdb1a543c13ef1a7daf73738e13441442ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 16:22:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4382de203fa242fea1170d2536453313\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b817c770506648d18cc4c3bced2b58df : SUCCESS in 9m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb2eef63c4e34fba974c763ec00ae34a : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6eea98c31680406896b0571aefd3700b : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/364722ce22e341fc85f9080a7fde061d : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d2eb4b6e0c7449696cd9566eb604d24 : SUCCESS in 7m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f888032c223d443d8c0a02b5fe1e0280 : SUCCESS in 10m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7f2f57ae1e1f43dd96698777ba4790d5 : SUCCESS in 23m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4e66bbe2ac714dcca8ffd1c49cbe0f62 : SUCCESS in 14m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5b29288a8de7446e8094ddbda7dc1dd0 : SUCCESS in 22m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fcb597ee6c7d4f01afaf58826232a312 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/660963a6b777459588c8dc58cd566d51 : SUCCESS in 8m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7f71342f40204d78bf111bfb756abaad : SUCCESS in 8m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9bcd65377b2e4ca1a89c7a087fe81b7a : SUCCESS in 22m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e8b1d16a3ae49758cc34427a376a74c : SUCCESS in 23m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2da7a4ccb37944adae7aa3fa7a9fb23a : SUCCESS in 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f91d505c9f841e68c2b00089cb6cc71 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4b6ebe26491a4d7286e36ad133feea03 : SUCCESS in 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"dcbf03558fe76d8844ccd2501f7c82c14bdfaadb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-12-01 18:28:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49f1ee80e3428e54ec59bd9e5ff1ae8acd3af60a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 18:28:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90f144959082b176ae11e51df01e8e013b7f29be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 18:51:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/921ca8db3e144aafa6c812ac70ac6d7c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/12b2c8b232e24e5fab34ff70dbc03ee7 : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"732a6698531846d9612d5152294c7461aaa6bb9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:49:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb4388b8d05144b9817dafaf7effe612\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7e89ab3a3a4c5797fa210a992ea59a : SUCCESS in 11m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6e1decf30e2f4a26b50121950113893d : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dea6ddcb13ab405baec6abab6bfa413c : SUCCESS in 22m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/108f84d162784a63a09cd9e777720fe3 : SUCCESS in 17m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/09e851046fd94bc0bc451cfbbd8d9ea6 : SUCCESS in 11m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/175e8d46ee33432ab5ff7ab6619c2968 : SUCCESS in 14m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6945d02191b34c8182d8224148c90541 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4d9a3f9c87ef49a0aabd438a27bbc163 : SUCCESS in 13m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c13eb3b7c1764018a8051b07fa586aea : SUCCESS in 16m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/12b0c1d6a1ff4f2599b2f0d061b1d636 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c52f56460fe84ab09a256473c0ead74e : SUCCESS in 8m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/722ed070685346359f33830eedcc979d : SUCCESS in 50m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/386fbfe7dfd249128f3dbfea9cebda52 : SUCCESS in 21m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b0c0580b4424375a405f20fb0e994c5 : SUCCESS in 28m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb4587e70bf4677b03954e04e49a3c8 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d1b325410764e8f969bfa271a1975fd : SUCCESS in 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ad0ced031f12ff20d9ee3dc372cd7d3142153c47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:49:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f39470b2f2e4898061159ca8e576b9031d4a7fbb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 19:51:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfff32cc3f1f48d688f0f034e0bb71b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4db07c743f4c53857d5d3be636973f : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/426f09fc758a47b5800c9f6b04bdc512 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac49394903da450f09ec13cd07b53a750bf305bb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-02 12:25:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca9db701c50a38b46fadb5a78be0869de9a0d378","revisions":{"ca9db701c50a38b46fadb5a78be0869de9a0d378":{"kind":"REWORK","_number":1,"created":"2025-12-01 15:37:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/04/969104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/969104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/969104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/969104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/969104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/969104/1"}}},"commit":{"parents":[{"commit":"e6f1b06c9388bfd4462dd4621027638d042c782b","subject":"test/unit/container: cleanup Timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6f1b06c9388bfd4462dd4621027638d042c782b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 15:27:13.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 15:36:51.000000000","tz":0},"subject":"account/test_reaper.py: cleanup Timestamp usage","message":"account/test_reaper.py: cleanup Timestamp usage\n\nChange-Id: I836cf1e1bdb48aab59b4f3d618f5e4f9320d80b5\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca9db701c50a38b46fadb5a78be0869de9a0d378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca9db701c50a38b46fadb5a78be0869de9a0d378"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
