)]}'
{"id":"openstack%2Fswift~969105","triplet_id":"openstack%2Fswift~master~I7bc1c60acbb0de9d999e6b33811f5e63e20bd4ac","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-01 21:52:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7bc1c60acbb0de9d999e6b33811f5e63e20bd4ac","subject":"cli/test_info.py: cleanup Timestamp usage","status":"MERGED","created":"2025-12-01 15:37:49.000000000","updated":"2025-12-01 21:53:36.000000000","submitted":"2025-12-01 21:52:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969105-p-timestamp-collision","meta_rev_id":"c0c3d19eb3e7644d42b68e79e555824e8415f91c","_number":969105,"virtual_id_number":969105,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-01 21:52:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-01 20:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-01 20:33:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 16:42:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-01 20:33:48.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"8bb8215246c4bfbef84093f80ab1bff39bbfaf38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-01 15:37:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80aa60a12611f7bfafe8b9e8e93a0422a77fc1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 16:42:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2560e53503064a40873212980d146bce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868e530a5e404a4e938688862f8030d1 : SUCCESS in 9m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d176d3f91f434508b788ea090597d35f : SUCCESS in 12m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78786900155540d5af4e20f2bca9fd59 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bda823612a6545fca9408a5f29b88d34 : SUCCESS in 20m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c105fac264c74c089e5e11fbe6e717b5 : SUCCESS in 4m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3b036dae7c204386bd86c8a905bb177e : SUCCESS in 24m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ffd4915144cf4e9cb498943586d76be5 : SUCCESS in 10m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/15418dc8147c49f4a547981de7be99b5 : SUCCESS in 28m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a03ec388cf674699a3eae89497a650fd : SUCCESS in 11m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f8eaf14500a14a7589a55498fec16ff1 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0f73d93a782e47368d00a0f16a1ae80c : SUCCESS in 8m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b42814632cf0431a8096906fae759b13 : SUCCESS in 12m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1f17e75348394284bdb31a771f39abdc : SUCCESS in 53m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f04a15b90029470a89c9c47ca84cb649 : SUCCESS in 22m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/209a7d59b7b9457e96434b3d7443df1b : SUCCESS in 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c73cbca6fe7459f8c46abd2fe4a7742 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6e87a9bcc854ca7b0a6aca451cab815 : SUCCESS in 18m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bd9c8f32dcd77b3655385d8afb83996a41731bc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 18:59:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa4d918d4b0f4ed88ffeea840c59dc86\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0882e3edbaf8468e91c27068b83675fc : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"381afb5669cedb1da494193a0ccc1057e203bc2b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-12-01 20:33:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"34c22a0f09c308e3872583f175afe3444433edad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 20:34:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfb7867bd7c732c5253ea8f790b0b44b8b25c2d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:52:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f11febd0ec3b44af978835854d151021\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abfa2532aabb4393954e0bbf1d6fded7 : SUCCESS in 10m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1d7fc726fe094ced8269bf4ed9847541 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ab6f31b725274bbfbb79fc958baf1fdc : SUCCESS in 23m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d46acf707c6a4a81a057096b3aff7075 : SUCCESS in 20m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3e03e17d1204494086d4ca47d4d845d6 : SUCCESS in 27m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c0baa3d3f8924563b43d902d4eb470c0 : SUCCESS in 11m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5ca6eea1ea5146ec8255fb78b964a34f : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7f0ed01a3b1142789da37db65ce8b0b1 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c195291932414255b9fa120fbeffaa4c : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d78299101cdf42ccb379c583efe4d555 : SUCCESS in 3m 47s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c1c2b45062c249ec9309156411e71107 : SUCCESS in 8m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/624e612a984a46bd9e73dec333407273 : SUCCESS in 38m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b169fbdd433f4fd6a9b4ef666fdbd703 : SUCCESS in 24m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e254a9fa189451899ff88f61c9ee525 : SUCCESS in 20m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/601030226f344b95b539c601d124afc6 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a79b2cbb81ae49f58b2ec922239c481b : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a1849d84c426613afd4302610ecab3068e72564a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:52:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c0c3d19eb3e7644d42b68e79e555824e8415f91c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:53:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25c45a7ff53e41468a2f1f1287712b9b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4921dc6d90d94453a087a0e32b9c0dfa : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e16b930549cd45799267c2acfb4796cd : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22605d21ee9c5cb25ad2f3b2f321c8593aefcbdf","revisions":{"22605d21ee9c5cb25ad2f3b2f321c8593aefcbdf":{"kind":"REWORK","_number":1,"created":"2025-12-01 15:37:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/969105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/969105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/969105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/969105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/969105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/969105/1"}}},"commit":{"parents":[{"commit":"ca9db701c50a38b46fadb5a78be0869de9a0d378","subject":"account/test_reaper.py: cleanup Timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca9db701c50a38b46fadb5a78be0869de9a0d378"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 15:27:45.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 15:36:51.000000000","tz":0},"subject":"cli/test_info.py: cleanup Timestamp usage","message":"cli/test_info.py: cleanup Timestamp usage\n\nChange-Id: I7bc1c60acbb0de9d999e6b33811f5e63e20bd4ac\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22605d21ee9c5cb25ad2f3b2f321c8593aefcbdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22605d21ee9c5cb25ad2f3b2f321c8593aefcbdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
