)]}'
{"id":"openstack%2Fswift~969106","triplet_id":"openstack%2Fswift~master~Ie0e0937c45238532682d390250d923e0d928e711","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-12-23 18:28:32.000000000","reason":"Change was abandoned"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-23 18:28:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie0e0937c45238532682d390250d923e0d928e711","subject":"WIP container/test_replicator.py: user jitterless timestamps","status":"ABANDONED","created":"2025-12-01 15:37:49.000000000","updated":"2025-12-23 18:28:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f757447ffc416fa622c77f93ba4083e95dfa900c","_number":969106,"virtual_id_number":969106,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-04 07:02:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 16:36:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc2363bbbf4fc9067767c90cf33a8be78362ef60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-01 15:37:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8685f8f88fab8e33e823a1f2c1c3e6d36b4d3ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 16:36:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44a5d162bfbb450da76313b3a5d4fd46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3a4804d9e994788b84f0103f4933cae : SUCCESS in 11m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7836adb7f0fe46b5b00a7b5fc28c584d : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76e938268d83416e99105612450291d9 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b32c213a0f334fbca0545066d422598a : SUCCESS in 22m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/090ee8a7c2bc4583a87cbe87d35ddca3 : SUCCESS in 9m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4848d963b5e64230a9b7d0832c17ddc7 : FAILURE in 15m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1b8c4e742e6e421db17ce5ccac131d29 : FAILURE in 12m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/086fa6bedb20453bb36c4cdc1c6166f2 : FAILURE in 30m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/04abacbcc57a414fbb920ba5c323156b : FAILURE in 16m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/189cceb362ee4290ae0ca54b27c91c51 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6296bdfa41d44d278bfff85fb166e885 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/93f5c6740b0b445c8a0f4987b07e2a20 : SUCCESS in 13m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1138d6c3a3a4133aba37a09a8f30d80 : SUCCESS in 50m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6d39023d580243bfa838fe5ac80013ff : SUCCESS in 51m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7f741a227ab43a2a2b41c0a829dca1d : FAILURE in 27m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea47d8a4f0d24bde850c9f68e45d0b90 : SUCCESS in 3m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36be1af41aa14accb7a8f3b2a79000d3 : SUCCESS in 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"24898e55e14518fee66433c7a5af96950e615cd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 21:45:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c8cf2193204726ad42e18cdb2bb822\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2fd4972fecb64c469f7ea5b6b1e17cdc : FAILURE in 33m 08s","accounts_in_message":[],"_revision_number":1},{"id":"206921c3a774601f5a21a27fdc223b7a109bcf77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-02 06:20:27.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce931bd13676e2a9a9dc3fddde216492930b1377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 07:14:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce1ef796647d44fa8e1fbd284f36b779\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7796e12af50846869a6eee8bdfd0011b : SUCCESS in 10m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc6ac265215f4b12b323717bff5528a0 : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b7fb1888ee843418668b8656a7f9259 : SUCCESS in 10m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/82fe74f00be9410aba961a080860a0b9 : SUCCESS in 18m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c13c4b4a485b4a57878cda1bea1925db : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ef6d252b5db5495ebb26e072e3a0e3ae : FAILURE in 11m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/18acd347b3f34aa3a88951f59f4deeae : FAILURE in 21m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ca3dc87c82de4cdd8622b755fa11bfaa : FAILURE in 32m 02s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e5284e4c3d8e43ad94a76a47eedba495 : FAILURE in 24m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/43155fd9f65245578fd3d22f63b74e04 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bba478d15a4a41a7b6030dc7d9929b5f : SUCCESS in 16m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/36912e0c959748b5bb08155775a40e2d : SUCCESS in 13m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0dfae6499b814f709f9a5ae5ed4f2514 : SUCCESS in 49m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d494c9e18fb447b8b252d2e3234c469b : SUCCESS in 36m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29fc6620beeb488399c24d7b0f328144 : FAILURE in 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e6fac0b47e84318a8f523ceccb8ed4e : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7d0324f38b24d0aa436f90da5400b0f : SUCCESS in 18m 47s","accounts_in_message":[],"_revision_number":2},{"id":"dbc79d66b11b9381c93c7cc86daf15f8a534c0ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 07:43:50.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfb3e191fbd24db894d593afc1dba471\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/48a18ee898a94387842fae66e246e0f6 : FAILURE in 1h 20m 31s","accounts_in_message":[],"_revision_number":2},{"id":"6fc6770ff8d498d33fbec68ee95da6e5a3b1cbd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-02 18:34:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1b5a15beab0bcf834c451aabefa3af1325c4b6cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 19:27:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b14368d21631409daf8038e0125f2b58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50eb47a92fd0490dbadb80a3b3fceb1f : SUCCESS in 9m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b413d7ff62b1460f96b36b96e6ab5acb : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e54f5df14e3d4d7baa4f023448caaffc : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8da1c368fe246a1bfeddcf3e8bd72d5 : SUCCESS in 21m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58220c4ed89b453d87201e2fefbc312f : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e9f2dd66e2df4e8791f19c00a58695d7 : FAILURE in 27m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e67423899a784be1a4efcc0d7d061db4 : FAILURE in 26m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/61906da85069426984bf052b469e65e3 : FAILURE in 39m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1237092416af41ac93d7e249eb098047 : FAILURE in 16m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/94f519e10418415cbf3c7b28347192ea : SUCCESS in 13m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e269da92f9b140ab9490bf4485731a4f : SUCCESS in 13m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ceb7e187238843b08e1d53ec7edeb50d : SUCCESS in 8m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b79fee489cf4231b4eee84bf81b31f4 : SUCCESS in 23m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/802347d8eaa045c990b2d79f53f35e90 : SUCCESS in 47m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bd3cd27fe224750b25d5b1dafe1e803 : FAILURE in 37m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd5f5fdd07a8405b867bab10e6f2ed5f : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7419f5b497b54e9a8103412b99cb4ace : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":3},{"id":"548712c0f0e3904a920e049032eb51f14d7503d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 23:33:22.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84937edb71ee428db68e493c13bcd898\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d87c6449e28d4feb88f2b8c80130de28 : FAILURE in 1h 42m 24s","accounts_in_message":[],"_revision_number":3},{"id":"11024edb27fd485931429ba4deefbc030840cbf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-02 23:57:02.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a19f83277743c8edff8a7e4c47401c2e4d5d0cf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 00:54:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8110039b0594d7fa856f772b5db2ad6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5959d976b0ae4f78b54526492f35f695 : SUCCESS in 10m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e0f4993ba4e4563bb81b7647aa9c1e5 : SUCCESS in 15m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59bc0270fe6b48ed875ece821dcacedc : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d517c67c9324233a98e7190e2a5a581 : SUCCESS in 11m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d30125b6198d424d8b043ff5094514b6 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/44c2fae84fe7464dbe367ce87b7500fb : FAILURE in 26m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3a944c212b7c4d038844dcb4c760f5c2 : FAILURE in 20m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1f2f40d053af415f8febc11b7a72efc7 : FAILURE in 25m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/339ee3364be04aa4b85fe2c3fceb253b : FAILURE in 32m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2f77d1ed56394cc78e4d549a6d5ec1a5 : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/61b940ce12294f7384b563c7dc1996bb : SUCCESS in 15m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/718e9853be914748b38823cf39439629 : SUCCESS in 13m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f77c61bd69424da4976554fc0906c030 : SUCCESS in 55m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92d411257ee945e099c48c2704d1b075 : SUCCESS in 50m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8758b6bb8c8d44fa89fcadd6a29fdd6e : FAILURE in 26m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ca9ba2acc64db9b47f74af25814d4e : FAILURE in 2m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d3dc8f884ace49529c710f660bd27bbe : SUCCESS in 24m 27s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_ssync_sender.py","accounts_in_message":[],"_revision_number":4},{"id":"ebdab79dfb46e5987e9485f1bfb6f35f325b4960","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 02:51:22.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82019c2cff2f43d98e104bd71bf86ddc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d2cbbc9790d44cfbf5eeb21c12a4dcb : FAILURE in 48m 06s","accounts_in_message":[],"_revision_number":4},{"id":"2f7e98afe4c95c94de314e99967e88e782b80e71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-03 05:19:52.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"50db82fd813481c43bac2cfc5e006a13d05a1287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 06:05:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b1a6265b3054a228b0accb9c016f781\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520a3618f6b143f4acccd7f4f17b1091 : SUCCESS in 10m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3be485dcd5a4baf87e241422da1b6dc : SUCCESS in 11m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cea83e0849504c7997080e37738c160f : SUCCESS in 12m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d36ebe71828420ab14b663771971afd : SUCCESS in 11m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dfd10ea9f5b04c4dba979207f06a931b : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6e42cdbd69ce4306a5d38b534364630c : FAILURE in 24m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2d5f86b2d2334219848aacf6773fc255 : FAILURE in 12m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/34389eeb0f3a42699880887511f54c4d : FAILURE in 16m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d8aab1e4ee6347879189d3fd5bf5befe : FAILURE in 30m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8f571c0834c943adbecffb11beb87c95 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4281e728379c44168526355ee85eb3cb : SUCCESS in 12m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/837d43f3037942cba980f58104f943d7 : SUCCESS in 14m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad0b09744179485abe37c91628f4d5b3 : SUCCESS in 32m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/27893d0fbed44160b78e29b83b1f4ca4 : SUCCESS in 24m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97cc768726224a4293950a40335debae : FAILURE in 39m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eacbc7fdbf4494c8455cecc4f2e49ad : FAILURE in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/161b43761ad54d1a833a1ae7e619d48e : SUCCESS in 17m 31s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_ssync_sender.py","accounts_in_message":[],"_revision_number":5},{"id":"339bed3984b4d9ef5db6a1f6d5a7e4d0286022a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-03 07:55:22.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0a110b936ec411555fb5faa716c947e718e41062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 08:51:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fb642db9ae34a5087dfc98fd67f8bf8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea2b80966ce4da1b8de25d0db7fdf52 : SUCCESS in 5m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60696edfc54c49e8a04420dffc0d7b41 : SUCCESS in 10m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7cb7e8b7bbd422f9a406f0cb8dfc24d : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5277e38c58ce41ceab769ab3d0636a4e : SUCCESS in 11m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87e6d6c213c24192b4709620855445a2 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1d367c5977d94600af7e22d61248ebaa : FAILURE in 26m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fe855859d1044fa1bb1fe2d26f9672fd : FAILURE in 11m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f8275f28c3414383be3e74885bf9b2ce : FAILURE in 37m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/70ca4b1e4dd74e99ad4b48a6b327b764 : FAILURE in 33m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d41874bbb6004b29843950425e28fba3 : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d0fa5333563e45f7985ee37b9d117412 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4562b37090874fcebce48d63382be4ff : SUCCESS in 13m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c3b923e596c499e9dfb030a33c7bbea : SUCCESS in 49m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0700180c5ef642719b5bfceba41616de : SUCCESS in 35m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df7efc63b0f64973bfa715baebc4b5d8 : FAILURE in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361e92df43974340bd744ff4e76c00f2 : FAILURE in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/db32369b8d7b4e158dc20e1744333062 : SUCCESS in 19m 35s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_ssync_sender.py","accounts_in_message":[],"_revision_number":6},{"id":"54011b9d54e54499cb958663c90d027f07a7ce72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 10:46:43.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11a1e58a0cef438ab3baae00e85cd361\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/11b665165b824b3185bafc809b5cff89 : FAILURE in 47m 07s","accounts_in_message":[],"_revision_number":6},{"id":"82619616f1f1deef64ac65f51e67cbbfdc152544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-04 06:07:12.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"15465f3fe86006e15df4e6d001a575c2ad1576d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 07:02:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b71f17ab1f641e2bb2b6c6e98d119a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68dd927c6c3c40deaa8efb56a2078045 : SUCCESS in 10m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2eb75a99528647489b4068b5f4cbc133 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5cdfa9efac44c28874e4bb65a0a68c0 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b9914cb7e064c2eae3c1ff22a9c58b3 : SUCCESS in 20m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77361da909c54e90a3c6c4585a63d722 : SUCCESS in 9m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/94041af3a9c74655b5dde4c56e49539f : FAILURE in 38m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d163711646b043e8a575568835133c54 : FAILURE in 21m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/872832c618a64810a94e0d421e108bdc : FAILURE in 36m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1d16456c416e46c4b2a1416c60826c24 : FAILURE in 27m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f1309545e93745fab75c72a0b1bf719c : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f903c9ef73f54e7fa040b37f12da4eb5 : SUCCESS in 16m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e65ba90a217a4f548ad8f0d5487fb21a : SUCCESS in 17m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6a6639731be0465c8c9eabcafc9340f2 : SUCCESS in 24m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/62959f99cc9b4f529176bddd59afd9ce : SUCCESS in 49m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ea5e98e44f44a8cbfa04cf27521c366 : FAILURE in 37m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4793b365fa9415986bf7098e5dbea61 : FAILURE in 2m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0d4968e10c684afd9b8684451bcb3cb5 : SUCCESS in 25m 20s\n\nWarning:\n  Comments left for invalid file test/unit/obj/test_ssync_sender.py","accounts_in_message":[],"_revision_number":7},{"id":"79bc7b553be922325b4baf74b4df467a20911b85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 07:07:08.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a2ade9114de4b259887558d227fc874\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e152ee23c4a64032980ffa83d48e1369 : FAILURE in 57m 16s","accounts_in_message":[],"_revision_number":7},{"id":"f757447ffc416fa622c77f93ba4083e95dfa900c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-23 18:28:32.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/967738","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"8162b08636ddd54caf5fe33505f71f1a8026f81f","revisions":{"c52d3cef135652f95ffb262633a7e7f7f4996b67":{"kind":"REWORK","_number":1,"created":"2025-12-01 15:37:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/06/969106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/969106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/969106/1"}}},"commit":{"parents":[{"commit":"fc2473f4d23853ef5d7d00019ea3bf944fb6505a","subject":"WIP Timestamp: add random part to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc2473f4d23853ef5d7d00019ea3bf944fb6505a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 14:36:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 15:37:39.000000000","tz":0},"subject":"WIP container/test_replicator.py: user jitterless timestamps","message":"WIP container/test_replicator.py: user jitterless timestamps\n\nSome tests suggest that jitter is being lost w.r.t. shard\nranges. These should be addressed by ensuring that ShardRanges only\nuse legacy timestamps.\n\nChange-Id: Ie0e0937c45238532682d390250d923e0d928e711\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c52d3cef135652f95ffb262633a7e7f7f4996b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c52d3cef135652f95ffb262633a7e7f7f4996b67"}]},"branch":"refs/heads/master"},"2f0c2f5c42292f56b90e4d86e6a078bfe0eed2df":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-02 06:20:27.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/06/969106/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/969106/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/969106/2"}}},"commit":{"parents":[{"commit":"c272d5a7bbcda63d22c1c273e34d04d3b9539fb6","subject":"WIP Timestamp: add random part to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c272d5a7bbcda63d22c1c273e34d04d3b9539fb6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 14:36:18.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 06:19:50.000000000","tz":660},"subject":"WIP container/test_replicator.py: user jitterless timestamps","message":"WIP container/test_replicator.py: user jitterless timestamps\n\nSome tests suggest that jitter is being lost w.r.t. shard\nranges. These should be addressed by ensuring that ShardRanges only\nuse legacy timestamps.\n\nChange-Id: Ie0e0937c45238532682d390250d923e0d928e711\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f0c2f5c42292f56b90e4d86e6a078bfe0eed2df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f0c2f5c42292f56b90e4d86e6a078bfe0eed2df"}]},"branch":"refs/heads/master"},"2898fc70d781f436036a8365b4c5d37df3c637ab":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-02 18:34:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/06/969106/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/969106/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/969106/3"}}},"commit":{"parents":[{"commit":"f21a4bd4a7ac14fca41b7be03290ec0121c2008a","subject":"WIP Timestamp: add random part to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f21a4bd4a7ac14fca41b7be03290ec0121c2008a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 14:36:18.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-02 18:34:21.000000000","tz":0},"subject":"WIP container/test_replicator.py: user jitterless timestamps","message":"WIP container/test_replicator.py: user jitterless timestamps\n\nSome tests suggest that jitter is being lost w.r.t. shard\nranges. These should be addressed by ensuring that ShardRanges only\nuse legacy timestamps.\n\nChange-Id: Ie0e0937c45238532682d390250d923e0d928e711\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2898fc70d781f436036a8365b4c5d37df3c637ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2898fc70d781f436036a8365b4c5d37df3c637ab"}]},"branch":"refs/heads/master"},"7adc80836f345c9b661e5a71869f7c7ee7d99520":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-12-02 23:57:02.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/06/969106/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/969106/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/969106/4"}}},"commit":{"parents":[{"commit":"f34a834132973d964138ae0e8fab2953f016a957","subject":"WIP Timestamp: add random part to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f34a834132973d964138ae0e8fab2953f016a957"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 14:36:18.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 23:56:40.000000000","tz":660},"subject":"WIP container/test_replicator.py: user jitterless timestamps","message":"WIP container/test_replicator.py: user jitterless timestamps\n\nSome tests suggest that jitter is being lost w.r.t. shard\nranges. These should be addressed by ensuring that ShardRanges only\nuse legacy timestamps.\n\nChange-Id: Ie0e0937c45238532682d390250d923e0d928e711\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7adc80836f345c9b661e5a71869f7c7ee7d99520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7adc80836f345c9b661e5a71869f7c7ee7d99520"}]},"branch":"refs/heads/master"},"0454726b6d5815d952a58296d3719f71667e9315":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-12-03 05:19:52.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/06/969106/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/969106/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/969106/5"}}},"commit":{"parents":[{"commit":"adea08ec32be88ad157e6ebe6f3e07aef07f13bf","subject":"WIP Timestamp: add random part to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/adea08ec32be88ad157e6ebe6f3e07aef07f13bf"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 14:36:18.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-03 05:19:39.000000000","tz":660},"subject":"WIP container/test_replicator.py: user jitterless timestamps","message":"WIP container/test_replicator.py: user jitterless timestamps\n\nSome tests suggest that jitter is being lost w.r.t. shard\nranges. These should be addressed by ensuring that ShardRanges only\nuse legacy timestamps.\n\nChange-Id: Ie0e0937c45238532682d390250d923e0d928e711\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0454726b6d5815d952a58296d3719f71667e9315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0454726b6d5815d952a58296d3719f71667e9315"}]},"branch":"refs/heads/master"},"b679d93c61ee7f483b59d19f4bdd9c6711e50f81":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-12-03 07:55:22.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/06/969106/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/969106/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/969106/6"}}},"commit":{"parents":[{"commit":"04d0f9216f2c4426ced707a6b15845d8bb91444f","subject":"WIP Timestamp: add random part to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04d0f9216f2c4426ced707a6b15845d8bb91444f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 14:36:18.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-03 07:55:03.000000000","tz":660},"subject":"WIP container/test_replicator.py: user jitterless timestamps","message":"WIP container/test_replicator.py: user jitterless timestamps\n\nSome tests suggest that jitter is being lost w.r.t. shard\nranges. These should be addressed by ensuring that ShardRanges only\nuse legacy timestamps.\n\nChange-Id: Ie0e0937c45238532682d390250d923e0d928e711\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b679d93c61ee7f483b59d19f4bdd9c6711e50f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b679d93c61ee7f483b59d19f4bdd9c6711e50f81"}]},"branch":"refs/heads/master"},"8162b08636ddd54caf5fe33505f71f1a8026f81f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-12-04 06:07:12.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/06/969106/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/969106/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/969106/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/969106/7"}}},"commit":{"parents":[{"commit":"3340a9c0c5b439c6af6b377992fe50bc2e10dff7","subject":"WIP Timestamp: add random part to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3340a9c0c5b439c6af6b377992fe50bc2e10dff7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-01 14:36:18.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-04 06:05:48.000000000","tz":660},"subject":"WIP container/test_replicator.py: user jitterless timestamps","message":"WIP container/test_replicator.py: user jitterless timestamps\n\nSome tests suggest that jitter is being lost w.r.t. shard\nranges. These should be addressed by ensuring that ShardRanges only\nuse legacy timestamps.\n\nChange-Id: Ie0e0937c45238532682d390250d923e0d928e711\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8162b08636ddd54caf5fe33505f71f1a8026f81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8162b08636ddd54caf5fe33505f71f1a8026f81f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
