)]}'
{"id":"openstack%2Fswift~969190","triplet_id":"openstack%2Fswift~master~I405e218f6e6d6f12818f611e08d5c7f8963ee8c6","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-12-05 18:11:27.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-05 16:03:36.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I405e218f6e6d6f12818f611e08d5c7f8963ee8c6","subject":"ssync: Refactor tests to use Timestamp and new assert","status":"MERGED","created":"2025-12-02 06:20:27.000000000","updated":"2025-12-05 18:12:55.000000000","submitted":"2025-12-05 18:11:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"969190-p-timestamp-collision","meta_rev_id":"3eb5cc7dcf8aa7793096caad4a6005730b2e513a","_number":969190,"virtual_id_number":969190,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 18:11:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-12-05 16:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-12-05 16:30:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-02 06:59:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-02 13:04:27.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-12-05 16:30:30.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"6ec09a92febdcec7e803abfe5cdcc68cf61449a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-02 06:20:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5704e4ed33af6f3a15a57e022a8fd58d6a5e9448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 06:59:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9448eedeb66347c0a4171c819e6780a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0624910c74124032acb93c4f8ad70c60 : SUCCESS in 5m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79fa408c8fe2479dbef9901544d04508 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7230f747c0934ed395e5bf3dacabbc82 : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/246e9404af5642faac1391c4f4b8a801 : SUCCESS in 10m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d631538a14e640739b090d15bbb81430 : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/805e834a49834fc286a45588a5ca1659 : FAILURE in 24m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2418cc2c23584c148681e4e730bfbb1c : FAILURE in 23m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f0263b02a4304b1b82d00090ec0bfbf3 : FAILURE in 18m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4de0df1dcfb04d29959514bbf98f3b63 : FAILURE in 12m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4a9f7baf22e942c6bcc1140109cecb56 : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4593849238b14de5b1c21d360c08d0e9 : SUCCESS in 8m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0560f41aa85c4afcb435bd02613a963b : SUCCESS in 14m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85054b43d8c94e408df410f002d664a5 : SUCCESS in 23m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/252e85f478d245428d59807cdecda5f1 : SUCCESS in 37m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ac2273a6b4f4bfb99216b8856578b83 : FAILURE in 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ced4e1ee724f4bf69e2a0e66091cd969 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/909c66927c5a44ad91c3a9d6e7796b2a : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f4bf5b491881e8362243ef75e205b5439d5a65d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 07:05:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e0bba18ac424fff9643d56480402861\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e6399e57de514ecfab93b30e95fe5a7b : FAILURE in 42m 18s","accounts_in_message":[],"_revision_number":1},{"id":"5c19bcbe58c06cc032f78c74144e1c4a65ebe7b6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-02 13:04:27.000000000","message":"Patch Set 1:\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"66a24d43d1ccdd8079376200d43d53ab6857702e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-02 23:57:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"90c874c109322715d728330ff28954228e007564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 00:33:14.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c5bb79fe09d48bca823673c6350da6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5743cb27b7b4790a468bb6ab8b60edb : SUCCESS in 10m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/909c1023c9954c1686b46ba02b214fc8 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8455a961d3684fdb8c674aa524e40bd1 : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/256dcc046f8644caa6dd61e65f39e8db : SUCCESS in 22m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b366aaba7f848c59961458e7947ab72 : SUCCESS in 4m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/df084ff0e4324ca286838ef2a14803f8 : FAILURE in 19m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/de1013c453ee4ea5800783750edcbc05 : FAILURE in 21m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f4a8899d3ba34bc9a443fd0e8ecedef6 : FAILURE in 29m 07s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3fc1159d465f40e695fd31da2d3b11b7 : FAILURE in 31m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a8af420b127e45e6820070b898220428 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7e99e63bbb8d45f98ad2bb802555bb05 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/eacd691e075d4b039652e2eb9b2e7acf : SUCCESS in 9m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3472636b86484c7f8ef955b2314a03d1 : SUCCESS in 23m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6456ec0d6c6e4a35b07076e549a9c802 : SUCCESS in 24m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a333ddee5514e95a4265484096210f7 : FAILURE in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f410d4e7fb564a6ab95ac9337d1810fb : FAILURE in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b7e7c276d7bf4672a9ed7bb0786314c6 : SUCCESS in 19m 29s","accounts_in_message":[],"_revision_number":2},{"id":"9ca55ec937e806ae238d1620dd82b71427231082","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 02:29:00.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/882dc69789684f6c9963b5b7d6214567\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94bb080613c04cdcbeeb4f5bfb307fa1 : FAILURE in 34m 54s","accounts_in_message":[],"_revision_number":2},{"id":"b87da87281bc5be3946a874dc2d03a84a05f8840","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-03 03:22:58.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"882ddd1fb797cf7078f2ac5b144c4f48782df0d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-03 05:19:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"af6d09331b95330ff7ddb6daf5aff6ac2685ce95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 06:25:47.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a0396d85f264f19bf444f3bfe8dfc98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c9f10a510294414a2d8c7aae9594316 : SUCCESS in 11m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f5224e608654fb6bceb09928896a9cb : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0613daf998f444b90f668f93a0b483b : SUCCESS in 11m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e71f481d01f442e83c8b7ff6fa5b335 : SUCCESS in 21m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7dbd8cf8e5844546bbdd71f7d958000b : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ed99f4027f174b958320c0570d201202 : SUCCESS in 27m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/62e78cb3e0cc4ddf81e6748bd2791488 : SUCCESS in 15m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/74c4e11c27634001bcd5fafc3f51d9fe : SUCCESS in 27m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/af2116128efb46cdb42014250ccbf43c : SUCCESS in 21m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/889c6608671c48228896b3d872c6ae65 : SUCCESS in 12m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/479fca96c4b647d5a19567ffd0f5ce32 : SUCCESS in 13m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2db5a8d4d5ee4c72aa8b35be235ea9a1 : SUCCESS in 16m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/829ac72ba28d4e5f8d2d8d3e10e94379 : SUCCESS in 47m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/29f854ff1f8b48db9a75816cd136db5c : SUCCESS in 51m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c4d7c1bb8f240e8b2c33ba3da3ae4e9 : SUCCESS in 28m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a86c6310b3e64724bcc823438a974777 : FAILURE in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00872e68619949e6929418c9e5bd8594 : SUCCESS in 18m 53s","accounts_in_message":[],"_revision_number":3},{"id":"a29cdb06a333e179d3397e785cd8255f78a61a8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 06:51:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/043cd2664d4444fb95a2f4e5f94d54a4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/adaca76e62c943abbf51185944806293 : SUCCESS in 1h 26m 31s","accounts_in_message":[],"_revision_number":3},{"id":"7a71d472c89dc2abdc7950ac1a4b2ab308c442ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 15:07:41.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4bd3c378f7a15cd403625d4bab8a79c066b07ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:42:21.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d202023dffd4ef4a5e5857dab0478a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20aa9fbf0044346a18c0cd754fdfd21 : SUCCESS in 5m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/535673ebf8f948a4b20af71aea7d690a : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2c55c5ae31f4cbca77f8c68d8d0133e : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e43fd225f5774ef9a92ac7c1fc4e554d : SUCCESS in 11m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bdb786dd56164221b4a5251c31c53126 : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f4a28abb49c0411993dcf038b130e77d : FAILURE in 24m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ff07e744ae6546b498e4a4674d110e58 : FAILURE in 8m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4314219572bf4312899303e8d15ff2af : FAILURE in 12m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8436bfdfd0f14430bcacbdf93fcea4db : FAILURE in 12m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9f58802157384db5bc1dc8c3ad419c2e : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9635ef54bb974665a4ddf076a1d20420 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/96af3ada0ec34b8d930bc971923c83de : SUCCESS in 9m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b099afd5c81c41edb7c728574ea6c503 : SUCCESS in 22m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6680020f263f479685e1f66cd50c1d5b : SUCCESS in 22m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aadd2fe1ac945e5a12c4781f22059ba : FAILURE in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad87a701afc64137838d1d458ab9772a : FAILURE in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a03bac5852be4cbb8b1fa263ce707701 : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":4},{"id":"b7d9902741320f0169f161b7bb937bf1f4cc75ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:47:03.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7fba3fa2c4c40eabc524525f144e9f9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/294d889b70ea4e4daad3a4df0b32c37e : FAILURE in 36m 16s","accounts_in_message":[],"_revision_number":4},{"id":"b3ea56d6b5b85e48d306ab132560d4e8f6c715b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 16:02:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"83db6a78bbf2dc9a50715316b26aeae7d343cd9d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 16:03:36.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"caa9f33c7560fa309b33a9845725fa724379ec0e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 16:30:30.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4cc5b72ceda835409a1e1a9f6668f8903aab207a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:00:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93e058d1646e46bd9315f8a94835f070\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb564a7c9b7646beac6e1ebd6d2a6664 : SUCCESS in 5m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c53e47dc4e0943f5af62da1a0138be77 : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e5ec2d7d76343c2916ad543a396f817 : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68800ad9cc5f49309286d52f79f63483 : SUCCESS in 22m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f1c446b829cf4a1299086cfabafbfa6d : SUCCESS in 8m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/776ed6bdf3f94fd4b3c7a72a158e83ba : SUCCESS in 11m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6fcf9c488ce54db5b3e03a84fe51819e : SUCCESS in 18m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/495354e0491442b7bfdd341a25154781 : SUCCESS in 22m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fc929f43ff554920bb628cf99b2f3bc7 : SUCCESS in 22m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/caca08e4b70c480695aa20353732a151 : SUCCESS in 8m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5ffb9d0815a448eba211ee7ee68ab380 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b678e1e7d19c410c87913e1b633bf2b5 : SUCCESS in 8m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ebe0f56c42f4482687eca9b886432532 : SUCCESS in 54m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1adc43a8767a4dd98c2e5812d1845817 : SUCCESS in 37m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21f239515e78409e8b3ded623337e069 : SUCCESS in 27m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14b3d2caf521482bb24b830d2076ed54 : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f94de726f4d48b5bbd9695698ff889d : SUCCESS in 17m 35s","accounts_in_message":[],"_revision_number":5},{"id":"57203fb221f6dba1a4cee6ad2005cc03e380fe49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:00:32.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4bde3b5ce4bbf88bdc3b87b5fb1f7deaaabf04ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:27:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08a25233823f4f85a35a4361ac85ad5e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f0b604ef55c4df7961168e67fca04d2 : SUCCESS in 1h 21m 11s","accounts_in_message":[],"_revision_number":5},{"id":"a840259515d37bcaee358bea3af4de8dc174f837","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:11:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e460d128ca9344c3b193089f8d4d5032\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b6fef854bb40789f363ddd86d0a473 : SUCCESS in 10m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/357dac198ad24f239db17593e555ed1a : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a8ef00b6415c48e4968d49b593008d17 : SUCCESS in 25m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/719bebc71f464a158045a071b7f93990 : SUCCESS in 8m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b385bffb201f43539c1ba0868f13493f : SUCCESS in 11m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e963dee598c147fcb1dd31618545e8fb : SUCCESS in 22m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/461eccb3adc64bcfbbc1b442ac38b324 : SUCCESS in 9m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b1fcee2a45f8477c8da4f64895fe92c5 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/af93b7f9df9649bd9a5b3758aeb877db : SUCCESS in 8m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a3d68c3b0d534d4bbafaf580aca23735 : SUCCESS in 4m 04s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bfbfd2f5798d41be8a05cc256a5e60ea : SUCCESS in 5m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b17909f7a844d178ccd91c9af96d375 : SUCCESS in 46m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40f3f57f336e4201b11d90706e33387c : SUCCESS in 24m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1eb42a5777140a28e7ccd0f08a152bb : SUCCESS in 29m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e8a5d660b0945ffb83fe76a59797f0a : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2049addf406744138923788ba90b9bc2 : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":5},{"id":"294612a6fc0ff4571c7a87f1a07182b54af9f172","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:11:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3eb5cc7dcf8aa7793096caad4a6005730b2e513a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:12:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bae3d456eaf94b01906fd9e7b1c0212e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1b882956a943a4819c2968fbf2e854 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1c53f6584cff40d6bb4219b125aa5bb8 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"259afd2a756bd300ffac0591ab81d2faa160fd18","revisions":{"8b3feb378f80378cd23875e3c6ab0a07a1fdec34":{"kind":"REWORK","_number":1,"created":"2025-12-02 06:20:27.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/90/969190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/969190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/969190/1"}}},"commit":{"parents":[{"commit":"cc27a177b908629c139367f9c939112aa161f1be","subject":"sharder: use correct Timestamp format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc27a177b908629c139367f9c939112aa161f1be"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 05:41:23.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 06:19:26.000000000","tz":660},"subject":"ssync: Refactor tests to use Timestamp and new assert","message":"ssync: Refactor tests to use Timestamp and new assert\n\nCreated a new assertConnectionMessages() assert that will take it a list\nof message strings (without the length and added \\r\\n), these are added\nand compared against the actual connection messages passed in.\n\nI.e:\n    self.assertEqual(\n        b\u0027\u0027.join(connection.sent),\n        b\u002717\\r\\n:MISSING_CHECK: START\\r\\n\\r\\n\u0027\n        b\u002715\\r\\n:MISSING_CHECK: END\\r\\n\\r\\n\u0027)\n\nBecomes:\n    self.assertConnectionMessages(\n        [b\u0027:MISSING_CHECK: START\u0027,\n         b\u0027:MISSING_CHECK: END\u0027]\n        connection.sent)\n\nAnd the hex lengths are automatically calculated. This because more\nuseful when you pass in variable messages and timestamps.\n\nChange-Id: I405e218f6e6d6f12818f611e08d5c7f8963ee8c6\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b3feb378f80378cd23875e3c6ab0a07a1fdec34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b3feb378f80378cd23875e3c6ab0a07a1fdec34"}]},"branch":"refs/heads/master"},"bfaf9873f760ddf70e9f8ce15b0c25f7afef1261":{"kind":"REWORK","_number":2,"created":"2025-12-02 23:57:02.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/90/969190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/969190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/969190/2"}}},"commit":{"parents":[{"commit":"655326a88578afa9de5ed6550bad7478cc01708a","subject":"sharder: use correct Timestamp formats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/655326a88578afa9de5ed6550bad7478cc01708a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 05:41:23.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 23:56:32.000000000","tz":660},"subject":"ssync: Refactor tests to use Timestamp and new assert","message":"ssync: Refactor tests to use Timestamp and new assert\n\nCreated a new assertConnectionMessages() assert that will take it a list\nof message strings (without the length and added \\r\\n), these are added\nand compared against the actual connection messages passed in.\n\nI.e:\n    self.assertEqual(\n        b\u0027\u0027.join(connection.sent),\n        b\u002717\\r\\n:MISSING_CHECK: START\\r\\n\\r\\n\u0027\n        b\u002715\\r\\n:MISSING_CHECK: END\\r\\n\\r\\n\u0027)\n\nBecomes:\n    self.assertConnectionMessages(\n        [b\u0027:MISSING_CHECK: START\u0027,\n         b\u0027:MISSING_CHECK: END\u0027]\n        connection.sent)\n\nAnd the hex lengths are automatically calculated. This because more\nuseful when you pass in variable messages and timestamps.\n\nChange-Id: I405e218f6e6d6f12818f611e08d5c7f8963ee8c6\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfaf9873f760ddf70e9f8ce15b0c25f7afef1261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfaf9873f760ddf70e9f8ce15b0c25f7afef1261"}]},"branch":"refs/heads/master"},"ceb0b7784080588f6445011eaa3e6dacb77a3144":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-03 05:19:52.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/90/969190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/969190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/969190/3"}}},"commit":{"parents":[{"commit":"b0e911f863fd6f1484b4b909ba474c10d1d9b8d1","subject":"sharder: use correct Timestamp formats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0e911f863fd6f1484b4b909ba474c10d1d9b8d1"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 05:41:23.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-03 05:19:39.000000000","tz":660},"subject":"ssync: Refactor tests to use Timestamp and new assert","message":"ssync: Refactor tests to use Timestamp and new assert\n\nCreated a new assertConnectionMessages() assert that will take it a list\nof message strings (without the length and added \\r\\n), these are added\nand compared against the actual connection messages passed in.\n\nI.e:\n    self.assertEqual(\n        b\u0027\u0027.join(connection.sent),\n        b\u002717\\r\\n:MISSING_CHECK: START\\r\\n\\r\\n\u0027\n        b\u002715\\r\\n:MISSING_CHECK: END\\r\\n\\r\\n\u0027)\n\nBecomes:\n    self.assertConnectionMessages(\n        [b\u0027:MISSING_CHECK: START\u0027,\n         b\u0027:MISSING_CHECK: END\u0027]\n        connection.sent)\n\nAnd the hex lengths are automatically calculated. This because more\nuseful when you pass in variable messages and timestamps.\n\nChange-Id: I405e218f6e6d6f12818f611e08d5c7f8963ee8c6\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceb0b7784080588f6445011eaa3e6dacb77a3144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceb0b7784080588f6445011eaa3e6dacb77a3144"}]},"branch":"refs/heads/master"},"faec6689184c809a614d152319d08c091fcb528a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-12-05 15:07:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/90/969190/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/969190/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/969190/4"}}},"commit":{"parents":[{"commit":"1411582080cc241d438807a9fda26ab02401715f","subject":"sharder: use correct Timestamp formats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1411582080cc241d438807a9fda26ab02401715f"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 05:41:23.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 13:03:59.000000000","tz":0},"subject":"ssync: Refactor tests to use Timestamp and new assert","message":"ssync: Refactor tests to use Timestamp and new assert\n\nCreated a new assertConnectionMessages() assert that will take it a list\nof message strings (without the length and added \\r\\n), these are added\nand compared against the actual connection messages passed in.\n\nI.e:\n    self.assertEqual(\n        b\u0027\u0027.join(connection.sent),\n        b\u002717\\r\\n:MISSING_CHECK: START\\r\\n\\r\\n\u0027\n        b\u002715\\r\\n:MISSING_CHECK: END\\r\\n\\r\\n\u0027)\n\nBecomes:\n    self.assertConnectionMessages(\n        [b\u0027:MISSING_CHECK: START\u0027,\n         b\u0027:MISSING_CHECK: END\u0027]\n        connection.sent)\n\nAnd the hex lengths are automatically calculated. This because more\nuseful when you pass in variable messages and timestamps.\n\nChange-Id: I405e218f6e6d6f12818f611e08d5c7f8963ee8c6\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/faec6689184c809a614d152319d08c091fcb528a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/faec6689184c809a614d152319d08c091fcb528a"}]},"branch":"refs/heads/master"},"259afd2a756bd300ffac0591ab81d2faa160fd18":{"kind":"REWORK","_number":5,"created":"2025-12-05 16:02:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/90/969190/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/969190/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/969190/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/969190/5"}}},"commit":{"parents":[{"commit":"22605d21ee9c5cb25ad2f3b2f321c8593aefcbdf","subject":"cli/test_info.py: cleanup Timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22605d21ee9c5cb25ad2f3b2f321c8593aefcbdf"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 05:41:23.000000000","tz":660},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 15:55:02.000000000","tz":0},"subject":"ssync: Refactor tests to use Timestamp and new assert","message":"ssync: Refactor tests to use Timestamp and new assert\n\nCreated a new assertConnectionMessages() assert that will take it a list\nof message strings (without the length and added \\r\\n), these are added\nand compared against the actual connection messages passed in.\n\nI.e:\n    self.assertEqual(\n        b\u0027\u0027.join(connection.sent),\n        b\u002717\\r\\n:MISSING_CHECK: START\\r\\n\\r\\n\u0027\n        b\u002715\\r\\n:MISSING_CHECK: END\\r\\n\\r\\n\u0027)\n\nBecomes:\n    self.assertConnectionMessages(\n        [b\u0027:MISSING_CHECK: START\u0027,\n         b\u0027:MISSING_CHECK: END\u0027]\n        connection.sent)\n\nAnd the hex lengths are automatically calculated. This because more\nuseful when you pass in variable messages and timestamps.\n\nChange-Id: I405e218f6e6d6f12818f611e08d5c7f8963ee8c6\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/259afd2a756bd300ffac0591ab81d2faa160fd18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/259afd2a756bd300ffac0591ab81d2faa160fd18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
