)]}'
{"id":"openstack%2Fswift~969292","triplet_id":"openstack%2Fswift~master~I59e140d1226a4b0c2c8f37678bc2dc87d9ffa631","project":"openstack/swift","branch":"master","topic":"p-timestamp-sharder","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-12-05 18:08:09.000000000","reason":"Reviewer/Cc was removed"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-16 23:13:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I59e140d1226a4b0c2c8f37678bc2dc87d9ffa631","subject":"sharder: use correct Timestamp formats","status":"MERGED","created":"2025-12-02 18:34:39.000000000","updated":"2025-12-16 23:14:27.000000000","submitted":"2025-12-16 23:13:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"969292-p-timestamp-sharder","meta_rev_id":"fda1cfbfc7610069dd2a62e177834c2451a65bb6","_number":969292,"virtual_id_number":969292,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-16 23:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"recommended":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"all":[{"value":1,"date":"2025-12-05 18:24:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":2,"date":"2025-12-12 01:08:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2025-12-16 11:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2025-12-16 22:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2025-12-16 22:00:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-02 18:34:39.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2025-12-02 20:04:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-03 05:19:52.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2025-12-05 15:07:41.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2025-12-05 16:02:50.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2025-12-05 18:08:09.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2025-12-05 18:24:57.000000000","updated_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"reviewer":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"state":"REVIEWER"},{"updated":"2025-12-12 01:08:33.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2025-12-16 22:00:03.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"a32b519a1a161a6d656a54a23b1776da964ae87c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-02 18:34:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b82ae5fc97fec14901456261f18e0b28bd5fb673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 20:04:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba4baf0a348943ee80fe5d25c1699466\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b243f0b42e243439e9ab09a72814f06 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/aae0da0aa9e64b3c8016df90bc5818e5 : SUCCESS in 31m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a2c408dbdb44dc09e74c2d126b9ef4f : SUCCESS in 29m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/accae649bee2456a9de256aa904e35db : SUCCESS in 53m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbf91d631b2b494ca295f6a95261f6eb : SUCCESS in 1h 08m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5654329d5e0c4ac282bf9ac15d5b1c8e : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e39a55506ab1489aa6e1ad0e599de30e : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d702ee18464447a5ae7106f54f0e92a9 : SUCCESS in 21m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/86e5087f76784cdda0c0a612acb20d0c : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1970b8a4f86a458dbaf0deaef1a0407a : FAILURE in 24m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4d3e4f0a389e4381908fc8fc60e32aac : FAILURE in 23m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7056305ae44b48c499a6467849d90e88 : FAILURE in 34m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9bd03f38f1ef4cdaae4b145685b7b86f : FAILURE in 28m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/857db02c7fe64bc7bba788430618dd49 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/939654e2681844498b543a4d36408767 : SUCCESS in 13m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fe19d088c50f46358e23b70e33b49ff2 : SUCCESS in 8m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b0a9ab4d65b34ff59bb211bcbcc50dd8 : SUCCESS in 8m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8afd9d11fa134035b97fa7b0ead0dd30 : FAILURE in 19m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0621e470410d41a7ae6dfd80e4897575 : SUCCESS in 8m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/17682b66e1bd448e9a9415682fad3f37 : SUCCESS in 55m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/caa2ad2436d348ca82e8e2997e0adbf3 : SUCCESS in 51m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4fa10867e4904b88b8a62311868a48e3 : SUCCESS in 54m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8aaf7c511c7438aa32a542db6f86fff : FAILURE in 27m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b36a42bf4c3436fa73daebd8b96198e : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/025057989585460aaa47e5ecc935ba95 : SUCCESS in 26m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6fb38435a8804193becf76b2c586d725 : SUCCESS in 50m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95a4cb3faab1455c84b5a58ce17702d4 : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c37aff33dfc6ceb53167e780dd7eb4edb4a7fce3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 01:21:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afc9e5d89f3e40f6872a9c53a06ed448\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b06e29185e94b55ba8ac84c7bf4bb06 : FAILURE in 1h 19m 47s","accounts_in_message":[],"_revision_number":1},{"id":"09e562f8376328f333381098abca53c6e8b572c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-03 05:19:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"606b946255f6bb04d8fecc3e86896e6388ef0534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 06:37:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4894d4d603064a7286c0951633f972a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7f01cef8a704ed087d17cc1d49c302e : SUCCESS in 11m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/5702b843a5a344b7a9ecb8ccb8447e24 : SUCCESS in 59m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2369c9a9bd914afdb69085b7f99953f2 : SUCCESS in 1h 00m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2ac36699c91540888a4681d920f202d8 : SUCCESS in 31m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1672d57e198b4e0ea478c175111aa747 : SUCCESS in 1h 01m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb6006742d4443cc9365f8a9fb18c9c7 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bb8972b63374b4da11892f36bfea29b : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10af474ffb454bb98e9b978c8dd31c11 : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a12a2714f48e49f4ab89be1f8eed403e : FAILURE in 5m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/557770754f234ffda657027aa0bc5554 : SUCCESS in 24m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce171579672b4ffe920747b204727026 : SUCCESS in 16m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/96855c8c1dc34a5f84a6aa7844883e03 : SUCCESS in 28m 06s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9c58063d81a44c3b9aef8d0377e511b1 : SUCCESS in 28m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0930f30bf1db424fa84ce3284bfa38cf : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f8cea995fe0640178ae4dbac6ab266ea : SUCCESS in 13m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a34ee3d86eaa4efb9e8dcb79d3d3ce5e : SUCCESS in 14m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ef4a3c2337a426fb4c0661cf59d5a1a : SUCCESS in 6m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/35da9f37a62e42dfac16e35234da420f : FAILURE in 10m 46s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/fa96792dfc614c56a5c6d8dae7a52456 : SUCCESS in 5m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a636d9ef02c4d07a385953f1d3107aa : SUCCESS in 54m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dbf6efcb9f444e59a606204859d5cc26 : SUCCESS in 24m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e11eb07833a4437a6f331cd721ea1d6 : SUCCESS in 21m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41b0578b27d3435aac3ba0cb15882ccf : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c80d8c78764736a9c1eedbe0d775e1 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11136f928e3c41d68ef4b299eab4da8d : SUCCESS in 24m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e7e200f14a54baea4e846f2b27d8f92 : SUCCESS in 48m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cce6c5f699d3415f971b31356e8fe51c : SUCCESS in 45m 30s","accounts_in_message":[],"_revision_number":2},{"id":"6c8c7807607c1cae8c8c9a0e4a112c1e59b9292c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 09:19:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25131925980c4525addc72c68f69d7c7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d65d1a27f0e43ebb04df84aacdf44d6 : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8ad5d8d7b321620311f6865d12e752496535e82d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 15:07:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"420fa04cef6200d667fa4681c9e6387170aadf33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:48:24.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/467d2cdc4e3d45f091a5afe8169f0317\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f3a76c1fbcc84f51bfe48785916d6c01 : FAILURE in 38m 02s","accounts_in_message":[],"_revision_number":3},{"id":"335be147f9508fee0e49d2a62b2c2b4b618157ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 16:02:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8c3eeb13229ee2f75f09f5ad75a1f3950f60c777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:17:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95d3a1f9a5a14fb8a77fb09b9ac37ee1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd24c62cd9ad46638fb8785b7a14b741 : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/678ca51d65044748b7e6aad14d9850b6 : SUCCESS in 32m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c38b703ce9e64b1f9ed5ae05bca87c52 : SUCCESS in 32m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b71467d3761f43eea521922d7df943a2 : SUCCESS in 38m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecca32baa01b483da8a7b9aa3847bd02 : SUCCESS in 1h 06m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/242821de4a0542ccaf6159b35bc2ff60 : SUCCESS in 21m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c167bf2b7a6a4d389d058d69c7c9059c : SUCCESS in 12m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03d625b0308240a9b2dda2bf5dbdb9e7 : SUCCESS in 11m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b59fc4fa6b9345d6bab9e64f5b99ef71 : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6fb299f38ad04703b5c9dc5edafe19f1 : FAILURE in 11m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46560058d68a42a9906201fdf2512bc7 : FAILURE in 9m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/82f4f544853c469d8f9d5e039c30ca12 : FAILURE in 28m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/95b86dd09f5c4e1b8c90b63182daf923 : FAILURE in 12m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9767a470e6bb4aa28a87940bbe1fda3c : SUCCESS in 8m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/65f1be56c58a449295dd14186a72acea : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/50a5fa2483b24cf298ef9fec4669a4cd : SUCCESS in 8m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5c050088a771465c82f6a1b35240669f : SUCCESS in 6m 49s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7916e5addb484e3b8334d61977b7bf51 : SUCCESS in 15m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7022ae164f194f888887357eff62f6a5 : SUCCESS in 11m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b96e23535e34dd89a01b7f8da40a239 : SUCCESS in 58m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3900f882b5df40cb9bf47c51438fff2d : SUCCESS in 49m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5b6e43ba3f4c4761accc4fb92523e0d7 : SUCCESS in 43m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d7a1a24cdd74bc1b1a258bf70c1ac16 : FAILURE in 22m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47246f015c8e45419c7eec46fd3b2e0e : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/15ae1f6d5f1b40f2bd084b2c23789763 : SUCCESS in 14m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/917263ca714d44739b501daf2ca84b1a : SUCCESS in 31m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05b9a7289f514581bd5f7932dc00e4cd : SUCCESS in 51m 58s","accounts_in_message":[],"_revision_number":4},{"id":"c881f6c13ebca492d2080dc2f8184448ac684c38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 17:29:30.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69920819a09408ea3808c0f369edcd2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/82c27fd7ff8b4c67bac6eb3868903f63 : FAILURE in 1h 23m 05s","accounts_in_message":[],"_revision_number":4},{"id":"763cbb0dee8ed372585bc8e1fad8e22a6911363f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-05 18:08:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"48224a343d6cdcf9d327b3893d2b96e843a280ec","author":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"},"date":"2025-12-05 18:24:57.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"466ab3b05662953e7d5d78fc11728b8c596a9063","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 19:07:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48bce52f7fa74e699e1a1e1f9e840576\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/de46c6e6beca456299c13472c8d78b2e : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":5},{"id":"69fa39e1c75d2dea685ffceacedc595b81911948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 19:15:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/563a1859e98a49b1ab3f68fef344df1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b7532b4ccfe48b1ad2468cb5ebb2bdf : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/9875c2ec8049451c8b6f8c2b59eb07a8 : SUCCESS in 59m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/442595ef24bb437ba8cc4ad1b04c6e57 : SUCCESS in 47m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ebf53b019e474c499b3ed936acf873ff : SUCCESS in 47m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8a80cfa3efd432086055b7444d91f2d : SUCCESS in 1h 02m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/692981da10964b75b0f9b3b045c328fb : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e70b120b65254c4c97f44539d6d78127 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a50d19b465d4bb5876844dd35a3172b : SUCCESS in 12m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de9a1ad2e7fa4e898589b573265f1d72 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c3570d3132542889ae9ec17bcd27d05 : SUCCESS in 11m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/be600e36134946d8984614f6dd8a21e3 : SUCCESS in 20m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/001145c0b7af428ebe310e66969ca137 : SUCCESS in 19m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4453dc6657184c8bb832f57ab997a4f8 : SUCCESS in 29m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6cc0eaa9927741998e0c83a613a803d8 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/707675a1e32b45e08ac50016798fab56 : SUCCESS in 13m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5e7e6b049ca344518548143738012e21 : SUCCESS in 12m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0d81dc8310294fd087d8f6ae517ff056 : SUCCESS in 6m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c2aff19d85dc431399bef2f650c505a9 : SUCCESS in 15m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9bc01f6b7f1495489d4856fada37e60 : SUCCESS in 7m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9e25d7dbc9f48b7a8230160de8cd93e : SUCCESS in 1h 01m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf46541a0737469396140aff51ecea8a : SUCCESS in 56m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99a83a36702a42e58270807cf840fe07 : SUCCESS in 24m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aec0cc1f497845dc80656b8c35a2f910 : SUCCESS in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70a466d80954815b7e9062a52725543 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/97bc067ef92642179f4bbf8cef09af5c : SUCCESS in 18m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06fca00421e84979b20ab3df8be72e01 : SUCCESS in 32m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41a25b968ae94e608469227ecacaeb10 : SUCCESS in 34m 24s","accounts_in_message":[],"_revision_number":5},{"id":"dfacb5378f16f19a6b16182e74dea23c3218b6ab","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-12-12 01:08:33.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0c3a5115cc588b5ef712a735ef17bb0653fafd10","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-16 11:56:28.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f5d38234272b019b4987889e1cb8dc3ab36be0a6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-12-16 22:00:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":5},{"id":"410731cb40296478b883e87d41c5105b26baffb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 22:00:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bfa8420578ffde8712decd4fd2946ab0b69ef8ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 23:13:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20a6084d7af948668d3e2ca01e335ded\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af7ef180020422a86e187599f8e5888 : SUCCESS in 6m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/cb5a5233af29416f8bc282a2108c60e9 : SUCCESS in 30m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6158a0335c4d4881b7748c3c5f8cc6a6 : SUCCESS in 43m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b3b8b6962324457b9b3c066a69ae34db : SUCCESS in 56m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bb0050447964dd98634cb9caa9cd6c1 : SUCCESS in 1h 07m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/713b8fa36e8249d2a5fd7f152ff15724 : SUCCESS in 10m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/69cd5cefd34e4aeebad3a78836219344 : SUCCESS in 19m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e78267d08d934870957c5de4f9c77a7b : SUCCESS in 20m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d7bc9e1717be41d0a41d96c31d145577 : SUCCESS in 30m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/15f29b025d95400290a3137499ffc11b : SUCCESS in 34m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/64e077fc44e74f33ba9ad143883e8208 : SUCCESS in 7m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0f6ebefcf960405a8b2907171e5f153e : SUCCESS in 12m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/25b41b9d61054a9e9885931f7f97a744 : SUCCESS in 16m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7071a836be414ddf9a25f918d73e4531 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e111c080e7564c1092d813c3f90d3e0e : SUCCESS in 7m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/25cf455ba7d0486ab64eebc7fc39c3a1 : SUCCESS in 34m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9921c49c9bd4eb4a5553a749f201657 : SUCCESS in 53m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/21468c589a7f4a209593464e3177df64 : SUCCESS in 38m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58d6e6b7d77a4c0d9b84c47e4f6ffca5 : SUCCESS in 28m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a4828448ec456e819511e021670487 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/44f4f0f0fe19488dafc1335ce04d6940 : SUCCESS in 25m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c1589733abf4667ad4e02973c9fab79 : SUCCESS in 52m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36270ad75f26438ca7f1e552c277fe6f : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":5},{"id":"4d9c4f429197ce08cb7452102467cc2e7c3741af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 23:13:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"fda1cfbfc7610069dd2a62e177834c2451a65bb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 23:14:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6617c0a80aa4163937f98aa1c755019\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71280b5a9fb4bd789d19940f8e06971 : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4aace7a96c9940638686cecab10be769 : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9abd8ae71ec385e7687a11b50d7473be65eb9341","revisions":{"655326a88578afa9de5ed6550bad7478cc01708a":{"kind":"REWORK","_number":1,"created":"2025-12-02 18:34:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/969292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/969292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/969292/1"}}},"commit":{"parents":[{"commit":"e536d1681d6f26a77a446db9f764816d4bb1797d","subject":"container/test_server: cleanup Timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e536d1681d6f26a77a446db9f764816d4bb1797d"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 14:36:40.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-02 18:34:21.000000000","tz":0},"subject":"sharder: use correct Timestamp formats","message":"sharder: use correct Timestamp formats\n\nUse Timestamp.internal for X-Timestamp header when making a PUT\nrequest to send shard range updates to the root container. The unit\ntest already asserts that the internal format is sent, but the test\nfailed to use a Timestamp where the internal and normal formats\ndiffered..\n\nDon\u0027t try to cast timestamp strings to floats when sorting\nCleavingContext metadata. It is safer to cast a Timestamp string to a\nTimestamp and then to a float.\n\nChange-Id: I59e140d1226a4b0c2c8f37678bc2dc87d9ffa631\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/655326a88578afa9de5ed6550bad7478cc01708a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/655326a88578afa9de5ed6550bad7478cc01708a"}]},"branch":"refs/heads/master"},"b0e911f863fd6f1484b4b909ba474c10d1d9b8d1":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-12-03 05:19:52.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/92/969292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/969292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/969292/2"}}},"commit":{"parents":[{"commit":"e9f9cf8efd2958372f4e441666f688d4d7bae295","subject":"container/test_server: cleanup Timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9f9cf8efd2958372f4e441666f688d4d7bae295"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 14:36:40.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-03 05:19:32.000000000","tz":660},"subject":"sharder: use correct Timestamp formats","message":"sharder: use correct Timestamp formats\n\nUse Timestamp.internal for X-Timestamp header when making a PUT\nrequest to send shard range updates to the root container. The unit\ntest already asserts that the internal format is sent, but the test\nfailed to use a Timestamp where the internal and normal formats\ndiffered..\n\nDon\u0027t try to cast timestamp strings to floats when sorting\nCleavingContext metadata. It is safer to cast a Timestamp string to a\nTimestamp and then to a float.\n\nChange-Id: I59e140d1226a4b0c2c8f37678bc2dc87d9ffa631\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0e911f863fd6f1484b4b909ba474c10d1d9b8d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0e911f863fd6f1484b4b909ba474c10d1d9b8d1"}]},"branch":"refs/heads/master"},"1411582080cc241d438807a9fda26ab02401715f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-05 15:07:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/969292/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/969292/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/969292/3"}}},"commit":{"parents":[{"commit":"7732815c55865817fd1696f064aa1b2a3731f263","subject":"container/test_server: cleanup Timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7732815c55865817fd1696f064aa1b2a3731f263"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 14:36:40.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 13:03:02.000000000","tz":0},"subject":"sharder: use correct Timestamp formats","message":"sharder: use correct Timestamp formats\n\nUse Timestamp.internal for X-Timestamp header when making a PUT\nrequest to send shard range updates to the root container. The unit\ntest already asserts that the internal format is sent, but the test\nfailed to use a Timestamp where the internal and normal formats\ndiffered..\n\nDon\u0027t try to cast timestamp strings to floats when sorting\nCleavingContext metadata. It is safer to cast a Timestamp string to a\nTimestamp and then to a float.\n\nChange-Id: I59e140d1226a4b0c2c8f37678bc2dc87d9ffa631\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1411582080cc241d438807a9fda26ab02401715f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1411582080cc241d438807a9fda26ab02401715f"}]},"branch":"refs/heads/master"},"ba4850e4b1c85d08e684ea9c201a545166465910":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-12-05 16:02:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/969292/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/969292/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/969292/4"}}},"commit":{"parents":[{"commit":"ee15aeaa91671c0a6a1d74a3338e2f2257e9b9b2","subject":"test_relinker: cleanup Timestamp usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee15aeaa91671c0a6a1d74a3338e2f2257e9b9b2"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 14:36:40.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 15:59:47.000000000","tz":0},"subject":"sharder: use correct Timestamp formats","message":"sharder: use correct Timestamp formats\n\nUse Timestamp.internal for X-Timestamp header when making a PUT\nrequest to send shard range updates to the root container. The unit\ntest already asserts that the internal format is sent, but the test\nfailed to use a Timestamp where the internal and normal formats\ndiffered..\n\nDon\u0027t try to cast timestamp strings to floats when sorting\nCleavingContext metadata. It is safer to cast a Timestamp string to a\nTimestamp and then to a float.\n\nChange-Id: I59e140d1226a4b0c2c8f37678bc2dc87d9ffa631\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba4850e4b1c85d08e684ea9c201a545166465910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba4850e4b1c85d08e684ea9c201a545166465910"}]},"branch":"refs/heads/master"},"9abd8ae71ec385e7687a11b50d7473be65eb9341":{"kind":"REWORK","_number":5,"created":"2025-12-05 18:08:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/969292/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/969292/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/969292/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/969292/5"}}},"commit":{"parents":[{"commit":"0eabd90388e88d5d45d8b3985f454c15798efcbb","subject":"Merge \"Adds labeled cleaving metrics to the sharder\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0eabd90388e88d5d45d8b3985f454c15798efcbb"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-12-02 14:36:40.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-05 18:04:42.000000000","tz":0},"subject":"sharder: use correct Timestamp formats","message":"sharder: use correct Timestamp formats\n\nUse Timestamp.internal for X-Timestamp header when making a PUT\nrequest to send shard range updates to the root container. The unit\ntest already asserts that the internal format is sent, but the test\nfailed to use a Timestamp where the internal and normal formats\ndiffered..\n\nDon\u0027t try to cast timestamp strings to floats when sorting\nCleavingContext metadata. Timestamp strings may have the form\n\u003cfloat\u003e_\u003coffset\u003e so it is safer to cast a Timestamp string to a\nTimestamp and then to cast the Timestamp to a float when necessary.\n\nChange-Id: I59e140d1226a4b0c2c8f37678bc2dc87d9ffa631\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9abd8ae71ec385e7687a11b50d7473be65eb9341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9abd8ae71ec385e7687a11b50d7473be65eb9341"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38368,"name":"Christian Ohanaja","display_name":"Christian Ohanaja","email":"cohanaja@nvidia.com","username":"cohanaja"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
