)]}'
{"id":"openstack%2Fswift~969505","triplet_id":"openstack%2Fswift~master~I4d1bcb1ceb129b0588695f98947afe1b868621d3","project":"openstack/swift","branch":"master","topic":"rebase-relink-err3","attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-12-03 18:47:52.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4d1bcb1ceb129b0588695f98947afe1b868621d3","subject":"wip: add new replication_delay option to replicator/reconstructor","status":"NEW","created":"2025-12-03 17:29:07.000000000","updated":"2025-12-03 19:34:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5f20e42dc673840e7d26db1d193b92733639b257","_number":969505,"virtual_id_number":969505,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-03 18:47:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":-1,"date":"2025-12-03 17:49:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 18:47:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46bf3921750fa8dd1e341d6fddd8af8cd32758df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-12-03 17:29:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64cd66d85fc99f1c03e7943fa902984b6f936a6c","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-12-03 17:49:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9601a652ccc71236694cdde693b0e213a3c2fee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 18:47:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26e260c7afe04dfc872bf2f2a601d0dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6504ca25506f4c828ce34ed0b9c89062 : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/f03599baf406414b8a968ff5bdc01ad2 : SUCCESS in 1h 03m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fe0f96d90e8b4b399a96ab5b325f3c29 : SUCCESS in 1h 10m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1deabbd28ba54025a4966f4d5ffea32d : SUCCESS in 41m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e73652b4fa2241d29da16b3cefcbdec8 : SUCCESS in 47m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/718d161a1b8348178dd93df3b6a560a2 : SUCCESS in 17m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57e7374847c24bc680d20810f01c1d5b : SUCCESS in 12m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f3857df1c9e4961807de3918c9cbdf4 : SUCCESS in 19m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7177c167559242368281b0b029284fd8 : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2dfcc2bc6f384d23873df071127a879e : FAILURE in 11m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4fcf7100a73f4607b64b5ab292980749 : FAILURE in 22m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/788aaeb731c84075bb211fbc915b5a4f : FAILURE in 30m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ed79780fe8af465db8252cc7b6f3c507 : FAILURE in 20m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/58a9ea42cd0644eb837d267983ed5b4d : SUCCESS in 13m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6123c44d33f74caa85b88202e322c0ce : SUCCESS in 15m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ea51dd10424d46baa462d823a0c23c82 : SUCCESS in 15m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/987aea3edea9452a9fa4447aa3d69404 : SUCCESS in 5m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c45b7633cfd54ff394abb6828e4b44e2 : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c3c72b30086d44109c266a7eab57d1d4 : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/817b43e9ecd748b7a111acf7b591d9e5 : FAILURE in 1h 01m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c149773a2a3a4174b9bd14748fed119d : SUCCESS in 39m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2d649bed006341a99f2b947b5b5ac29e : SUCCESS in 46m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a409d247ca6432c8751494c68480eb5 : FAILURE in 22m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c9fa5d0c85547dc967fd86450c78076 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23042eb9aa3247c896605d6711ca5a26 : SUCCESS in 18m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e1cfe058a8e4ecba05c312a1de69aed : SUCCESS in 38m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5098f977b25149a7b3097953245e63fc : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5f20e42dc673840e7d26db1d193b92733639b257","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 19:34:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38e6240201b04e67b144957b4069b6bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a488ea98dd634b4480d1c8656d79366c : FAILURE in 34m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e566967c122a353af5b6b06d488da28e5a5dac3","revisions":{"4e566967c122a353af5b6b06d488da28e5a5dac3":{"kind":"REWORK","_number":1,"created":"2025-12-03 17:29:07.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/05/969505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/969505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/969505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/969505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/969505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/969505/1"}}},"commit":{"parents":[{"commit":"d64e4efb9f08b8337769433755f2d5a69ea89a11","subject":"Only 500 if we race with another iv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d64e4efb9f08b8337769433755f2d5a69ea89a11"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-12-02 17:26:59.000000000","tz":-480},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-12-03 17:28:02.000000000","tz":-480},"subject":"wip: add new replication_delay option to replicator/reconstructor","message":"wip: add new replication_delay option to replicator/reconstructor\n\nChange-Id: I4d1bcb1ceb129b0588695f98947afe1b868621d3\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e566967c122a353af5b6b06d488da28e5a5dac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e566967c122a353af5b6b06d488da28e5a5dac3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
