)]}'
{"id":"openstack%2Fswift~970975","triplet_id":"openstack%2Fswift~master~Ibe48a9193cd0a062cbf629f2ff0d950881380f48","project":"openstack/swift","branch":"master","topic":"p-timestamp-merge","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-12-16 11:13:49.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-12-16 14:22:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibe48a9193cd0a062cbf629f2ff0d950881380f48","subject":"OldestAsyncPendingTracker accepts Timestamp.internal","status":"MERGED","created":"2025-12-13 23:22:44.000000000","updated":"2025-12-16 14:23:24.000000000","submitted":"2025-12-16 14:22:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"970975-p-timestamp-merge","meta_rev_id":"e613f0aad2d29a93fe340158e347842938de7876","_number":970975,"virtual_id_number":970975,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-16 14:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-12-16 11:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-16 11:12:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-12-16 11:13:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-13 23:54:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-14 00:33:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-15 21:20:37.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"a9997f607498584461df924b6263a1aa0b3b8732","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-13 23:22:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1ed03ba68e4f6c4241273463a34fbd6598fb36e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-13 23:54:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa146a8efa18492ab61ca64a1681892f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52aacee13c47429faa00355686ee40f9 : SUCCESS in 30m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6e721fa2765b3580d59686257cdd6fe55243ef7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 00:33:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61ab694ca40f4311aa0678c33c113443\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4343384c75404ed3872a64d6c697e157 : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/3dc08d42949444eab68622eb49fa9d56 : SUCCESS in 1h 04m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b660aa174a414008a7922000b23df47b : SUCCESS in 1h 03m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/774d692e3b4446e292ad787f28cfbc6d : SUCCESS in 58m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/175341dd1e914cb591bbd95101dd9a0b : SUCCESS in 1h 01m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6266371558664ff09a3226b397df0939 : SUCCESS in 13m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65d7a1a9bdcb4726ab93430a94b9688c : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd16bccb31ac426aa8166ac0c3f5f3ef : SUCCESS in 19m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7c09bd03f68343e7a5abf40246716f4e : FAILURE in 4m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4901f1383d7c408ca2088662656d2449 : SUCCESS in 24m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e7914133869444f3a0d2f24d8f8f147e : SUCCESS in 13m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7a760f8d6234420d8825678c624aa80a : SUCCESS in 27m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f62bf95dd93e45afbfeaed2cc585dcca : SUCCESS in 12m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dc5a02d7cf0d406b8672a02f00f60630 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/194e2024505b4f40b0c32e1b01db8569 : SUCCESS in 16m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/311d6da5a061452a8900b66b564af714 : SUCCESS in 11m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bc41ebb9db0c42b38faa72ae429dc2e8 : SUCCESS in 5m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b5e899acbc884c4d83fb884ef90e2b92 : SUCCESS in 7m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b5e90de9273840ccab331ba2c7e508f4 : SUCCESS in 8m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a73041ac29cd4345b0eec9d550e8e898 : SUCCESS in 54m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24495d932f964332b3faec8c5fe16b4f : SUCCESS in 57m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4ad228b5c5be4b69bb33d9ae990f1704 : SUCCESS in 23m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9c2fdd1a1da4bc097af5ba3c6a47901 : SUCCESS in 27m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fa77e315fd64f689bc9f1554405ad5b : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/15812bdb595f491cb1d38e4633e404ae : SUCCESS in 17m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d080a58e8b047e0a28f22440d10049d : SUCCESS in 50m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ee2dcabde5e464c851f4e874fc70cf9 : SUCCESS in 32m 50s","accounts_in_message":[],"_revision_number":1},{"id":"38c32ddd3f0b6f6b9a525e148dcb1d67bfc44020","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-14 18:47:57.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bdcb53bc8ac339ddac7283f047771f42e417f7a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 19:34:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96c4fce0414a44fb94165fa9f2f270fe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aad5cd9a55b7438f874f84a2e62ede21 : SUCCESS in 44m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c41009bfc1947fec445bb046a0bdfe0a042ec052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-14 19:55:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24bb8cf457a9438ca4e4e9a325987472\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da0b51e0bcb49baa60c7d40b8bb5b23 : SUCCESS in 10m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/c50320ab11e440c5b841fbe8f2be49fa : SUCCESS in 54m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7ea4cc4abaa14ac799e84d0fdaeaf8cc : SUCCESS in 1h 00m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bfda21836e3543379d20345a8acbeab9 : SUCCESS in 52m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6579491ef19f4dd8bdebd2f48deed143 : SUCCESS in 1h 00m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e36e382d19164367851f1eb84c538814 : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02370b31374e4abf9db691ec3126f05d : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4e6e428c21f4100ad8422a5b7888021 : SUCCESS in 17m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/907db5b63d8d4658a2d51f5175769eeb : FAILURE in 4m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0d4e6e27986740a682793ebf81bc6d3e : SUCCESS in 11m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6b50577a8e59470ea7863d795c7e9289 : SUCCESS in 19m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e7da6948add1429081fac6822142b7f2 : SUCCESS in 19m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/49795b02fa8c414c9e55c1b278d66f6c : SUCCESS in 29m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2bac1961aa1146cca10b0ee73ada7a06 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4a0f3ee97869400587bca2b32ec53b39 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/972d918049f14e299787c7ad3a08d3f5 : SUCCESS in 13m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89903cd18c824c05b7523576500ad0a2 : SUCCESS in 5m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/db56969ffc5547959a229c88ea95ca43 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dfe1a34bf74e44c4b77d20fa9b7415da : SUCCESS in 8m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/becccf69f0e64a82b0a066162c0d2489 : SUCCESS in 43m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c90418a36794c1996a5de53ad29858c : SUCCESS in 33m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f4428c451ef2453a8db6b1b85f9baaaf : SUCCESS in 56m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b73e4d6cf49244a2b704aeb324f67084 : SUCCESS in 27m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/981e943d28c5412ca716deeac4992bcb : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc4676c1059e425bb6e3425b51d10120 : SUCCESS in 24m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5344e3214dbb4cb3855d0a2ec0a88d2f : SUCCESS in 1h 01m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d78d5e99211640e98fed6c66797b0595 : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":2},{"id":"e80a4dfddd701155d42179288e95401f635a44ab","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-12-15 21:20:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"92f007a40258ed34fb45204df2d20731ec7f57c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-16 11:12:12.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"fdfabc31b107c871b27487e4e9fcd0c0e6153f8a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-12-16 11:13:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5fcbd256c522a9e35827c798a248451a325b5f0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:19:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d6a870a417a4a81af2a9ebec22215bc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7eba74b4a1624bad98cec91ebbc02fc6 : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":3},{"id":"ebe471e66cd4876a63cc26d5e0c0196d12a7eeb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:20:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12e579e1d7ca4a82bb43c0ed56ef8621\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4173b6130e094c2098fc36df2d07c958 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/52974dbab8e442da9865fb726af03958 : SUCCESS in 1h 00m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c5e7201b74843c08ce63482e51f1b08 : SUCCESS in 52m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ecff35765ca4c8c9d38c34a9e245bf1 : SUCCESS in 52m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c16ab585454044c9bcfd04a7d0f35c73 : SUCCESS in 34m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49172c72d78547479f38edd0187534ee : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cacc54db2a524533befe6883d279d43c : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f5264a3a7f146848a88112ce2c870b1 : SUCCESS in 10m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5cc2310a1ed541d097139d0a96dff465 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c7b5abf38a324b95912d3b0aed8e3381 : SUCCESS in 11m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/47b0605cdfee4b648e0238d42bb4f8f3 : SUCCESS in 9m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8fefd58d0f6b440195a48bf73b73412d : SUCCESS in 33m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9a288d986a5040d0a5499f19a5546092 : SUCCESS in 28m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bc3a545659f841d5bfe327dfcd6c1837 : SUCCESS in 8m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/24a952e740c14393bcf5acc2d95eae61 : SUCCESS in 14m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3f7f19a2aff84fd1a3bf7ad7da3cfd32 : SUCCESS in 20m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b36355089c6c49148d4d7d4d1fb934a1 : SUCCESS in 6m 05s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d12d7e6a07b54732a76312d295bace52 : SUCCESS in 14m 44s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/16208e8f941349afb3b99fb9677c2e3b : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/680fda0989794ccf8bc19eea7bb7d8b4 : SUCCESS in 55m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4939e8bc2ab14c2ba9fea0a99b507edc : SUCCESS in 39m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f478a80d32524655b0844d5324ab6e25 : SUCCESS in 55m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99bf3bec17424553abe6eacc99dd3bc7 : SUCCESS in 24m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d3ebc3821c454995d92f45fed38dec : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd91790cd77b4f69b69ad2927033a06e : SUCCESS in 20m 27s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b4b7b423218465e9739b100bb448f42 : SUCCESS in 31m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02c7cd37a0f64ba48dc45b7885d07e13 : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":3},{"id":"25245a0336b04a25f7c6d54e1703336881063010","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:21:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"baf7ed0812b9ec222c0e3ebcbb2c1b0beca5affa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 14:22:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c8706e61da04067a9d190b4f300a947\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a017063d5a44edaaf38ba2ce8421a98 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/921964e5f6454e1c81446edde57d2a91 : SUCCESS in 1h 02m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6df5d0e986a74e75abdb786f1b0f4bcb : SUCCESS in 59m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8466655b5bb74884973c4b0185d1672f : SUCCESS in 49m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd1322b882b4489294874e230fe5ed17 : SUCCESS in 1h 06m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/983368b04bcc4f47a08335d102cff22c : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b5c5d4a3b0ca41f19b23cf7030699352 : SUCCESS in 25m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2efd12f564a94c73bea33a2aa2642b5c : SUCCESS in 8m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/265395a3dc724112816ea5fe57c2e9ca : SUCCESS in 23m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/682f77e90309490ba3d39ffe51b3e5bd : SUCCESS in 11m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/580b1e26653144ca80cc8279d6533219 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d4f8df8496f24ac08d366b90bbcca4fc : SUCCESS in 8m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f782deddb48b468f8058963c32669bcd : SUCCESS in 16m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e1442c1dcddf481cb1bb2962a9598078 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9e1cbbebe5614eefa8518b1c8dce50dc : SUCCESS in 11m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7788b18666d44d8875c52d80445b630 : SUCCESS in 59m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7df6c6f2f8a746be950acbb686aebf52 : SUCCESS in 1h 05m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/065399f7dc6142548dfb1dae9db92122 : SUCCESS in 47m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6fc245a38b548fe8fabaa0632740b99 : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5d306996b8a49dda065acf29b3aa91e : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7de626712334e77b61142e4ad5f6552 : SUCCESS in 24m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63433718c656461c92bda38e03d6898d : SUCCESS in 42m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00cab4824e6e458bacff3d362ac9acc5 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":3},{"id":"7c86e6a6a4390eff71cacd20bb6c3ac928ebe154","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 14:22:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e613f0aad2d29a93fe340158e347842938de7876","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 14:23:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1683fa91ffe4ebba3d9c597449db2b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04121fb423943958d3d4d10fd151934 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/de4d2718a7b04118b95571b0e4ad41d4 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1698fc32489572787af45cab2bb9b8774ab1fb90","revisions":{"f13d2e5fcc6fd36196e7b22e26d24c84256f7130":{"kind":"REWORK","_number":1,"created":"2025-12-13 23:22:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/970975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/970975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/970975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/970975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/970975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/970975/1"}}},"commit":{"parents":[{"commit":"48809274d369254c1b600ab4c22d5573058de64c","subject":"test_relinker.py: fix delta timestamp setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48809274d369254c1b600ab4c22d5573058de64c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-13 23:16:50.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-13 23:22:09.000000000","tz":-360},"subject":"OldestAsyncPendingTracker accepts Timestamp.internal","message":"OldestAsyncPendingTracker accepts Timestamp.internal\n\nThe OldestAsyncPendingTracker.add_update() method is passed a\ntimestamp string which is derived from the name of the async_pending\nfile. This string may take the internal form of a Timestamp which\ncannot be directly cast to a float; the string should be cast to a\nTimestamp instance and then cast to a float.\n\nChange-Id: Ibe48a9193cd0a062cbf629f2ff0d950881380f48\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f13d2e5fcc6fd36196e7b22e26d24c84256f7130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f13d2e5fcc6fd36196e7b22e26d24c84256f7130"}]},"branch":"refs/heads/master"},"61d084626d422321900f57252ea566a7c1ca8434":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-14 18:47:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/970975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/970975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/970975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/970975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/970975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/970975/2"}}},"commit":{"parents":[{"commit":"23180fc29ed02541f2cb80fd5917ba3a75f3df8f","subject":"test_relinker.py: fix delta timestamp setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23180fc29ed02541f2cb80fd5917ba3a75f3df8f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-13 23:16:50.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-14 17:42:17.000000000","tz":-360},"subject":"OldestAsyncPendingTracker accepts Timestamp.internal","message":"OldestAsyncPendingTracker accepts Timestamp.internal\n\nThe OldestAsyncPendingTracker.add_update() method is passed a\ntimestamp string which is derived from the name of the async_pending\nfile. This string may take the internal form of a Timestamp which\ncannot be directly cast to a float; the string should be cast to a\nTimestamp instance and then cast to a float.\n\nChange-Id: Ibe48a9193cd0a062cbf629f2ff0d950881380f48\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61d084626d422321900f57252ea566a7c1ca8434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61d084626d422321900f57252ea566a7c1ca8434"}]},"branch":"refs/heads/master"},"1698fc32489572787af45cab2bb9b8774ab1fb90":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-16 11:12:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/75/970975/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/970975/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/970975/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/970975/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/970975/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/970975/3"}}},"commit":{"parents":[{"commit":"bf1e370b2892b7615a96517d2e156f57efea8933","subject":"test_relinker.py: fix delta timestamp setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf1e370b2892b7615a96517d2e156f57efea8933"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-13 23:16:50.000000000","tz":-360},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-12-16 11:10:54.000000000","tz":0},"subject":"OldestAsyncPendingTracker accepts Timestamp.internal","message":"OldestAsyncPendingTracker accepts Timestamp.internal\n\nThe OldestAsyncPendingTracker.add_update() method is passed a\ntimestamp string which is derived from the name of the async_pending\nfile. This string may take the internal form of a Timestamp which\ncannot be directly cast to a float; the string should be cast to a\nTimestamp instance and then cast to a float.\n\nChange-Id: Ibe48a9193cd0a062cbf629f2ff0d950881380f48\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1698fc32489572787af45cab2bb9b8774ab1fb90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1698fc32489572787af45cab2bb9b8774ab1fb90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
