)]}'
{"id":"openstack%2Fswift~971351","triplet_id":"openstack%2Fswift~master~If05e097fa32e7a332f01a40530bfa0d0307eddd0","project":"openstack/swift","branch":"master","topic":"s3apilabels","hashtags":[],"change_id":"If05e097fa32e7a332f01a40530bfa0d0307eddd0","subject":"test: s3 checksum header classification is weird","status":"NEW","created":"2025-12-17 16:16:52.000000000","updated":"2025-12-17 17:14:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1fd23367e638b5deb8bbfef594485e762b0af76b","_number":971351,"virtual_id_number":971351,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-17 17:14:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-17 16:51:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-17 17:14:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"119ec38ab3fed1ccf63684f53a13bf6f1ecfe554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-12-17 16:16:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63077143ec7b612d9af401a6738993645d6ae1aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 16:51:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd97099bdd6145c8aba53f030d311b3e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/575a53c47525433b88ef28be03b2ce52 : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1fd23367e638b5deb8bbfef594485e762b0af76b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 17:14:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e79aee40b004e48a8f0f873d8de223d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2c642ff82144faaa3e1280ab81d11c : SUCCESS in 4m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af523be20b344afd9f3747a15e31ab1a : SUCCESS in 11m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d980c68d3024a568042be840af3bc62 : SUCCESS in 10m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a49712307ae64d889c1549e13aef57e8 : SUCCESS in 10m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4a04fad558414fb9ba958722fa01c50a : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cb2b6aa1bc4d44dfaa9fae6f8fbe08b3 : SUCCESS in 25m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/095142d2f9604832a46c733d37c3b740 : SUCCESS in 19m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d97da48372044688a340e50651e61a5e : SUCCESS in 27m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/464117bd530d4412a5a4e2872061423a : SUCCESS in 11m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/99f5473b07a9452d9efa73decaa36d26 : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dc369f2bc036450b8195b6cce9f5cffb : SUCCESS in 9m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0856606a919447219f04c3f162d25297 : SUCCESS in 8m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c87841225aa4d3d9812bedb68825fd6 : SUCCESS in 24m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ca0bcc0ca04043fba59c703f36532bdb : SUCCESS in 51m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/553813246b194235af117db490b2b294 : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389383f7ade54e8cb42ef52099bebadc : SUCCESS in 2m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29214275f73141ef9ef1dfd196be12a1 : SUCCESS in 18m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64297042c9aedd674e1d41fda932e160c1609d06","revisions":{"64297042c9aedd674e1d41fda932e160c1609d06":{"kind":"REWORK","_number":1,"created":"2025-12-17 16:16:52.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/51/971351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/971351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/971351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/971351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/971351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/971351/1"}}},"commit":{"parents":[{"commit":"01d7eef889c23e11a8befa2bd2fecbef9906c135","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/01d7eef889c23e11a8befa2bd2fecbef9906c135"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-12-17 16:16:07.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-12-17 16:16:44.000000000","tz":-360},"subject":"test: s3 checksum header classification is weird","message":"test: s3 checksum header classification is weird\n\nChange-Id: If05e097fa32e7a332f01a40530bfa0d0307eddd0\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64297042c9aedd674e1d41fda932e160c1609d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64297042c9aedd674e1d41fda932e160c1609d06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
