)]}'
{"id":"openstack%2Fswift~972217","triplet_id":"openstack%2Fswift~master~Ia09ddea386bd57519c788e1e31543769b7e53b28","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-02-27 15:35:09.000000000","reason":"Change was abandoned"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-02-27 15:35:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia09ddea386bd57519c788e1e31543769b7e53b28","subject":"sq? Timestamp: default with_jitter\u003dFalse","status":"ABANDONED","created":"2026-01-05 16:17:50.000000000","updated":"2026-02-27 15:35:09.000000000","total_comment_count":7,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"f579dcc81d236e223358257d00ce254c625d4273","_number":972217,"virtual_id_number":972217,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-14 16:14:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-05 17:54:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-07 01:08:57.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"}],"messages":[{"id":"7c7ac0bc648a2176e0fe409368f200e82eb9355b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-05 16:17:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc8d8dbb073ae9de13f5dee9e09f4d9c8e545e70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 17:54:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf9ffe7908874d1499a7e98e9ca54605\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e8623ed7a34efa9067623c39b7ae60 : SUCCESS in 9m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/a96548e78ed34b5fb749b92e426726cc : SUCCESS in 47m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9dbd5196dd9a4df9b56c4473248f2147 : SUCCESS in 29m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e2ba7af8f464d0892c775f739166e59 : SUCCESS in 34m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a4b7b33879f4a4e83dcf1a6841ed077 : SUCCESS in 1h 00m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4d823d578a747d98100f0831a396ff8 : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3d9a6056ba14ebc9038cc79ccb61604 : SUCCESS in 10m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b882c00f1323439da344572f1205f1b7 : SUCCESS in 17m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/05c9c8e6095b4d2f8619f7ca06df4ee4 : FAILURE in 3m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a073ccf0da3f4cacaf9125b491c367b1 : SUCCESS in 27m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4bff681543ea4b4bbe0ef19be71cd9ed : SUCCESS in 16m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7a48c655f6c7448b9d92d9fb174932c8 : SUCCESS in 30m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7b72e6459bef4f0ab204168d2eef567e : SUCCESS in 31m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bbe3cd4580874fccaf96780175f104db : SUCCESS in 12m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/93dd836416b1437ba2e1dce1f1a781fa : SUCCESS in 16m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5ddc1f6faf8645509c481fcc66cf2b92 : SUCCESS in 15m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92a56e52fca9420080ff856dab4b6e8e : SUCCESS in 5m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/477c32d19d0744ad962a0f1fc1e2f616 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/45fdbc3958184b5eb11b8cab4d09868b : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e707a2d9fdd240b68adc7cb461f5662b : SUCCESS in 36m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/daff7219b9b24f629caf47abf5d0d81c : SUCCESS in 49m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/69fe4b7e86c64adbba49bad42912acf9 : SUCCESS in 56m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a84c8d178d14ee1873a3d166e48b46f : SUCCESS in 29m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2f58918eee54008a8b02818af882667 : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d349734b310546718584105e7caebd28 : SUCCESS in 25m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/081dd1cf97394e2685bbeb35ef58725a : SUCCESS in 51m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc3bd3a789c34ecda9ddf58795e6130d : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4075a8d21a7a7a073747d051a3b95e3726d54521","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 18:46:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0905bcdb339940dbae0cb28fb9e1940b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f777184cf1b64e72a5b9959abe9c88c0 : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":1},{"id":"48bb7b9e515ba9d6f31b3524389ece7b5778aa9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-06 17:43:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0f8cd8b5fb4697aa2a6afcb65b9bfde9d07991a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-06 18:04:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bac2ffba1bcab4dbe36c440fcc944b77c2bc83d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 18:44:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b7974d27e80465399a89671ba76d0d4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fa6271bcbe1a4a0fb117c383103007f6 : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":3},{"id":"747b8f643a07e9fd501349bcc9e8e9ed036db30c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 19:12:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ba3d7ed9f584a0f88a11c80dcc2de85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc31c735fc65463ca7a16b3609721749 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/3efcbe66cd174003847ef90dbec3f573 : SUCCESS in 30m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/54e2e090fe9f4d6c9f8de066f0f725b3 : SUCCESS in 54m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/952cc59b88a5406f945fa9043bb6d404 : SUCCESS in 41m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18ad5b92d2ef46d6b859493e0e9855bb : SUCCESS in 33m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db06ad103c6e4e9181ebd1947086a919 : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/770a9182f19d43bcb9795f1227ba57ed : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2c9f62939cc4637ace08b481b7d6c93 : SUCCESS in 19m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/60cff2ac9e99420b940c895141891c14 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/47ba190d9b394760899a2c190905a62c : SUCCESS in 21m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8bb259889efa43da83e243e8425240e4 : SUCCESS in 23m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5a1fed85e1aa4ec4893fd80ce94933e0 : SUCCESS in 14m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d681cf9bb42a48638a31e1c90d28cb08 : SUCCESS in 31m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/435752df30404f5d9dda95da9a44eebe : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0fcc68fd4f0048ae8ba1fd1b35753cf5 : SUCCESS in 13m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1258a880e6e449e38b356c94d87533a3 : SUCCESS in 14m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9acb0d71b8d4f819c904458153d9fec : SUCCESS in 4m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/341ba0da2f3e4ae0b8c418660d632321 : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7fd5fcf433c94ce09e1ea054e4da473c : SUCCESS in 7m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fa90c97bd9540f9b11960c342f5ec63 : SUCCESS in 50m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6942518f1c3141579217b36549ceddce : SUCCESS in 50m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/826aa06bd00a454a81988bf383b2fbd5 : SUCCESS in 48m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85d0c779ec5d40d2b97368d0dfd4c7b6 : SUCCESS in 29m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8f8e9019f64cc3a6b12ebafb8832e7 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/177be72e560646c8bf8d08d95fb60171 : SUCCESS in 19m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47d0779f63464dad8bf1491400580254 : SUCCESS in 47m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acba2a490a1341458b5effb6859e3d69 : SUCCESS in 34m 27s","accounts_in_message":[],"_revision_number":3},{"id":"0587da43dbe239795d99f33fc2442939af121cfc","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-01-07 01:08:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c42ed3a3728d95b8ba052ce015ecae21c9efd179","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 10:33:03.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5c4d1f0ce915487a439f80c67250245d920db9c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 12:12:42.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"647ea5bd69314e2cea4b6b0766d7e9b97f7405ad","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 12:17:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"32edc8bec35a516fb6eb03dae5217083a7262f15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 12:30:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e1afbe6a068b3b6db93ea99a1ecfdecf179343de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 13:40:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa4cd9b5a9c8436b91ff7cb6f613fc1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6579d01b55348b4b7e46b085ea0819c : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/e69327ad0b9047ac80ffceb7eb633bf0 : SUCCESS in 51m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d01b87bece614c4684b76b6c4f69b0dc : SUCCESS in 1h 01m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fab06ed93dac47339399802fa23870b0 : SUCCESS in 41m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f25ffc448f914482b6d61d091394f9b4 : SUCCESS in 1h 05m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8196287513374ca2912d8fe8681f7cec : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e695a4cd4854a8f9570d7b19e5539a4 : SUCCESS in 10m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be3fb0b15af44dce9507e1271563579a : SUCCESS in 11m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c25e66ba67643dcac7a0b43df328afb : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/faa92a84c1034611b7a00abf5ea547eb : SUCCESS in 10m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e8652656b67498984b79094816a4019 : SUCCESS in 20m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0fe1095bbb6c49b896e453ed6e209cb8 : SUCCESS in 19m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8b47d2f11d4c4b39ba7ff42e58b8c0a8 : SUCCESS in 33m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4000497a28cf4439a34148359d42b553 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/78659f3589064d049514f9ce0240eba0 : SUCCESS in 13m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d35e7bd52bcd4bcaaa54a4b342d6938f : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a4616052f767475498e542d884439048 : SUCCESS in 5m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a83884e8a1a42dea78b245389498599 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/12e2fbad33d94466be8113ec755be6d8 : SUCCESS in 4m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/32ded0e507f641e882a64055d4757df5 : SUCCESS in 1h 00m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/373f762859134eb99fb3de786fdcb44a : SUCCESS in 57m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/089212abfb7a4d2ca57c1f7377b5963a : SUCCESS in 49m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7e8c806c0654e669da054541d14c63b : SUCCESS in 26m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e8f013ee7743d3b636710c2a669e80 : SUCCESS in 2m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/03e89bce1a7046048f66d69f49254f94 : SUCCESS in 16m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0099865490249c4a0ba05491877681d : SUCCESS in 44m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/302ab699b4da479f923923fd68a67d92 : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":5},{"id":"5addf055ea1bee45558b5f30f8801d5fe7b85f86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 14:10:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e756e0f4cf3462d9b7655584e256177\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/71a847073da645b7a3cc187ebf0343bb : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":5},{"id":"323fcb681245f48a45a095d80a3f5aa8afbfe127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 18:16:52.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f495d483c8f3117969943ee126eb5ca8ff6d6965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 18:21:49.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"95c0616558aa222b65b8bc55cc52d500546f4bbd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 18:29:03.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ec1187ea642b158b14db8a957faf6f40a759dd3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 20:34:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d4ff5e743b448dbacffce9cc5d3249d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd5b018ba1874ed2954a8470eb38073f : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/ab8862b9d90e4e5c8145b476a5bc0921 : SUCCESS in 57m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/011a64485f334582bcc5724e96998de1 : SUCCESS in 1h 02m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ad68ed642ef455da33cb713df0eb787 : SUCCESS in 52m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dc8f38c6d92457085511c6b7bf53ed8 : SUCCESS in 1h 50m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/518061acda024e17a7b3541ed7dd96d2 : SUCCESS in 15m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/053bc12ecc9a47f98051d3426ed29cc8 : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b21161e50c742c0a41245ecf6abacea : SUCCESS in 19m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a411df3616d49958118de222f200b2a : SUCCESS in 11m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a405473cebb040eb846430dab3d777f8 : SUCCESS in 26m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/08f0b37a5f824f6fbd40c57288ec723b : SUCCESS in 24m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e427138182134e7c91178693ea5c26a4 : SUCCESS in 29m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ede3175db08e4a6e8a9dd3db0a86a324 : SUCCESS in 34m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f4a9d4d511f3415a8c1cde8100c26f71 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c15ff7a9fe4348da9afd0f7b1b76243c : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fa8096218c6a44cca76dc86909cae65f : SUCCESS in 13m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa2286d0176a4832a137fa71f9919d4e : SUCCESS in 6m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6366b591a7054d88834f27806470e016 : SUCCESS in 13m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e973199a9dcf4c9694ed9978a51493d2 : SUCCESS in 9m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0cb71119ff34dddab755624fceaa204 : SUCCESS in 55m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd9ca88a9804451389eb057b0f3c9ba3 : SUCCESS in 49m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6da11391dd8c49068a395355c3a6ea03 : SUCCESS in 37m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24a392b468964a75ad7312181bf14129 : SUCCESS in 27m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e3fde3b405c494d879ef32a0c0dbf2a : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88130c7a781f4358bc6171a5b648d40a : SUCCESS in 16m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c8df473c7e443eea94078abfa0939ae : SUCCESS in 51m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/391e43d103bd4e0e9851121ef8345948 : SUCCESS in 1h 09m 29s","accounts_in_message":[],"_revision_number":7},{"id":"b2ba6fc451bd7ab26502da8a784ceb2ffc82a73a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 21:16:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39aa2c52028848a9a0abe58b46eb046d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a34b48195c0144a09baf2a68b308ea51 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":7},{"id":"cd5e302655aef4c369c633313f530e823bb0ff80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-13 14:18:58.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d657198a2af89b97e7bc8f0c354f22ef4cca0ab6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:14:18.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/441d80b253d546538fee32d4868fa04c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/465939a8ebf64b9b9e716a546c52abbb : SUCCESS in 40m 26s","accounts_in_message":[],"_revision_number":8},{"id":"56d4c7f1b1ae2690b0f8f325e6ded57a18b64830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:26:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ccb7e6e187948aab28ce116cd9754ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f871a0e08743c181f579acffb3b7d6 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/07c6ea0c2e1b4b4189176fefc33c3347 : SUCCESS in 59m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/034f41f55db842028c4aec82ce505dbf : SUCCESS in 1h 00m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f6f9cc15a3d4dc1a15c9105ed993e9e : SUCCESS in 52m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf55bf119f8b49d0bab897a8ee9a9eec : SUCCESS in 56m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b8e776323f34087b5a854b8f27fdcf6 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d7cfb16433a4ff2a2e9a7161617b32e : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cfdc8f9151194abfb2cebe238d7ce77e : SUCCESS in 19m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b4013341ca4e46a2a8988051c0588141 : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb0024db755f497a95431e25a281fa27 : SUCCESS in 25m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b46a08adfe440208ab9922376ac57bb : SUCCESS in 22m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4b386ac6b39948a4ab32b11f1dd216df : SUCCESS in 29m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bb6c40c3eeb243869b6cca4815fc0dbb : SUCCESS in 30m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/340da072b96b4a3588a407c61e008691 : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c41f4606b52b43bc905817e884a71117 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2ad52262aeb34316ae26075d59b13c13 : SUCCESS in 13m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/301c560baa8f46a1ba17744e75f2be14 : SUCCESS in 6m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eca910e697f5432e834623089586199e : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/70b8c23d6590473cbc48b0bb08537dcd : SUCCESS in 8m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c89511183d654816a891e07e9a3109d3 : SUCCESS in 58m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/43b831452eff4c22b9f51df71cf352e6 : SUCCESS in 49m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e335d02a20eb47e3a754f6def532f2c6 : SUCCESS in 43m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecc07c144a3f4a81b47b7cdf71f66a3d : SUCCESS in 31m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3bf2a5354543c09394939ce823a674 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b65d0231ba07402bbf637c12f82660de : SUCCESS in 16m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24d5934b01584ddf90e0ff8863b49b1f : SUCCESS in 37m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2954ad1db7484db7a1a082bcdf7235ad : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":8},{"id":"fa92af296cda733d9a253598e4821d586361cb33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-14 14:51:16.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"ff3a75453f1efbd2ba77b89558b5ee4ccb7881c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 15:48:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90adbadc4c4b43daacfeee3108c324d3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9412b0c8e2ba429885256cdced11906d : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":9},{"id":"d6a8bc679ca22f2e93bd049d99a10828c4814c97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 16:14:16.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c491d86738ce4e82ac439520f6b8815c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70b9910c4e78457a98b3ed9f2d11231e : SUCCESS in 10m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/e2c7e0ad9f2d48239167e54a1bf3c65b : SUCCESS in 59m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c043fc48c8a8410e9efb37f4e38e313f : SUCCESS in 37m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3e1ea9b919c4e4291b25a3a7d371e45 : SUCCESS in 51m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de95b068d75141678445d7d40f93ed39 : SUCCESS in 1h 05m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84f1f837e39c47c4a8b1e7d6fcc9465c : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07ec2c81b01b4db2ade1ed7135db47ee : SUCCESS in 11m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b5cf10ec8ef49d28f3083497b0177da : SUCCESS in 21m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/528226851404451abbdbc4a14aebfdc4 : SUCCESS in 8m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f10a0ceb9e14266935ca164f32a89b0 : SUCCESS in 26m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/371146fa35514facaf306225ac7d0486 : SUCCESS in 8m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2fe5eaefec2142a9a52e349434d51155 : SUCCESS in 11m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fffe207521744c6288b2ef4681c9e7e4 : SUCCESS in 32m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ffe9cde1479e49c6bb36a06aef8898a2 : SUCCESS in 10m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/014b4db2c0054a23bd9b2cbcec1ccadc : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5f3a24845a1341de87e556437fcd66a0 : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fb042a7aae74ae3b96dd10ab5f3bb60 : SUCCESS in 7m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f0738c42d2684054ac793e8f7ccef0db : SUCCESS in 10m 27s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/11b74cd2140b43b7bc0c8cc017bc75bb : SUCCESS in 10m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/83beaf0042ec436a9360b0c9f2a92834 : SUCCESS in 55m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/974f203c536e4c0f8fa732860e8e4d3a : SUCCESS in 56m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c3f78b165d884f9680187df41b0a9f63 : SUCCESS in 56m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ffe5b2a15594f149db40cf7eb65f700 : SUCCESS in 31m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c1b4656c0e4cba83ad58fb664ee01a : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/53f105f5b4f64e6f832749dc1ba9469e : SUCCESS in 18m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f12c41e93aba485baafdede4f8e4f42c : SUCCESS in 55m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1102314f699e4774a6c42de7812bdf72 : SUCCESS in 1h 04m 35s","accounts_in_message":[],"_revision_number":9},{"id":"f579dcc81d236e223358257d00ce254c625d4273","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-27 15:35:09.000000000","message":"Abandoned\n\nwe decided that jitter should always be added to newly minted Timestamps","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"32dcdd80b7b28201793ebe7dca6e89e699278cdc","revisions":{"8d4b9cd9c96f54e88ffe4dc82e2979cf538f4cd5":{"kind":"REWORK","_number":1,"created":"2026-01-05 16:17:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/972217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/972217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/972217/1"}}},"commit":{"parents":[{"commit":"84386b50c583146290109e454994767049f06f9f","subject":"WIP Timestamp: add random jitter to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84386b50c583146290109e454994767049f06f9f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 13:49:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 14:35:28.000000000","tz":0},"subject":"Timestamp: default with_jitter\u003dFalse","message":"Timestamp: default with_jitter\u003dFalse\n\nBy default, init a Timestamp without jitter.\n\nTimestamp.now() uses with_jitter\u003dTrue.\n\nChange-Id: Ia09ddea386bd57519c788e1e31543769b7e53b28\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d4b9cd9c96f54e88ffe4dc82e2979cf538f4cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d4b9cd9c96f54e88ffe4dc82e2979cf538f4cd5"}]},"branch":"refs/heads/master"},"388993326100247478443fcf74e7c1cc4663ba30":{"kind":"REWORK","_number":2,"created":"2026-01-06 17:43:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/972217/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/972217/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/972217/2"}}},"commit":{"parents":[{"commit":"dbb873462c5a8fcd1d194e043a1393a61782ee50","subject":"WIP Timestamp: add random jitter to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbb873462c5a8fcd1d194e043a1393a61782ee50"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 13:49:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-06 15:28:57.000000000","tz":0},"subject":"Timestamp: default with_jitter\u003dFalse","message":"Timestamp: default with_jitter\u003dFalse\n\nBy default, init a Timestamp without jitter.\n\nTimestamp.now() uses with_jitter\u003dTrue.\n\nChange-Id: Ia09ddea386bd57519c788e1e31543769b7e53b28\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/388993326100247478443fcf74e7c1cc4663ba30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/388993326100247478443fcf74e7c1cc4663ba30"}]},"branch":"refs/heads/master"},"32f5dd158a399651d3c11ac75bd95a69c3714525":{"kind":"REWORK","_number":3,"created":"2026-01-06 18:04:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/972217/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/972217/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/972217/3"}}},"commit":{"parents":[{"commit":"3690ca3f77179340acd80225bdb739d5cf6017c9","subject":"WIP Timestamp: add random jitter to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3690ca3f77179340acd80225bdb739d5cf6017c9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 13:49:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-06 18:03:56.000000000","tz":0},"subject":"sq? Timestamp: default with_jitter\u003dFalse","message":"sq? Timestamp: default with_jitter\u003dFalse\n\nBy default, init a Timestamp without jitter.\n\nTimestamp.now() uses with_jitter\u003dTrue.\n\nChange-Id: Ia09ddea386bd57519c788e1e31543769b7e53b28\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32f5dd158a399651d3c11ac75bd95a69c3714525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32f5dd158a399651d3c11ac75bd95a69c3714525"}]},"branch":"refs/heads/master"},"95b84a6f41518b8e417d09f43edf1c44558a8924":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-07 12:12:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/972217/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/972217/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/972217/4"}}},"commit":{"parents":[{"commit":"a6120b81b6240390f8f8b395d6c42f7862645d76","subject":"WIP Timestamp: add random jitter to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6120b81b6240390f8f8b395d6c42f7862645d76"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 13:49:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 12:03:30.000000000","tz":0},"subject":"sq? Timestamp: default with_jitter\u003dFalse","message":"sq? Timestamp: default with_jitter\u003dFalse\n\nBy default, init a Timestamp without jitter.\n\nTimestamp.now() uses with_jitter\u003dTrue.\n\nChange-Id: Ia09ddea386bd57519c788e1e31543769b7e53b28\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95b84a6f41518b8e417d09f43edf1c44558a8924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95b84a6f41518b8e417d09f43edf1c44558a8924"}]},"branch":"refs/heads/master"},"529d0e86552a4fd486988fbde98fd4fd2c1c60ec":{"kind":"REWORK","_number":5,"created":"2026-01-07 12:30:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/972217/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/972217/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/972217/5"}}},"commit":{"parents":[{"commit":"78c947335ad3a0b797798ce9b98a6786ddb5219c","subject":"WIP Timestamp: add random jitter to offset field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78c947335ad3a0b797798ce9b98a6786ddb5219c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 13:49:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 12:28:08.000000000","tz":0},"subject":"sq? Timestamp: default with_jitter\u003dFalse","message":"sq? Timestamp: default with_jitter\u003dFalse\n\nBy default, init a Timestamp without jitter.\n\nTimestamp.now() uses with_jitter\u003dTrue.\n\nChange-Id: Ia09ddea386bd57519c788e1e31543769b7e53b28\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/529d0e86552a4fd486988fbde98fd4fd2c1c60ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/529d0e86552a4fd486988fbde98fd4fd2c1c60ec"}]},"branch":"refs/heads/master"},"99c2d2c3a363a52c46132470e764d06ae9237d7d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-01-07 18:16:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/972217/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/972217/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/972217/6"}}},"commit":{"parents":[{"commit":"3c7eca03717c88f4ff5523b774eb95bc519566ce","subject":"tolerate but round meta timestamp jitter in backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c7eca03717c88f4ff5523b774eb95bc519566ce"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 13:49:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 18:15:45.000000000","tz":0},"subject":"sq? Timestamp: default with_jitter\u003dFalse","message":"sq? Timestamp: default with_jitter\u003dFalse\n\nBy default, init a Timestamp without jitter.\n\nTimestamp.now() uses with_jitter\u003dTrue.\n\nChange-Id: Ia09ddea386bd57519c788e1e31543769b7e53b28\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c2d2c3a363a52c46132470e764d06ae9237d7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c2d2c3a363a52c46132470e764d06ae9237d7d"}]},"branch":"refs/heads/master"},"44d093c61c4da005806e8293cdb8a6d1b5172740":{"kind":"NO_CHANGE","_number":7,"created":"2026-01-07 18:21:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/972217/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/972217/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/972217/7"}}},"commit":{"parents":[{"commit":"c6e0be8060ffff486feea245c3cec5b8b8234b49","subject":"tolerate but round meta timestamp jitter in backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6e0be8060ffff486feea245c3cec5b8b8234b49"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 13:49:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 18:18:58.000000000","tz":0},"subject":"sq? Timestamp: default with_jitter\u003dFalse","message":"sq? Timestamp: default with_jitter\u003dFalse\n\nBy default, init a Timestamp without jitter.\n\nTimestamp.now() uses with_jitter\u003dTrue.\n\nChange-Id: Ia09ddea386bd57519c788e1e31543769b7e53b28\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44d093c61c4da005806e8293cdb8a6d1b5172740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44d093c61c4da005806e8293cdb8a6d1b5172740"}]},"branch":"refs/heads/master"},"40081b857cbe8ab3cd84eebef25d2d0c1a8fb49e":{"kind":"REWORK","_number":8,"created":"2026-01-13 14:18:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/972217/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/972217/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/972217/8"}}},"commit":{"parents":[{"commit":"9c0fae691219333cd7348b6d341e3de99561ddb4","subject":"tolerate but round meta timestamp jitter in backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c0fae691219333cd7348b6d341e3de99561ddb4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 13:49:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 14:18:11.000000000","tz":0},"subject":"sq? Timestamp: default with_jitter\u003dFalse","message":"sq? Timestamp: default with_jitter\u003dFalse\n\nBy default, init a Timestamp without jitter.\n\nTimestamp.now() uses with_jitter\u003dTrue.\n\nChange-Id: Ia09ddea386bd57519c788e1e31543769b7e53b28\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40081b857cbe8ab3cd84eebef25d2d0c1a8fb49e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40081b857cbe8ab3cd84eebef25d2d0c1a8fb49e"}]},"branch":"refs/heads/master"},"32dcdd80b7b28201793ebe7dca6e89e699278cdc":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-01-14 14:51:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/972217/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/972217/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/972217/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/972217/9"}}},"commit":{"parents":[{"commit":"eaa55727b384b11b89edb4e7e05629483329d452","subject":"tolerate but round meta timestamp jitter in backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eaa55727b384b11b89edb4e7e05629483329d452"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-05 13:49:50.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-14 14:25:24.000000000","tz":0},"subject":"sq? Timestamp: default with_jitter\u003dFalse","message":"sq? Timestamp: default with_jitter\u003dFalse\n\nBy default, init a Timestamp without jitter.\n\nTimestamp.now() uses with_jitter\u003dTrue.\n\nChange-Id: Ia09ddea386bd57519c788e1e31543769b7e53b28\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32dcdd80b7b28201793ebe7dca6e89e699278cdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32dcdd80b7b28201793ebe7dca6e89e699278cdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
