)]}'
{"id":"openstack%2Fswift~972536","triplet_id":"openstack%2Fswift~master~Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-01-19 19:46:06.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc","subject":"tolerate but round meta timestamp jitter in backend","status":"NEW","created":"2026-01-07 18:16:52.000000000","updated":"2026-03-02 06:52:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0a5492a9611976c4a061f96a5874cb55c22d3241","_number":972536,"virtual_id_number":972536,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-27 19:11:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-07 19:48:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-02 06:52:30.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"402838e34428a56d0f287d18e37eaf2461f2123e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 18:16:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02aadf12d876dbda145c4cc5f68672c46eaf66bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-07 18:21:49.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ba7654747afaba477d15e34cb9ff00c714f92086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 19:48:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc39cc6e99714c1681e087832fbda17d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a4ddfac340464390d5555691a65cc0 : SUCCESS in 10m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/62a78d45c5304103aad2a4c3508497ca : SUCCESS in 54m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d094890e11cf43fcaa5c9440e8da5368 : SUCCESS in 51m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/34c9fbdd5ea644f096aa0ce07fd7f844 : SUCCESS in 54m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e78efa4994a4eb6b24528e22ad3d357 : SUCCESS in 55m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0554a9fe51d6459bb2324d373c209a75 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3226fa1304e148d0b07fb97e51d097f8 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58da9036da0c449b816ec1110825e5c2 : SUCCESS in 18m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bea41f9418b540a4bb02a4ba5bffb996 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5ec28eb772c1417987aa2dbc42de741d : SUCCESS in 24m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4bdc0f72cecc4263adf2d08324587701 : SUCCESS in 21m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e0a69a00a4a242c982cdb523dfb467d5 : SUCCESS in 32m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/49932cde951542c68d672166ab0a3ca8 : SUCCESS in 34m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/34a7c7d1deee46ff94f077fe2ec510ac : SUCCESS in 12m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/00a5b0ce87e84807bd4cd8d24dbb7d12 : SUCCESS in 13m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f4bf680b67064f6a8cad39f2a5a0d9e3 : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/efcac5a78ed049ad9c4f4408ffa216b1 : SUCCESS in 8m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9943276c21d74d4faee8c3bf886eaedf : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e1481ad6811c40c6b292a5f9375e11b1 : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e373370394146dca9b1b29d5e1a6b19 : SUCCESS in 59m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fffefe7142d446a80edb65eb9c4d9a6 : SUCCESS in 37m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce28dc922eff4f2e80f0a5a2d11b52ae : SUCCESS in 51m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e090d5d0947437293ae9f8ded199479 : SUCCESS in 28m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a2fc09bfca040469986067d9ac81822 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38c94632f721496b873773526677c93f : SUCCESS in 19m 03s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f557b0eed204705acac4285e570aa59 : SUCCESS in 42m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f06b7f4eae27484daaefec02d5d9c4ab : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c6b23c5054aee244539f414646d362f1de5a8726","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 21:14:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ef6ae9a5e5f40f1ae6317a2a4d5e742\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7295177eb0c745cbb4cb4306974a9a86 : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":2},{"id":"fa8ea5350df6c9f94d8ac0c2f8d25a61fcaa2940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-13 14:18:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"292842233c6b87619affbbba483a1339580c037c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:19:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b65a092bb1ea4650a5e5b29d5799b36f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/78ea5fceeec144b3bb46f53794d97ae9 : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":3},{"id":"c91c4ff047402a6d5ae3676156178d684711b6ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:29:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42634ccda5514364b4ea63d5dd1f771c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1747c9d682b548e18146ec875b2081d7 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/e9fe20c6e4b64ede970109b1f8c9f031 : SUCCESS in 56m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/136b457704c74a07ba966a007fde0f5f : SUCCESS in 55m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb7b0286a29346299a7b3abf64c1e925 : SUCCESS in 40m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c900051eba11407193b5765c0f5954cd : SUCCESS in 1h 01m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/619878ad40d749a0b6bce7adedfa85e4 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b6b99953cf5649cfa0b691db6cb8e7e9 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0334eaf7096646b0bdf4afcf84370505 : SUCCESS in 18m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/95fe1430e79d4291985aa752b070f25f : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8e5034667255431ab8c3c2a143bdb64c : SUCCESS in 29m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e1f43fb0d4e44d2e8ce5597738001a1d : SUCCESS in 12m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8fa0a2df41a847a2a76dba936482b68d : SUCCESS in 15m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a860009e3fa742b8ad08466d7a8d6f9d : SUCCESS in 34m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e5332019ed9944a4a132c3ee5628e4dc : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/614fe515ee454747964b4f4df93b9eb0 : SUCCESS in 13m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/babb3eaa4b464d7bba74369d05f8ddea : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/675b2598e8a841b38950e015bd38115d : SUCCESS in 5m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/08b9e45197754b4d9ece894ce55e2c14 : SUCCESS in 16m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dd94db8b52964d77985e3a573461f597 : SUCCESS in 9m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dffb431580534864bc61f2d15cbe00fa : SUCCESS in 55m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a0b14fd38dc948a7a7950e1e8e202a2f : SUCCESS in 48m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b04993113a1a4cc082ac41224d85da4f : SUCCESS in 35m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c6d9692d71e46d79c38db89c98d5f78 : SUCCESS in 27m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e142095d7c7346b8bf0e8ecc10c5f925 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/006cd4681d434539a176aafc39a1d581 : SUCCESS in 18m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66b511168f5d406eb1ba5b764685a160 : SUCCESS in 54m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6ea5aad9fad4de1a3b6891f461c4654 : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":3},{"id":"b989e0d4471fc682526d663d7be4b7e9c1eff3e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-14 14:51:16.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"0cc1ad55d6c034beb8c4cdf0238a90a81a31af48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 16:09:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/007c330865a24b168314540f770d9a04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311e1cd5a1184b469cafdd8953450070 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/7c69ac470f194cb2aff9a0c29d08cc6a : SUCCESS in 1h 00m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d429f682973415394563e7f18caf818 : SUCCESS in 1h 07m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dde343de7be541d58f50b698d8f263d8 : SUCCESS in 53m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a21775f1fcb14893a209875758854059 : SUCCESS in 39m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/780dc91c2a9e4d72a6bc604e63d27bd8 : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99e5c78c728d4d2e9201cd3f3ec8e267 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e50109b2b337486f8a606c13607d6a48 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd354e80eace4d34b7aca0b0da4dd123 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0474007b63c84e39bb92d993c96a7713 : SUCCESS in 26m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d07abc15a7a740bdb779391bab830cd8 : SUCCESS in 9m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9d6f6f9f79674019bb9de22bde18a2e5 : SUCCESS in 15m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a1c916a055024904a6258665a2f054ea : SUCCESS in 33m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0c2b30255fc7408bba36dc86765926a4 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/88b860e26e1c40a086772802ac2e4c2f : SUCCESS in 8m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/59eec4258e6c4bb6a4fe3597c15cdc4e : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2fec63f4df6e416fae904294b4b3c95d : SUCCESS in 4m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fdea8ea561c5467e98661730a5e8af93 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d25c8585f3c405aacf4ecb402863314 : SUCCESS in 5m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfcb06c0544d4090b0ac83ddd35c4b07 : SUCCESS in 59m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e460d76bafbd458a8899bb378d76a96d : SUCCESS in 46m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f9e20b6e7ea4c6da60682b8fd702c03 : SUCCESS in 37m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ce2441083fc49d6a871fbe935f5a658 : SUCCESS in 29m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7903eaf7decb4176a10caf32a5453844 : SUCCESS in 2m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02c9ead003c94fca82e4f22dca63cfb5 : SUCCESS in 19m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ba2054d57944c238c050aacf675f399 : SUCCESS in 56m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b3e334d5ab948c9b24828137c0e6124 : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":4},{"id":"b29c458816c644bca4996e529861d9e6ee0fe21b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 16:54:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25fbe9a7802e497ea563f08b093c3305\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/72e67e066415450db15132360f82b1a9 : SUCCESS in 1h 01m 48s","accounts_in_message":[],"_revision_number":4},{"id":"767bf204d67c7f0026c461dd260bac93f538f267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-16 17:43:41.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0b11427cdf3082da49b783142309c5b884658806","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 18:39:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e68091e2ed24725bc01eb5db9fddf69\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d18eedbce7346c7ac87e14b72973cd3 : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":5},{"id":"cf547aea851a94d56b981d722f6040eff1a99de8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 19:02:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9de54b47b23c42f7a5bf4f84f5cf4b3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88301cdd49fb49adade69030e89499d3 : SUCCESS in 5m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/28ce4befe3934651b567fab8268cd1a0 : SUCCESS in 1h 02m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/598ff12fd6b94da4a3a01ccd95f2a0bc : SUCCESS in 1h 06m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c36f464a64d3458ea421ff4f49498e08 : SUCCESS in 35m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f54de750bed549e7acb10ffa460281f1 : SUCCESS in 38m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6eae7e77e1e4683ae91bfb66cb1aa38 : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf557197e32549b59b1c10574219ffea : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80458ff4b5a845308b280fbe55ec505b : SUCCESS in 12m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8390faf575f64d72afbd9ff4381e69f5 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/13be7498a1dc4665897b21d043d8eebe : SUCCESS in 25m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eab095f1e89a40c4b6c5707a1786fc5b : SUCCESS in 25m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d7f0215700534da18062cbc3f2ae8a29 : SUCCESS in 14m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2c74e4a8442846aa9536e141cc08f164 : SUCCESS in 20m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/66a2a244cef04cc0b8781f7af5765ab6 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5632c64a9cf443d6b732bd62cdddff49 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0ca7649dbde942daa1f5ba90de6c2222 : SUCCESS in 14m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b1116918dc44ddaa2c5643c57f9cf50 : SUCCESS in 9m 57s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5033c1aa5fcb415a9c0be3b704aca179 : SUCCESS in 14m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/17feddcb8f4d41f2a8882439e729acba : SUCCESS in 8m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ac82cd998964267aa6de6254eb79a87 : SUCCESS in 54m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/954cf56a385041e7b6755631a0c5efe0 : SUCCESS in 54m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/09d969ffe79c481ebe3c07f3f3683d7d : SUCCESS in 25m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/330a6a87119f4e60bd397ed911485940 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df0b8cd517714948a39a8fac6a1229f6 : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/703186f74cef4aa7b073e28421b3bc9c : SUCCESS in 20m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/386c75b805c844e886e75a984194a57f : SUCCESS in 34m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da11282391fe4aa2aa766ec8123b6601 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":5},{"id":"8f6d8e130db61c8cdad058ca277fb6e474861eb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-19 18:25:13.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"5008b5568dfd869161bd837230a512cb778bbf0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 19:46:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3eb8700c04f475bb1c7d9767880d98e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9afe2e892c69419aaac5d5c741a73d18 : SUCCESS in 9m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/34c2729ca2b64a179dc5467165e103f7 : SUCCESS in 59m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b1e0e172f81841d9968f32dfc686264c : SUCCESS in 59m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/12b10d43e9ca4cee9791d3ba79b554c7 : SUCCESS in 55m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/520f6cedafa54115883147c800d6e880 : SUCCESS in 1h 04m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd6a985bf81947e5bb4206cd5719a845 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15c01e4f8092485b97aa5a17f7c7e418 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da9d4ed4fe9643ea8b344701fd5079c2 : SUCCESS in 16m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0eec0475c878482f92da867f5ac51288 : SUCCESS in 4m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a8ebc72d08a2495bb24c8dc0f21a7ae2 : FAILURE in 27m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/22a396c307cb4c619395521e456814d4 : FAILURE in 23m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cb0d677feb354783ad09d663de02e645 : FAILURE in 21m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a1884ac5a6374dffb18b93b66d8c7979 : FAILURE in 25m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/44f713db087f4fb899be6166f1b7c5d2 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f27d281f28d0428d99e6c58f7870cbfa : SUCCESS in 14m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f07608a2fe2947949c7881aa0586b566 : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7482f521feda49f4b89f42769f8a7c12 : SUCCESS in 6m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/902b7b696ed448ffa19f7eb1b0568c16 : SUCCESS in 14m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4c205858a2134f30a4e50fb5116da7d8 : SUCCESS in 9m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ab1544eecea4775ac76e67a9e58542b : SUCCESS in 54m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0783e9db150d4565951c4fff2b0d21d4 : SUCCESS in 50m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2c6adac2b27343d39d671e63bc57e746 : SUCCESS in 37m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f40c42550e14091a14911eba393cd38 : FAILURE in 29m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56a9e7efeb664d36bcea711b9ca6ee02 : FAILURE in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7e2ca0a56c14c8a999753e395c37d37 : SUCCESS in 18m 53s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/531b7ae5596c4d5187d09fb8ada68dcb : SUCCESS in 50m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddbc8ce58fe744679ffa97dafbeb6e81 : SUCCESS in 44m 44s","accounts_in_message":[],"_revision_number":6},{"id":"0d19d6cf31ca3ab9ecfba6004326fe4f35852f4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 20:45:20.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fd25f835ba347dfaa53088c4abe326d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/790e7d7e7b1d48f094e6a9556731ec4c : FAILURE in 1h 03m 02s","accounts_in_message":[],"_revision_number":6},{"id":"158279daf2733439217cb1808fdc5ae54d9c3235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-20 17:05:12.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"1fe627e7b8f5d58422f006de92da885a7c64531e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 18:20:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9af2e923988c4315993197b2bf744009\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b55e5b9d4784c54992a1e2482de9f7f : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/970f4812d416499da59c549248ae9817 : SUCCESS in 31m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eacb80b8a76f4e8fa0962e2a821027f9 : SUCCESS in 59m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2fef5a8959944a3bbbb94bcb88cd0c50 : SUCCESS in 36m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c035b78fe1f46d9980e0b2c4daf157c : SUCCESS in 37m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3cbd33f79e64a04afe6a52425f9063e : SUCCESS in 11m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5cd370a9698c4828b7eacaa6948cb171 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57307b08a8a3432385107325eea4353b : SUCCESS in 20m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de5bb96f0c7644398af101e5d7999de3 : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/08d1144e6a8546a4ab57e66e2f86988d : SUCCESS in 11m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a79e2d70cefd4b7f9f0e6d6429934764 : SUCCESS in 22m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/767d7603f63a48798242293e5feb1c63 : SUCCESS in 40m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6ba839000520456ab9c6092bafee0503 : SUCCESS in 15m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a187534d07704055b90ec575262c77fa : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8628bafb0a9e4f44b7ae50f6a39b3c10 : SUCCESS in 13m 39s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ac8774510b30459ab74b45a0fb8732e1 : SUCCESS in 8m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dc33ae6b2fb54eaa869248b101bccc34 : SUCCESS in 12m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/67d95abbf5194896bae2378afdb3c15f : SUCCESS in 10m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9ac27d93e3a24749acd5da299f6a2f25 : SUCCESS in 8m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/94c408431a6445a88525a5cad283661f : SUCCESS in 1h 07m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f04695c6e68d4db19f474d33c31fe1b5 : SUCCESS in 21m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fd00bead8e574bf98e1d9d8c08c49442 : SUCCESS in 44m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5316d968638d44d0aff654eb074f7d97 : SUCCESS in 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/065f0fd6fa0b4f89a06861ed6a580eee : FAILURE in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a64fc2b90824eb2bf463e0ea5b135cc : SUCCESS in 18m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1d1adbdceac4c9d9e5c09301723c9ce : SUCCESS in 40m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/751d10dec7604132bd34646b2856b005 : SUCCESS in 1h 08m 46s\n\nWarning:\n  Comments left for invalid file test/unit/container/test_sharder.py","accounts_in_message":[],"_revision_number":7},{"id":"6057fe2c5b0864ed656b40496f8646a1b8369ad9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 19:35:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33107b6b26144760b7c15dea0ee0a6c1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50c1b981fdcd4e91b858122a3c949d96 : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":7},{"id":"efab00f02a6981c8fa3adb1ea277b8f7974a7f77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-01-21 10:11:20.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"52c0852668887c7fba4ba336e787163799beb402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 11:27:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f5caa97ebc04589b126fc05eb204902\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8690437d1b134ee88726a675ab638b16 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/fb7cde336b6144c5a9f2a6c68759b9d9 : SUCCESS in 58m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f965f14f72814d0d8b15ac2704b543c8 : SUCCESS in 1h 08m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/254b72d6bce1483cbfca2eabef0199a7 : SUCCESS in 46m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00cb765d280144ef8cca2bae39d146c7 : SUCCESS in 55m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e480777881e420fa5097a26ce9b2c78 : SUCCESS in 17m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6000a18a58d4a44a9a6d37268119d54 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0210142f29c6458690a19fe268426ff9 : SUCCESS in 14m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/937b9430a249420dbf3305d72d6059e0 : SUCCESS in 10m 47s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/378e6fde4e794b8288bfb3d51cc84988 : SUCCESS in 21m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8eaea7e739eb4d169e8244cf5908d19e : SUCCESS in 19m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/29324ec4834a44d0a2eae2f5ec983020 : SUCCESS in 22m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/124d012f0f014363a8eb1964abcdd4e3 : SUCCESS in 32m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8cdc8834d6c84f4b8aaa274ba350d759 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e19d4938bde4490fbc78002f16680e78 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dd83fd782cba4a2b8bd1ec7e336a9099 : SUCCESS in 14m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ca6b9e945ea4bc8ae05b3f96729670c : SUCCESS in 7m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8f102f2ccd9d4d3f8c1a70793d3b4f06 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d1b5cae911ed4eb49a92801d02930074 : SUCCESS in 6m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/39172eae2a7a4ed083ff6e08e8d89f2d : SUCCESS in 57m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/604dc6440d2d436e90a1573150815027 : SUCCESS in 49m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/22457579ed74414c90225a60f7b7dfed : SUCCESS in 55m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfe4e71de66d4b0aa855ee07181dbfbc : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5719b6a9e60b4cc19a41b8c072246b3a : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8bbf5bcafa974e1cb052f3ef3793856f : SUCCESS in 13m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b12b1f29c6444b5694877d4b8e89469d : SUCCESS in 52m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9a4aa5f720b42d8807db679b5e82a92 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":8},{"id":"7483e4cac1030321b4bd67d590e6694996b75d49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:10:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24114717de0e49e9b61ee7c07f010cbd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e30bd28e1a2b4e2aa28b90fce619176c : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":8},{"id":"928bd9a692af7d5d69770cfd6a5ee46a080e914f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-01-29 03:56:58.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"b0a3d96167a3bcc9917e724f7c1ba6fd2b3f41e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 05:10:50.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b7625f74786425992e97e4ef2c56871\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171b7b57749d448d8b8b2df21796b5d5 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/b83200eb68f5400a83332795fdaf94aa : SUCCESS in 52m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ecde316f606418bbb8ad74de9dc8d8f : SUCCESS in 1h 00m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b7676260dc1b4cac98ab2275e92bdfc0 : SUCCESS in 35m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8f41d3011f74eb4a1a1d31b13fc2f6d : SUCCESS in 37m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d980c699425f4210a86f7ba7bb0b55de : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a2ede937bf746d2a023c6369e1ab5dc : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/451d87fe3b224649a7b71ad1da0443b0 : SUCCESS in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/84378b94c0de4eab8b9ff11cee60e876 : SUCCESS in 7m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0a24161946164a39a6dc7093aa56ae56 : SUCCESS in 26m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4526293825b64f26af6d623504ec6ba4 : SUCCESS in 17m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/64a0a2cc7e264cddb62fe2a410548231 : SUCCESS in 40m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4b12adfc822645fe9832259cb097219e : SUCCESS in 33m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aa0cfc0120494a33bcb4e21d312c4c94 : SUCCESS in 13m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a336a66d3dcc49f1a3d5af6614a577d7 : SUCCESS in 16m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c4a8a7d3765a4b5aa160440337e0dacf : SUCCESS in 12m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e54f39746504de281ce8b4f1e81ae50 : SUCCESS in 6m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a17ce9683665425f882cac5436ccc271 : SUCCESS in 14m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5e29fd6d68bf41f19454174a604b1e79 : SUCCESS in 8m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc2e5df564ce4706b8f1f28feef9ed08 : SUCCESS in 50m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b27d5d650bbf4beb95cc3b87215ac44d : SUCCESS in 51m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c443ebf96f424c59b8d08bb27238be00 : SUCCESS in 56m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48ea0058ca644fcb86eb0d0e4ec52977 : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb8f627a2b5b4923a4ed64dd95d208dd : SUCCESS in 2m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ae7a56a845b4a61be9bb200aa361597 : SUCCESS in 17m 33s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d09c2bada42e4a6891256ad4cc21c8a5 : SUCCESS in 59m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a7a90225273461da01560ae54b6c70a : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":9},{"id":"c9954216ee1a9feda46ff13e92ab8d691cdfa31f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 16:44:07.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad11fb5288cd45a59377ad5be8b0f07c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4299ddc5a96e4f269b28dd7545cd2920 : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":9},{"id":"a24123865e6f02c38c789f629830456d29a53a9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-02-27 18:03:04.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"942ce1c9209963c37558b983bd92fe9c31de9ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 19:11:39.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c22a22cba92440888f32388b665935c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa0866dbcfca4297b81c447b94e092d7 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/3015485d03b240c3b6d4c58be110cc37 : SUCCESS in 58m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/536562b69bc44f87abfb22b17cd6caf5 : SUCCESS in 1h 01m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5bbb61d8ce3f4f33bd7f8992e4a3902e : SUCCESS in 56m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5293c5cb199402986fe217e7dfaddc8 : SUCCESS in 37m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db527de5940f409ca270b0f4596ccfe6 : SUCCESS in 11m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93c594d74dd54cb69f1853e05c00986c : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3aed7c5cbd3449d8323719847c7d14b : SUCCESS in 18m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/72a83ea23175487b9ae17eed383ea9e7 : FAILURE in 4m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6baa1b9ba9a746569e9a22e5b59babc6 : SUCCESS in 17m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1312541eb9dc4e3c9f25392604f1a05b : SUCCESS in 10m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5dff55e167354c129e382c3275c910c4 : SUCCESS in 11m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8ae0f054744346a1a3b769d9a0160b39 : SUCCESS in 11m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6e024092c2024f70986731ec99b206ba : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/62431e60c11c4c858b734e86a5dde299 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ebf2296e6df1433382330f833384256a : SUCCESS in 11m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57fc83b96ce34200aba3362f1b6602f2 : SUCCESS in 4m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff9af53bc4a64473b34fc6734b858657 : SUCCESS in 14m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d0b1dd17b674dd69dd7d0b1ecfe2653 : SUCCESS in 5m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5776afb59144bc9ade9aaf09715c98f : SUCCESS in 48m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a57cf3f60214ffea3236b59f762d4e2 : SUCCESS in 1h 00m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e688b4728ba74cc59f3f484912cd7e0d : SUCCESS in 51m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/244d211cce4b47059ec6fc8e1614b3a2 : SUCCESS in 32m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b691161103b4436b952985c1c5e7361 : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c8cc37e761af474ead765796fea97bdb : SUCCESS in 27m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75fb803e623644909c778c91b54a3ada : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":10},{"id":"ba3e0617935c4cb4b996ab82e2e4426d309bbeb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-27 19:57:12.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b1b2f356dfb4635830d8235410d0a0a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/01d51a38adc54e92a571b49282263885 : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":10},{"id":"0a5492a9611976c4a061f96a5874cb55c22d3241","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-03-02 06:52:30.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"8ba8093d94dc4b7a93b3e45c78fe7f243b53c8c6","revisions":{"3c7eca03717c88f4ff5523b774eb95bc519566ce":{"kind":"REWORK","_number":1,"created":"2026-01-07 18:16:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/972536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/972536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/972536/1"}}},"commit":{"parents":[{"commit":"78009e87e1103e1fd4bb1e6488e381efe1200b43","subject":"object_server: return 400 if POST timestamp has jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78009e87e1103e1fd4bb1e6488e381efe1200b43"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 13:44:54.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 18:15:44.000000000","tz":0},"subject":"tolerate but round meta timestamp jitter in backend","message":"tolerate but round meta timestamp jitter in backend\n\nModify object and container servers to tolerate object POST requests\nwith jitter, and therefore meta timestamps with jitter. The jitter\ncomponent is not always preserved, so concurrent POSTs in the same\ndeca-microsecond that have different metadata may succeed in\npersisting different metadata on different replicas.\n\nOn object servers content-type timestamps in the same deca-microsecond\ncannot be differentiated in on-disk .meta file names.\n\nOn container servers, meta anc content-type timestamps in the same\ndeca-microsecond cannot be differentiated in object table rows.\n\nChange-Id: Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c7eca03717c88f4ff5523b774eb95bc519566ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3c7eca03717c88f4ff5523b774eb95bc519566ce"}]},"branch":"refs/heads/master"},"c6e0be8060ffff486feea245c3cec5b8b8234b49":{"kind":"NO_CHANGE","_number":2,"created":"2026-01-07 18:21:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/972536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/972536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/972536/2"}}},"commit":{"parents":[{"commit":"dbfce274ba28d0c5adeed1f738067bdabb4319c4","subject":"object_server: return 400 if POST timestamp has jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbfce274ba28d0c5adeed1f738067bdabb4319c4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 13:44:54.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 18:18:58.000000000","tz":0},"subject":"tolerate but round meta timestamp jitter in backend","message":"tolerate but round meta timestamp jitter in backend\n\nModify object and container servers to tolerate object POST requests\nwith jitter, and therefore meta timestamps with jitter. The jitter\ncomponent is not always preserved, so concurrent POSTs in the same\ndeca-microsecond that have different metadata may succeed in\npersisting different metadata on different replicas.\n\nOn object servers content-type timestamps in the same deca-microsecond\ncannot be differentiated in on-disk .meta file names.\n\nOn container servers, meta anc content-type timestamps in the same\ndeca-microsecond cannot be differentiated in object table rows.\n\nChange-Id: Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6e0be8060ffff486feea245c3cec5b8b8234b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6e0be8060ffff486feea245c3cec5b8b8234b49"}]},"branch":"refs/heads/master"},"9c0fae691219333cd7348b6d341e3de99561ddb4":{"kind":"REWORK","_number":3,"created":"2026-01-13 14:18:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/972536/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/972536/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/972536/3"}}},"commit":{"parents":[{"commit":"190706f0958cfc55e746ffc109faba207f9aa246","subject":"object_server: return 400 if POST timestamp has jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/190706f0958cfc55e746ffc109faba207f9aa246"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 13:44:54.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-13 14:18:11.000000000","tz":0},"subject":"tolerate but round meta timestamp jitter in backend","message":"tolerate but round meta timestamp jitter in backend\n\nModify object and container servers to tolerate object POST requests\nwith jitter, and therefore meta timestamps with jitter. The jitter\ncomponent is not always preserved, so concurrent POSTs in the same\ndeca-microsecond that have different metadata may succeed in\npersisting different metadata on different replicas.\n\nOn object servers content-type timestamps in the same deca-microsecond\ncannot be differentiated in on-disk .meta file names.\n\nOn container servers, meta anc content-type timestamps in the same\ndeca-microsecond cannot be differentiated in object table rows.\n\nChange-Id: Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c0fae691219333cd7348b6d341e3de99561ddb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c0fae691219333cd7348b6d341e3de99561ddb4"}]},"branch":"refs/heads/master"},"eaa55727b384b11b89edb4e7e05629483329d452":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-14 14:51:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/972536/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/972536/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/972536/4"}}},"commit":{"parents":[{"commit":"a0c39580d0f6f3cdfbfafbe561cbec16a6d4092e","subject":"object_server: return 400 if POST timestamp has jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0c39580d0f6f3cdfbfafbe561cbec16a6d4092e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 13:44:54.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-14 14:25:24.000000000","tz":0},"subject":"tolerate but round meta timestamp jitter in backend","message":"tolerate but round meta timestamp jitter in backend\n\nModify object and container servers to tolerate object POST requests\nwith jitter, and therefore meta timestamps with jitter. The jitter\ncomponent is not always preserved, so concurrent POSTs in the same\ndeca-microsecond that have different metadata may succeed in\npersisting different metadata on different replicas.\n\nOn object servers content-type timestamps in the same deca-microsecond\ncannot be differentiated in on-disk .meta file names.\n\nOn container servers, meta anc content-type timestamps in the same\ndeca-microsecond cannot be differentiated in object table rows.\n\nChange-Id: Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eaa55727b384b11b89edb4e7e05629483329d452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eaa55727b384b11b89edb4e7e05629483329d452"}]},"branch":"refs/heads/master"},"82874d553cef2dcd22eaeadb46bf665639ef2549":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-01-16 17:43:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/972536/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/972536/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/972536/5"}}},"commit":{"parents":[{"commit":"641dc5038956a990fc96a04d6d0e2c11a66de763","subject":"object_server: return 400 if POST timestamp has jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/641dc5038956a990fc96a04d6d0e2c11a66de763"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 13:44:54.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-16 17:42:42.000000000","tz":0},"subject":"tolerate but round meta timestamp jitter in backend","message":"tolerate but round meta timestamp jitter in backend\n\nModify object and container servers to tolerate object POST requests\nwith jitter, and therefore meta timestamps with jitter. The jitter\ncomponent is not always preserved, so concurrent POSTs in the same\ndeca-microsecond that have different metadata may succeed in\npersisting different metadata on different replicas.\n\nOn object servers content-type timestamps in the same deca-microsecond\ncannot be differentiated in on-disk .meta file names.\n\nOn container servers, meta anc content-type timestamps in the same\ndeca-microsecond cannot be differentiated in object table rows.\n\nChange-Id: Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82874d553cef2dcd22eaeadb46bf665639ef2549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82874d553cef2dcd22eaeadb46bf665639ef2549"}]},"branch":"refs/heads/master"},"4b5a0ef3f40c0f6178e584c485b6274f5ecb2ab4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-01-19 18:25:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/972536/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/972536/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/972536/6"}}},"commit":{"parents":[{"commit":"710b726efb1918ec5e975a4b32306c2e975daa82","subject":"object_server: return 400 if POST timestamp has jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/710b726efb1918ec5e975a4b32306c2e975daa82"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 13:44:54.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-19 18:25:00.000000000","tz":0},"subject":"tolerate but round meta timestamp jitter in backend","message":"tolerate but round meta timestamp jitter in backend\n\nModify object and container servers to tolerate object POST requests\nwith jitter, and therefore meta timestamps with jitter. The jitter\ncomponent is not always preserved, so concurrent POSTs in the same\ndeca-microsecond that have different metadata may succeed in\npersisting different metadata on different replicas.\n\nOn object servers content-type timestamps in the same deca-microsecond\ncannot be differentiated in on-disk .meta file names.\n\nOn container servers, meta anc content-type timestamps in the same\ndeca-microsecond cannot be differentiated in object table rows.\n\nChange-Id: Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b5a0ef3f40c0f6178e584c485b6274f5ecb2ab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b5a0ef3f40c0f6178e584c485b6274f5ecb2ab4"}]},"branch":"refs/heads/master"},"fc7649d048c710da9d3dc1c923b4d486604eac77":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-01-20 17:05:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/972536/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/972536/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/972536/7"}}},"commit":{"parents":[{"commit":"1dbf3f91195dc993ce9cb1d8ccfba65290aeca04","subject":"object_server: return 400 if POST timestamp has jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1dbf3f91195dc993ce9cb1d8ccfba65290aeca04"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 13:44:54.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-20 15:25:21.000000000","tz":0},"subject":"tolerate but round meta timestamp jitter in backend","message":"tolerate but round meta timestamp jitter in backend\n\nModify object and container servers to tolerate object POST requests\nwith jitter, and therefore meta timestamps with jitter. The jitter\ncomponent is not always preserved, so concurrent POSTs in the same\ndeca-microsecond that have different metadata may succeed in\npersisting different metadata on different replicas.\n\nOn object servers content-type timestamps in the same deca-microsecond\ncannot be differentiated in on-disk .meta file names.\n\nOn container servers, meta anc content-type timestamps in the same\ndeca-microsecond cannot be differentiated in object table rows.\n\nChange-Id: Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7649d048c710da9d3dc1c923b4d486604eac77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7649d048c710da9d3dc1c923b4d486604eac77"}]},"branch":"refs/heads/master"},"8fa7680db585255b1ee0571354891d36b6aece5c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-01-21 10:11:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/972536/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/972536/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/972536/8"}}},"commit":{"parents":[{"commit":"6a92daf4bcc9418b24b9963f81a392dc81d5fc50","subject":"object_server: return 400 if POST timestamp has jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a92daf4bcc9418b24b9963f81a392dc81d5fc50"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 13:44:54.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-21 10:10:11.000000000","tz":0},"subject":"tolerate but round meta timestamp jitter in backend","message":"tolerate but round meta timestamp jitter in backend\n\nModify object and container servers to tolerate object POST requests\nwith jitter, and therefore meta timestamps with jitter. The jitter\ncomponent is not always preserved, so concurrent POSTs in the same\ndeca-microsecond that have different metadata may succeed in\npersisting different metadata on different replicas.\n\nOn object servers content-type timestamps in the same deca-microsecond\ncannot be differentiated in on-disk .meta file names.\n\nOn container servers, meta anc content-type timestamps in the same\ndeca-microsecond cannot be differentiated in object table rows.\n\nChange-Id: Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fa7680db585255b1ee0571354891d36b6aece5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fa7680db585255b1ee0571354891d36b6aece5c"}]},"branch":"refs/heads/master"},"c7e181a0fc417305c069595195dd635d4c23c331":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-01-29 03:56:58.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/36/972536/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/972536/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/972536/9"}}},"commit":{"parents":[{"commit":"04dc8bc486379413a6c14a530745fb21e0174e96","subject":"object_server: return 400 if POST timestamp has jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04dc8bc486379413a6c14a530745fb21e0174e96"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 13:44:54.000000000","tz":0},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-01-29 03:55:28.000000000","tz":660},"subject":"tolerate but round meta timestamp jitter in backend","message":"tolerate but round meta timestamp jitter in backend\n\nModify object and container servers to tolerate object POST requests\nwith jitter, and therefore meta timestamps with jitter. The jitter\ncomponent is not always preserved, so concurrent POSTs in the same\ndeca-microsecond that have different metadata may succeed in\npersisting different metadata on different replicas.\n\nOn object servers content-type timestamps in the same deca-microsecond\ncannot be differentiated in on-disk .meta file names.\n\nOn container servers, meta anc content-type timestamps in the same\ndeca-microsecond cannot be differentiated in object table rows.\n\nChange-Id: Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7e181a0fc417305c069595195dd635d4c23c331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7e181a0fc417305c069595195dd635d4c23c331"}]},"branch":"refs/heads/master"},"8ba8093d94dc4b7a93b3e45c78fe7f243b53c8c6":{"kind":"REWORK","_number":10,"created":"2026-02-27 18:03:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/972536/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/972536/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/972536/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/972536/10"}}},"commit":{"parents":[{"commit":"7c24d02e0e2ba050ced0707377d133e8ca8be938","subject":"object_server: return 400 if POST timestamp has jitter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c24d02e0e2ba050ced0707377d133e8ca8be938"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-01-07 13:44:54.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-02-27 18:02:07.000000000","tz":0},"subject":"tolerate but round meta timestamp jitter in backend","message":"tolerate but round meta timestamp jitter in backend\n\nModify object and container servers to tolerate object POST requests\nwith jitter, and therefore meta timestamps with jitter. The jitter\ncomponent is not always preserved, so concurrent POSTs in the same\ndeca-microsecond that have different metadata may succeed in\npersisting different metadata on different replicas.\n\nOn object servers content-type timestamps in the same deca-microsecond\ncannot be differentiated in on-disk .meta file names.\n\nOn container servers, meta anc content-type timestamps in the same\ndeca-microsecond cannot be differentiated in object table rows.\n\nChange-Id: Iaef8c0b63fb9b406d7c5e736c30075412bcc3fbc\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ba8093d94dc4b7a93b3e45c78fe7f243b53c8c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ba8093d94dc4b7a93b3e45c78fe7f243b53c8c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
