)]}'
{"id":"openstack%2Fswift~972802","triplet_id":"openstack%2Fswift~master~I932e10f6d2867fb0ad57443a72f734eea64dfd6b","project":"openstack/swift","branch":"master","topic":"swift-conf-d","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-01-09 03:16:35.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I932e10f6d2867fb0ad57443a72f734eea64dfd6b","subject":"use readconf in storage_policy","status":"NEW","created":"2026-01-08 21:58:13.000000000","updated":"2026-01-09 03:16:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"942536ffed5498a27fe728228e2bccdbd535ad16","_number":972802,"virtual_id_number":972802,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-08 23:27:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 23:27:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-09 03:16:35.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"b24510b66471c0cf9c7e0c10abfa4851f3031782","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-01-08 21:58:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c65af681e9237bc63a255113c0e6e8adb7e2a2b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 23:27:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5192ce3ef6e94fdba653ac58e98e0964\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffe109b6e3fe4dd9aa3edd2206c3ebfa : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/9887be2837f146288410aea282b72454 : SUCCESS in 46m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4893a1d4250b46579d09944a8c18785d : SUCCESS in 1h 07m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/153a04e877d9498982481cee75e289ab : SUCCESS in 1h 00m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56e85d4abc584b2cae668b21c1912e68 : SUCCESS in 41m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c6412e9b80b4ed490aec4ea38c9afd5 : SUCCESS in 10m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c001242d5fa4cc8a4e282dcba59f530 : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b16f941c6f3049c49702d696efe4beac : SUCCESS in 11m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0308bab86e9b426fa2283179165b469a : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d38dba5ee9ac49098c105550f237417c : SUCCESS in 28m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/79a4cf7ab36e4adc99ff674cca433728 : SUCCESS in 18m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f0e458f735854a71805a747924379d2e : SUCCESS in 26m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b49af2e351884cbbb9d5f0903f78b349 : SUCCESS in 34m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1252cfc903db4c4c96201a67f5f997a5 : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4a172e7732f449a48d78e84ccc186181 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aaf0e7f303cc4c84a0a83f418069326b : SUCCESS in 14m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b81e05f02e644197b7d4b6ba6c7d1583 : SUCCESS in 5m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9b8cb2b6ab224e8d95b743efcb5a96c5 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5499f359eff9432da5706bdc11ae5f31 : SUCCESS in 7m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/74fbd541105440a5bba8920bfbf2e9b6 : SUCCESS in 47m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13467f36124841cc9a3298f0d1370014 : SUCCESS in 44m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6f408335d7a4c02b0c2530f6ae03dcf : SUCCESS in 51m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d13889b327642398dfa78f0bd94fcf1 : SUCCESS in 26m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096fd9b17b2e4861978f4db68d1c9f18 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf9e3ab4e2a545dca0c31f8e8ed1d510 : SUCCESS in 15m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8eeaab41b9148d788b0f7d41934ec43 : SUCCESS in 30m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4c32e50101646abb265006ef78e1249 : SUCCESS in 48m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9f1a43b09008751c2804de4cc82ee796ee579424","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 23:38:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f535812f36d42afade89f929fc47dd4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d1810a21fa14db7bf2d14492dc65b1b : SUCCESS in 42m 21s","accounts_in_message":[],"_revision_number":1},{"id":"942536ffed5498a27fe728228e2bccdbd535ad16","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-01-09 03:16:35.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3558512027a2cdeeedf7d3bddcf5a3894fa41bc","revisions":{"e3558512027a2cdeeedf7d3bddcf5a3894fa41bc":{"kind":"REWORK","_number":1,"created":"2026-01-08 21:58:13.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/02/972802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/02/972802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/972802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/972802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/02/972802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/02/972802/1"}}},"commit":{"parents":[{"commit":"184768a97984ce530a9ec3ff680a837e9b0f6a9d","subject":"use readconf for swift-hash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/184768a97984ce530a9ec3ff680a837e9b0f6a9d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-08 20:19:00.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-08 21:57:35.000000000","tz":-360},"subject":"use readconf in storage_policy","message":"use readconf in storage_policy\n\nChange-Id: I932e10f6d2867fb0ad57443a72f734eea64dfd6b\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3558512027a2cdeeedf7d3bddcf5a3894fa41bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3558512027a2cdeeedf7d3bddcf5a3894fa41bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
