)]}'
{"id":"openstack%2Fswift~972803","triplet_id":"openstack%2Fswift~master~Ie6036d80dd89ea7b1f162f6ae31620f3d12ca98f","project":"openstack/swift","branch":"master","topic":"swift-conf-d","hashtags":[],"change_id":"Ie6036d80dd89ea7b1f162f6ae31620f3d12ca98f","subject":"use readconf for swift-constraints","status":"NEW","created":"2026-01-08 21:58:13.000000000","updated":"2026-01-08 23:17:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2729ea1101b900f02d67c21aef48783a0643ece3","_number":972803,"virtual_id_number":972803,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-08 23:17:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 23:17:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21ba93c32d17fae14e07a9bfcd2a569812622d9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-01-08 21:58:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2729ea1101b900f02d67c21aef48783a0643ece3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 23:17:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64e1905ee2af4ccd82aab8f0ce06c01f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ecfc7dbb2a4a14a0be29c4267eeb8c : SUCCESS in 10m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/6e1275c5445f4a449bf1ae4a23047cd3 : SUCCESS in 30m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f76155564e4497b837b259770d221dc : SUCCESS in 30m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/927dc2152cd745c49ae5d62352aba2b3 : SUCCESS in 53m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e77dd0b938b4e0d9dfd9facf59ae0f1 : SUCCESS in 1h 07m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61633729e1ac4247bb6eb90fbaf923b0 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/194af4cd25254da29fb8090ffcbf6d9d : SUCCESS in 10m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0316361d949a467c8750e50724fbbd71 : SUCCESS in 16m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4fe1f9a2e19b46959da5b8d09f7cd795 : SUCCESS in 7m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a678cad536ea45ea9e6472d85397b14d : SUCCESS in 24m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/06bc72cd0ed441db8223731b1d09544f : SUCCESS in 20m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a69060804209400593172262d9c4df6a : SUCCESS in 28m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/220cefcda4ab44729b25a133161952bb : SUCCESS in 23m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f0c296d5e93c4dd0a34f876565e03a5e : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6e4ba28f8236483d93b94271ebc8ff44 : SUCCESS in 14m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/35958b84c5074646b6729de278229c04 : SUCCESS in 13m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/851ffcc12dab47a3944afcf1f1d1d588 : SUCCESS in 6m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d75d900a01c044908a0c5ef96e29e56c : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9c9610cbf44f46eaa218c5134fde88a8 : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b012ae550df14ae8b824a2e0022ae98c : SUCCESS in 52m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/894186bb12a347d6abeca1fe685b47ce : SUCCESS in 52m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e27ba7359bc46d099c7342d87f6fab9 : SUCCESS in 38m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c708170ae614b64929ab5eda53023f1 : SUCCESS in 27m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0fe4d7788f240138e9d152d77971987 : SUCCESS in 3m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88642c6f1f9640e5b926f726879e3cec : SUCCESS in 16m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/720d5d4e62624e67b241771557d82519 : SUCCESS in 45m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d81ca062ed5c44e5bb473c97853b520c : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5cee48c29c433283e60c212485b8021447805363","revisions":{"5cee48c29c433283e60c212485b8021447805363":{"kind":"REWORK","_number":1,"created":"2026-01-08 21:58:13.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/03/972803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/972803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/972803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/972803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/972803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/972803/1"}}},"commit":{"parents":[{"commit":"e3558512027a2cdeeedf7d3bddcf5a3894fa41bc","subject":"use readconf in storage_policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3558512027a2cdeeedf7d3bddcf5a3894fa41bc"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-08 21:14:51.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-08 21:57:35.000000000","tz":-360},"subject":"use readconf for swift-constraints","message":"use readconf for swift-constraints\n\nChange-Id: Ie6036d80dd89ea7b1f162f6ae31620f3d12ca98f\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cee48c29c433283e60c212485b8021447805363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cee48c29c433283e60c212485b8021447805363"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
