)]}'
{"id":"openstack%2Fswift~972804","triplet_id":"openstack%2Fswift~master~Iea1c8dc1cfc182fdd7e0f4c3b89060e06e7e7206","project":"openstack/swift","branch":"master","topic":"swift-conf-d","hashtags":[],"change_id":"Iea1c8dc1cfc182fdd7e0f4c3b89060e06e7e7206","subject":"Support swift.conf.d","status":"NEW","created":"2026-01-08 21:58:13.000000000","updated":"2026-01-08 23:10:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9070937319319ad1f54adaf1c8277c9df88c5abd","_number":972804,"virtual_id_number":972804,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-08 23:10:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 23:10:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6fde137b346423e06880d9286eef0f118eb9a72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-01-08 21:58:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9070937319319ad1f54adaf1c8277c9df88c5abd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 23:10:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/694a202f404a4d09ab2284868cc9969e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94d14e5d611445fda74f9d11163cb641 : SUCCESS in 5m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/418c1b777a904fba8ae1fc941c1c4811 : SUCCESS in 52m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1f36712720a34700aebcb34a8bb66ad5 : SUCCESS in 47m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/acb89f4edd224fb1a51e0617de64afbf : SUCCESS in 38m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b214881d0699472a95afe07961cdaea7 : SUCCESS in 34m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89c10c6d5ab843289ad3e3c4a0815163 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f7c01613ae64804acddf4d1925fdd78 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07af3bc046374ff998064329cfae66f5 : SUCCESS in 17m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9663ffc752354e5798dbf1abf2d484a9 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9621c7257c544155915e6cc660c72f80 : SUCCESS in 18m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6bb7c1b5987348dba80789f0502c0ebe : SUCCESS in 21m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3e79aacdb81e4849b48c47c99e72362b : SUCCESS in 22m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3b883e06ba54438b9e887d57156787ff : SUCCESS in 28m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/86cca1d4be8b4da99303f1f6d3579b40 : SUCCESS in 12m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a7bd0659e32747b38515eedbbe88e8d4 : SUCCESS in 13m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/af2c832bdede4f88a483ce2b48cfadea : SUCCESS in 8m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/96ded885b4f84c05bd22054f56dc1f13 : SUCCESS in 8m 22s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8dfe22b4c38245c3a1a8df2028e964ec : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/60ac5e560bb84c5caf4172ed8e623031 : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0b3fcecce464736a0faead2f30f645b : SUCCESS in 1h 00m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57bffc34703e4a4fb4ade5d18d3be2e3 : SUCCESS in 36m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fed58553d4ae424aaf917c52c165a3b5 : SUCCESS in 56m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37bb540f98664b45b0a4725949237f93 : SUCCESS in 25m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eece171eef3439fb41644283b131225 : SUCCESS in 2m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc0a51e88d974a21aba4a93bc4da8f27 : SUCCESS in 19m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbfb9687dd7445709c101d75c065024a : SUCCESS in 46m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bdd770a8e8d429c9d81a5f958aa3e6e : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a82a53465b0f1a42059584fd6645784ee2677d5c","revisions":{"a82a53465b0f1a42059584fd6645784ee2677d5c":{"kind":"REWORK","_number":1,"created":"2026-01-08 21:58:13.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/04/972804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/972804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/972804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/972804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/972804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/972804/1"}}},"commit":{"parents":[{"commit":"5cee48c29c433283e60c212485b8021447805363","subject":"use readconf for swift-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cee48c29c433283e60c212485b8021447805363"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-08 21:52:03.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2026-01-08 21:57:57.000000000","tz":-360},"subject":"Support swift.conf.d","message":"Support swift.conf.d\n\nConsistently read swift.conf and allow override configuration\nfrom swift.conf.d\u0027s .conf files in all the places we load swift.conf\n\nTODO: tests for readconf_and_dir_override\n\nChange-Id: Iea1c8dc1cfc182fdd7e0f4c3b89060e06e7e7206\nSigned-off-by: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a82a53465b0f1a42059584fd6645784ee2677d5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a82a53465b0f1a42059584fd6645784ee2677d5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
